)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~756587","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~Iaf92e78758c159db7e969a89a889c4d897f5045c","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","hashtags":[],"change_id":"Iaf92e78758c159db7e969a89a889c4d897f5045c","subject":"Include libpython and rsync for centos in lxc base image","status":"MERGED","created":"2020-10-07 20:57:59.000000000","updated":"2020-10-08 16:48:48.000000000","submitted":"2020-10-08 16:46:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756587-1602175620092-2bbe566a","meta_rev_id":"939b7b0accda25302143b3a27757c5dc7cd35c25","_number":756587,"virtual_id_number":756587,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"date":"2020-10-08 14:13:57.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-08 16:46:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-10-08 15:00:31.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2020-10-08 12:22:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":2,"date":"2020-10-08 12:53:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":1,"date":"2020-10-08 12:53:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-08 12:22:17.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2020-10-08 12:53:49.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"},{"updated":"2020-10-08 14:13:57.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-10-08 16:46:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4e944799327f6a3fd7b435ee3e672c15d9831425","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-07 20:57:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e6fea27c877df9097a6eac6d05c41d8adeff6f1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 21:27:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/285656bbbd174de69bb797bfa6d01146 : SUCCESS in 11m 05s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/ec18cd4878e3487fa8c9cc2251ae816f : SUCCESS in 12m 29s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/d53a083688f043a48cd65d1c9cfc9552 : SUCCESS in 15m 31s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/344f4bffd79b43869fd582d2e355e146 : SUCCESS in 12m 00s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4cb570f2711d4ff1815ebed9ff5efec0 : SUCCESS in 12m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b3f6ab7555345dd9ade2187142094e5 : SUCCESS in 5m 46s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/4f44ee68c2ed45a98d1fdc4c9206efd6 : SUCCESS in 20m 58s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6e485e3e3e72453d967966f7124b803d : SUCCESS in 20m 20s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b1ee4a4b5f764bf9aa9974b0a75c7a64 : SUCCESS in 14m 52s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0dc5376a24214ca98c31f4ae59b0603e : SUCCESS in 14m 54s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/566a5a769ceb4b7cb0732fea3335c168 : SUCCESS in 19m 31s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/921cbae79cbf46b7a43ae641b543c535 : SUCCESS in 14m 44s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36625ffb96be4bc4afd7686271781b04 : SUCCESS in 15m 41s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9427f4496ae8408098626c477a8d3271 : SUCCESS in 16m 27s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d829c0258f444d1af6dc951d76b564d : SUCCESS in 15m 50s","accounts_in_message":[],"_revision_number":1},{"id":"b04a9c4c08ae37cbadd37ddaf493212e28713d73","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-10-07 21:45:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3720d3b727962abe5526c21d3437f2f4d59c9a74","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-08 06:40:53.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"b61f6416d897e2076fe8723d01e8704a32991e78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 07:14:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8ea8aacf7c1b438fadb332d68edb3dd3 : SUCCESS in 8m 25s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/65c79dc119224a9d93eec88808293c21 : SUCCESS in 10m 15s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/4f4d754fe819457593ed5fd790e432ea : SUCCESS in 19m 27s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6fabf7786ae5474c9a08bc9245473079 : SUCCESS in 12m 11s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6f12aed98b247cdbb254fd8c088e832 : SUCCESS in 11m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f17d8f96ce684f6b8d27d1469b82ec66 : SUCCESS in 7m 04s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/5d0a80a57c49431fafb94214ba55c4c0 : SUCCESS in 25m 21s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/edc6d54192064930b7ed36f1f8d7db84 : SUCCESS in 15m 57s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6db79316302b45d1936d1a4d98a62a17 : SUCCESS in 20m 28s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/96d7eb03a52c4b13a78b3084a2af9979 : SUCCESS in 15m 52s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a299d5afbc9049acb78d10bdb757bfb4 : SUCCESS in 18m 48s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e0387ada009c42548ee8913fba36ab69 : SUCCESS in 22m 25s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8888f6c3f294818a2881732b8d0076e : SUCCESS in 16m 01s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/13332fe803b24963b2e5298af20cb1aa : SUCCESS in 17m 56s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6bb88a500ebe4963b92a2117ae345b73 : SUCCESS in 19m 19s","accounts_in_message":[],"_revision_number":2},{"id":"36998f2c37f91825454fa2681ee5bc5802421219","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-10-08 12:22:17.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"420945bd52ca2a413d87388d0bbe2c3ca21cbc8a","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-10-08 12:53:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6d255aee7ee3f1340b8c56e5301e880785bbc350","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 12:54:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"61c340b3a65697b756115220b7d4a8e47131ced4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 13:13:17.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0e15148ebf464d549cbf738ceedcc612 : SUCCESS in 12m 16s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/941ea0f765414365b7756d43eb1c94d9 : FAILURE in 9m 37s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3b9ab58deab3468696dd8618c439b555 : FAILURE in 11m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4958410b92844473ab4c97be4b8976d6 : SUCCESS in 7m 16s","accounts_in_message":[],"_revision_number":2},{"id":"33c7c8d1794e8c3ed2eb2e0888be99fac1cc9c0c","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-10-08 13:15:25.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1ad9d01478b164267229c3df7dcd584271162f7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 14:00:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/eb6ac27725fe4355ae08977dbcc459ee : FAILURE in 8m 18s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/840447170e124c2fa90e0d104049acdc : SUCCESS in 15m 06s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/2b52bd5ba865450cbd5b8afcb9181d90 : SUCCESS in 15m 49s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7e543ce613da4ddd8a325a5d3ebd1026 : SUCCESS in 15m 07s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/acbd950f5d36464ab169b4db01b3c2e7 : SUCCESS in 18m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9998f1c87f3a421187f798eb965224e1 : SUCCESS in 5m 21s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/8d08e0c1b60f4fd4883b7138b8093b2d : FAILURE in 10m 22s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6f2e82b6d5984cf3aefebb5604265fef : SUCCESS in 15m 01s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4f136f1399043f38873e6808fb75f9a : SUCCESS in 15m 55s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4e574d6ce8fb42f79c73837182b44702 : FAILURE in 9m 05s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/647d9cd6e2db4aaaaf8179c69740aec3 : SUCCESS in 16m 28s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/25e61e89fead42d58db2e7388a56aafd : SUCCESS in 16m 33s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/232ec7324b0c4e909b828d687be75c1b : SUCCESS in 16m 28s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/07743ba8a98d4885bce5807fcb2a51f7 : FAILURE in 8m 06s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e54a67ff9b9a4400967ce516daea7044 : SUCCESS in 15m 48s","accounts_in_message":[],"_revision_number":2},{"id":"344b1359f5b85038f00cbedab829fb8ac9c4cf29","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-08 14:13:57.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7e10913883fda07eebe065c73c9c6e8e826fecf0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 14:57:32.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/60acdfab1b2444b6b197e1b99ae2e4a9 : SUCCESS in 10m 12s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/38d6355c0aa74ebba3b565bd5534a3eb : SUCCESS in 10m 56s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/cb16fe307549491bbe1a4fcc972b59e3 : SUCCESS in 15m 02s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b0a2e10319d74a32bf55a0276f53afb8 : FAILURE in 8m 09s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/58c1ceee0c1249e4bd8b610e3efdbd25 : SUCCESS in 16m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62b77d44ff9444539b3724dbace37e0d : SUCCESS in 6m 48s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/9221a293f5be4dfb97939337b5c0fd06 : SUCCESS in 21m 54s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fde1a0988a244b57a462c4dab6c486a0 : SUCCESS in 17m 22s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b38b93e39cab4dc18037957456ab88bb : SUCCESS in 16m 20s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7aac8f08bb4447b585159708f5ba53ad : SUCCESS in 17m 36s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c57a2dd5f3449819d4a12e442e7ea0e : SUCCESS in 17m 00s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/18a063d305454c2188ac810b5c80ddbb : SUCCESS in 15m 58s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c01a457f7c54d6a88ba466d7bf26b8a : SUCCESS in 18m 27s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a0b9a6b15d964b31b94b482e5763a02c : SUCCESS in 16m 31s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa857f89d9a043699bbc8b8aed2be642 : SUCCESS in 21m 19s","accounts_in_message":[],"_revision_number":2},{"id":"cc84a39fb4e6e65772d42fbb7c9d895315b86b0e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-08 15:00:31.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f08d3bf320b70610dbf17ca38357b13585af655b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 16:26:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/83b1e36b828542bd949595f4fc6d067b : SUCCESS in 11m 32s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/53aca2d4c02d4caab6959da766991451 : SUCCESS in 9m 52s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/df949337467b4d14a78348b81e5d74c8 : SUCCESS in 31m 48s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b516b6c390f34f66aaf5027b46608730 : SUCCESS in 10m 36s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/64eedae0a3154f79b9bce41e44f56c8e : SUCCESS in 11m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b25220d9b36241bcbb56b9d4b6205db5 : SUCCESS in 7m 15s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/813093f20c94458398a38dcd8926ae3d : SUCCESS in 23m 42s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5d1d074d65cd4c7893873f91a616aa17 : SUCCESS in 18m 16s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/63b0efdc8e834b92838457f62713eed2 : SUCCESS in 14m 55s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/421928f36e304c8791e1bdb2c0ba4f69 : SUCCESS in 16m 08s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a6ef79a75324ef7ae0f9e71928d3a68 : SUCCESS in 15m 56s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/05895f16c17d4cc1a3820ab80be9feed : SUCCESS in 17m 36s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/81155c27bcc54fd3bbbbb64485e4a486 : SUCCESS in 15m 47s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a38f3e7bd8084a14894cfd6ac2f6c4ef : SUCCESS in 14m 58s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c71e2b163e14999aa9946ead99717dc : SUCCESS in 18m 25s","accounts_in_message":[],"_revision_number":2},{"id":"486efed8c00c6001f9262557fdbc8814b01ac46c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 16:26:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c890d0c51debbcdcd58acc66a7151c2d29442900","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 16:46:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ce19ea661e084766b9b7d233564378c1 : SUCCESS in 8m 35s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/fb9fc6e2b3a34a199b5fa9d8346f136f : SUCCESS in 9m 56s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/337643e72a33455083bc3d1c155a03b9 : SUCCESS in 11m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56b0ed66f6a04f1d9f797845d0e36dcf : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":2},{"id":"771bbe209309a8db1de62d8b78168b85a9ca101c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 16:47:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"939b7b0accda25302143b3a27757c5dc7cd35c25","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 16:48:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98c82d281d7540898cb137a0b05d9efd : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"676f309eb9f959a2ceae079284453be84add7e66","revisions":{"ae6c4492ab8a3b38a37c668a5071ff0f04c7cbb8":{"kind":"REWORK","_number":1,"created":"2020-10-07 20:57:59.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/87/756587/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/87/756587/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/87/756587/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/87/756587/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/87/756587/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/87/756587/1"}}},"commit":{"parents":[{"commit":"26d4a8066e1ddba4858319cbaf1a2e544cf0423a","subject":"Merge \"Replace some base packages removed when minimising the base image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/26d4a8066e1ddba4858319cbaf1a2e544cf0423a"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-07 20:57:06.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-07 20:57:06.000000000","tz":60},"subject":"Include libpython in lxc base image","message":"Include libpython in lxc base image\n\nSome but not all roles use ansible-role-uwsgi to install libpython.\nThose which do not require libpython to be in the container base\nimage.\n\nChange-Id: Iaf92e78758c159db7e969a89a889c4d897f5045c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ae6c4492ab8a3b38a37c668a5071ff0f04c7cbb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ae6c4492ab8a3b38a37c668a5071ff0f04c7cbb8"}]},"branch":"refs/heads/master"},"676f309eb9f959a2ceae079284453be84add7e66":{"kind":"REWORK","_number":2,"created":"2020-10-08 06:40:53.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/87/756587/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/87/756587/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/87/756587/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/87/756587/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/87/756587/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/87/756587/2"}}},"commit":{"parents":[{"commit":"26d4a8066e1ddba4858319cbaf1a2e544cf0423a","subject":"Merge \"Replace some base packages removed when minimising the base image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/26d4a8066e1ddba4858319cbaf1a2e544cf0423a"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-07 20:57:06.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-08 06:40:47.000000000","tz":0},"subject":"Include libpython and rsync for centos in lxc base image","message":"Include libpython and rsync for centos in lxc base image\n\nSome but not all roles use ansible-role-uwsgi to install libpython.\nThose which do not require libpython to be in the container base\nimage.\n\nSome roles use rsync for copying rootwrap files (cinder as example) and\nin the long term this should be added to the distro packages for those\nroles. In the meantime we add it back to the container base for centos\nto allow other patches to merge.\n\nChange-Id: Iaf92e78758c159db7e969a89a889c4d897f5045c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/676f309eb9f959a2ceae079284453be84add7e66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/676f309eb9f959a2ceae079284453be84add7e66"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
