)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~757612","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I730ea2b7e6d913af020a81993f0b0e37c90a1a59","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","hashtags":[],"change_id":"I730ea2b7e6d913af020a81993f0b0e37c90a1a59","subject":"Add rsync to container base image","status":"MERGED","created":"2020-10-12 16:11:27.000000000","updated":"2020-10-13 07:25:52.000000000","submitted":"2020-10-13 07:23:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"757612-1602573835590-0f5752a4","meta_rev_id":"a4ab00bc2a261e2e570725a8e235bcfb9a42b873","_number":757612,"virtual_id_number":757612,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-13 07:23:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2020-10-13 07:04:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2020-10-12 16:15:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2020-10-13 07:04:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-12 16:15:29.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-10-13 07:04:15.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2020-10-13 07:23:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5b5c5ea04e181d3e961ec20ba35ea38e761b98b0","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-12 16:11:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cc8c7dabef286d07b5f97906ae12a38fbda23bf8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-12 16:15:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"84547c00030692a68cca9c78d9d8cf31519a6f10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 17:08:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a019c3656d5a45369fc91c28c3e66686 : SUCCESS in 7m 31s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/7cc7d61b89304affb39dba18854b5410 : SUCCESS in 10m 30s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/bcf86fe9f7df4c838bd601cfd9c60235 : SUCCESS in 13m 23s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fea81517038248488f3161df934758bd : SUCCESS in 10m 52s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0e14111cc95470ebe1ca47228dd9d3e : SUCCESS in 11m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2548a23fad9409cbbdcdf0040770549 : SUCCESS in 6m 27s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/a239c5e0089d4dd9927235f6cdede7be : SUCCESS in 21m 06s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5281407543184551a493b4dca8550c1c : SUCCESS in 16m 45s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e3e02f463c0c4197817e4f99f199f0a2 : SUCCESS in 16m 40s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8e3960c91c2b4ccc84d5bbdcf233656c : SUCCESS in 18m 08s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2117083e5a414e41a9a5c77bf2fcfa1b : SUCCESS in 13m 36s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f30a32e01842432481293565dd960120 : SUCCESS in 16m 23s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/008dd87f3eb14dce88335ed7f15414b0 : SUCCESS in 18m 12s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/11673f2fde574968b21d06d9e1ffd238 : SUCCESS in 16m 35s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/88a9d90c7b2b4e138e55472fe7ea8be5 : SUCCESS in 18m 44s","accounts_in_message":[],"_revision_number":1},{"id":"3d05e278f71fae89ac05c671e852237d85887cf3","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-10-13 07:04:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"18c479904c482c38084fe36cd56b1f60b3da3488","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 07:07:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b1e21443529236077086d9ea8188a443c7d99e14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 07:23:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/100a38259b0e46bd9aed0ce4c70d7285 : SUCCESS in 9m 36s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/90f0bd08f5654e1cb61b17973cd4fec3 : SUCCESS in 13m 14s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/705077f84b2d42b19ec8ff8f134bf21e : SUCCESS in 12m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d217b669760491dbbbf4f2a532b24ce : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":1},{"id":"4a2ce602be77c8a08f653fcd45a7aee402cc97e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 07:23:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a4ab00bc2a261e2e570725a8e235bcfb9a42b873","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 07:25:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec4771696e664aec9eded3b61ab0112d : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fd0ba694b545216e1ec92915bf14cd7200c5a344","revisions":{"fd0ba694b545216e1ec92915bf14cd7200c5a344":{"kind":"REWORK","_number":1,"created":"2020-10-12 16:11:27.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/12/757612/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/12/757612/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/12/757612/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/12/757612/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/12/757612/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/12/757612/1"}}},"commit":{"parents":[{"commit":"676f309eb9f959a2ceae079284453be84add7e66","subject":"Include libpython and rsync for centos in lxc base image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/676f309eb9f959a2ceae079284453be84add7e66"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-12 16:10:20.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-12 16:10:20.000000000","tz":60},"subject":"Add rsync to container base image","message":"Add rsync to container base image\n\nKeystone requires rsync to be already present in the container image\nbecasue it runs serial\u003d1 and uses rsync between multiple keystone\nhosts. rsync will not be present on the target host in this situation.\n\nChange-Id: I730ea2b7e6d913af020a81993f0b0e37c90a1a59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/fd0ba694b545216e1ec92915bf14cd7200c5a344"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/fd0ba694b545216e1ec92915bf14cd7200c5a344"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
