)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~758364","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~Ib3e850ba79658a42995cd782a11342aca6858342","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","hashtags":[],"change_id":"Ib3e850ba79658a42995cd782a11342aca6858342","subject":"Increase amount of MaxSessions","status":"MERGED","created":"2020-10-15 08:25:47.000000000","updated":"2020-10-19 13:38:07.000000000","submitted":"2020-10-19 13:36:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"758364-1603114593967-0113b11c","meta_rev_id":"35a1f97a1e74e0acb3bbc23fcf88ba3e12e67efa","_number":758364,"virtual_id_number":758364,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"date":"2020-10-19 06:52:39.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-19 13:36:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2020-10-19 08:39:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-19 09:10:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-19 09:10:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-19 08:39:29.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2020-10-19 09:10:26.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-10-19 13:36:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c359388aaef9c5f8ec06d41261fdd87605489837","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-15 08:25:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f2493a309669ad048c55223d3f09a535cc0cccb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 08:58:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/892affdb6b25411c936fa35d571d04aa : SUCCESS in 8m 09s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/d238373363d443c7afff042af40dc6e0 : FAILURE in 13m 47s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/bebc7da28f414ead8258396a81ab96c1 : FAILURE in 9m 20s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f4fbf2ab409b44fe8037976c1f6995be : FAILURE in 10m 00s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a2174ea3c8e45e9961e41e50129dce4 : FAILURE in 10m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bd6c9051b9c479ba604f99bff0401ab : SUCCESS in 5m 57s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/826f4fdd5dda46bead0ea7cb3fe787e6 : FAILURE in 22m 44s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/5a434d5f9ed14619bf63f22cf1358705 : FAILURE in 18m 06s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/771042b772954b49847b1333f76cfbea : FAILURE in 18m 50s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/e1350564b2054b1984aee39f7e475352 : FAILURE in 13m 24s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2782154d6d2e41efbb0f37d3dc9d3630 : FAILURE in 11m 49s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10a8fe1c389c4f089a5bf287aa47a63e : FAILURE in 12m 04s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9b6ef9a57e084b62a7ac3fafbd0437b8 : FAILURE in 12m 12s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/827581f3c5fe4024a4f4ec8665481c33 : FAILURE in 12m 01s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9a8b40fd6852426ca3b28376a2914ec2 : FAILURE in 10m 10s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab10fb52194c494eb41a7ee592d267fe : FAILURE in 10m 12s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/af4c2cef70ae4d31b83653006241a79d : FAILURE in 17m 55s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/250125dd018b4c339ab17081a9f1a238 : FAILURE in 12m 35s","accounts_in_message":[],"_revision_number":1},{"id":"68baabf0cc966e6318f10c852af28bb2bb965cf7","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-15 09:31:06.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"4d03c7b420547c4ba58b9ccfd29ab2fd3278ad85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 11:48:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/dd08bedb32f443e1b695fcda2e912c75 : SUCCESS in 11m 08s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/7784334c06f1491eb15ab87ad3716519 : SUCCESS in 9m 42s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/4b04f4e9753042dab00b3a83ef9ea470 : SUCCESS in 13m 13s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1de0963d3dff4d25887f52a9119278a4 : SUCCESS in 11m 21s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d15a9e2d04444e6b8644a63f6844973e : SUCCESS in 10m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/291ef82696e64830865479bc26ba606f : SUCCESS in 5m 40s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/bf48764d1acd450998f5921344f097bd : FAILURE in 1h 09m 24s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/d7413109698045c99170ef68f6ae1c93 : FAILURE in 31m 56s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e19301b25c04a74bd16ba74e4551491 : FAILURE in 2h 08m 58s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/26ad3305957548ddaf280b4a57951936 : SUCCESS in 18m 53s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e6a65098217e40c1b3eed6c096b6716b : SUCCESS in 16m 15s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c21317d17764d8a9c65d8830d8aebe0 : SUCCESS in 15m 17s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/726bbaf9553440f487e1775b2dcdcc9a : SUCCESS in 17m 04s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8109d08ad27a4142a8e2a017d3676f8d : SUCCESS in 17m 46s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f297dd84b4dc483dba6d3d5ab6f4ff1c : SUCCESS in 20m 41s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a2c5fba899046d8a216f5a9602a870a : SUCCESS in 18m 32s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/36cb747c02b64499874ea31eefe20734 : SUCCESS in 17m 39s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/04c8c334a9b44f08bd2d58e84eeea838 : SUCCESS in 17m 33s","accounts_in_message":[],"_revision_number":2},{"id":"334f89432b6379d55a5e8d2b9bf6f0b6b173ac0f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-15 13:05:13.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"14aef4f33b8405eb18848c5cfc7dc7c6c63dad6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 14:53:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/df9007ade74d438e9385db08e796c7f5 : SUCCESS in 8m 54s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/60ec0511abcd40f3bf813951b1fb9c40 : SUCCESS in 10m 43s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/3b77c034ddc64fbe93714923064ed265 : FAILURE in 12m 33s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dae05c44447243e9ad5e4f4010cf2c2a : SUCCESS in 12m 43s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51f882739801452bb7598138c107ece0 : SUCCESS in 14m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9bc6aa2d9754f459b0870102f574747 : SUCCESS in 6m 53s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/d924c019286341b0ae28b0cf0f4357aa : RETRY_LIMIT in 10m 44s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/e699a35726e042e0a97b8bb56fdb9748 : FAILURE in 45m 02s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de8568db8503477492079be7b7779d95 : FAILURE in 32m 55s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/efea7b63e12b473e9e5229bb3753e4e2 : FAILURE in 9m 26s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/73de209f22594894830550fec82a9cf1 : SUCCESS in 15m 12s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/daf2d9ab6095471e8dc51c704c2103f4 : FAILURE in 12m 04s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/53b417fc57a1407a8416c3bdb3d212b1 : FAILURE in 9m 27s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eede2bda22eb499587aafae529768179 : SUCCESS in 19m 31s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2763de6604e04cb89b05724822ea537d : FAILURE in 10m 34s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e97c54e63f9b4ae0a086884076d3be0f : SUCCESS in 18m 51s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ed149e3c60024fbd91ce955b79a56301 : SUCCESS in 16m 06s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8647a9690534d96b1aa3ef94cb35377 : FAILURE in 8m 47s","accounts_in_message":[],"_revision_number":3},{"id":"75901cc272f5ad1a2d8e3f0454dbf48168ce7005","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-19 06:52:39.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"df73a94c9ec6d897c4dd8a3fc253807aa6235956","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 08:19:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1a5ef45c3e374734a87c8088104e303e : SUCCESS in 13m 19s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/1908a384233a4c2b90474e3b8f8974f6 : SUCCESS in 13m 22s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/d0322d4b69f4462d9bb7a5e14f4ed82b : SUCCESS in 16m 28s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f2ba6219989e4f8682b09108edfaa281 : SUCCESS in 15m 54s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57b888d6e19a44dabfae17278e657d6d : SUCCESS in 12m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/872751e3366b4379abe135bc071f0598 : SUCCESS in 7m 07s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/ba649ae648dd4ca08353b38fd76a20d4 : SUCCESS in 1h 14m 49s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/c89a845c9d7842b6b01da1b728c47416 : SUCCESS in 1h 04m 30s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/154c6fac6f634649a837585d5b47ca29 : SUCCESS in 1h 02m 05s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/8ae79f6bb9ab4e59a5705e5a08620b6e : SUCCESS in 23m 56s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b6ce9755a5284f2ba31113259e35d037 : SUCCESS in 17m 36s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47da928ee75345119e2896e88d640b4a : SUCCESS in 17m 28s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0cfcc7736db240adbd732fc520991d2e : SUCCESS in 17m 45s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a108b0a898124fccb2efd63cd42028ae : SUCCESS in 19m 47s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bcb8e41c1d5642399db5fdba61154a8a : SUCCESS in 21m 10s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/722856ff0a99483d92597bd236950142 : SUCCESS in 21m 40s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b51487347c364e78a9d7dc49db2480b8 : SUCCESS in 22m 08s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b056e174ffbf48269098bc31b1806b58 : SUCCESS in 22m 09s","accounts_in_message":[],"_revision_number":3},{"id":"247c88380afc87d90a948c8870ef71ec1a4c8ac6","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-10-19 08:39:29.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"67fffb798802f53530c8c1e79559087827bb77f1","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-19 09:10:26.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4ff93647b6bfe6e48a267d5168ec400609f98f18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 09:10:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e9251b08c1d4f90d8b1c8ce74d994a570ea9b4a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 13:36:33.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/83ff93eae7b54598b677b3e6304bee81 : SUCCESS in 15m 21s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/e39e60f8c93a4e70980efa9f195d727e : SUCCESS in 17m 15s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f844e03f7cb549db8ad7ffdc804c8513 : SUCCESS in 15m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47807a59fa2a4cde9ffdb2961745ebe3 : SUCCESS in 10m 32s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/49092ca340d24ea58d532f8c6feb1632 : SUCCESS in 1h 19m 54s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/b400b62f10a4419fa9d9ae0678b8fd13 : SUCCESS in 1h 11m 46s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bfa333ece80442489f4180249cf9d571 : SUCCESS in 1h 05m 13s","accounts_in_message":[],"_revision_number":3},{"id":"08296616e7fce3c8791de8f3208cf354643631b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 13:36:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"35a1f97a1e74e0acb3bbc23fcf88ba3e12e67efa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 13:38:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/149609744e394bbab138e00624021eb6 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a6476c3f5b6b4707c3a177cd1ed68ef6f804de51","revisions":{"cfb86f74dc1cc19989c74fb2b78762c6dd3941e0":{"kind":"REWORK","_number":1,"created":"2020-10-15 08:25:47.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/64/758364/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/64/758364/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/64/758364/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/64/758364/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/64/758364/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/64/758364/1"}}},"commit":{"parents":[{"commit":"b6685491e9cab5620bf2c761b3bca8c59275b674","subject":"Merge \"Add infra scenario CI jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b6685491e9cab5620bf2c761b3bca8c59275b674"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-15 08:25:36.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-15 08:25:36.000000000","tz":180},"subject":"Increase amount of MaxSessions","message":"Increase amount of MaxSessions\n\nFor running bigger amount of ansible forks, we need to increase\nssh MaxSessions parameter for lxc hosts, since\nall connections to lxc containers occur through hosts\n\nChange-Id: Ib3e850ba79658a42995cd782a11342aca6858342\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/cfb86f74dc1cc19989c74fb2b78762c6dd3941e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/cfb86f74dc1cc19989c74fb2b78762c6dd3941e0"}]},"branch":"refs/heads/master"},"540db10b7e9ec63a1ec8629482e734509140c964":{"kind":"REWORK","_number":2,"created":"2020-10-15 09:31:06.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/64/758364/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/64/758364/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/64/758364/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/64/758364/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/64/758364/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/64/758364/2"}}},"commit":{"parents":[{"commit":"b6685491e9cab5620bf2c761b3bca8c59275b674","subject":"Merge \"Add infra scenario CI jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b6685491e9cab5620bf2c761b3bca8c59275b674"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-15 08:25:36.000000000","tz":180},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-15 09:30:58.000000000","tz":0},"subject":"Increase amount of MaxSessions","message":"Increase amount of MaxSessions\n\nFor running bigger amount of ansible forks, we need to increase\nssh MaxSessions parameter for lxc hosts, since\nall connections to lxc containers occur through hosts\n\nChange-Id: Ib3e850ba79658a42995cd782a11342aca6858342\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/540db10b7e9ec63a1ec8629482e734509140c964"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/540db10b7e9ec63a1ec8629482e734509140c964"}]},"branch":"refs/heads/master"},"a6476c3f5b6b4707c3a177cd1ed68ef6f804de51":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-10-15 13:05:13.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/64/758364/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/64/758364/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/64/758364/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/64/758364/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/64/758364/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/64/758364/3"}}},"commit":{"parents":[{"commit":"b6685491e9cab5620bf2c761b3bca8c59275b674","subject":"Merge \"Add infra scenario CI jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b6685491e9cab5620bf2c761b3bca8c59275b674"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-15 08:25:36.000000000","tz":180},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-15 13:05:11.000000000","tz":0},"subject":"Increase amount of MaxSessions","message":"Increase amount of MaxSessions\n\nFor running bigger amount of ansible forks, we need to increase\nssh MaxSessions parameter for lxc hosts, since\nall connections to lxc containers occur through hosts\n\nDepends-On: https://review.opendev.org/758399\nChange-Id: Ib3e850ba79658a42995cd782a11342aca6858342\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a6476c3f5b6b4707c3a177cd1ed68ef6f804de51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a6476c3f5b6b4707c3a177cd1ed68ef6f804de51"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
