)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~758992","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I049e2105267cf0df213f48c9695f9fa736382f5b","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"update-image","hashtags":[],"change_id":"I049e2105267cf0df213f48c9695f9fa736382f5b","subject":"Update ubuntu base images to 18.04.5 and 20.04.1","status":"ABANDONED","created":"2020-10-21 10:13:14.000000000","updated":"2020-10-28 09:40:02.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"845e79b904d16081a59184ee61a0b02c94cf75d8","_number":758992,"virtual_id_number":758992,"owner":{"_account_id":21785,"name":"Namrata","email":"namrata.sitlani@citynetwork.eu","username":"namrata"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29881,"name":"Jonathan Rosser","email":"jonathan.rosser@gmail.com"},{"date":"2020-10-26 10:01:09.000000000","_account_id":21785,"name":"Namrata","email":"namrata.sitlani@citynetwork.eu","username":"namrata"},{"date":"2020-10-21 20:09:54.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-10-26 11:16:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29881,"name":"Jonathan Rosser","email":"jonathan.rosser@gmail.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21785,"name":"Namrata","email":"namrata.sitlani@citynetwork.eu","username":"namrata"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"date":"2020-10-26 12:22:22.000000000","_account_id":29881,"name":"Jonathan Rosser","email":"jonathan.rosser@gmail.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":21785,"name":"Namrata","email":"namrata.sitlani@citynetwork.eu","username":"namrata"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":29881,"name":"Jonathan Rosser","email":"jonathan.rosser@gmail.com"},{"_account_id":21785,"name":"Namrata","email":"namrata.sitlani@citynetwork.eu","username":"namrata"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21785,"name":"Namrata","email":"namrata.sitlani@citynetwork.eu","username":"namrata"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":29881,"name":"Jonathan Rosser","email":"jonathan.rosser@gmail.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-21 20:09:54.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-10-26 11:16:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-26 12:22:22.000000000","updated_by":{"_account_id":29881,"name":"Jonathan Rosser","email":"jonathan.rosser@gmail.com"},"reviewer":{"_account_id":29881,"name":"Jonathan Rosser","email":"jonathan.rosser@gmail.com"},"state":"REVIEWER"}],"messages":[{"id":"fe518c7cd812949ab4830bd8032f9c9f3f42e971","author":{"_account_id":21785,"name":"Namrata","email":"namrata.sitlani@citynetwork.eu","username":"namrata"},"date":"2020-10-21 10:13:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bd34e50dba3d50cd4784788406d943947559d2db","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-21 10:39:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"61021687db34e9f70c37bc25f0550d355bd96dd0","author":{"_account_id":21785,"name":"Namrata","email":"namrata.sitlani@citynetwork.eu","username":"namrata"},"date":"2020-10-21 10:43:31.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 2e8af003e0f9e1c90dc395bee3b27f275b3cb7ec","accounts_in_message":[],"_revision_number":1},{"id":"79759910697c60967b29c7dab374598dcaaa0cda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 12:10:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/44078b6e56174fd6940d0a18f1455abe : SUCCESS in 9m 24s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/5307f1c6656b4e648269a9fd1bf988ff : SUCCESS in 11m 03s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/739ae003833447ec98d7e092675f68b7 : FAILURE in 9m 50s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/790923f362d449fcb82db256ed72846f : SUCCESS in 12m 36s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f3de6b2181a4895b94858dc85208cb6 : FAILURE in 13m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7e0b6d3d58549cc9befce17e51d1c72 : SUCCESS in 6m 40s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/7a326cde2b324255afb2753ef80ce1cc : SUCCESS in 1h 18m 57s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/4d7ac25749304817ae84eab0c99be85b : SUCCESS in 1h 08m 56s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3bbd12772a0e4123a740ed8ab6f64675 : FAILURE in 19m 48s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/237d45457b1142b59351c681235538e9 : FAILURE in 16m 48s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/863e550a671d4f849dae4368f3fa14ab : FAILURE in 13m 15s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bbe912bc3cfb453ba173cbe704c1735f : FAILURE in 9m 12s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c0c3fd9b2f2e4bfcaca8843916f7c420 : FAILURE in 12m 38s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21f0bf8efc034c279d72e38c08c10500 : FAILURE in 13m 23s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a227521133874c649caddbdd4509a19b : SUCCESS in 16m 51s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed9dfea3cd834d39a32ce91118dd9661 : FAILURE in 12m 11s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/84f2f4a70e3e4d90af388f904c072ad7 : FAILURE in 15m 17s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18fb1047b61c4fb588b775ddccde4af3 : FAILURE in 15m 03s","accounts_in_message":[],"_revision_number":1},{"id":"88fb730d8a320811b9344ba4122308423b11d7ae","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-21 13:03:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6afed81670d1746e05f0280774c76b9d1d45e362","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 15:07:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/005e339898a146d5a6733dffd559f217 : FAILURE in 11m 51s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/0413dc8985fa4305bf84dc3f6b1e7715 : SUCCESS in 12m 37s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/361b8845077541db84b9f4556aae9ab4 : SUCCESS in 17m 17s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ba32813c6e7a4738886d60fbe8a175a9 : SUCCESS in 13m 14s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ff1215fa5f2412caccda8c6777df68b : FAILURE in 11m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f516d5ed07b24d208618c35270b75b18 : SUCCESS in 7m 35s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/dbcf72bed7b64938982d6965d51fbe99 : SUCCESS in 1h 22m 59s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/4a1a915af55d4253b49c1a0c6bef2ea6 : SUCCESS in 1h 08m 11s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/150bab432e944c8db5012b06dfced043 : FAILURE in 19m 33s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/075b74e0476b401e9cdaa921983c7cd3 : SUCCESS in 21m 50s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1380c74c2ab04dffab1271146ceff761 : SUCCESS in 18m 37s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c838a9264d394676b2121312070883da : FAILURE in 16m 09s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8a4c23f4b32644d2b972ff5bd948028f : FAILURE in 12m 37s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a0ce0dabdcd45c891eaa7a9e510d251 : FAILURE in 17m 17s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d53f9e43fc4548b08288299bbd037539 : SUCCESS in 18m 34s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94ab9489aced46d6b374a255ec8e49cb : FAILURE in 14m 34s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7206b0d70dea457db0eb2c530970e58a : SUCCESS in 17m 24s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b37fcf134fc84b4e80b895e3f172e5a6 : FAILURE in 13m 26s","accounts_in_message":[],"_revision_number":1},{"id":"76990afff58020e2cf906c435d1c2782257bbb69","author":{"_account_id":21785,"name":"Namrata","email":"namrata.sitlani@citynetwork.eu","username":"namrata"},"date":"2020-10-21 15:20:25.000000000","message":"Patch Set 1:\n\nRecheck","accounts_in_message":[],"_revision_number":1},{"id":"b2aa8587ce84a6ab6242a7e53c10bb419a895315","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 16:44:07.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/192e855f0e6c451cb1d458f00a030fa1 : SUCCESS in 11m 22s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/94aafa5cf036431294f166fbf3238a83 : SUCCESS in 13m 42s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/5f967b2f9fce442884254df8c91f126e : FAILURE in 9m 38s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9f222024ade6465a886ea030ba47e532 : SUCCESS in 12m 40s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25016dd1b16b490aaaf67a1813f1e0dd : FAILURE in 12m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43a5f81541b74dc2b5f0cb20a668ca80 : SUCCESS in 7m 49s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/5835d848f6d04c15a011ff41d4841b9e : SUCCESS in 1h 11m 26s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/b654bc8ed37546469ab13916cbf76f41 : SUCCESS in 1h 01m 59s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0f648918c80431fa269677cb867e032 : FAILURE in 19m 08s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/f9d17e648d6144baabcff5f28136aeac : SUCCESS in 24m 14s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/52ede975b170408cba16ca6418bacc0b : SUCCESS in 17m 48s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8319a026e29b4adf8d371f8f23f5f02f : FAILURE in 14m 07s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6d8a43d158b84d03ae2e61b1f311e0e3 : SUCCESS in 17m 36s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f4eaa156ffd4e5983d4f6060d95856c : FAILURE in 13m 57s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/36de594b6a2e492a94e9dc5a8d7a059d : SUCCESS in 17m 58s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22dc6bbc00224d18863deaacc0263f7a : FAILURE in 14m 17s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a20a80a436a7415db424093b37a72dea : SUCCESS in 22m 06s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3be1b3fa3d6544d1aa5bb3ada824b7db : FAILURE in 15m 01s","accounts_in_message":[],"_revision_number":1},{"id":"732291b2bee0ead1687167b2ca8fd2b2701225c4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-21 16:50:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0fa2125fbfa2131ecd0a9866ef6e5eb3a5f680ce","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-21 19:33:22.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"2f4bfa81e86ac9e964dad221edf0f36801d71871","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-21 19:35:09.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"3ffc046e6d74c5c513129b191537d13f26b1031a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-21 19:55:38.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"fce1852ccc8234585e4ec314b3fe9dc7efbce2d4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-21 20:09:54.000000000","message":"Patch Set 4:\n\nnot sure, maybe it\u0027s worth switching to https://cloud-images.ubuntu.com/minimal/releases/bionic/release/ ?","accounts_in_message":[],"_revision_number":4},{"id":"52579b145644880690c2a3624c5d638fbbbcbcfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 21:29:18.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c287acd57a6143349f8857bf45fccf47 : SUCCESS in 10m 38s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/8d73793fff6248c9a288bf4675f45450 : SUCCESS in 12m 32s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/a1fbb250be2c4b9aad167aed81147103 : SUCCESS in 16m 26s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b9b9623a16b64a01a344de9b0448de22 : SUCCESS in 13m 15s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1be144fba7fe45609b9f08a79059bae2 : SUCCESS in 14m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19977f836a07440ab65cdd74b46a2982 : SUCCESS in 8m 51s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/b92e0f2692dc479ba7661f5354341ea7 : RETRY_LIMIT in 16m 48s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/98673e0d7e314f4aa1351cefd36a3e15 : SUCCESS in 1h 09m 54s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56491c64793c483aa838a5479c42c3a4 : SUCCESS in 1h 25m 51s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/e6ca4038c2844de093e16984e0f84b21 : SUCCESS in 22m 54s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/87093d9517034299afdbbd5e927c3d5d : SUCCESS in 17m 15s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/71b9a7a588b2434f9e69bfd0f8f824a7 : SUCCESS in 17m 26s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/37999f68dd5a496397e0c69919673301 : SUCCESS in 21m 45s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b5324316f5c4a4bbbaaecf334c0ada4 : FAILURE in 13m 21s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a1253f7abf7c4bf6824111fd5a1ce4b9 : SUCCESS in 21m 43s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5b95fecef73457e87bde97823466e44 : FAILURE in 15m 13s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/515c7bc9762845b994b0392914381de1 : SUCCESS in 18m 54s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a345835be855495facb56cd59f70179f : SUCCESS in 18m 44s","accounts_in_message":[],"_revision_number":4},{"id":"9a5d3a352e622e6b9110e9b2c3144f624a7ef9a8","author":{"_account_id":21785,"name":"Namrata","email":"namrata.sitlani@citynetwork.eu","username":"namrata"},"date":"2020-10-26 10:01:09.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"f041b9d396083f8c38d82ab517f7c86b8a707f2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 11:16:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0ee10fe924704acfa2ceca4769b5039c : SUCCESS in 8m 54s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/bcd913525a094f2095c2c0d97be1b0ca : SUCCESS in 10m 49s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/d6c8114883cb481b929aaa2745213c59 : SUCCESS in 12m 56s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a13952585eb54b57875f011dd7083650 : SUCCESS in 10m 46s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ff6da7267584db980ff88d7b0b503fb : SUCCESS in 12m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4660e983b1984a3d90366aa2828d8e2c : SUCCESS in 4m 56s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/b01170903fe9466dade4a828d09f4943 : SUCCESS in 1h 06m 12s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/c479d4fde24b494690ac170b89ac1321 : SUCCESS in 1h 01m 27s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a05b883fb585453aa02d0a506e4c5f71 : SUCCESS in 54m 29s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/cafdcddae22a4b7b8f42fcee5034c840 : SUCCESS in 19m 52s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ef1962b91faf4594997cc2afa5cfda90 : SUCCESS in 16m 58s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0c814c0688641edb85a842f427e209a : SUCCESS in 22m 22s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ba431a7082af46019ea503ba1d5f5168 : SUCCESS in 17m 15s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d9470fc8880428ba8179d21b84dacad : SUCCESS in 14m 32s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/616822b5fb774d069db6dab6a18f3295 : SUCCESS in 16m 12s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65e777c097aa48efa438bcd04564e995 : SUCCESS in 14m 55s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/941af3a4ce7f46b39fa096c6ea44fa6f : SUCCESS in 23m 12s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b768d64a38048579ca8da90060e01c9 : SUCCESS in 16m 45s","accounts_in_message":[],"_revision_number":4},{"id":"6a9a3144341b66363ca4869f82865b7a59e4c2a4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-28 09:40:02.000000000","message":"Abandoned\n\nin favor of https://review.opendev.org/#/c/759229/","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"0a4c9c10e580738193d60bc14b056d42c75f25ac","revisions":{"bf99c46d5cfd3f0ac05842ebf902b1733f4144bf":{"kind":"REWORK","_number":1,"created":"2020-10-21 10:13:14.000000000","uploader":{"_account_id":21785,"name":"Namrata","email":"namrata.sitlani@citynetwork.eu","username":"namrata"},"ref":"refs/changes/92/758992/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/92/758992/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/92/758992/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/92/758992/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/92/758992/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/92/758992/1"}}},"commit":{"parents":[{"commit":"a6476c3f5b6b4707c3a177cd1ed68ef6f804de51","subject":"Increase amount of MaxSessions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a6476c3f5b6b4707c3a177cd1ed68ef6f804de51"}]}],"author":{"name":"Namrata Sitlani","email":"namrata.sitlani@citynetwork.eu","date":"2020-10-21 10:08:14.000000000","tz":330},"committer":{"name":"Namrata Sitlani","email":"namrata.sitlani@citynetwork.eu","date":"2020-10-21 10:13:07.000000000","tz":330},"subject":"Update ubuntu 18.04 lxc image to 18.04.5","message":"Update ubuntu 18.04 lxc image to 18.04.5\n\nChange-Id: I049e2105267cf0df213f48c9695f9fa736382f5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/bf99c46d5cfd3f0ac05842ebf902b1733f4144bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/bf99c46d5cfd3f0ac05842ebf902b1733f4144bf"}]},"branch":"refs/heads/master"},"0c764f55951777b9b628ee9943fdc6d5e714a628":{"kind":"REWORK","_number":2,"created":"2020-10-21 19:33:22.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/92/758992/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/92/758992/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/92/758992/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/92/758992/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/92/758992/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/92/758992/2"}}},"commit":{"parents":[{"commit":"a6476c3f5b6b4707c3a177cd1ed68ef6f804de51","subject":"Increase amount of MaxSessions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a6476c3f5b6b4707c3a177cd1ed68ef6f804de51"}]}],"author":{"name":"Namrata Sitlani","email":"namrata.sitlani@citynetwork.eu","date":"2020-10-21 10:08:14.000000000","tz":330},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-21 19:33:20.000000000","tz":0},"subject":"Update ubuntu 18.04 lxc image to 18.04.5","message":"Update ubuntu 18.04 lxc image to 18.04.5\n\nChange-Id: I049e2105267cf0df213f48c9695f9fa736382f5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/0c764f55951777b9b628ee9943fdc6d5e714a628"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/0c764f55951777b9b628ee9943fdc6d5e714a628"}]},"branch":"refs/heads/master"},"ae636c3f69c61a6625c29e1feaed79d636122f98":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-10-21 19:35:09.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/92/758992/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/92/758992/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/92/758992/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/92/758992/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/92/758992/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/92/758992/3"}}},"commit":{"parents":[{"commit":"a6476c3f5b6b4707c3a177cd1ed68ef6f804de51","subject":"Increase amount of MaxSessions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a6476c3f5b6b4707c3a177cd1ed68ef6f804de51"}]}],"author":{"name":"Namrata Sitlani","email":"namrata.sitlani@citynetwork.eu","date":"2020-10-21 10:08:14.000000000","tz":330},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-21 19:35:07.000000000","tz":0},"subject":"Update ubuntu base images to 18.04.5 and 20.04.1","message":"Update ubuntu base images to 18.04.5 and 20.04.1\n\nThe originals are no longer available at the previous URL.\n\nChange-Id: I049e2105267cf0df213f48c9695f9fa736382f5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ae636c3f69c61a6625c29e1feaed79d636122f98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ae636c3f69c61a6625c29e1feaed79d636122f98"}]},"branch":"refs/heads/master"},"0a4c9c10e580738193d60bc14b056d42c75f25ac":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-10-21 19:55:38.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/92/758992/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/92/758992/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/92/758992/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/92/758992/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/92/758992/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/92/758992/4"}}},"commit":{"parents":[{"commit":"a6476c3f5b6b4707c3a177cd1ed68ef6f804de51","subject":"Increase amount of MaxSessions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a6476c3f5b6b4707c3a177cd1ed68ef6f804de51"}]}],"author":{"name":"Namrata Sitlani","email":"namrata.sitlani@citynetwork.eu","date":"2020-10-21 10:08:14.000000000","tz":330},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-10-21 19:55:32.000000000","tz":0},"subject":"Update ubuntu base images to 18.04.5 and 20.04.1","message":"Update ubuntu base images to 18.04.5 and 20.04.1\n\nThe originals are no longer available at the previous URL.\n\nCloses-Bug: #1900884\nChange-Id: I049e2105267cf0df213f48c9695f9fa736382f5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/0a4c9c10e580738193d60bc14b056d42c75f25ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/0a4c9c10e580738193d60bc14b056d42c75f25ac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
