)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~758998","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2Ftrain~I049e2105267cf0df213f48c9695f9fa736382f5b","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/train","topic":"update-image-stable/train","hashtags":[],"change_id":"I049e2105267cf0df213f48c9695f9fa736382f5b","subject":"Update ubuntu 18.04 lxc image to 18.04.5","status":"ABANDONED","created":"2020-10-21 10:43:31.000000000","updated":"2020-10-28 09:40:32.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"00c2b4a1f2a7edfc6ebf9b41539398edfacec4a9","_number":758998,"virtual_id_number":758998,"owner":{"_account_id":21785,"name":"Namrata","email":"namrata.sitlani@citynetwork.eu","username":"namrata"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-10-21 10:46:21.000000000","_account_id":21785,"name":"Namrata","email":"namrata.sitlani@citynetwork.eu","username":"namrata"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-10-21 11:20:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21785,"name":"Namrata","email":"namrata.sitlani@citynetwork.eu","username":"namrata"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":21785,"name":"Namrata","email":"namrata.sitlani@citynetwork.eu","username":"namrata"},{"value":-1,"date":"2020-10-21 10:44:32.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":21785,"name":"Namrata","email":"namrata.sitlani@citynetwork.eu","username":"namrata"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21785,"name":"Namrata","email":"namrata.sitlani@citynetwork.eu","username":"namrata"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-21 10:44:32.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-10-21 11:20:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8c2e462dd5da1f02427b30800dd586da4e6c23f1","author":{"_account_id":21785,"name":"Namrata","email":"namrata.sitlani@citynetwork.eu","username":"namrata"},"date":"2020-10-21 10:43:31.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"778220a46623053350657b395853d49b9950b89d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-21 10:44:32.000000000","message":"Patch Set 1: Workflow-1\n\nwe can do cherry-picks only when master is merged","accounts_in_message":[],"_revision_number":1},{"id":"527270ca6089c329cf06be4001b1d44ea2cb6ce7","author":{"_account_id":21785,"name":"Namrata","email":"namrata.sitlani@citynetwork.eu","username":"namrata"},"date":"2020-10-21 10:46:21.000000000","message":"Patch Set 1:\n\n@noonedeadpunk ACK!","accounts_in_message":[],"_revision_number":1},{"id":"1583c6e292f7855f2c6fe409041599909649131d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 11:20:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/886a13e7530e4c6a91c954ec875436d3 : SUCCESS in 9m 28s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/929354c2a6bb402589144a26675c1ac3 : SUCCESS in 11m 28s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/955d2c50ded14a1cb65419963b25264a : SUCCESS in 14m 17s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/793b73eb5c184c57b51770d5460ee737 : SUCCESS in 11m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d39c2644a184873b4bfbf0ed25b4ba3 : SUCCESS in 7m 51s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/7b2607ae0b0c4ae7bc467b80f84a12bd : SUCCESS in 16m 30s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b4cb721915894c7b9f6ec71847ee4447 : SUCCESS in 14m 56s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e47bd47cf4d44e7d958b85b3cf7c7985 : SUCCESS in 22m 58s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d125ac9d71f64a1098fa5de13a79cc79 : SUCCESS in 14m 40s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/85bc03f992614d329b1e5eff90e17212 : SUCCESS in 18m 27s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/40ca253b3790417e95c3ef19e47969ea : FAILURE in 17m 49s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0407ccdc497e43a29923e994b3837313 : SUCCESS in 16m 06s","accounts_in_message":[],"_revision_number":1},{"id":"ff0777384b8593a11cf3f3fb8f62afc2291c67dc","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-28 09:40:32.000000000","message":"Abandoned\n\nin favor of https://review.opendev.org/#/c/759299/","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2e8af003e0f9e1c90dc395bee3b27f275b3cb7ec","revisions":{"2e8af003e0f9e1c90dc395bee3b27f275b3cb7ec":{"kind":"REWORK","_number":1,"created":"2020-10-21 10:43:31.000000000","uploader":{"_account_id":21785,"name":"Namrata","email":"namrata.sitlani@citynetwork.eu","username":"namrata"},"ref":"refs/changes/98/758998/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/98/758998/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/98/758998/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/98/758998/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/98/758998/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/98/758998/1"}}},"commit":{"parents":[{"commit":"daaffb02b31c48f8ac4e2fdd7de41b617b17dc9c","subject":"Merge \"copy the actual keyring\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/daaffb02b31c48f8ac4e2fdd7de41b617b17dc9c"}]}],"author":{"name":"Namrata Sitlani","email":"namrata.sitlani@citynetwork.eu","date":"2020-10-21 10:08:14.000000000","tz":330},"committer":{"name":"Namrata","email":"namrata.sitlani@citynetwork.eu","date":"2020-10-21 10:43:31.000000000","tz":0},"subject":"Update ubuntu 18.04 lxc image to 18.04.5","message":"Update ubuntu 18.04 lxc image to 18.04.5\n\nChange-Id: I049e2105267cf0df213f48c9695f9fa736382f5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2e8af003e0f9e1c90dc395bee3b27f275b3cb7ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2e8af003e0f9e1c90dc395bee3b27f275b3cb7ec"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
