)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~759229","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I362a15397fee891d967ce7150819b02312200d2e","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"bug/1900884","hashtags":[],"change_id":"I362a15397fee891d967ce7150819b02312200d2e","subject":"Determine latest base image available","status":"MERGED","created":"2020-10-22 09:26:56.000000000","updated":"2020-10-27 16:12:48.000000000","submitted":"2020-10-27 14:05:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"759229-1603807512741-2ca14753","meta_rev_id":"f0816f255e38b4369cb96e533e4186bcaa8e9a22","_number":759229,"virtual_id_number":759229,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31792,"name":"Marcus Klein","email":"marcus.klein@open-xchange.com","username":"marcus.klein"},{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-27 14:05:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":0,"_account_id":31792,"name":"Marcus Klein","email":"marcus.klein@open-xchange.com","username":"marcus.klein"},{"value":2,"date":"2020-10-26 08:33:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-26 10:42:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":31792,"name":"Marcus Klein","email":"marcus.klein@open-xchange.com","username":"marcus.klein"},{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"date":"2020-10-27 11:11:31.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-26 10:42:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31792,"name":"Marcus Klein","email":"marcus.klein@open-xchange.com","username":"marcus.klein"},{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":31792,"name":"Marcus Klein","email":"marcus.klein@open-xchange.com","username":"marcus.klein"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-22 15:08:38.000000000","updated_by":{"_account_id":31792,"name":"Marcus Klein","email":"marcus.klein@open-xchange.com","username":"marcus.klein"},"reviewer":{"_account_id":31792,"name":"Marcus Klein","email":"marcus.klein@open-xchange.com","username":"marcus.klein"},"state":"REVIEWER"},{"updated":"2020-10-26 08:33:32.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2020-10-26 10:42:07.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-10-27 14:05:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"83ce939223bea04acf85eb0751fd45c61d5b5087","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-22 09:26:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a6b48f597b4d14756bb99623f6d9e347e52b8e13","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-22 10:27:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"263f98c4fffd5d97b6e4afa8b6b1ccb1d4fe7b2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 10:52:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a0003217b4ef42a59ace55aac6414c0f : SUCCESS in 10m 51s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/f551c05215544aae9db97ae71c6c17d9 : SUCCESS in 12m 06s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/c1073fd83d5e4898b31abd2b48267a2b : FAILURE in 13m 42s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/638e18dfa0c8421eb51dbcb70fe6fe45 : SUCCESS in 12m 48s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/15eab6434b254d219ed3c2f1129e6ac8 : SUCCESS in 13m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/268669f24c5e4ec684a218f05b366bf0 : SUCCESS in 7m 42s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/67fe9ab418914585ac407f0f4d02481c : FAILURE in 22m 33s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/9d911332e7194ed5a8a8e7386e4046ee : SUCCESS in 1h 15m 31s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a30e2148097c49e48225a3bcd48b406b : SUCCESS in 1h 11m 50s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/89d794528b804e58a62e7edda76e90d8 : FAILURE in 18m 20s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f1ca6d9e3df74359b8908ad04b5b3bed : SUCCESS in 16m 53s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91c25142b9e5419bb23569eeb238ea4e : SUCCESS in 17m 29s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0d8f605e77ee44d19c05f01ef98a6b64 : SUCCESS in 18m 53s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f35b904cdf894320b623210bfd237c97 : SUCCESS in 18m 15s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fb019810dd5d45eb95bb311d9e0cc827 : SUCCESS in 17m 28s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2510555fa894df59ba74122f1f56c78 : SUCCESS in 19m 11s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0fd501083b4749898b5481b194686e29 : SUCCESS in 18m 43s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d9f18411dd54971b484a397d146ad82 : SUCCESS in 16m 08s","accounts_in_message":[],"_revision_number":1},{"id":"a57d7ca0d1d233bcb071ef27469a3df7c7428077","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-22 10:55:49.000000000","message":"Patch Set 1:\n\nseems centos 8.2 is different a bit...","accounts_in_message":[],"_revision_number":1},{"id":"fc7b8d33a35c81ea96c1f29110ce4665a2e70ffc","author":{"_account_id":31792,"name":"Marcus Klein","email":"marcus.klein@open-xchange.com","username":"marcus.klein"},"date":"2020-10-22 14:46:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"21310c249432de62ee76b892ef048fa91d6e09b8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-22 15:08:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c714a9b4b1fecb74a0a045e048c2bb933dee596d","author":{"_account_id":31792,"name":"Marcus Klein","email":"marcus.klein@open-xchange.com","username":"marcus.klein"},"date":"2020-10-22 15:08:38.000000000","message":"Patch Set 1: Code-Review+1\n\nworks successfully for me downloading the Ubuntu 18.04.5 base image on a new 21.1.0 deployment.","accounts_in_message":[],"_revision_number":1},{"id":"376e1b9535809ae9568b1ddcf924663d4e25143d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-22 15:30:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"04ccfa3e06b5f8e43e0a77b27559ec3d223c73ba","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-22 15:30:54.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"953cb4f98ed025120c202001a5699dc5618f228b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-22 15:50:16.000000000","message":"Removed Workflow-1 by Dmitriy Rabotyagov (noonedeadpunk) \u003cnoonedeadpunk@ya.ru\u003e\n","accounts_in_message":[],"_revision_number":3},{"id":"c06b81e3730670a478555e920339601f2ee7daed","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-22 15:53:05.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"9986da366c73cf9a0354e5720798c78c1bb759d3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-22 17:12:36.000000000","message":"Topic changed from find_lxc_image to bug/1900884","accounts_in_message":[],"_revision_number":4},{"id":"5eda0dcb66b9ae1ff6a4a90455053c704036de52","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-22 18:04:33.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"2ec2ea21d30016e09362c4eadbb8878269d3718d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 19:17:12.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a7ba1c6f11a14f1b80e467e45f778d1f : SUCCESS in 10m 30s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/8141d2cbe45646cc949a7a3943ad383d : SUCCESS in 12m 09s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/2a076cfac0c8401e8e92d631e16bb3c0 : SUCCESS in 14m 48s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4aff3147492b4427bbb6d548e008d1dd : SUCCESS in 13m 20s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e6aaefa3a1c34ab5b46cf30a6c0a9653 : SUCCESS in 13m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4076c921b12e451d92afc4bfe3ff64ea : SUCCESS in 6m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f765bd42c1e1402b8ec92162f4f874ab : SUCCESS in 4m 57s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/ec65952591a4416ba29d292dbb056511 : TIMED_OUT in 3h 07m 20s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/d135fda0dee1441893f5e78d183ae9c4 : SUCCESS in 1h 04m 27s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b715f99f9944692a905645b92473b28 : SUCCESS in 1h 12m 10s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/09de7da4507f412383efe126f98e8a37 : SUCCESS in 24m 18s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2dc3c1c257d847a087e3a09506e6cc0e : SUCCESS in 21m 16s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/328d298f053646bbb01607055a0f61de : SUCCESS in 19m 11s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/524e1db48dd5438c9b63df975b21ce05 : SUCCESS in 19m 50s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a69b957e20e430c8126f8a6511c85d6 : SUCCESS in 17m 48s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/91443bec681440229adff62a6ff020e7 : SUCCESS in 17m 38s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b30dd3a8ec94c4e8adb298f34bc0ca5 : SUCCESS in 20m 46s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ee4d5127086a4618acbf65a90ca056ec : SUCCESS in 17m 10s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1c44d621ed804ed0acf0b3916b05304e : SUCCESS in 19m 25s","accounts_in_message":[],"_revision_number":4},{"id":"c8f4cb6d54c9f63578c4bd4a95adbbf802c75828","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-22 20:55:09.000000000","message":"Removed Workflow-1 by Dmitriy Rabotyagov (noonedeadpunk) \u003cnoonedeadpunk@ya.ru\u003e\n","accounts_in_message":[],"_revision_number":4},{"id":"6ba3c0090daf98554d06e8e4cc064814339d1d06","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-22 20:55:15.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"5742bb22f6a870740cced5bf6b8258bcd0f50c45","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-22 22:03:13.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"122c5def9346e5760b32e3c65bb58d96f33eff30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 00:19:23.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9370c294fa4d47e1b2f9aba207a72e4e : SUCCESS in 13m 27s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/c7523824f98449ed933b5384fb4a0c6e : SUCCESS in 19m 20s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/c7ef158b292c4e10a536e642033792f2 : SUCCESS in 15m 27s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7ab8229f0b9244daafae6b574ce3e5a3 : SUCCESS in 12m 02s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1dd8b4484e54198a09408f8b54f174d : SUCCESS in 13m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54127252888a49d0b073b4f86c986a7e : SUCCESS in 6m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e32d8132bf8b4bafbe9a14841bea6c08 : SUCCESS in 5m 18s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/45e8852612244fbb9c931eee7d29c222 : SUCCESS in 1h 18m 49s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/7a628cacbeb24849bd96275fcb22330f : SUCCESS in 1h 04m 48s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf9c6c5bfc184a9bb35123dada69e79d : TIMED_OUT in 3h 06m 19s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/fad90cbea1df4e96b1ce4e4ef271d166 : SUCCESS in 23m 16s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6843732ca99b4d9890dc77b65d6b5739 : SUCCESS in 21m 49s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0960c35878ba47d4bd06e75baf026750 : SUCCESS in 18m 25s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ad806dd9ba974d0a84504875efa51667 : FAILURE in 12m 51s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91739608b9134a98b5ec7c44ae3dba21 : SUCCESS in 18m 23s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/31ebb2043a22435e9b4e46d472809b6c : SUCCESS in 16m 23s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/82401c3a2f7645ae88821c0496f36bfb : SUCCESS in 18m 26s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d6b7072553094acea5e2035a065d8a67 : SUCCESS in 18m 14s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52a25fc8c12f442188e51cf3f0fb72f6 : SUCCESS in 17m 54s","accounts_in_message":[],"_revision_number":4},{"id":"409b731ce38b05722db7631a6b3b6639dd36e135","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-23 06:11:55.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"e09d8c5e44cb62dbc8080aa2b489a5d605abf85c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 09:51:22.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ffddccd540d4497d980810fa898026f1 : SUCCESS in 10m 44s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/30040eb1a2874f6e8ecebd7e856c22ac : SUCCESS in 11m 51s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/45029c5d2ce94ebc9ccff0dd0716a34a : SUCCESS in 15m 28s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/89a5a6a9dec54b60a976b7c067b8579b : SUCCESS in 13m 44s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0697c503d5c4fd0a7167f30bbb0917d : SUCCESS in 11m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75c32bb598c444a5bec0a2aa339e98ea : SUCCESS in 8m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23ae3e83cce640689d969e67213f8a72 : RETRY_LIMIT in 3m 10s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/7c2a078f85f043559dd4a20c90ab3e5e : TIMED_OUT in 3h 09m 07s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/52f97cfac0ba4c448d49b8e719e56f8e : SUCCESS in 1h 07m 59s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70c7785dfa8b4a3aad2395e0a50f6473 : SUCCESS in 59m 28s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/ac4c5f2f503b414485a3b63347f56776 : SUCCESS in 23m 20s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/afa84d91125249dc93a54a3f90a24781 : SUCCESS in 17m 05s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/79e61561954b41869ae1a0ee62fd1258 : SUCCESS in 17m 56s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/496af20b64fb41aca32bbb19e8b49284 : TIMED_OUT in 1h 32m 50s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5b5a5ab82234a2abf1f03af3018cad0 : SUCCESS in 16m 46s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/37e6bbb9bce14db687cda3fbef033e83 : SUCCESS in 18m 37s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91410147ec2c42b2a4b9bc0aaedd07eb : SUCCESS in 18m 44s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aa7a10d5bf6f4ab38615a4a0d470ce2e : SUCCESS in 19m 56s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03a7ca6d79b843b8a40c2cc432d70c89 : SUCCESS in 18m 32s","accounts_in_message":[],"_revision_number":4},{"id":"0404b0f90f5df92ce5c9de92035e5b1ec01b4bb6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-23 10:34:10.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"ebdb3562409057f9b223297d9a6cf4bb36f093ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 12:27:07.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5cc731d6278c45eabdd4c7e79235d033 : FAILURE in 5m 06s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/0a3c2f7f220946e1944064136c14c3ea : FAILURE in 4m 45s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/54d0cd73b0ff4677aa363c0fbef8b43a : SUCCESS in 15m 38s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/39d9c86d792142188514aa116b27409b : FAILURE in 4m 51s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/74de10db3c4b4a949ee3cf26074b5575 : FAILURE in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/005d1658abd843a5b259bc1ced4a4830 : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6685f2e2373f4e9c80b1f82a765cf18d : SUCCESS in 4m 52s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/3c54503576874178a929221e10b6ddda : SUCCESS in 1h 28m 14s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/8a19d8b316c84dd1aafa877eb017b44c : SUCCESS in 57m 31s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f10c84f95ad4f35ab4b70223cfa0d73 : RETRY_LIMIT in 6m 42s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/c43657ae795241d89aedcde847636f45 : SUCCESS in 24m 23s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d6400b7c75e3465db23bfa1c4a1bb3fc : SUCCESS in 17m 54s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da73a94ab5b5443597027c6ef25a91ec : FAILURE in 5m 08s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ad8f071e5f8b4dd38b739cfbe59ab462 : SUCCESS in 20m 27s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10d1b76c486549fdbdfd2897cd31749d : SUCCESS in 18m 48s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/23109d60c49041d38500ba3e611ddb42 : FAILURE in 5m 02s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7cb0b9755a044017beb7490aa5c121d1 : FAILURE in 5m 02s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4bbe2dd63d1148a1a6b080fb76b91700 : FAILURE in 5m 19s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7f2edbf4a994128bf1a373f669868c7 : FAILURE in 5m 14s","accounts_in_message":[],"_revision_number":4},{"id":"7dc7714254fe373ffaa046336189ca0f96d9d929","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-23 12:30:58.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"c6772c8fd5ab39e2bb530af6b35904386877abd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 19:00:28.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/68e369d68fd04fc4acdc4eb4e51dca41 : SUCCESS in 10m 52s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/d6c2dd4a40f942f3858974d6b8cd7351 : SUCCESS in 12m 19s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/cf02df8bb0de4a6189d5134bc23dbfc5 : SUCCESS in 15m 32s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/62925b8e53b94d53879e50d448a45341 : SUCCESS in 11m 49s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22e779886d0a4b87998ca1d5c20f072d : SUCCESS in 14m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d9bd4715b9847659685c928bc696e22 : SUCCESS in 7m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/16edec48c02542439ee5ea1b517702d3 : SUCCESS in 4m 35s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/7d96bd3672f9449dbb712035fea63869 : TIMED_OUT in 3h 08m 24s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/63d0740b41a342b0977b5f5fd62b45e5 : SUCCESS in 1h 08m 37s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02dc125a41e94565bdb2e30716e450bd : TIMED_OUT in 3h 08m 58s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/1b1fa8cd687743df95dad60a823df7c7 : SUCCESS in 24m 20s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d15ca19abe1d4190817ccffb84cb1e39 : TIMED_OUT in 1h 32m 59s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b12d93399eb647fe863620e66f555b63 : SUCCESS in 17m 19s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0d5b43fd71e94053af27803c72b36021 : SUCCESS in 18m 04s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b00ac9580c44baab8b8c074abb74293 : SUCCESS in 17m 58s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f50a233c1a59453dbb2e65a36ad211f8 : SUCCESS in 18m 52s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e857cbcbd1f4e04adc23173fc7088e8 : SUCCESS in 18m 18s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/33f3fb59a3834ef881fd103de9cc47e8 : SUCCESS in 18m 40s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11022655ad674384b24b5127fe4aed0c : SUCCESS in 19m 03s","accounts_in_message":[],"_revision_number":4},{"id":"3c49850a814f46a8804b78bd47391ae505debb5f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-23 20:29:18.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"3973b31272822f0747d49f8b68eefbb23f939d69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 22:37:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/713757f7f66c47468d88ef431c7befc1 : SUCCESS in 10m 57s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/63086ab7fab848e788f3c0f370118055 : SUCCESS in 32m 03s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/30c08809fa9a42e1be45ca1f6d0b6bfe : SUCCESS in 13m 50s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ab896d1c58a1433aace7d2184ca7b859 : SUCCESS in 16m 09s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd876057e19c489e98a562c3da623802 : SUCCESS in 16m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11e9348d59a048aca7741ebfc49bb8d5 : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04754ad525954cfab3c69843af6171ab : SUCCESS in 4m 57s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/51c09e3540b542b9b53b2b82e191ddb2 : SUCCESS in 1h 11m 39s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/eabac24cedc04c84b4d9baed65c142b6 : SUCCESS in 59m 46s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7747cbdab9f849e99b17c57ba90db8d5 : SUCCESS in 1h 59m 14s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/45f9fdfd3d9b4efda333d4a21db2366a : SUCCESS in 56m 55s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4d1d600e3fb945629cac2996d9fd2c2c : SUCCESS in 46m 25s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e299b78aa3b4361be3e570622eda1de : SUCCESS in 19m 04s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/79eeab0221084af182ca45c378f9ce19 : SUCCESS in 47m 16s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ad41d41cbb542ab8002c4393fe04780 : SUCCESS in 34m 44s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dc491737c8664d4e9317aaf734337c85 : SUCCESS in 46m 45s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ee3b879615f46e78fae8655f839872b : SUCCESS in 17m 49s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/20751561f9714f24a6511157cb336e98 : SUCCESS in 16m 35s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c246b11c8cc245cea37c8cbcc6b7ee67 : SUCCESS in 35m 18s","accounts_in_message":[],"_revision_number":4},{"id":"adc3b7f4bd73f63e238feb8d66b0d94d3416715d","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-10-26 08:33:32.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"c6835c77cfc11ad6aa7c66254676eb6ea3f31086","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-26 10:42:07.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"01a6ca405d63986e804f8cba39ec26a9017c16af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 10:42:18.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"fabde0f4344963a57bbe8af9261879ae24022986","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 11:45:12.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a57cc313891748e0b61b9377648c8d2d : SUCCESS in 7m 11s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/50d25d2d35bb478f89ddf5198babc9b6 : SUCCESS in 9m 05s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/38248e89da7e445f8adbda3c044da064 : SUCCESS in 9m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f172c3090ef4be7996f7f45c7b877e3 : SUCCESS in 4m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf7832bb32724a60949cd8ef2b53e537 : SUCCESS in 3m 12s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/cbb2baa1d0dd43ce8c0b80f5b1b28aed : FAILURE in 58m 07s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/5b8251db63284815bca68dd145e66ad7 : SUCCESS in 58m 19s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ee8b025c9c34f368b6d5535a73bd0e5 : SUCCESS in 57m 40s","accounts_in_message":[],"_revision_number":4},{"id":"c8546cad563e84869e4959315f943b169ec3749f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-26 12:36:26.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"7ba1270e14ddff4c420812712568e61b4f449ad3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 14:34:36.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2fd4052926e1497987656be4d7a679d8 : SUCCESS in 8m 09s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/a479e8038edc42faa52e44a91eff9f32 : SUCCESS in 14m 42s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/b8b6693cf0d04051a55233ae0c1fc7e0 : SUCCESS in 12m 30s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d6309656ec8c4242b05ad88758246400 : SUCCESS in 10m 27s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9fb61af68f7b4175a7fb9e7341befcf6 : SUCCESS in 16m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac80313469a04bb7b3098e251aef86bb : SUCCESS in 4m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b2cb3946bf84d018d795e67db4e3f5c : SUCCESS in 12m 43s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/49eabaa0a3614108afc92383901e82a1 : SUCCESS in 1h 10m 45s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/669e00b5c32a4197a1fba52e2f3b9630 : SUCCESS in 1h 50m 14s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/907beafe96ff45578790ecf0447f3567 : SUCCESS in 54m 17s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/933e56d90f6e4360870926aba765c6f8 : SUCCESS in 20m 32s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/85808be407b34910873118cbf9c11993 : SUCCESS in 16m 10s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/26f3350aaa0347f88aba2d46bcb3154c : SUCCESS in 15m 27s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4ab64ee60b3d483e98a77d16fe38367d : SUCCESS in 47m 08s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4019ab0983744345aed620b99f3d7769 : SUCCESS in 17m 12s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f8419089d8de42a9855fc94e064123cb : SUCCESS in 17m 10s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e2e57cd8c1046249a76e89f29ff6663 : FAILURE in 22m 01s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dd2a71cb4693487bb0cfc0ea2da35b71 : SUCCESS in 18m 11s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c59f0f3b6f674d1187f5e7caf43634ad : SUCCESS in 15m 18s","accounts_in_message":[],"_revision_number":4},{"id":"8b1966c46ca5c5b33e3afdcc4840044028c4b12b","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-10-26 14:38:31.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"557afc84bec8ccb87b51d450ef949372282fec96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 15:56:41.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f78b90b8814a4230a842930c1d0931bd : SUCCESS in 25m 03s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/554a9c1f3bb142ccb4127406d64fe6b9 : SUCCESS in 9m 24s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/561f7feceb2e431199d7db3e7ab89f92 : SUCCESS in 13m 36s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5faec6649ceb43c99388b98c12d812f5 : SUCCESS in 10m 25s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad5b80e6bcda4700bcfe59faf57e8275 : SUCCESS in 10m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc09f638775e48b1849d9f78871bddf6 : SUCCESS in 11m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9135123010b940e9bf57353137c31c8a : SUCCESS in 3m 27s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/316a194cdc374c7bb53899865076208f : SUCCESS in 1h 10m 10s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/a8605206f9274b1b931b242856272bb2 : SUCCESS in 1h 07m 08s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f60e19edeaf49419faab51847ac0683 : SUCCESS in 58m 12s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/6d531262f6444c568536b6cc59d947b5 : SUCCESS in 20m 42s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/123fdba08d1344e0b0da80b58347cb09 : SUCCESS in 15m 01s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d30aac3784ec4d32904264509560dea7 : SUCCESS in 16m 17s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/29d5101759914bf98df3ed3cabc04f62 : SUCCESS in 35m 42s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84857366da7f4ae4bdb608327c007d76 : SUCCESS in 19m 53s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ea95c6c628194ecf8a0ae1779b6fefd6 : SUCCESS in 15m 50s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e89444f7654249d2b1447e4cbf32f1fd : FAILURE in 6m 19s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/898c9205b7d54333a14559071f7be019 : SUCCESS in 16m 28s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d17d352e10344018c0291ec52cfae34 : SUCCESS in 20m 43s","accounts_in_message":[],"_revision_number":4},{"id":"b1d08e3c4a5db106db66b5468eddd4834989c95b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-26 16:10:54.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"d15c9de6542f84432200564a8127217864a3fb7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 18:37:17.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/54367e7823354cc69cb735a7af0b9ab8 : SUCCESS in 8m 27s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/57efdfaccd504de9bf3354d114541019 : SUCCESS in 11m 36s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/9b445a91961947359410be619ca7f3f7 : SUCCESS in 15m 04s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bbcc6058d1784c0cb3f7ca0dfa6c625f : SUCCESS in 11m 43s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e38d4c67cceb4aae9a498938d6a885cd : SUCCESS in 11m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b7293dc846941bb9e0eac01994cd848 : SUCCESS in 4m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d6dc06228c145ecae01c0decd1e6b46 : SUCCESS in 3m 44s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/433bc9e805e945378c8782e4ac79de97 : SUCCESS in 1h 10m 47s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/39295cee0bee4b3e9a26595252e638b3 : SUCCESS in 49m 28s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc63339611da4bc987db06d910298978 : SUCCESS in 2h 15m 03s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/f565b2c2ed1f46b99bab6822ef55a908 : SUCCESS in 23m 29s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/78e2244d1eb346c78168e25b50f20fb4 : SUCCESS in 18m 39s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd140283cd4a4e828a8c7e27d64fe299 : SUCCESS in 18m 47s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/55c6747260494a2290089a4e443c3772 : SUCCESS in 15m 55s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/471f7f2c7e4f4bc58b48c7edb0535c98 : SUCCESS in 17m 28s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8e7c05609c264e57878580a2e2a368b3 : SUCCESS in 44m 02s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/436dff1b35aa4b05a792acb197512515 : FAILURE in 28m 52s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9a0e424e9bb54ee19d12d13da6825c35 : SUCCESS in 18m 00s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9fab9dcbb34943ffbda4a81a0645515e : SUCCESS in 18m 48s","accounts_in_message":[],"_revision_number":4},{"id":"0634aac356f2b7c725691b293f460c44dacc8237","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-26 18:58:11.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"e199b4a87c0b44c09572f19fb9f7110fa9cdbc68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 21:57:17.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1836a528316842bc8a18ce9d39574c40 : SUCCESS in 27m 01s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/78c13e55a99f4aef9262db393ae770b6 : SUCCESS in 31m 42s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/ee28e5777d164952a1a35d6a1410c99a : SUCCESS in 14m 08s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5d1421076c8f435abb440af396750035 : SUCCESS in 11m 56s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31f50ee8562040afb363d42dde403695 : SUCCESS in 25m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26e60011caf64273a6d089f4ba97dfd9 : SUCCESS in 4m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42db7aacfa6c4cf79af1b9addbd9ef7c : SUCCESS in 3m 33s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/c2c9e0e8a8744350bda5317722db0c08 : SUCCESS in 2h 42m 10s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/ea3427c6ba9d474fbedb0108c4fa05cf : SUCCESS in 1h 01m 50s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fcff4500ed064414a3bede6b482e210e : SUCCESS in 2h 42m 14s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/01da8950c903468dac61824f4cffd83c : FAILURE in 9m 08s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/74b7565e6f1943318c3f2379bc0065d1 : SUCCESS in 17m 16s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bce38f45b5b843cb9ea74eee5dfd2df4 : SUCCESS in 15m 11s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/878fc51de58a48c0a654ac5aa1010089 : SUCCESS in 16m 21s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8395b8b66bd41e5983001adc50a3df5 : SUCCESS in 16m 23s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ead0dc87fdce4f9f840441a70d031cdf : SUCCESS in 18m 04s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51b3658e4e7546ba959e4ca86c9c74ae : FAILURE in 6m 00s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/765f8b7f2b98446d9b7938ebb1eb3513 : SUCCESS in 16m 24s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8cb308155b134948afbd7f60ab1976c1 : SUCCESS in 15m 50s","accounts_in_message":[],"_revision_number":4},{"id":"5196638ef15c9a325f7442eb2558f6a4db05447d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-27 11:11:31.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"2895551c99a715fe71f9f72c4f95d0e376b9a3a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 12:46:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/287369385561489d93932b522509621a : SUCCESS in 8m 18s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/2f5b65a325bc4d91b2fa797185df602a : SUCCESS in 10m 16s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/ae4c8df294ac4b94802f15668697c47a : SUCCESS in 11m 48s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/913931c0f9794066a2d90b65cb14deb5 : SUCCESS in 11m 43s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dffaf5b751b34daab6b7dcf3715d2725 : SUCCESS in 11m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/add9fffd892e4239abf628e2ff422ec5 : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1ebc1b8b47b49c9ab90c2dc826c8319 : SUCCESS in 4m 10s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/da27ca7f04ee4f4493a0459ad55c69f8 : SUCCESS in 1h 16m 33s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/63afcf18ab5349b3a64a0041bd77c30d : SUCCESS in 1h 07m 42s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d6c7b0bae0144458c9d14c6a7e93487 : SUCCESS in 1h 11m 07s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/4cc259964a584601a319dcf97e453244 : SUCCESS in 22m 58s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a14d96929ff24aefa5848eb01b2eff1a : SUCCESS in 15m 18s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77e97f0977df419cbe701f05a88624f7 : SUCCESS in 15m 41s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f27eab67c5e2428786a41567a296a0c8 : SUCCESS in 15m 19s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb643311294142a295a3352278632578 : SUCCESS in 18m 15s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2d2ee2d9e1074236bc51cf66ff1f64d8 : SUCCESS in 17m 02s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/876f65a7015641328fed7ed4b56aa1b7 : SUCCESS in 16m 50s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/92394647bde3493386ea674c8dff62a4 : SUCCESS in 15m 16s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5c2380a05d646b7b9f75cd396607b04 : SUCCESS in 16m 33s","accounts_in_message":[],"_revision_number":4},{"id":"1b07464caca6e5569344e796cb0e092ee7ee908a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 12:47:07.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"04d2d686c49502f0bab8250d97eeae0aad5af0bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 14:05:12.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f542a491ccf84044ab55abd4fa7df0ea : SUCCESS in 9m 03s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/a99216cb19d84e65969fd4f7e69c5af7 : SUCCESS in 10m 40s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/90c80f3b4eea4cd5b638e644f062c202 : SUCCESS in 13m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edc20b035b684106a93636539182715c : SUCCESS in 7m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/095e5009444045f694483ef545ef4972 : SUCCESS in 3m 48s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/ae678ff87cd44fc1a95379237668778d : SUCCESS in 1h 09m 39s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/106c6b746ce8481ca408cdea35315995 : SUCCESS in 59m 58s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/789d08a4e5b24112994699b80c4fcb5f : SUCCESS in 1h 00m 28s","accounts_in_message":[],"_revision_number":4},{"id":"706fa9b043fdb859e0ec82c24f64ebbc983672cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 14:05:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"c614a857abde28321d26eea9865fc40f9f85540b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 14:06:49.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be9f097befff455ea10265ec0a200c42 : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e083f273ea6a49ff92d820364ffa2fd6 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":4},{"id":"f0816f255e38b4369cb96e533e4186bcaa8e9a22","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-27 16:12:48.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit bced619ebd01b533554a156b0f5cd97886433698","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"46e10d21f15fa7ec50c35e800125278e270f46b6","revisions":{"b069a8fb87b38f7eb33c333edbd868c1e1ed055c":{"kind":"REWORK","_number":1,"created":"2020-10-22 09:26:56.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/29/759229/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/29/759229/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/759229/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/759229/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/759229/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/759229/1"}}},"commit":{"parents":[{"commit":"a6476c3f5b6b4707c3a177cd1ed68ef6f804de51","subject":"Increase amount of MaxSessions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a6476c3f5b6b4707c3a177cd1ed68ef6f804de51"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-22 09:26:44.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-22 09:26:44.000000000","tz":180},"subject":"Determine latest base image available","message":"Determine latest base image available\n\nChange-Id: I362a15397fee891d967ce7150819b02312200d2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b069a8fb87b38f7eb33c333edbd868c1e1ed055c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b069a8fb87b38f7eb33c333edbd868c1e1ed055c"}]},"branch":"refs/heads/master"},"0a07c2a8bdef8b57b9beef4b15f2693a70408ebc":{"kind":"REWORK","_number":2,"created":"2020-10-22 15:08:22.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/29/759229/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/29/759229/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/759229/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/759229/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/759229/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/759229/2"}}},"commit":{"parents":[{"commit":"a6476c3f5b6b4707c3a177cd1ed68ef6f804de51","subject":"Increase amount of MaxSessions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a6476c3f5b6b4707c3a177cd1ed68ef6f804de51"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-22 09:26:44.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-22 15:08:04.000000000","tz":180},"subject":"Determine latest base image available","message":"Determine latest base image available\n\nChange-Id: I362a15397fee891d967ce7150819b02312200d2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/0a07c2a8bdef8b57b9beef4b15f2693a70408ebc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/0a07c2a8bdef8b57b9beef4b15f2693a70408ebc"}]},"branch":"refs/heads/master"},"5f18ba0d83cabea77a7861c9254b2aefbf4e140f":{"kind":"REWORK","_number":3,"created":"2020-10-22 15:30:35.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/29/759229/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/29/759229/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/759229/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/759229/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/759229/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/759229/3"}}},"commit":{"parents":[{"commit":"a6476c3f5b6b4707c3a177cd1ed68ef6f804de51","subject":"Increase amount of MaxSessions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a6476c3f5b6b4707c3a177cd1ed68ef6f804de51"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-22 09:26:44.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-22 15:30:18.000000000","tz":180},"subject":"Determine latest base image available","message":"Determine latest base image available\n\nChange-Id: I362a15397fee891d967ce7150819b02312200d2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/5f18ba0d83cabea77a7861c9254b2aefbf4e140f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/5f18ba0d83cabea77a7861c9254b2aefbf4e140f"}]},"branch":"refs/heads/master"},"46e10d21f15fa7ec50c35e800125278e270f46b6":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-10-22 15:53:05.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/29/759229/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/29/759229/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/759229/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/759229/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/759229/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/759229/4"}}},"commit":{"parents":[{"commit":"a6476c3f5b6b4707c3a177cd1ed68ef6f804de51","subject":"Increase amount of MaxSessions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a6476c3f5b6b4707c3a177cd1ed68ef6f804de51"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-22 09:26:44.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-10-22 15:53:01.000000000","tz":0},"subject":"Determine latest base image available","message":"Determine latest base image available\n\nCloses-Bug: #1900884\nCloses-Bug: #1900975\nChange-Id: I362a15397fee891d967ce7150819b02312200d2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/46e10d21f15fa7ec50c35e800125278e270f46b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/46e10d21f15fa7ec50c35e800125278e270f46b6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
