)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~759299","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2Ftrain~I616c7ea903114b084408b52dce28696b13dcbe5d","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/train","topic":"bug/1900884","hashtags":[],"change_id":"I616c7ea903114b084408b52dce28696b13dcbe5d","subject":"Determine latest base image available","status":"MERGED","created":"2020-10-22 16:50:52.000000000","updated":"2020-10-27 17:08:25.000000000","submitted":"2020-10-27 16:59:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"759299-1603817972135-1c0c9753","meta_rev_id":"6836bb943dbe05958cbeee46c50556731b963b3a","_number":759299,"virtual_id_number":759299,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-27 16:59:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2020-10-27 14:25:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-27 16:16:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"date":"2020-10-22 18:04:52.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-27 16:16:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-27 14:25:32.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2020-10-27 16:16:19.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-10-27 16:59:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8998efda5c61079a8148dfdf6844a50404b90cb0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-22 16:50:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"530aad54b76fb7f7d725ec5f083d44c7aa1516e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 17:21:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/548bdd906e0a4b089c0bb687bf7a1c80 : SUCCESS in 8m 47s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/9ed670de96ea4d1ab831021503d6dcdd : SUCCESS in 11m 53s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/a66c7df56d48469295cd727aaa0c7825 : SUCCESS in 13m 11s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/97b0bf9761f84659a08c9a36b9fc171c : SUCCESS in 9m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8596ac8ad76e45e4aa77cc0ebe7a0a82 : SUCCESS in 8m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e982bffa688c458aacae12294e3a0ff0 : SUCCESS in 4m 30s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/ea2793cfae554fa587386e9a2f4276ef : SUCCESS in 16m 32s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eb6726f445ed4b39b91f3a041d11dd42 : SUCCESS in 12m 54s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c3a5dcfaa3f14295871b6cc5b94bfcce : SUCCESS in 17m 36s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/449782cc14d24f55bc31a30a795122bf : SUCCESS in 16m 32s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/fd59d8ff9af744e79a63f070255a3bea : SUCCESS in 19m 40s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff7a13bac5cc4bb7b7cdc396d45da40d : SUCCESS in 17m 39s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bab2d40f1604490e930303604ae9175f : SUCCESS in 13m 11s","accounts_in_message":[],"_revision_number":1},{"id":"4e554a0d9865bf31d024f0cc7ad3c4f9355decc3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-22 18:04:52.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"1e1269d89d338336b5b2b79262939a79aac5737d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-26 10:44:32.000000000","message":"Removed Workflow-1 by Dmitriy Rabotyagov (noonedeadpunk) \u003cnoonedeadpunk@ya.ru\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"cfc5ff7b5ceae2bffd95251ae08e262e6f5b6ef5","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-10-27 14:25:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"259f0f7b54bbfe7919ad29a735688fd0f35971fc","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-27 16:16:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"064b90b617f0b05896d2245f4d36395519398704","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 16:16:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"32ceda5cbbcdd4db7b4c3b6ae84564c7aaa992e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 16:59:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6fe4e8ad63964b21ac3fe9a434a581f5 : SUCCESS in 16m 52s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/82c6f786d00e4277b5f9c226dd82f76a : SUCCESS in 20m 51s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/22a2d8acbf3048378eea0ede563f7f47 : SUCCESS in 15m 07s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bf34d0ed2d924827bfdf6df22865408e : SUCCESS in 10m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0e6d3d86bc8460db03b5139868c8aec : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/068d6a2263fc43939730d6266f1b4130 : SUCCESS in 8m 40s","accounts_in_message":[],"_revision_number":1},{"id":"d1c73c0489fcf3287dd329d2aded53da448bdd9e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 16:59:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"039d04795a9a44124eb13975366da972c1299d3e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 17:08:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b91bbcff17a454c90ee609c4edc8cbd : SUCCESS in 1m 17s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c26962b8e512412580d528005d5b6d17 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"88d196453b52ec5b211c236f11e7be1b18aa1d92","revisions":{"88d196453b52ec5b211c236f11e7be1b18aa1d92":{"kind":"REWORK","_number":1,"created":"2020-10-22 16:50:52.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/99/759299/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/99/759299/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/99/759299/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/99/759299/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/99/759299/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/99/759299/1"}}},"commit":{"parents":[{"commit":"daaffb02b31c48f8ac4e2fdd7de41b617b17dc9c","subject":"Merge \"copy the actual keyring\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/daaffb02b31c48f8ac4e2fdd7de41b617b17dc9c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-22 16:50:40.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-22 16:50:40.000000000","tz":180},"subject":"Determine latest base image available","message":"Determine latest base image available\n\nChange-Id: I616c7ea903114b084408b52dce28696b13dcbe5d\nCloses-Bug: #1900884\nCloses-Bug: #1900975\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/88d196453b52ec5b211c236f11e7be1b18aa1d92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/88d196453b52ec5b211c236f11e7be1b18aa1d92"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
