)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~759310","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~Ib926e930ba105eeb49b7cf0031c77c7dc4c97b62","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"bug/1900884","hashtags":[],"change_id":"Ib926e930ba105eeb49b7cf0031c77c7dc4c97b62","subject":"Use lxc_image_cache_server_mirrors as image source","status":"ABANDONED","created":"2020-10-22 18:02:25.000000000","updated":"2020-10-27 19:12:07.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2501e1e822b40e3f73420221a9217f6d8570e88d","_number":759310,"virtual_id_number":759310,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-10-23 22:05:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":-1,"date":"2020-10-23 17:44:30.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-23 22:05:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6bea5e0091c184178300524d443ee53b62b8c370","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-22 18:02:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fbd7354adc7e9a26f35d1cc82e465ce87b4e8a54","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-22 18:13:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"394569f8fd58e7b9c1de429283fbbbdf42bdc8f7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-22 18:35:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"717cf12a3937e6f10e97482a2fe6690333f79610","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-22 19:17:07.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0e00511488fcb007bd3cd8fb0e306c09f6821e6b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-22 19:40:49.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"bbfec8cc0fe0573e3a9611437bfb5317d1f1c13e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-22 20:05:16.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"7a025c0a24279a58d88e342ead9863076a5c0df0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-22 21:57:19.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"a0c60be5804df833e00688294de7b4e8fb08cf20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 23:15:41.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/863382c26a554ec3af58630c45dc15a2 : SUCCESS in 10m 50s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/1d3562de448449a88cb57cb48097c9c9 : SUCCESS in 11m 59s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/e99befd98739420daf16f7f613f89068 : FAILURE in 11m 27s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/71198c98caf64e6a91057f374ec5ddf6 : SUCCESS in 12m 09s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f226ea810969472a95a81990aafbf2a4 : SUCCESS in 13m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/468c6b9e080b4a0da0f70755597db5d6 : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0baae063c422482c9e886925ee3449c9 : SUCCESS in 5m 36s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/63851134df4d4d75aa050f38e9ad1b6a : SUCCESS in 2h 21m 10s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/75b6979e3c2a4b1599734f37dba8b21f : FAILURE in 19m 47s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be4414a435d24530928462c42a130211 : SUCCESS in 1h 05m 59s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/e28aa426987146509aed06a47f326477 : SUCCESS in 34m 33s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/496f1e2a6c924583b7d34f4f04e66e89 : SUCCESS in 19m 33s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ad83511694a45199f47c970c408bace : SUCCESS in 19m 43s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d9c2fc1e73b34af385f0145d569d8582 : FAILURE in 10m 06s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ee56e7d092e46cd9d4057c6efdb7481 : SUCCESS in 19m 31s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f408acc04cb94a42b628ca730c769c26 : SUCCESS in 18m 43s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e63232f4fc94783b3c7e216c3a37121 : FAILURE in 9m 52s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/696c15e7300c4181b98b4ebefdc04ae5 : SUCCESS in 20m 42s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6d436bd187b4d76bbc65eceef32d3b7 : SUCCESS in 20m 16s","accounts_in_message":[],"_revision_number":6},{"id":"c31346047a5d92ce2f3636b7e1b76550f8c6fe7d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-22 23:16:15.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"61006ebc771ac1eae93d098b351aa3bf6a0e407f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 02:24:12.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/de2e8ccf04944664bc66d053e439eb81 : SUCCESS in 11m 01s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/9ed11176405f41468445774307bc1d92 : SUCCESS in 14m 38s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/4675f2e83b0e411285c7ecbab72e91c9 : SUCCESS in 16m 18s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8e13d0097f494ce091bcf831e4ae2fb3 : SUCCESS in 12m 30s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3f1bf1f36c745f9a9920cb7c40585fa : SUCCESS in 13m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/132742d4e7644a39bfbf44f88c5c712f : SUCCESS in 6m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/334751d7ccc64f5ebd13e4652a130c1e : SUCCESS in 5m 22s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/d07400384d434f6296ab123279ad5934 : SUCCESS in 2h 27m 12s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/cc2c89c34c82446aae77d4d54a9b7312 : FAILURE in 23m 16s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/856a868fc5474304981030a0132f0c3e : SUCCESS in 1h 02m 56s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/cbe5c959b73c4350aa9ddde991cd12f9 : SUCCESS in 36m 28s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/91c4c20af1ee4541998187f25ae1d0f0 : SUCCESS in 19m 50s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c5333db6209d4da6ab6117a1f505fdaf : SUCCESS in 18m 01s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e779a73634054ccd8f1e514d7b7b72d5 : SUCCESS in 19m 46s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0cf9a46228a406b9c33fe3e324af1ea : SUCCESS in 18m 29s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d450b420667642c1a96411536be35aba : SUCCESS in 19m 53s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d96d60a8fc01490ab2730cb425648e2c : SUCCESS in 20m 08s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5d61059f89124142a91e6e3ba83270f3 : SUCCESS in 24m 35s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0993da7e8d3c4234bf406f7d2a0492fb : SUCCESS in 23m 00s","accounts_in_message":[],"_revision_number":6},{"id":"63ea0f4ef9f74149d9189ab81eb7d555583b6ef3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-23 06:23:07.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"3bb6ec80617d97c0aebdc5026b10899c772267b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 10:15:33.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b7dfa44481734350aad828c180e0c1da : SUCCESS in 9m 11s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/8e09ce2eae3c425ba55c92a93368884d : SUCCESS in 10m 44s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/2aeb408a1a7e4dd2bbb44d07775032c9 : SUCCESS in 12m 43s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1c8430bc1301481ead95f295ea71dfc7 : SUCCESS in 13m 04s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91cdf098b9924fc7bd02bfb5bf5e4a80 : SUCCESS in 11m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e231d607a547409596afbdcabecd5ee2 : SUCCESS in 7m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0319f6709afb428882508a17d3f9148b : SUCCESS in 7m 33s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/ce51c6668b264c0899df1ca88e90d4b8 : SUCCESS in 2h 21m 32s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/6dddf01d87c84c7ca9ff543d94067cf4 : SUCCESS in 1h 07m 42s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6d86ce52abb437bb89cf92faf6be95f : TIMED_OUT in 3h 06m 42s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/de2d89c169744436adc7dd39e5fb7cbe : SUCCESS in 38m 38s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6db0a6695ce243eea881113370cb16f1 : TIMED_OUT in 1h 33m 24s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d96dc8f1b5f9471aba835205b336f462 : SUCCESS in 19m 18s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8fa14e8570fc403087556152b6ab3c3e : SUCCESS in 18m 55s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a3d62387be242dd98eefeb6f9507960 : SUCCESS in 20m 00s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dc8ab03e61574f3d9f6361be631e73c6 : SUCCESS in 18m 10s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/feb5a22a712d434eb690790cb70ea34b : SUCCESS in 19m 05s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3b7d01159e5c46178000196ae6e997bd : SUCCESS in 17m 55s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6bd87ac88d9b4f22b30e0acff32e35af : SUCCESS in 19m 06s","accounts_in_message":[],"_revision_number":7},{"id":"3681f3c26d68b966ea57df0254c105c63df6c3f0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-23 10:44:43.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"c1abb1108842bb3b53ede3a767ceaf8c2d549b0b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-23 10:45:37.000000000","message":"Patch Set 7:\n\ncentos execution speed really concerns me. that is most likely due to the new image, but not sure what has changed there.","accounts_in_message":[],"_revision_number":7},{"id":"0208d94ec223a7e82fa72968f17a7d0e0f904474","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-23 12:30:52.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"d91600c8c11f6194965e22514dea1e4dcb0b9490","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 13:18:29.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ba45bb5c1e8649eb948c1cca85add6fa : FAILURE in 5m 16s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/b3718b225c47476dadade457c5168f35 : SUCCESS in 10m 51s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/2c4be60b49854946b55f4e2e6d16e349 : SUCCESS in 14m 08s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/232c1b3925d7466c9e24951132db3361 : SUCCESS in 12m 59s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e43b0bee8d04a0896c254172da378bc : SUCCESS in 13m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23ebd7b7be794366b52c36c80b917778 : SUCCESS in 6m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f69a28d8d744936bb9272d89b76d1b3 : SUCCESS in 7m 15s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/d552d156c0a346bb892c1dd19c296103 : SUCCESS in 2h 23m 53s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/103f9c3fc41f439b82fba3447437b7f7 : SUCCESS in 1h 04m 49s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57274fbdc35c4885aecc7c4e263e1068 : SUCCESS in 1h 06m 27s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/490deeff767c4e5897d919cba11436f7 : SUCCESS in 35m 31s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f766b48bc0984b14a7717bd615f16df5 : SUCCESS in 18m 49s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be131b23738d4daa90e6ffd87670d179 : FAILURE in 4m 41s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5c2216d4bb304c5fa56f80207ebdbb35 : FAILURE in 5m 10s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/479f9c8fbfa04fe0a1be3b3a7f51d9fb : FAILURE in 5m 19s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/414ad8e7d6004d1289e1e37e602b1b8d : SUCCESS in 18m 10s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da5eabe244074faa86e7cb65a437cb0a : SUCCESS in 19m 43s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c820c25f243c40da8692780abcb51aa5 : SUCCESS in 19m 39s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dde58ae4b2464fb1a54ef0fbe4ce113c : SUCCESS in 19m 31s","accounts_in_message":[],"_revision_number":7},{"id":"2e88cdb95abe09009aeca359ebc015f60f199116","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-23 13:51:50.000000000","message":"Patch Set 7: Workflow-1\n\ncentos image is real.... Tested lxc-containers-create.yml on single VM with containers removal:\n\nold image:\nreal    13m27,345s\nuser    5m13,063s\nsys     1m49,110s\n\n\nnew image:\nreal    23m20,550s\nuser    7m5,946s\nsys     2m15,115s","accounts_in_message":[],"_revision_number":7},{"id":"700026290a6a8969ab5c9ea7d08c4e04482b3e21","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-23 17:44:24.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"7574c1ae08227b88d540cd5f9bda3231010a5c54","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-23 17:44:30.000000000","message":"Patch Set 8: Workflow-1","accounts_in_message":[],"_revision_number":8},{"id":"40f711a9b8d3a6dd7dbd53271294210fad633310","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 22:05:45.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6416473eff7b4890bd5295d1c70861dd : SUCCESS in 14m 18s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/91edd8619212461c8eb416f933583a6a : SUCCESS in 11m 45s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/134a775eb67b4b21b0a3164b29a9698d : SUCCESS in 20m 26s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9caadb233a42414e8484569225c41fee : SUCCESS in 17m 32s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a30796c1f4874fdd83703c0f1e6dcef7 : SUCCESS in 13m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52c2f7299a234ddc88e6d4789fd4ebb3 : SUCCESS in 8m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/48f3be4211d54afbad5a4a0d5d996cdc : SUCCESS in 6m 05s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/c0819fc2ce364e63acc7db59a5df9799 : SUCCESS in 3h 02m 18s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/e3052d49d4074f1c98df5dc67767aa42 : SUCCESS in 1h 19m 33s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/438d6818297b4f2daf9e286aec05c575 : SUCCESS in 59m 54s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/019edc86618f4e4fb085ba1e76ca5841 : SUCCESS in 39m 42s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/41277b008a634cc0ac5c62981cdb2335 : SUCCESS in 21m 54s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fbd27ae7efa847c083f2b6419c44123b : SUCCESS in 18m 10s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2e7c92c96f3c4c12a909910f075527cf : SUCCESS in 19m 46s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/900fdd623ac94bafa229b7dcafd1b1d2 : SUCCESS in 19m 47s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6b77187df32b4129bbbf8b7d5d7d8357 : SUCCESS in 19m 41s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b55f71e346049e5b2665783e103947e : SUCCESS in 23m 44s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5ae286b3a23247c495bb85855d8eb822 : SUCCESS in 20m 15s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f29b07962204a3b8df383830bdbbc04 : SUCCESS in 23m 40s","accounts_in_message":[],"_revision_number":8},{"id":"5d3e480fc8be769f22a9b5b15e1bb8ff78a60807","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-27 19:12:07.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"927fa71d450f1355cd449d9ca514363c4a476ee2","revisions":{"6238a49073761d71518b17d9a7b80d1d6b4ccf70":{"kind":"REWORK","_number":1,"created":"2020-10-22 18:02:25.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/10/759310/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/10/759310/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/10/759310/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/10/759310/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/10/759310/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/10/759310/1"}}},"commit":{"parents":[{"commit":"a6476c3f5b6b4707c3a177cd1ed68ef6f804de51","subject":"Increase amount of MaxSessions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a6476c3f5b6b4707c3a177cd1ed68ef6f804de51"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-22 18:02:11.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-22 18:02:11.000000000","tz":180},"subject":"Determine latest base image available v4","message":"Determine latest base image available v4\n\nChange-Id: Ib926e930ba105eeb49b7cf0031c77c7dc4c97b62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6238a49073761d71518b17d9a7b80d1d6b4ccf70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6238a49073761d71518b17d9a7b80d1d6b4ccf70"}]},"branch":"refs/heads/master"},"de03c9ad665a78a46fadbdeffd3d922f28a14abe":{"kind":"REWORK","_number":2,"created":"2020-10-22 18:13:16.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/10/759310/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/10/759310/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/10/759310/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/10/759310/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/10/759310/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/10/759310/2"}}},"commit":{"parents":[{"commit":"a6476c3f5b6b4707c3a177cd1ed68ef6f804de51","subject":"Increase amount of MaxSessions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a6476c3f5b6b4707c3a177cd1ed68ef6f804de51"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-22 18:02:11.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-22 18:13:02.000000000","tz":180},"subject":"Determine latest base image available v4","message":"Determine latest base image available v4\n\nChange-Id: Ib926e930ba105eeb49b7cf0031c77c7dc4c97b62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/de03c9ad665a78a46fadbdeffd3d922f28a14abe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/de03c9ad665a78a46fadbdeffd3d922f28a14abe"}]},"branch":"refs/heads/master"},"d2aa0c9ccf14902d9eaf0488f2d80abd8f22dff6":{"kind":"REWORK","_number":3,"created":"2020-10-22 18:35:39.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/10/759310/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/10/759310/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/10/759310/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/10/759310/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/10/759310/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/10/759310/3"}}},"commit":{"parents":[{"commit":"a6476c3f5b6b4707c3a177cd1ed68ef6f804de51","subject":"Increase amount of MaxSessions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a6476c3f5b6b4707c3a177cd1ed68ef6f804de51"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-22 18:02:11.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-22 18:35:29.000000000","tz":180},"subject":"Determine latest base image available v4","message":"Determine latest base image available v4\n\nChange-Id: Ib926e930ba105eeb49b7cf0031c77c7dc4c97b62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d2aa0c9ccf14902d9eaf0488f2d80abd8f22dff6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d2aa0c9ccf14902d9eaf0488f2d80abd8f22dff6"}]},"branch":"refs/heads/master"},"6a21ebcd0258011c6d6486cffdfb24724fcc4b56":{"kind":"REWORK","_number":4,"created":"2020-10-22 19:17:07.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/10/759310/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/10/759310/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/10/759310/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/10/759310/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/10/759310/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/10/759310/4"}}},"commit":{"parents":[{"commit":"a6476c3f5b6b4707c3a177cd1ed68ef6f804de51","subject":"Increase amount of MaxSessions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a6476c3f5b6b4707c3a177cd1ed68ef6f804de51"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-22 18:02:11.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-22 19:15:57.000000000","tz":180},"subject":"Determine latest base image available v4","message":"Determine latest base image available v4\n\nChange-Id: Ib926e930ba105eeb49b7cf0031c77c7dc4c97b62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6a21ebcd0258011c6d6486cffdfb24724fcc4b56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6a21ebcd0258011c6d6486cffdfb24724fcc4b56"}]},"branch":"refs/heads/master"},"2ea92cf36c93860e4780557c347dbc4ab0c04938":{"kind":"REWORK","_number":5,"created":"2020-10-22 19:40:49.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/10/759310/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/10/759310/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/10/759310/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/10/759310/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/10/759310/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/10/759310/5"}}},"commit":{"parents":[{"commit":"a6476c3f5b6b4707c3a177cd1ed68ef6f804de51","subject":"Increase amount of MaxSessions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a6476c3f5b6b4707c3a177cd1ed68ef6f804de51"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-22 18:02:11.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-22 19:40:37.000000000","tz":180},"subject":"Use lxc_image_cache_server_mirrors as image source","message":"Use lxc_image_cache_server_mirrors as image source\n\nBecause of ubuntu has dropped their LTS base images and\nCentos does not ship not docker layered images,\nwe decided to return back to use default LXC images\nby default. In the meanwhile, in case deployer\ndefined lxc_hosts_container_image_url - this url will\nhave prescedence and URL detection will not happen.\n\nCloses-Bug: #1900884\nCloses-Bug: #1900975\nChange-Id: Ib926e930ba105eeb49b7cf0031c77c7dc4c97b62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2ea92cf36c93860e4780557c347dbc4ab0c04938"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2ea92cf36c93860e4780557c347dbc4ab0c04938"}]},"branch":"refs/heads/master"},"6b1877a238ff4cbc6de115e12bd91408f3a8ba23":{"kind":"REWORK","_number":6,"created":"2020-10-22 20:05:16.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/10/759310/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/10/759310/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/10/759310/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/10/759310/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/10/759310/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/10/759310/6"}}},"commit":{"parents":[{"commit":"a6476c3f5b6b4707c3a177cd1ed68ef6f804de51","subject":"Increase amount of MaxSessions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a6476c3f5b6b4707c3a177cd1ed68ef6f804de51"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-22 18:02:11.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-22 20:04:01.000000000","tz":180},"subject":"Use lxc_image_cache_server_mirrors as image source","message":"Use lxc_image_cache_server_mirrors as image source\n\nBecause of ubuntu has dropped their LTS base images and\nCentos does not ship not docker layered images,\nwe decided to return back to use default LXC images\nby default. In the meanwhile, in case deployer\ndefined lxc_hosts_container_image_url - this url will\nhave prescedence and URL detection will not happen.\n\nCloses-Bug: #1900884\nCloses-Bug: #1900975\nChange-Id: Ib926e930ba105eeb49b7cf0031c77c7dc4c97b62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6b1877a238ff4cbc6de115e12bd91408f3a8ba23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6b1877a238ff4cbc6de115e12bd91408f3a8ba23"}]},"branch":"refs/heads/master"},"6abc6e2d5ee4f4bbc5fbf1780b2ac2956a618ed1":{"kind":"REWORK","_number":7,"created":"2020-10-23 06:23:07.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/10/759310/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/10/759310/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/10/759310/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/10/759310/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/10/759310/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/10/759310/7"}}},"commit":{"parents":[{"commit":"a6476c3f5b6b4707c3a177cd1ed68ef6f804de51","subject":"Increase amount of MaxSessions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a6476c3f5b6b4707c3a177cd1ed68ef6f804de51"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-22 18:02:11.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-10-23 06:23:01.000000000","tz":0},"subject":"Use lxc_image_cache_server_mirrors as image source","message":"Use lxc_image_cache_server_mirrors as image source\n\nBecause of ubuntu has dropped their LTS base images and\nCentos does not ship not docker layered images,\nwe decided to return back to use default LXC images\nby default. In the meanwhile, in case deployer\ndefined lxc_hosts_container_image_url - this url will\nhave prescedence and URL detection will not happen.\n\nCloses-Bug: #1900884\nCloses-Bug: #1900975\nChange-Id: Ib926e930ba105eeb49b7cf0031c77c7dc4c97b62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6abc6e2d5ee4f4bbc5fbf1780b2ac2956a618ed1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6abc6e2d5ee4f4bbc5fbf1780b2ac2956a618ed1"}]},"branch":"refs/heads/master"},"927fa71d450f1355cd449d9ca514363c4a476ee2":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2020-10-23 17:44:24.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/10/759310/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/10/759310/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/10/759310/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/10/759310/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/10/759310/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/10/759310/8"}}},"commit":{"parents":[{"commit":"a6476c3f5b6b4707c3a177cd1ed68ef6f804de51","subject":"Increase amount of MaxSessions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a6476c3f5b6b4707c3a177cd1ed68ef6f804de51"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-22 18:02:11.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-10-23 17:44:21.000000000","tz":0},"subject":"Use lxc_image_cache_server_mirrors as image source","message":"Use lxc_image_cache_server_mirrors as image source\n\nBecause of ubuntu has dropped their LTS base images and\nCentos does not ship not docker layered images,\nwe decided to return back to use default LXC images\nby default. In the meanwhile, in case deployer\ndefined lxc_hosts_container_image_url - this url will\nhave prescedence and URL detection will not happen.\n\nCloses-Bug: #1900884\nCloses-Bug: #1900975\n\nDepends-On: https://review.opendev.org/759498\nChange-Id: Ib926e930ba105eeb49b7cf0031c77c7dc4c97b62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/927fa71d450f1355cd449d9ca514363c4a476ee2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/927fa71d450f1355cd449d9ca514363c4a476ee2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
