)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~768723","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~Ieb290bba4bfbd03f1000249102b4cf6ed67b57ec","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","hashtags":[],"change_id":"Ieb290bba4bfbd03f1000249102b4cf6ed67b57ec","subject":"Use proxy if any proxy_vars are set","status":"ABANDONED","created":"2020-12-29 10:59:10.000000000","updated":"2021-03-04 09:18:24.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d605ba5800e68aca15faf9683fc49154dff85734","_number":768723,"virtual_id_number":768723,"owner":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-12-29 13:03:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"all":[{"value":-2,"date":"2021-01-21 22:58:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"all":[{"value":-1,"date":"2021-01-21 22:58:16.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-29 13:03:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"99699d1a418d0dccb0857d5f3cefb4db7ff99611","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2020-12-29 10:59:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ca218ff67e2f8b4b71a6599afd16b60b658054b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-29 13:03:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/acb50bbdedd345198f5da590b72e484b : FAILURE in 7m 10s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/aeeb21a89fc04f20894fb03fd3103c8b : FAILURE in 6m 39s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/d27ae38949914ccd986c3ce2e7002b1c : FAILURE in 8m 32s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/06e05fba1c5841528c8ebd5d5b250d37 : FAILURE in 8m 21s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6745ca75dd374ccbb7feed87c0f3498f : FAILURE in 7m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c08fee3275d4820906e780a312720b1 : SUCCESS in 5m 10s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/8fce6462ca3c489a8026622e75e16599 : SUCCESS in 1h 18m 34s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/d7c9d185b8304e2c8cdcb70e2cb1e7f3 : FAILURE in 36m 34s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/037cd643ee22434da828117300322d6a : SUCCESS in 1h 26m 04s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/ce7f63b474964d649bd1215b6b103c99 : FAILURE in 9m 30s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ec233e7faf7c4aea9425c3c2f433dce9 : FAILURE in 8m 27s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5becb273411b479e8341887f78e539d4 : FAILURE in 6m 54s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/20457865b9804d30a8ec8edd2dd9fdfc : FAILURE in 6m 33s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67dc0a95724c4eb6b166e50ca2a10d82 : FAILURE in 9m 31s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/23fb53dd942f4f1781492024194e8045 : FAILURE in 7m 37s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99c9fc9c6f934b6aa23e7029fead7c72 : FAILURE in 9m 12s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a0d30c2fcbe848b8a68e3dfcb0adc3ae : FAILURE in 7m 49s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cb59e4f044dd42a7a4b11685b1896115 : FAILURE in 9m 20s","accounts_in_message":[],"_revision_number":1},{"id":"dd7408914b22baf3a3d3feb9d849e0190de472fc","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2021-01-21 22:58:16.000000000","message":"Patch Set 1: Code-Review-2 Workflow-1\n\nThis should really happen automatically, and seems to do so if you\u0027ve got it set up in /etc/environment.","accounts_in_message":[],"_revision_number":1},{"id":"d605ba5800e68aca15faf9683fc49154dff85734","tag":"autogenerated:gerrit:abandon","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2021-03-04 09:18:24.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ca0b5e22c4a4a68c006dec8cefc4efd18ef327e0","revisions":{"ca0b5e22c4a4a68c006dec8cefc4efd18ef327e0":{"kind":"REWORK","_number":1,"created":"2020-12-29 10:59:10.000000000","uploader":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"ref":"refs/changes/23/768723/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/23/768723/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/23/768723/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/23/768723/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/23/768723/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/23/768723/1"}}},"commit":{"parents":[{"commit":"b3bff3289ac2e9510be81f562f6d35500ac47723","subject":"Install sysctl package in container for debian","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b3bff3289ac2e9510be81f562f6d35500ac47723"}]}],"author":{"name":"Erik Berg","email":"openstack@slipsprogrammor.no","date":"2020-12-29 10:42:23.000000000","tz":0},"committer":{"name":"Erik Berg","email":"openstack@slipsprogrammor.no","date":"2020-12-29 10:59:08.000000000","tz":60},"subject":"Use proxy if any proxy_vars are set","message":"Use proxy if any proxy_vars are set\n\nIf you have an environment where your openstack hosts have no routes to\nthe internet, and just granting them basic access via the proxy_env_url,\nyou might have the same case for your deploy host. Set use_proxy to true\nin the lookup function if http_proxy or https_proxy env are provided.\n\nChange-Id: Ieb290bba4bfbd03f1000249102b4cf6ed67b57ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ca0b5e22c4a4a68c006dec8cefc4efd18ef327e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ca0b5e22c4a4a68c006dec8cefc4efd18ef327e0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
