)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~773914","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2Fvictoria~I0f12760840d81c9d73118d63bc0927eab78b9df2","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/victoria","topic":"bug/1908544-stable/victoria","hashtags":[],"change_id":"I0f12760840d81c9d73118d63bc0927eab78b9df2","subject":"Fix lxc_hosts_container_image_url condition","status":"MERGED","created":"2021-02-03 17:31:47.000000000","updated":"2021-02-04 12:24:15.000000000","submitted":"2021-02-04 12:22:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"773914-bug/1908544-stable/victoria","cherry_pick_of_change":773781,"cherry_pick_of_patch_set":1,"meta_rev_id":"c9c4df8b3d2a9a90f37a5a9130afc63bd9d8b995","_number":773914,"virtual_id_number":773914,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-04 12:22:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2021-02-03 17:32:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-03 19:17:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-03 19:17:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-03 17:31:47.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2021-02-03 18:50:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"16ff45149005e6937fcb84afe8d1e8205f19afc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-03 17:31:47.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"bb42a5955d21a7a45a27123b25a4ebb310633a0b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-03 17:32:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c2be9bee174e85a70a2cc7d0f755bc2e9305aa3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 18:50:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8a21c74eee6642a7b749ae4217464b46 : SUCCESS in 9m 43s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/ad7989c28c2c4be2bfbd5bece551941d : SUCCESS in 9m 35s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/40565cdf9c614b88bdc01243e1501fa9 : SUCCESS in 13m 21s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/64b78fbe36584e8f987273c2d4bb67de : SUCCESS in 11m 31s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61a1e1900e0745199b7afb3ede26c8d9 : SUCCESS in 12m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a6478a2ce4643c5b1c9405405adb5d2 : SUCCESS in 5m 49s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/6a01bf2a75bf485f82332baeac90e4e5 : SUCCESS in 1h 09m 40s (non-voting)\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/5ab47cad494a4f1e97a6ff29bbb990b4 : SUCCESS in 1h 02m 44s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08d249a2e2e846f3929228db797d0ee8 : SUCCESS in 59m 47s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/26f0f84f44764d909f1d3b324bc46ace : SUCCESS in 21m 37s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4162cf6c188141d08798f7c081486d67 : SUCCESS in 21m 31s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90e2eb60791a4cff96a1c614b10d4fbb : SUCCESS in 18m 37s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/479ea2f415304b7b8e62dc14d9077557 : SUCCESS in 15m 10s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6866cf027fd4e45afb197da43890aad : SUCCESS in 19m 38s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d1d287b0dc7f406cbd12f71308f4c807 : SUCCESS in 15m 51s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22778cc62de24f9db5af858bcaa8008d : SUCCESS in 18m 21s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b9bbd2fc50954aa08cb178de45a2c508 : SUCCESS in 19m 20s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6ad856d54514886a0321b88a67bca6a : SUCCESS in 17m 54s","accounts_in_message":[],"_revision_number":1},{"id":"2dc6b0d75201e3bbac9e0631af30cce28829da20","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-03 19:17:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d8e7556e0c61a2ed37a3d8ebbda39f4d4dab8107","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 19:19:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d4dfd0b83359ed7edc434d4a3d8873d502279903","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 20:49:41.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f163ab9c9e884302b050809a18083f41 : FAILURE in 7m 28s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/4814d90b72f9423c8030f8be652f5cc4 : SUCCESS in 9m 24s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d2ebe59d49244747b66c2d955760e68a : SUCCESS in 9m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e61a24ef46d45c19b22210230000685 : SUCCESS in 4m 48s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/96b2e1fde5594af69b0d6f3a1f7ac0e5 : SUCCESS in 1h 15m 19s (non-voting)\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/f7eaf13a7efe4905a133403981860cbb : SUCCESS in 58m 09s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d041b0db6c414c14994d96562f4d8bbd : FAILURE in 25m 36s","accounts_in_message":[],"_revision_number":1},{"id":"679a579f27eb25970dab8b1669f9cf9d82022ad2","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-04 07:32:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"77cbd144a4c9a297d5191ef468cb60b9a194939f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 09:03:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6175af8abc704928a91ccff1ca6a1425 : SUCCESS in 14m 36s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/6d04d62d33174f9ea04bcf5f039900e5 : SUCCESS in 14m 11s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/05cd8d6b2cbc401697f22e092c8e2953 : SUCCESS in 19m 30s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a825bebbef484b3abdd69f6b70a9d543 : SUCCESS in 15m 43s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b64ce962f4f44c0ae8614eeed9ef62b : SUCCESS in 15m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbf37ae3eefc4647a859583feea9212e : SUCCESS in 6m 47s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/85db26fcafac4713bb3561b631f88aa6 : SUCCESS in 1h 21m 23s (non-voting)\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/001398766424470187e108c72ec68845 : FAILURE in 49m 52s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d97703b7ed5249d0a350941a85888d76 : SUCCESS in 1h 02m 29s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/8b3fc4fe6cc7406f8bd9f625d9506fd4 : SUCCESS in 29m 03s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8f179262c56b45619eba4ca6e5f1a679 : SUCCESS in 19m 39s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae47b335e5fd42d2be815f27db8ed121 : SUCCESS in 15m 27s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cd20c59b89304325840828cf853e77df : SUCCESS in 20m 06s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02cb28b3a5c1489c86a8e06a68d8909b : SUCCESS in 17m 43s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7f5f6cfd68194fb1a0c666de8473573d : SUCCESS in 17m 25s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8c2f838e2d8482cafe8a02fa3c19ee0 : SUCCESS in 16m 15s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/de3cc10e9102451aa6013d369f35bab2 : SUCCESS in 17m 19s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bef05593783e42c6b70063e5d7ab828b : SUCCESS in 16m 15s","accounts_in_message":[],"_revision_number":1},{"id":"02e5a474672d220b5c0ec40d9127739e09f70a99","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-04 09:08:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"67d542133949a4491cede9835e20ba06eb09ebec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 10:28:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9ceb2533ca3343c3b31b01f379940621 : SUCCESS in 11m 44s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/db004e4d72a74da691462145bf82211a : SUCCESS in 13m 20s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/3eb99f3a1fd94d14a04dfc197db59371 : SUCCESS in 16m 33s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/04b60102598d4943a3cda455cda308d4 : SUCCESS in 11m 23s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/86318fc3f37f4a0385958848e793a229 : SUCCESS in 11m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/093286a174e444818292ab90fd60a03d : SUCCESS in 4m 40s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/7a9a0ad986974294a6fc2db06d579435 : SUCCESS in 1h 12m 41s (non-voting)\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/224c165bbd7a4a3e8144b7e55df82650 : SUCCESS in 58m 33s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/983a4942daa549bd9d371582518e581d : SUCCESS in 54m 46s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/939de59ae422435bace8c92f83663b5c : SUCCESS in 16m 13s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/68dbe0c703a844b0af3188656001eab6 : SUCCESS in 15m 41s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e1e4823485714cf9b7aa9fc91fdf5f19 : SUCCESS in 16m 08s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ce20bdf781fe4cc893d2cee5ffcfc776 : SUCCESS in 16m 39s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/622f1ebae89c4adda20e6de8b04e2ea9 : SUCCESS in 17m 50s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/12d8d3e565874898950fc3ac4e8b9bbd : SUCCESS in 18m 27s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a884ff23fbbd4ec09c69ef02c454ee4b : SUCCESS in 17m 17s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7f558ff2f13443f6a322de3132ca9f93 : SUCCESS in 16m 48s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c5b054fa04e4567955497ee9149932f : SUCCESS in 17m 54s","accounts_in_message":[],"_revision_number":1},{"id":"0b9481bdb2f72fb7b98ba2daaf2d175ab202fbf2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 10:29:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d0eb5d539ee585e0cf8b951ddb8655baca3d2155","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 12:22:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a7c1c47a659d481b874f6b854396914b : SUCCESS in 9m 05s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/2a9f95df74904fb3934cc249a8a6b463 : SUCCESS in 9m 11s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d34b1767fc9c426e937407befae056f3 : SUCCESS in 15m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77c6c825b0bd4cc898c3e5d654740814 : SUCCESS in 5m 15s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/0624684b885547ea9b621da016aafada : SUCCESS in 1h 16m 34s (non-voting)\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/bf402960db1d4b52ad836f2f1eb46f55 : SUCCESS in 1h 01m 01s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/453e4f3ed680411788f1861490b411a1 : SUCCESS in 1h 40m 15s","accounts_in_message":[],"_revision_number":1},{"id":"db866a34ff396f3c315cd0813df211c1ac98ebf5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 12:22:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c9c4df8b3d2a9a90f37a5a9130afc63bd9d8b995","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 12:24:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94a5d0835b434312b28dff772896bf4e : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c1cff6dfdcbd21f71679f880cdeafb2bc336f6d9","revisions":{"c1cff6dfdcbd21f71679f880cdeafb2bc336f6d9":{"kind":"REWORK","_number":1,"created":"2021-02-03 17:31:47.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/14/773914/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/14/773914/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/14/773914/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/14/773914/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/14/773914/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/14/773914/1"}}},"commit":{"parents":[{"commit":"57025530f0b7ac0d8c0550ec52288005b2817a9a","subject":"Add trusted.gpg to the copy_from_host list for 20.04","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/57025530f0b7ac0d8c0550ec52288005b2817a9a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-02-02 18:07:21.000000000","tz":120},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-03 17:31:47.000000000","tz":0},"subject":"Fix lxc_hosts_container_image_url condition","message":"Fix lxc_hosts_container_image_url condition\n\nlxc_hosts_container_image_url is string, so can\u0027t be passed through bool\nfilter as it will always result as false.\n\nChange-Id: I0f12760840d81c9d73118d63bc0927eab78b9df2\nCloses-Bug: #1908544\n(cherry picked from commit bd40afa00ad88c35b7916c9f9bf8f25c5fef0647)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c1cff6dfdcbd21f71679f880cdeafb2bc336f6d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c1cff6dfdcbd21f71679f880cdeafb2bc336f6d9"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
