)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~775218","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2Fussuri~I0f12760840d81c9d73118d63bc0927eab78b9df2","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/ussuri","topic":"bug/1908544-stable/ussuri","hashtags":[],"change_id":"I0f12760840d81c9d73118d63bc0927eab78b9df2","subject":"Fix lxc_hosts_container_image_url condition","status":"MERGED","created":"2021-02-11 16:47:40.000000000","updated":"2021-02-11 19:22:42.000000000","submitted":"2021-02-11 19:20:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"775218-bug/1908544-stable/ussuri","cherry_pick_of_change":773781,"cherry_pick_of_patch_set":1,"meta_rev_id":"96079c00281d8fa5a07af2b08c5912d62dda9230","_number":775218,"virtual_id_number":775218,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-11 19:20:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2021-02-11 18:31:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-11 19:02:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-11 19:02:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-11 16:47:40.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2021-02-11 17:16:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bd13a0b7642d12bbb49251bc971015c01f573a06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-11 16:47:40.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"41c91fe3b12c289179fbe149b5610da2443d2c43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 17:16:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1cb0d10a1be544b7aac20af0ef02e74f : SUCCESS in 6m 51s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/ca86891bab3b4e17819c25a97e79b9e8 : SUCCESS in 8m 53s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/8aefd26ce50b41ed95553f46afdcb6d7 : SUCCESS in 14m 05s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/144f8f73431e45bfac01d45b3a0bd3d3 : SUCCESS in 15m 25s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/00afd829e5f44e14956628e54e3e4f10 : SUCCESS in 8m 44s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36ae4525f0054fcebe190ca0ae2040ff : SUCCESS in 12m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4066951a23842dd873afb1d56689fd9 : SUCCESS in 6m 47s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/1f5c91c4dc2d45c1b5e5d7ca09d68dc6 : SUCCESS in 18m 16s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/24121ac424b64122bc1369790233c2df : SUCCESS in 19m 20s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54febad609e546e2baffe49d10e8fe91 : SUCCESS in 13m 21s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b15beb61e5a54be8a29aecb5bb9d2960 : SUCCESS in 13m 43s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a8dcd5530814d1abcdd90fe12497cbc : SUCCESS in 15m 29s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4e5c527b772d4762bd45e8c09e88aa81 : SUCCESS in 14m 59s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c18289120e974edc8e8690e7f02b560f : SUCCESS in 21m 50s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bfadcafd8c334c28aeda01e418c08367 : SUCCESS in 14m 27s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6386f580849d408b938ccb2079af9ec2 : SUCCESS in 16m 45s","accounts_in_message":[],"_revision_number":1},{"id":"0827e763947587be097780fbd4c9ecfb7d0e7990","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-11 18:31:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ea7e60cedc3263906d608260763e708969fb0479","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-11 19:02:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b4f3c2ae745f1d787556fac936d7b6eed8865d56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 19:02:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"74bfa49fe3d74b221f06a8c94e877537739bcf15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 19:20:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b4c149c9a51d492483083daf85a632ce : SUCCESS in 7m 24s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/6aeea2b453364b40852fe93fc8f4d942 : SUCCESS in 10m 49s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/0f973e4611e04184abc4cc6cbe084464 : SUCCESS in 10m 41s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b0a2a96ab37b43309a777d094353c92e : SUCCESS in 10m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2982fb4b8b474f02a9747614024e37d9 : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":1},{"id":"27ee8986258d10ce82e8d10cd96eee7c23cc1fa7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 19:20:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"96079c00281d8fa5a07af2b08c5912d62dda9230","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 19:22:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ea1edb709b244339f82ad0dbb52a6ae : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a7943972a3bdde4f8e35544c571c14a242a9aa86","revisions":{"a7943972a3bdde4f8e35544c571c14a242a9aa86":{"kind":"REWORK","_number":1,"created":"2021-02-11 16:47:40.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/18/775218/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/18/775218/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/18/775218/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/18/775218/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/18/775218/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/18/775218/1"}}},"commit":{"parents":[{"commit":"e247e9544ae91bb28fed4755f05dfb99f3be0769","subject":"Add trusted.gpg to the copy_from_host list for 20.04","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/e247e9544ae91bb28fed4755f05dfb99f3be0769"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-02-02 18:07:21.000000000","tz":120},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-11 16:47:40.000000000","tz":0},"subject":"Fix lxc_hosts_container_image_url condition","message":"Fix lxc_hosts_container_image_url condition\n\nlxc_hosts_container_image_url is string, so can\u0027t be passed through bool\nfilter as it will always result as false.\n\nChange-Id: I0f12760840d81c9d73118d63bc0927eab78b9df2\nCloses-Bug: #1908544\n(cherry picked from commit bd40afa00ad88c35b7916c9f9bf8f25c5fef0647)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a7943972a3bdde4f8e35544c571c14a242a9aa86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a7943972a3bdde4f8e35544c571c14a242a9aa86"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
