)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~775778","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I443f5b9ecef1c142dcbcac91a0f36cb28035f3f6","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","hashtags":[],"change_id":"I443f5b9ecef1c142dcbcac91a0f36cb28035f3f6","subject":"Update Centos-8 LXC image to 8.3","status":"MERGED","created":"2021-02-16 09:24:42.000000000","updated":"2021-02-24 11:14:53.000000000","submitted":"2021-02-24 11:13:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"775778","meta_rev_id":"015606e7da23ff8a0083d45b32bfe4927561740f","_number":775778,"virtual_id_number":775778,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-24 11:13:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2021-02-24 09:01:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2021-02-23 16:34:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2021-02-24 09:01:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-16 10:38:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-16 12:02:52.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2021-02-23 16:34:22.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2021-02-24 09:01:17.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"52e5194d3fff1c0165be1896bc8fb24184ec87cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-16 09:24:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c5176f83f27f49ee6a86102ec7d9cd8595a16135","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 10:38:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5fb6b52a1a29438eb16cfce9a66dfccb : FAILURE in 8m 43s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/3e1c5c6282da4eb1a3f27a5daae4774f : SUCCESS in 10m 06s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/5a8ab12a546042269b65373a17245386 : SUCCESS in 10m 26s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e845f9299b92496db573906b4b7e3d21 : SUCCESS in 12m 26s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4ae59c6bf3c045579b8b14914dc4ff92 : SUCCESS in 12m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9979f4e1cc414f30acfa85c25949f5ff : SUCCESS in 5m 33s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/185880b56c5440fbba713730d1ed5f3e : FAILURE in 37m 32s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/1805a961e8744673878d96ac3050320a : SUCCESS in 1h 02m 09s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cebfd086c51f4ef0b581fd323d32632f : SUCCESS in 57m 48s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/77ab015e4fb24b2da0955b847fa13064 : SUCCESS in 20m 22s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6afd7c5633e346f2885afb5d2f7ca96b : SUCCESS in 17m 20s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6471ade21e5146059004528aaab67f44 : SUCCESS in 15m 37s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b8ee16b12b044b23b5535329666c1774 : SUCCESS in 16m 53s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1164c20ebaca43a8821dd09834263d7e : SUCCESS in 16m 16s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7a4e75b361044178bf7cf645d4ea1960 : SUCCESS in 16m 08s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/58cd5e72ea3244548294ff053ab84403 : SUCCESS in 18m 37s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/82f9763065e7404898e182a89d444d77 : SUCCESS in 16m 01s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e684434e51db4d92b4384f945412b8b2 : SUCCESS in 17m 27s","accounts_in_message":[],"_revision_number":1},{"id":"c1bcf1c6170242672ab42bd9fa91b2e3002fe860","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-16 11:31:48.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a147a95e2a5d403aa03f4cdf01a72e71222705bd","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-16 12:02:52.000000000","message":"Patch Set 2:\n\nNot sure it\u0027s worth effort atm considering there are no even layered images for Stream...","accounts_in_message":[],"_revision_number":2},{"id":"e679b808352b8ea7a070891305dbe241903c645a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-16 12:06:02.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e Not sure it\u0027s worth effort atm considering there are no even layered images for Stream...\n\nI did this because with the 8.1 image on an 8.3 CI node, the image gets completely new repo config when it does the dirst dnf update, makes it use external mirrors rather than the infra ones.\n\nAttempt really to reduce CI failures where possible.","accounts_in_message":[],"_revision_number":2},{"id":"6a8f9bea3d51ffaf8600c68fb163a0272de3427a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-16 12:10:04.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e \u003e Patch Set 2:\n\u003e \u003e \n\u003e \u003e Not sure it\u0027s worth effort atm considering there are no even layered images for Stream...\n\u003e \n\u003e I did this because with the 8.1 image on an 8.3 CI node, the image gets completely new repo config when it does the dirst dnf update, makes it use external mirrors rather than the infra ones.\n\u003e \n\u003e Attempt really to reduce CI failures where possible.\n\nOh, hm, ok then. I thought that we replace mirrors while preparring image to the needed ones, so container should not have some issues, but yeah, I see the point.","accounts_in_message":[],"_revision_number":2},{"id":"269d32e4c45b3cf515b682bc8d1653bcfc0dedcb","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-16 12:19:05.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e \u003e Patch Set 2:\n\u003e \u003e \n\u003e \u003e \u003e Patch Set 2:\n\u003e \u003e \u003e \n\u003e \u003e \u003e Not sure it\u0027s worth effort atm considering there are no even layered images for Stream...\n\u003e \u003e \n\u003e \u003e I did this because with the 8.1 image on an 8.3 CI node, the image gets completely new repo config when it does the dirst dnf update, makes it use external mirrors rather than the infra ones.\n\u003e \u003e \n\u003e \u003e Attempt really to reduce CI failures where possible.\n\u003e \n\u003e Oh, hm, ok then. I thought that we replace mirrors while preparring image to the needed ones, so container should not have some issues, but yeah, I see the point.\n\nLots of \u0027rpmnew\u0027 files end up in here https://f1ee1119f5b089211b7b-35c8bbbd1478f534398bbd564f0c210c.ssl.cf2.rackcdn.com/775598/2/check/openstack-ansible-deploy-aio_lxc-centos-8/49a8219/logs/etc/openstack/aio1_cinder_api_container-679afad1/yum.repos.d/index.html\n\nI think they come from upgrading the centos repos package from 8.1 to 8.3","accounts_in_message":[],"_revision_number":2},{"id":"8135c92f5d7d43e9a8648425c2c72c19abc21be3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 12:59:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a7146e552f814a7b8f387e9e32e05091 : FAILURE in 11m 24s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/4578b38aabcb414a9e6566eda7a3d05c : SUCCESS in 14m 14s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/ddaf04db4fc54102abff5bcfb87ab4bc : SUCCESS in 23m 32s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c52fde9f21d84af5b01cfebc22044865 : SUCCESS in 13m 17s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f279c013f594ec99fb0f789e22a563c : SUCCESS in 11m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4175bae6928948aba8b0f859af17f450 : SUCCESS in 5m 29s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/f7e5bdad9f9044789063ec55bdaa42c1 : SUCCESS in 1h 11m 11s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/9d38e915d10b4ea4bb0dfe711f91326c : FAILURE in 29m 19s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d609d4a297ad434aaa64640612bfa640 : FAILURE in 29m 23s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/35aa0a0cae944b6e8885be5059592340 : SUCCESS in 19m 35s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/943503a732c14ee384361689df5e0432 : SUCCESS in 17m 51s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e2d85aea05c246a68d493e8728f7dec3 : SUCCESS in 16m 16s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9b758850d87f499b8dfa428c63c0f11e : SUCCESS in 17m 40s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/caff45a4e89d44fe80a4d5a31474b42b : SUCCESS in 18m 22s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/76653029675e4e4ab25ccc07368adda8 : SUCCESS in 18m 21s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a1a28219b344129b97c34b8d2e1ef1f : SUCCESS in 16m 59s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/10204c0713b5449aaef5284df58809ab : SUCCESS in 15m 26s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c8e949fee244e4ea5644e8cbe05a7e7 : SUCCESS in 17m 50s","accounts_in_message":[],"_revision_number":2},{"id":"28b15a14dff9d6dd0a959999fae59d7bbb639b8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-16 13:30:43.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"b787c5942de767623e94c3f9dd0f523f7361e85b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 13:52:47.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"3e21e89666f7b936fd3701e56893778b2c687d41","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-16 14:25:17.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"83a14b1434b181b0b349c1bedebb072086fc2767","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 15:08:27.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6f67c25fe42640d8bb6fef399e82bfe8 : SUCCESS in 8m 46s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/4a2f95ef997140e19747fd809111a179 : SUCCESS in 10m 21s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/827b454af46844538c77ff7ad09e65af : SUCCESS in 12m 16s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2daef720cf724f6f975c444a31fc46b7 : SUCCESS in 12m 58s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bcc7ef545d5d49c1a09749c9c034758a : SUCCESS in 10m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/468ac141d8d74afb9ddb976c6b70272a : SUCCESS in 5m 53s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/6bf5d9a201494a0d940a38ffa3d89076 : FAILURE in 34m 11s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/7c93ceb1983f4b03aa4865d73aadc4a9 : FAILURE in 26m 59s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7b5314d4dbe42a6a296fad7b7845ad4 : FAILURE in 27m 19s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/0a4cfe93a409494ba2ea71f01a28c687 : SUCCESS in 18m 11s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/deb065ce88864b3b9f1e6439ac305a8b : SUCCESS in 15m 27s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0844a391048243fbab986c0f4c5d39b6 : SUCCESS in 13m 43s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f6db3b9f66864d7ba1da722d52598ff0 : SUCCESS in 20m 04s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b4af44d19f914d9bb2018b2f8928b561 : SUCCESS in 19m 55s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/65455e9fafc847deae8d216b34607b7f : SUCCESS in 18m 58s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84d81f8827134695a606235981220862 : SUCCESS in 17m 23s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/649c2182211b45ed9c1fa651e3a241b5 : SUCCESS in 17m 19s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e762455c08624e9591572177169e6675 : SUCCESS in 19m 48s","accounts_in_message":[],"_revision_number":3},{"id":"0498332ecc9152956a674dfc905ac7d831d4d340","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-18 10:05:20.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"2c381d395862ccedf4d7ed4627dff873a7059a22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 11:31:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/72ff0b892c09499dac7b63604da411e5 : SUCCESS in 9m 03s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/1997e9678b5a4a5cab45dc679819af59 : SUCCESS in 10m 19s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/0ba29bd5e46e4d2ab1af15b82ea80b33 : SUCCESS in 12m 58s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1ee9112dd9b648cb9437103566efa26c : SUCCESS in 12m 33s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0dab8cbbebc14e17a0e21c81cbf543db : SUCCESS in 11m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/434c738aac924ce6ab626f3ce294498b : SUCCESS in 5m 33s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/231ff0ae57444db99bba0ec593e21004 : SUCCESS in 1h 17m 37s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/28c040c3fd584405b2b434fca32fe7d5 : SUCCESS in 1h 00m 18s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5dd84f75fbca45dfbc4a95c810c9e98e : SUCCESS in 1h 10m 13s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/20270f71855a4158a24229064684b234 : SUCCESS in 18m 23s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aaf18a56ec12453193934c6913fb986d : SUCCESS in 15m 20s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31ae217867e14910abc5579cbd4f0138 : SUCCESS in 17m 22s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2acb9a21a5704684b706ae76d08c2224 : SUCCESS in 16m 52s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9c5989907ff4a61adeb4d676ab0db28 : SUCCESS in 18m 05s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b4362f4b77ab401892de63b850df31ec : SUCCESS in 17m 20s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69363fb5fba04385bea2393d9a41ebd3 : SUCCESS in 21m 02s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c6fb9c803e054e77ba73eb61860446f3 : SUCCESS in 18m 44s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6b2c2f535a44270a7e6e538d86deae5 : SUCCESS in 16m 41s","accounts_in_message":[],"_revision_number":3},{"id":"aa9cd21a969a571f083d9c342e04bc47418411a9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-23 16:34:22.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"f02e2db8150541946324adaafb10776105ddf549","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-02-24 09:01:17.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"416c55cdc9e23dad71eba074e517957e34771681","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 09:01:32.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3ec2361e7c42c1a9cd20602311dabc45a93f355c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 11:13:08.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/dee3cccac3834383b0570d8a7e2b22ba : SUCCESS in 10m 00s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/835a35fec480465da36e40bdff25687c : SUCCESS in 12m 10s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c438a4c427c44c07b01d36a63efe8d83 : SUCCESS in 12m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94a8b5e1dfab45a08675e19c6beb90ef : SUCCESS in 7m 04s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/4ee72ecdc5d041b8b290f36d449e28d6 : SUCCESS in 1h 04m 32s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/66f1f12947b6471e900090b66922afe7 : SUCCESS in 55m 33s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce6b3049315246efa83d22ed6fe910ce : SUCCESS in 1h 00m 15s","accounts_in_message":[],"_revision_number":3},{"id":"ae7295e2685674a3d8b7aaf3ffbd9530a8ce41a4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 11:13:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"015606e7da23ff8a0083d45b32bfe4927561740f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 11:14:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/757a8412c3a140cfa0962c01360a4b66 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"172e3e4bd09dcd4a0c7fea222596f0ddef19f8df","revisions":{"31b34207c0ddfc381230b4bcde2d9aad9ba5fd8f":{"kind":"REWORK","_number":1,"created":"2021-02-16 09:24:42.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/78/775778/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/78/775778/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/775778/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/775778/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/775778/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/775778/1"}}},"commit":{"parents":[{"commit":"bd40afa00ad88c35b7916c9f9bf8f25c5fef0647","subject":"Fix lxc_hosts_container_image_url condition","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/bd40afa00ad88c35b7916c9f9bf8f25c5fef0647"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-16 09:22:47.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-16 09:22:47.000000000","tz":0},"subject":"Update Centos-8 LXC image to 8.3","message":"Update Centos-8 LXC image to 8.3\n\nThis requires extra ansible tasks to unpack the nested tar file\nstructure of the 8.3 container image.\n\nChange-Id: I443f5b9ecef1c142dcbcac91a0f36cb28035f3f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/31b34207c0ddfc381230b4bcde2d9aad9ba5fd8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/31b34207c0ddfc381230b4bcde2d9aad9ba5fd8f"}]},"branch":"refs/heads/master"},"0844d21a664df912095b4cb645602978cab66ba0":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-02-16 11:31:48.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/78/775778/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/78/775778/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/775778/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/775778/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/775778/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/775778/2"}}},"commit":{"parents":[{"commit":"bd40afa00ad88c35b7916c9f9bf8f25c5fef0647","subject":"Fix lxc_hosts_container_image_url condition","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/bd40afa00ad88c35b7916c9f9bf8f25c5fef0647"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-16 09:22:47.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-16 11:31:48.000000000","tz":0},"subject":"Update Centos-8 LXC image to 8.3","message":"Update Centos-8 LXC image to 8.3\n\nThis requires extra ansible tasks to unpack the nested tar file\nstructure of the 8.3 container image.\n\nDepends-On: I31880ca995735b737d33532eaa4c29be02523117\nChange-Id: I443f5b9ecef1c142dcbcac91a0f36cb28035f3f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/0844d21a664df912095b4cb645602978cab66ba0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/0844d21a664df912095b4cb645602978cab66ba0"}]},"branch":"refs/heads/master","description":"Edit commit message"},"172e3e4bd09dcd4a0c7fea222596f0ddef19f8df":{"kind":"REWORK","_number":3,"created":"2021-02-16 13:30:43.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/78/775778/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/78/775778/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/775778/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/775778/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/775778/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/775778/3"}}},"commit":{"parents":[{"commit":"bd40afa00ad88c35b7916c9f9bf8f25c5fef0647","subject":"Fix lxc_hosts_container_image_url condition","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/bd40afa00ad88c35b7916c9f9bf8f25c5fef0647"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-16 09:22:47.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-16 13:30:38.000000000","tz":0},"subject":"Update Centos-8 LXC image to 8.3","message":"Update Centos-8 LXC image to 8.3\n\nThis requires extra ansible tasks to unpack the nested tar file\nstructure of the 8.3 container image.\n\nDepends-On: I31880ca995735b737d33532eaa4c29be02523117\nChange-Id: I443f5b9ecef1c142dcbcac91a0f36cb28035f3f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/172e3e4bd09dcd4a0c7fea222596f0ddef19f8df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/172e3e4bd09dcd4a0c7fea222596f0ddef19f8df"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
