)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~776200","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I5e5ea241308a11b79864885584944037dfecc6bc","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"lxc-hosts-cleanup","hashtags":[],"change_id":"I5e5ea241308a11b79864885584944037dfecc6bc","subject":"Remove duplicate lxc volume tasks","status":"MERGED","created":"2021-02-17 13:50:21.000000000","updated":"2021-03-02 01:01:46.000000000","submitted":"2021-03-02 01:00:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"776200-lxc-hosts-cleanup","meta_rev_id":"d7193acff9172e06e9a2419dd9eb79dcc4dc43a5","_number":776200,"virtual_id_number":776200,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-02 01:00:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2021-02-24 08:53:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2021-02-23 10:41:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2021-02-24 08:53:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-17 15:42:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-23 10:41:28.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2021-02-24 08:53:54.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"433ea50026e4751b0a9d2a7797ec9fa6865298c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-17 13:50:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d19a01cc74c212e5da7cd5cc15cc9c7c93f9f04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 15:42:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e3626e5abe8f405ea6448a80a248f0fd : SUCCESS in 9m 35s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/4e21d31cd79046d0abdba2e44d967ba0 : SUCCESS in 11m 19s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/be7969a4b5db45008dfc573d03d33075 : SUCCESS in 13m 30s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/85f1e72fc2bc4a11bf4dd97a7480c264 : SUCCESS in 13m 48s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc5d5d8577ce4f1385738b8325129216 : SUCCESS in 11m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f14f18cf3d14ee195eb862c6011f8e7 : SUCCESS in 9m 28s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/10faab116e924dd0afccac6d68a74f52 : SUCCESS in 1h 16m 35s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/efa53525e8d641ee982e55b4a6699790 : SUCCESS in 59m 13s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ff8a55b1271437488e3f97fe003d5cb : SUCCESS in 57m 23s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/e2339a45680040c0bebc18c2819edfea : SUCCESS in 21m 15s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fc387c885a2646adabce3d30281bb879 : SUCCESS in 14m 59s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02c471d8018a47388d5ce5b7f1e75490 : SUCCESS in 17m 13s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b05f826b6e724c49a537a4c4680e4df1 : SUCCESS in 17m 10s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fcfacacd22e44b1ba5360428ad350133 : SUCCESS in 16m 52s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bc249674b6b54d6cb49052fdfd1c59f6 : SUCCESS in 17m 09s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59cd6d140272431980e9d5f865851d57 : SUCCESS in 16m 14s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4343922a0ddf4695b43b225140d7f2b4 : SUCCESS in 19m 13s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59a534c7979f4862b0f9e8b2d035d04f : SUCCESS in 20m 24s","accounts_in_message":[],"_revision_number":1},{"id":"fb4803fb7b8047f9af6c943fd9210d14e096dfee","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-22 23:19:59.000000000","message":"Topic set to osa-stream","accounts_in_message":[],"_revision_number":1},{"id":"5ce13ab827782876bd5624fc8cb700a5e54eaee1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-22 23:23:31.000000000","message":"Topic osa-stream removed","accounts_in_message":[],"_revision_number":1},{"id":"71fbfc81d34bdd6c7207e1cf3f0eccc9d8956759","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-22 23:23:39.000000000","message":"Topic set to lxc-hosts-cleanup","accounts_in_message":[],"_revision_number":1},{"id":"d7ac606334672d721e7ad4dcd3f0966e3b2b4d53","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-23 10:41:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dbbb7b7d12e01884c2ab35e977af4c3653226e35","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-02-24 08:53:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a9dfc8d6cff1e294c1d8de89b5ad94c864a3411e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 17:51:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"00be3e94a5dc478fc59373c60c23032ed9aebf9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 21:01:05.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"15af51acdf4b947092c8b4b83208efc0bbf44d18","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-01 22:25:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b4039c4203896f046545719d25963614ec669e7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 23:43:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a7ec5a7992df4921986d25a2d5e755f5 : SUCCESS in 10m 32s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/531ce6e8e52c403f809be11f795c7c79 : SUCCESS in 11m 55s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/440b6ee775e34697b8b7dedc4b745c0e : SUCCESS in 13m 33s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e5c301e86d9b40afb3f7b19cf65dbd15 : SUCCESS in 12m 21s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/073f40d2ea6947c7bf24938669522e02 : SUCCESS in 15m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25043d6bc43d40cb828f86004558ca9b : SUCCESS in 5m 32s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/fb6bc7fa354d4fce9d270cf27a99d681 : SUCCESS in 1h 08m 23s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/e556ff3eb79748ea8c23e963b8f1167f : SUCCESS in 1h 02m 40s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f82b9b1000ee4f6fbdda3fb367e981d3 : SUCCESS in 1h 05m 33s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/38eaa1afc85b4140b151b20c3cee9180 : SUCCESS in 19m 48s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0bc163ade69e421aab09aebdb0519b2a : SUCCESS in 19m 59s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/071535db12a5411b8f5c9368c0ccd604 : SUCCESS in 20m 11s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/124b66a82c8b4cc8a3459b83dca48fd2 : SUCCESS in 18m 52s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0d8f53509934d1ba729329ec7c77f4e : SUCCESS in 20m 16s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cece72d89878480d96c9fc15b582d04c : SUCCESS in 16m 58s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a167a2a9f27459da114f293ce71123b : SUCCESS in 18m 20s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/78bc34d6a7594f6cb50085099f27d8e6 : SUCCESS in 19m 27s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/352de0dbbbbf4b419d9a5ff45198d07e : SUCCESS in 19m 52s","accounts_in_message":[],"_revision_number":1},{"id":"ff4ce2e01792486a09e64447402cb784eea6a8e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 23:45:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b7736fb317fff7d2cb6f38ada3579caf29ef1b40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 01:00:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/51dba16aecf545bfa809bb12d9f9589b : SUCCESS in 9m 24s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/985e3689d54045408df91503b9f3f8e3 : SUCCESS in 11m 15s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/40a32c5902c447678faf6c0b643f0dc4 : SUCCESS in 12m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e90b79026361418b90a2d6d5dad3ab4c : SUCCESS in 5m 34s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/6c30b976c8734b12872d8fdf68661e13 : SUCCESS in 1h 07m 27s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/5ba42394e6e94e2ea30cd7f61a900644 : SUCCESS in 1h 08m 52s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/045d3315f42343469f9c0ecc1c34e07d : SUCCESS in 56m 58s","accounts_in_message":[],"_revision_number":1},{"id":"723ecc1c826fd494d265f519feb4451bbdae9af5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 01:00:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d7193acff9172e06e9a2419dd9eb79dcc4dc43a5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 01:01:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c5903dae57546348394129a3b9d61bc : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"608254f0a136eb0de4abab904716ae84418b48d8","revisions":{"608254f0a136eb0de4abab904716ae84418b48d8":{"kind":"REWORK","_number":1,"created":"2021-02-17 13:50:21.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/00/776200/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/00/776200/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/00/776200/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/00/776200/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/00/776200/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/00/776200/1"}}},"commit":{"parents":[{"commit":"369c807b90e0be7b604c3e336811ac5decf9ead8","subject":"Move lxc cache facts setting to where those facts are used.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/369c807b90e0be7b604c3e336811ac5decf9ead8"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 13:35:08.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 13:35:08.000000000","tz":0},"subject":"Remove duplicate lxc volume tasks","message":"Remove duplicate lxc volume tasks\n\nThese tasks are duplicated in both lxc_cache_preparation_simple.yml\nand lxc_volume.yml, so we remove one of the duplicate set of tasks.\n\nChange-Id: I5e5ea241308a11b79864885584944037dfecc6bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/608254f0a136eb0de4abab904716ae84418b48d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/608254f0a136eb0de4abab904716ae84418b48d8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
