)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~776201","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I1ddfe36259610b25e86b69d64d1d7f32a56c0e4d","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"osa-facts","hashtags":[],"change_id":"I1ddfe36259610b25e86b69d64d1d7f32a56c0e4d","subject":"Optionally build container image as a local chroot","status":"MERGED","created":"2021-02-17 13:50:21.000000000","updated":"2021-04-20 11:36:49.000000000","submitted":"2021-04-20 11:34:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":6,"has_review_started":true,"submission_id":"776201-osa-facts","meta_rev_id":"500752d425caac00d4c18651e35ffdac3bf01c8b","_number":776201,"virtual_id_number":776201,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-20 11:34:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2021-04-20 08:57:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2021-04-20 08:59:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2021-04-20 08:59:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-17 17:07:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-23 10:42:32.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2021-04-19 07:23:17.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2021-04-20 08:51:39.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"CC"},{"updated":"2021-04-20 08:57:45.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"c4752a644e3f3fae5fb0469d7dcc23503a3fe229","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-17 13:50:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e0429ac7b745ea425298a3334e651aec55b2ed6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-17 15:46:43.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"660d1907b0c5a7f92dd06c04fee52440777422db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 17:07:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4facdaaaee054244b53e7bb8427137d9 : FAILURE in 9m 07s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/463f870e2624452facb8eb59d0676c4f : SUCCESS in 10m 42s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/4cda31080c5d494891f03930833c3201 : SUCCESS in 15m 33s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0fbb54619ffb4cc0b9c449c42eca8d90 : SUCCESS in 14m 48s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3dbc9817fd848fc9f10ac38e28366f5 : SUCCESS in 14m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67c21f09f4a84d9a87ca6c702ce34dc5 : SUCCESS in 6m 19s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/6e6b53e0961040a383f9ebabc728dcf7 : FAILURE in 18m 29s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/c9c779a2858b4b09b444e4eb3d6a49f4 : SUCCESS in 1h 02m 32s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e387c5454aa47d387960c1b7f2395c5 : SUCCESS in 57m 42s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/4f13e56d731840b3997c2b9d76e79faa : SUCCESS in 34m 36s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8a979258670847ce8e09abb83784420d : SUCCESS in 16m 35s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f6181e6778f4a029f8a74861fb7e959 : SUCCESS in 16m 58s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/44e02b9a0969401dbae51d074a97d866 : SUCCESS in 15m 38s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90bfbe0ba7424bef94948a617cf7ebe3 : SUCCESS in 20m 47s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e5b2f3c2ff6a4772b37d6d323ebfe478 : SUCCESS in 16m 05s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/efb1c0f97e2f46fba19ed90a27bdedd4 : SUCCESS in 21m 12s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9bb3a6f0b018407e8e8f88c3c4a8d79a : SUCCESS in 17m 35s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f405d08b3aa42e788f76bdad1c3e0b9 : SUCCESS in 17m 37s","accounts_in_message":[],"_revision_number":2},{"id":"9b4bb31616dd6dfc28c94ef709b5570b330e088a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-17 17:11:31.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"3f4f9de8e5af9505bbc752b6439ce326a524a397","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 18:38:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/54609ca84d4444c7bfa52c824345dd2d : SUCCESS in 9m 12s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/bbd2400012cd4b7e85c6efa26f7c0e9a : SUCCESS in 9m 19s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/dfe412f5b44e4e1fb35ab0a870733319 : SUCCESS in 16m 59s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6e39a7de8adf494b8a7c4ea01dbce9e8 : SUCCESS in 12m 39s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9fb112b7e0f4500bddaf688d9c68e49 : SUCCESS in 14m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3bd59d34f8e4d879a915f1d00952e9a : SUCCESS in 5m 30s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/22d8c74aa3544d2799a93a0dc4c6df0a : FAILURE in 53m 19s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/71e8bfc1455e4ad4a8c1432a71e4d0a3 : SUCCESS in 1h 03m 13s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e74cb5a03104bdba3a0e5e33ca5a747 : SUCCESS in 59m 54s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/10188439fb7f45dd817c49b2849fbe22 : SUCCESS in 35m 36s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2201a7bb2f1a4659999b59f3d5c34712 : SUCCESS in 15m 27s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d5dddea79594392836f4d416b1048a5 : SUCCESS in 17m 34s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f2f5f178cead422b946f39f8e8ec6bfd : SUCCESS in 16m 56s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2fc2048912042ad84d62573dc5991df : SUCCESS in 17m 16s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/de74f13d6cb84defbcdeb740ce2d0a4f : SUCCESS in 22m 09s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f23fcb5708f4476ac5852963e0d76e3 : SUCCESS in 16m 50s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/66a4666e1d964a5fa5eebaf755363e62 : SUCCESS in 17m 15s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f640fb6b318421da397c7829e9dbe05 : SUCCESS in 21m 23s","accounts_in_message":[],"_revision_number":3},{"id":"ff8240cabb36917232a5d2f8558c33444938da01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-18 08:39:30.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"050239ae154651c34e3510aece7ac4ff67c5e2e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 10:29:34.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d771ae2a259548f6b1504bcccfda1423 : SUCCESS in 9m 10s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/c7dcbfe83970462dab3ab43d6d7c68f1 : SUCCESS in 9m 13s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/59164993f790424f98e67f5a65e36137 : SUCCESS in 15m 20s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9b8a4ffdcbb640d1923f67c37b75a44c : SUCCESS in 11m 06s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6aba059318c741d48bb89b0562fea52e : SUCCESS in 14m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65b65fe7efa54ecea2c67b4cf2db85a3 : SUCCESS in 7m 04s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/eedfe186575e4bcc9a77e6173a193d13 : FAILURE in 1h 42m 20s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/e960056fb9884cd5a427b18e7aeabfa5 : SUCCESS in 1h 09m 16s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c31359fc769423f879752a2b03e17ff : SUCCESS in 1h 02m 54s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/c1cfeaf9dae84b5e84cdaead11f942d0 : SUCCESS in 39m 10s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2643601e48824ce69cba6158907ef957 : SUCCESS in 14m 30s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fac7afc7bf56427eb8af4710eaa9b7e7 : SUCCESS in 17m 56s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b67bb991b71a4dc9ad9858c35b42b106 : SUCCESS in 17m 12s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/969254f000424b29a07366294159ffd8 : SUCCESS in 20m 12s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9d2c8f968e09456e9b35d08140e895d7 : SUCCESS in 16m 49s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a73ab83c2224d988fce789500ca533a : SUCCESS in 20m 11s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ae2c39ca172446cb966ff776df1fbcb6 : SUCCESS in 20m 31s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/37be880d32dc4a20871fab82326c0e2a : SUCCESS in 17m 15s","accounts_in_message":[],"_revision_number":4},{"id":"1dd78eefdef643d4d9b3884a14112f4604f8d6f6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-22 23:20:20.000000000","message":"Topic set to osa-stream","accounts_in_message":[],"_revision_number":4},{"id":"d784ff5c5e9f1e226638927df1d3fb77cb886b27","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-23 10:42:32.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ac64f4d7aa991c6f45376a802f3d77a30b705392","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-02 08:54:35.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"71c7eed3b55105c6a614fbfc73bbee8260ac9239","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 08:54:57.000000000","message":"Patch Set 5: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"b0edceee83cf3afbbeb76c3188465e5e7fb0d004","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-02 22:49:01.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2cc436327d68d33335fd14db2ffc40b70b9a70d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 01:05:58.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4790ce35f05f4c7b9c76e425879012dc : SUCCESS in 7m 59s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/92936ed247ea48389f16056d600a9ae9 : SUCCESS in 9m 51s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/a09228f1f3624fb7930ad363f3e72c78 : SUCCESS in 15m 00s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7e332469cfdf4a9ba13703e15889d916 : SUCCESS in 12m 53s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f808f54fab64cbca9ac1b9f73f3816d : SUCCESS in 12m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74eee90094c74d0692b7c95219329704 : SUCCESS in 5m 35s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/201edf8fe43540da87e64817e120b82d : SUCCESS in 2h 13m 43s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/e56fed1f9e4340f1b433845eb978c3d8 : FAILURE in 53m 01s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56c6580921a64108a9ec44dc11008e9b : SUCCESS in 1h 02m 50s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/9dfe6019341c4a84b377f50abbf91bb8 : SUCCESS in 40m 07s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7f1947f3965548bdb2176356a5f97e16 : SUCCESS in 16m 06s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/505b3ce97f5f42838000b036d420b4e9 : SUCCESS in 14m 32s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4e33acbe19d24af5a4ede23b888a4ce9 : SUCCESS in 17m 29s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/851d1121e0b44440b5d2e122bd55e86a : SUCCESS in 17m 56s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/916f847e3cdd424fbdf71dd5af73fed0 : SUCCESS in 16m 51s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95a3501af9734876b792bdce8dc63fd3 : SUCCESS in 17m 29s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cd4ab22e37bc4e1fa1150ab626cad874 : SUCCESS in 18m 52s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dde49fefef274f4c8086adebde731893 : SUCCESS in 17m 44s","accounts_in_message":[],"_revision_number":6},{"id":"cf9cb54963a141a04b29a929c084605ecf67cd2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-04 14:50:23.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"f0c832c7887f73c5a493270631dad493db0c4adc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 17:32:58.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3fa16c84938e49a5a6000cca0392affb : SUCCESS in 14m 18s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/7b0b75c679c142c39077eb2ec2059c30 : SUCCESS in 15m 44s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/41406a55e9db478b9991e05497f57ea6 : SUCCESS in 18m 31s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/30575b021a19453f8e40dae6d278f488 : SUCCESS in 16m 17s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a212e4b6135d4c2fad6666974ac41daa : SUCCESS in 13m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4128675458604982b14067615d821d6c : SUCCESS in 8m 41s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/ad62d1bc9efa4be8b1402495fad1d0a6 : SUCCESS in 2h 12m 36s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/b7cf7524bc7a4823a131ecf5bbc890b6 : SUCCESS in 57m 43s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/839713f7a52940fdb9059fff90eaae9e : SUCCESS in 1h 05m 45s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/d62b871c034c4470877d7e4182f3a928 : SUCCESS in 41m 38s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1723963e1b4346a19c0abd46e039e191 : SUCCESS in 19m 19s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b317704dd7c940f08b9d2c011c8a59ff : SUCCESS in 19m 17s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/526299803756477c9deacb16091efe49 : SUCCESS in 21m 10s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8d0c4891dec400a8f3995a4a22f2954 : SUCCESS in 20m 34s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/29c578f5e50e43e498178aa9e1f70a8e : SUCCESS in 19m 48s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e90f09957a1d4a5f8055354425f6679a : SUCCESS in 28m 49s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e82d835f569a4d768740e0a3136f7d28 : SUCCESS in 18m 23s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0c4d1a18384463cb11b0740745acab4 : SUCCESS in 21m 02s","accounts_in_message":[],"_revision_number":7},{"id":"686a5ecddcfbed68444a58e9cf54794538c34a22","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-04 18:20:22.000000000","message":"Patch Set 7: Workflow-1","accounts_in_message":[],"_revision_number":7},{"id":"8aaf40bdfbc1bd312c7b99fae7ada5163392d062","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-04-15 09:33:33.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"97ebcf38a6f02b54d3e1b48a916eaa42e38385f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 10:43:17.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4850bc1ed5a24c93a9d28809518fd300 : SUCCESS in 9m 23s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/f2af1d901a654f28b07158db1d32f912 : SUCCESS in 12m 13s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/deb61fa38ea345f28655d9ab0c26f837 : FAILURE in 12m 43s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b4769fa1957d403b8e4bd1e72bc5784e : SUCCESS in 13m 09s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96998f775a5b480fb695328c80eaf9ee : SUCCESS in 12m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb92f17fa63f4a839b628e219977ec19 : SUCCESS in 5m 41s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/ddd8dda3c69d493bb060cf2552148db6 : FAILURE in 19m 12s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/9f7852a0c9994dc3b52bb4ca3ef57ab2 : SUCCESS in 1h 03m 01s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a2d99d37fb99423c8fc0e7ba11806cc9 : SUCCESS in 59m 57s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/7f5b054fbb55464eb7fef1efe5e48031 : FAILURE in 14m 33s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b594f79aab734e5a9bc1a8946c8b86bf : FAILURE in 10m 01s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1fe418d43d64ee896b213d048b8c138 : FAILURE in 9m 30s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/15e12d70e1cf483592ececfc6ee9fdab : FAILURE in 10m 27s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/38350f72372b4dca99c5c8f07e783777 : FAILURE in 9m 50s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/64a814968db243efb46eb4394baed895 : FAILURE in 9m 43s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/73d1ecca41904877ba9f4a21266f829a : FAILURE in 13m 11s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c8d276ee2e51465b93aa550c0164cb34 : FAILURE in 10m 10s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25a0a170b81c42ed98c3b978ffe053a3 : FAILURE in 10m 26s","accounts_in_message":[],"_revision_number":8},{"id":"b61ee6dc0bf621e5850e5944be934f184ad029e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-04-15 11:06:16.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"4bd7112aee7c9eb8f6a14a537e812026e37a5456","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-04-15 12:06:28.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"d55c53ed0a03f68671a3c90e4473b4eff3e9e59b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 13:29:14.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/16dcae5652834cd1b88dd48d3aca7762 : SUCCESS in 10m 04s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/5868b62306ec44748cf30e0c43d7ae0d : SUCCESS in 12m 43s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/3078f7d62f7e4ed6816157357d1a163f : SUCCESS in 13m 50s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e39ca36147404149b695c6ae27d570f1 : SUCCESS in 12m 08s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d846750930f42728ee38c36d9935bc3 : SUCCESS in 13m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c1657c68963447a90984c5fd8adc8f3 : SUCCESS in 6m 14s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/4b073cbb137844d6a99d475a3112a1d9 : SUCCESS in 1h 06m 27s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/8af814a04d4a44fe8bb3590633c53edd : SUCCESS in 1h 01m 36s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6dc40583f1b44de0b642057f816ca51f : SUCCESS in 1h 16m 27s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/a6b58bfb610543318739b1e1b7f170c5 : FAILURE in 11m 48s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bd1ea0a11c594f10806e361ca4ff5ee3 : FAILURE in 11m 46s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2186980c79bb46b78f552090dae79a2c : FAILURE in 9m 58s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/60169d267a9744e5bd41a42257c291cb : FAILURE in 11m 25s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d17c465d4fd43ad98dc10657bfb322e : FAILURE in 11m 16s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aa0655bf278547e487bbc8c763f9368f : FAILURE in 9m 49s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d3bb37dce8a49e681328d2341b4c8d2 : FAILURE in 13m 01s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/13c106adfc9d4644a0289a29c732de23 : FAILURE in 9m 28s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd20acb50f7b4b329150fe0906a9a2c4 : FAILURE in 10m 18s","accounts_in_message":[],"_revision_number":10},{"id":"d9102eefcd39e3d42d3f44e05f109b952d9fb4cf","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-17 16:46:53.000000000","message":"Patch Set 10:\n\nAh, I guess it just needs https://review.opendev.org/c/openstack/openstack-ansible-tests/+/786396 as well","accounts_in_message":[],"_revision_number":10},{"id":"a0cd90c6803350436414c52b6e508408da3b3a88","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-04-17 21:25:14.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"d30ed9d9c90f8aaaef2a448289208087297dcfa3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-17 22:43:44.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7351726440c949908adfb8f89566cb89 : SUCCESS in 10m 04s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/64c1bc53ba214163830e5b279718a564 : SUCCESS in 10m 51s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/ade57a9487b54f68902336c02e1b786b : SUCCESS in 13m 13s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3815c5e22d534b7d8344981be9afb787 : SUCCESS in 15m 10s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c2f8dbc2a16440e8c0029e67e9995d1 : SUCCESS in 13m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e1f305363ec46efb98e8c119be3f14b : SUCCESS in 5m 44s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/de562f68de864782932c1699675aa268 : SUCCESS in 1h 11m 21s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/adc979d286e448d3a3bec74dabc688db : SUCCESS in 1h 01m 28s (non-voting)\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/983f2b23784040ff80ab2083ff6126f7 : SUCCESS in 1h 02m 20s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/d0c97302a2904f8293e2e0d3fb9fb313 : SUCCESS in 23m 06s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cf5112b8087a4dad81cad10069c9df5d : SUCCESS in 17m 37s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba632595663443a6bfb9af2176a34276 : SUCCESS in 18m 14s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1ac6e49a2358479c8f70b50838046e54 : SUCCESS in 16m 47s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0fb9ceed7d3d4367980ea5e8c40ef76c : SUCCESS in 18m 18s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b260a2d2a65e4af082ed797bc4702041 : SUCCESS in 17m 24s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ed56c5be3cb4654af09292a50193f3d : SUCCESS in 18m 19s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4701114964b24d0ba104d9b1d61016d0 : SUCCESS in 19m 25s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c848a2f19a474962aeb02904a2eeb1c7 : SUCCESS in 17m 56s","accounts_in_message":[],"_revision_number":10},{"id":"7177a7e1eca80cdf3c732fc4270fb780a3b90ec2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-18 07:47:06.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"93d135b3acf6ea316a12cbe0a9102727e711ed74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-04-18 12:16:09.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"50804a272bca4b8235cdccaeea93d0b8e7a140e5","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-04-18 12:17:36.000000000","message":"Patch Set 11:\n\nYes we should make this patch tidy as we\u0027ll probably need to backport this if we decide to switch any stable branches from centos-8 to centos-8 stream, this is the smallest necessary change for image building just for centos.","accounts_in_message":[],"_revision_number":11},{"id":"f2f032130584658bf17e9baa026d3e96373032da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-18 13:32:01.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8652ccc3f40e4af79dc2d16d8d2e8820 : SUCCESS in 11m 12s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/22dafe1af6cb4e34bec6ec237d9537a3 : SUCCESS in 11m 52s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/c5b7ab7c8d6e4d1f9343ca2f0b3ab5ad : SUCCESS in 13m 27s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9b6d797d592540199f67ce08db9888d5 : SUCCESS in 13m 03s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e79dffa46abb4d2798771ece9b0b8778 : SUCCESS in 12m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6abcd1c8e8934ad9a567d86ccbe59436 : SUCCESS in 5m 18s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/3e56c7ebee434159acedfecd16e3ed87 : SUCCESS in 1h 08m 52s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/cc2a1aa36dde4634a9fbd2d3feca91c8 : SUCCESS in 1h 01m 47s (non-voting)\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bcb2aabc7f8a464ab0231207e486c223 : SUCCESS in 1h 02m 14s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/18169cbb91b8454b910115a4f385515d : SUCCESS in 21m 09s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0c0090010b594131bf500edaba0ddd86 : SUCCESS in 16m 53s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9fb24ef6427e415ea90ff8b284d8f976 : SUCCESS in 16m 06s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e55d4fdb2d3c442d8e40c6400268322f : SUCCESS in 18m 00s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1aab4c8d66e946f989349ed1aaedce48 : SUCCESS in 16m 06s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5896d8133a074d5b92497ec49f5a638f : SUCCESS in 16m 44s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/200877bdf5b04776b4c96c7866181d41 : SUCCESS in 17m 42s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1055a678059b491f9be670e496453bc2 : SUCCESS in 16m 40s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db54106a50964e958e3f40e9dfa637eb : SUCCESS in 17m 21s","accounts_in_message":[],"_revision_number":11},{"id":"11199f3a3a6cad0a5b5d05b414ca35273bbe4e98","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-18 14:48:44.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"9da8b32755ebdc500fd667679a9bafbb5d9c3290","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-04-18 15:53:45.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"7c47aa96e9e0082cb66a334efd9cb1ff916a603c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-04-18 15:54:22.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"f1eedb3c4cf301b903d8320d3880340ea3660734","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-18 16:38:30.000000000","message":"Patch Set 13:\n\n(3 comments)","accounts_in_message":[],"_revision_number":13},{"id":"bf7f004d11dd24f8d20d68f7a570a7405e9b1b36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-04-18 16:58:42.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"20ed021be38d7a3507b88370925e88df49d1eea8","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-04-18 16:58:56.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"d1f117f69de01b69195481529c56270b0d311bfe","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-18 17:12:00.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"6c4124c6719533135e4abed04454e944b37f88e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-18 18:08:27.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9e1440cc12644f5388f06b35b7b770dd : SUCCESS in 9m 49s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/3026671892be4816aec577818b38b179 : SUCCESS in 12m 26s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/ea7c215c7eec401d96ba38491ed0134f : SUCCESS in 20m 02s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4a5b65bcbef14406b8239a0adab5afd6 : SUCCESS in 20m 14s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/929dea56091b40709db860c490f6d48b : SUCCESS in 13m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0495f6d1d98a4878a542ae7be1d15e9a : SUCCESS in 5m 39s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/1f870d25b7e44e79b80541133b808fe0 : SUCCESS in 53m 14s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/66fdc569007449aaa54444208691de1c : FAILURE in 1h 02m 22s (non-voting)\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce253a9cbe174b34a2e98ae6bf8ea1dc : SUCCESS in 1h 06m 44s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/1181f0506c354a9d9cdaf21bd59bdacc : SUCCESS in 32m 29s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8abcd4a8cdaa4b0480a34c286f800977 : SUCCESS in 21m 15s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61234a1bac9f46aaba2ef702ea2dbfaa : SUCCESS in 20m 35s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/de91f8c6606a4197814a6e48c3e792bf : SUCCESS in 19m 35s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52e8b610fb914f6ea94e7ff4075688dd : SUCCESS in 26m 36s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b869ef98681b4a3c88fde188ed21d46e : SUCCESS in 23m 09s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ad5a39e8bc947c29aa4a565971a9af7 : SUCCESS in 20m 47s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6f7043d1dc8b41088c7d2f7998011149 : SUCCESS in 19m 17s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f87c909cb0a64e67babdeaa8bef3d035 : SUCCESS in 18m 53s","accounts_in_message":[],"_revision_number":14},{"id":"ca74941a7daf7397f8096f56f78c5f7f9c2c90ee","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-04-18 19:26:08.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"21280f4f0dc55b74ba9c23a513334b9604a52357","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-18 20:28:16.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"d8d8b9cbe977949fca6b82d254c224db0db9c816","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-04-19 06:39:07.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"5d1780ecefa2e9a5b0065e629e10e8ad5f561183","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-04-19 06:45:08.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"da3c9c341da686361645671a8ec443ba7160a5c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-04-19 06:49:31.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"40105520af4ee5198fd621c5ea5a1847443910db","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-19 07:23:17.000000000","message":"Patch Set 17: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"2c400357af56c1957f70432f02c4659f29b1ea51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-04-19 07:37:34.000000000","message":"Patch Set 18: Published edit on patch set 17.","accounts_in_message":[],"_revision_number":18},{"id":"e3a01bffe0f35f31806b02f2dc6300a2047cf609","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-04-19 07:37:51.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"cba738ee4ee6bacab9856a04e9845cb95bb8201a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 08:50:59.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4c2dd9e3319541b5bef4114642567245 : SUCCESS in 7m 39s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/974798122b8b4143a966d11b8dc40d12 : SUCCESS in 12m 42s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/361d162b97d3444e9f5ca898b1dfcc16 : SUCCESS in 14m 12s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1af6b91030d045c7aa8555fbd9e7bfd7 : SUCCESS in 12m 42s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a2e6e718bae44b61bd40b49789953caf : SUCCESS in 13m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba44ea4dfffa4974b71721563a194798 : SUCCESS in 6m 25s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/9fc355dc869648adab0cf73a938e6404 : SUCCESS in 1h 10m 34s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/e4eb15ef7f1043119d93c90372d32ece : SUCCESS in 1h 04m 30s (non-voting)\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b4b54f7e7c124606bf6ddead631689a0 : SUCCESS in 1h 06m 54s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/17252e1f7c3548d2a58c9960302522d1 : SUCCESS in 21m 15s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3a2147cde5324ee098a393a9c7c44e79 : SUCCESS in 17m 55s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a85ec02eedc54af49aaa528dc9e2a477 : SUCCESS in 17m 51s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1955d6765dc64b138416f09431327d03 : SUCCESS in 17m 48s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1c19d697a7b4ae791c6f52bb95efc7e : SUCCESS in 18m 35s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9429818c2a26423085aa6f995abe9f5d : SUCCESS in 17m 42s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/591826c26399402eabb7adbd59b9c1f7 : SUCCESS in 17m 43s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/19b1035d856040d99bce8c3e1ef4688c : SUCCESS in 18m 35s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dfe9783be70c4550814b61bf4fcffb1f : SUCCESS in 18m 46s","accounts_in_message":[],"_revision_number":18},{"id":"a5a838961cbcd149f67102d67fa87288a791512b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-19 14:47:09.000000000","message":"Patch Set 18: Code-Review+2","accounts_in_message":[],"_revision_number":18},{"id":"c9bc29cd5cf84ee5724de73d7e6144477f3cacd5","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-04-20 08:51:39.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"21ec539d2b0abe201a36d025bd0a1dffb2101756","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-04-20 08:56:26.000000000","message":"Patch Set 19: Published edit on patch set 18.","accounts_in_message":[],"_revision_number":19},{"id":"5d532bfeb8a0941eab3c485816d1d222a52c718f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-04-20 08:56:44.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"108fb24915a6d8fbda3d8f5ff35e31090bba0e6c","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-04-20 08:57:45.000000000","message":"Patch Set 19: Code-Review+2","accounts_in_message":[],"_revision_number":19},{"id":"48a89a8308346ccd47f02b074437fe4a3f5f9189","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-20 08:59:43.000000000","message":"Patch Set 19: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"691731a6937f1985bbe0882d5fa4ba859ddff29f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 10:16:35.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/687c8b43e851465b85af712109b35834 : SUCCESS in 9m 41s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/a922931254834d399ef4c1d13a417da6 : SUCCESS in 11m 10s (non-voting)\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/932f8d789ccf4e31a79e6e75026d7202 : SUCCESS in 14m 45s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2926b29d731344e2bc7a8048a75aad3b : SUCCESS in 12m 42s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee096beae05445529b145722473a865c : SUCCESS in 12m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b25d95c14d894f609f62dacb1d673417 : SUCCESS in 8m 26s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/0d3c895bce0f484ca1f5931ed84de448 : SUCCESS in 1h 09m 17s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/ca3333ef8565440da6c9a4b7367f67b1 : SUCCESS in 1h 09m 04s (non-voting)\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab1353724a9b4d81b90f48d142ec496b : SUCCESS in 1h 07m 54s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/5bcfceba5067407c982838e53701b5a9 : SUCCESS in 22m 02s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1dd0de383bf04c3a84f930573b756830 : SUCCESS in 19m 18s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1e317da6e314a03a6f90f6786e8bf3f : SUCCESS in 14m 51s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/efb68b7384ce435fb4a0840199103fac : SUCCESS in 20m 39s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/347febbfe858498abaf1b013806b18c9 : SUCCESS in 16m 54s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7d7a5839affa4191a7cac397d7caa271 : SUCCESS in 23m 54s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5751c6ad314d4445992e8e4ed203d357 : SUCCESS in 19m 43s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aab48c99558a48888b642999af53812c : SUCCESS in 19m 35s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ee7b22a27914579bc9f1efe58fa794d : SUCCESS in 20m 47s","accounts_in_message":[],"_revision_number":19},{"id":"95e220671c452320b9eeae958f678b2a4a85fb01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 10:16:52.000000000","message":"Patch Set 19: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":19},{"id":"35ae75ca5332a3d415c9db897aa4844b5485499e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 11:34:47.000000000","message":"Patch Set 19: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/89e04991e1f347e58ca566bb9ae666e1 : SUCCESS in 9m 44s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a880c7a31dde4d01a5070d9afeae5730 : SUCCESS in 13m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e80d89cc2cee42caa9cd00d17992610b : SUCCESS in 5m 51s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/d18b3193f5ef4eea92f2e0c84ce7361c : SUCCESS in 1h 10m 17s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e833576922fe446f900bc631191cecd7 : SUCCESS in 1h 04m 01s","accounts_in_message":[],"_revision_number":19},{"id":"8ad65c81165ccd5b7ae1d19518b67059613c3f65","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 11:34:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":19},{"id":"500752d425caac00d4c18651e35ffdac3bf01c8b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 11:36:49.000000000","message":"Patch Set 19:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0edd3b703dfc4b869149396be1c0fee5 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":19}],"current_revision_number":19,"current_revision":"791b9e813ed29c1187803c53ad46e369b24eaad4","revisions":{"ffce7c1b5d4b91f90177ffa7a80e6a97e17c1228":{"kind":"REWORK","_number":1,"created":"2021-02-17 13:50:21.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/01/776201/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/01/776201/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/1"}}},"commit":{"parents":[{"commit":"608254f0a136eb0de4abab904716ae84418b48d8","subject":"Remove duplicate lxc volume tasks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/608254f0a136eb0de4abab904716ae84418b48d8"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 13:48:13.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 13:48:13.000000000","tz":0},"subject":"WIP - Build centos 8 container image with dnf rather than download","message":"WIP - Build centos 8 container image with dnf rather than download\n\nThis change uses dnf to build the container image with dnf locally\nrather than rely on an external image that is downloaded and\nunpacked.\n\nThe image prestage commands are made conditional, and an operating\nsystem specific command can be provided via role variables to\nbuild a chroot in /var/lib/machines.\n\nChange-Id: I1ddfe36259610b25e86b69d64d1d7f32a56c0e4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ffce7c1b5d4b91f90177ffa7a80e6a97e17c1228"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ffce7c1b5d4b91f90177ffa7a80e6a97e17c1228"}]},"branch":"refs/heads/master"},"ec6c367c95dd51a207ec5603840b0307fa05ff84":{"kind":"REWORK","_number":2,"created":"2021-02-17 15:46:43.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/01/776201/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/01/776201/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/2"}}},"commit":{"parents":[{"commit":"608254f0a136eb0de4abab904716ae84418b48d8","subject":"Remove duplicate lxc volume tasks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/608254f0a136eb0de4abab904716ae84418b48d8"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 13:48:13.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 15:46:37.000000000","tz":0},"subject":"WIP - Build centos 8 container image with dnf rather than download","message":"WIP - Build centos 8 container image with dnf rather than download\n\nThis change uses dnf to build the container image with dnf locally\nrather than rely on an external image that is downloaded and\nunpacked.\n\nThe image prestage commands are made conditional, and an operating\nsystem specific command can be provided via role variables to\nbuild a chroot in /var/lib/machines.\n\nChange-Id: I1ddfe36259610b25e86b69d64d1d7f32a56c0e4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ec6c367c95dd51a207ec5603840b0307fa05ff84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ec6c367c95dd51a207ec5603840b0307fa05ff84"}]},"branch":"refs/heads/master"},"34a940c9433fa82ca8c4c74322cc90edc9ee85af":{"kind":"REWORK","_number":3,"created":"2021-02-17 17:11:31.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/01/776201/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/01/776201/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/3"}}},"commit":{"parents":[{"commit":"608254f0a136eb0de4abab904716ae84418b48d8","subject":"Remove duplicate lxc volume tasks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/608254f0a136eb0de4abab904716ae84418b48d8"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 13:48:13.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 17:11:28.000000000","tz":0},"subject":"WIP - Build centos 8 container image with dnf rather than download","message":"WIP - Build centos 8 container image with dnf rather than download\n\nThis change uses dnf to build the container image with dnf locally\nrather than rely on an external image that is downloaded and\nunpacked.\n\nThe image prestage commands are made conditional, and an operating\nsystem specific command can be provided via role variables to\nbuild a chroot in /var/lib/machines.\n\nDepends-On: I74f02669b013b8580d3469a8ffe214d88cd0f525\nChange-Id: I1ddfe36259610b25e86b69d64d1d7f32a56c0e4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/34a940c9433fa82ca8c4c74322cc90edc9ee85af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/34a940c9433fa82ca8c4c74322cc90edc9ee85af"}]},"branch":"refs/heads/master"},"f76a593b899c0565a7729e957ed23b4ba8e53d2c":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-02-18 08:39:30.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/01/776201/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/01/776201/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/4"}}},"commit":{"parents":[{"commit":"608254f0a136eb0de4abab904716ae84418b48d8","subject":"Remove duplicate lxc volume tasks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/608254f0a136eb0de4abab904716ae84418b48d8"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 13:48:13.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-18 08:39:30.000000000","tz":0},"subject":"WIP - Build centos 8 container image with dnf rather than download","message":"WIP - Build centos 8 container image with dnf rather than download\n\nThis change uses dnf to build the container image with dnf locally\nrather than rely on an external image that is downloaded and\nunpacked.\n\nThe image prestage commands are made conditional, and an operating\nsystem specific command can be provided via role variables to\nbuild a chroot in /var/lib/machines.\n\nDepends-On: I31880ca995735b737d33532eaa4c29be02523117\nDepends-On: I74f02669b013b8580d3469a8ffe214d88cd0f525\nChange-Id: I1ddfe36259610b25e86b69d64d1d7f32a56c0e4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f76a593b899c0565a7729e957ed23b4ba8e53d2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f76a593b899c0565a7729e957ed23b4ba8e53d2c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"70f336670246ca232f62a2beb00bfe202439621d":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2021-03-02 08:54:35.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/01/776201/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/01/776201/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/5"}}},"commit":{"parents":[{"commit":"608254f0a136eb0de4abab904716ae84418b48d8","subject":"Remove duplicate lxc volume tasks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/608254f0a136eb0de4abab904716ae84418b48d8"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 13:48:13.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-03-02 08:53:39.000000000","tz":0},"subject":"Optionally build container image as a local chroot","message":"Optionally build container image as a local chroot\n\nThis change uses dnf to build the container image fr Centos-8\nusing dnf locally rather than rely on an external image that is\ndownloaded and unpacked. This is because there is no container\nimage available for Centos-8 Stream and rather than rely on an\nexternal image, we can create a rootfs locally.\n\nThe existing image prestage commands are made conditional, and\nan operating system specific command can be provided via role\nvariables to build a chroot in /var/lib/machines.\n\nDepends-On: I31880ca995735b737d33532eaa4c29be02523117\nDepends-On: I74f02669b013b8580d3469a8ffe214d88cd0f525\nChange-Id: I1ddfe36259610b25e86b69d64d1d7f32a56c0e4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/70f336670246ca232f62a2beb00bfe202439621d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/70f336670246ca232f62a2beb00bfe202439621d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"dd4eefe673b7dbf95a3d8c582ee562df986278d5":{"kind":"REWORK","_number":6,"created":"2021-03-02 22:49:01.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/01/776201/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/01/776201/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/6"}}},"commit":{"parents":[{"commit":"ba1788dd2f884e91bb09c5fc073b8ed954bf307f","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ba1788dd2f884e91bb09c5fc073b8ed954bf307f"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 13:48:13.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-03-02 22:48:29.000000000","tz":0},"subject":"Optionally build container image as a local chroot","message":"Optionally build container image as a local chroot\n\nThis change uses dnf to build the container image for Centos-8\nusing dnf locally rather than rely on an external image that is\ndownloaded and unpacked.\n\nThe existing image prestage commands are made conditional, and\nan operating system specific command can be provided via role\nvariables to build a chroot in /var/lib/machines.\n\nDepends-On: I31880ca995735b737d33532eaa4c29be02523117\nDepends-On: I74f02669b013b8580d3469a8ffe214d88cd0f525\nChange-Id: I1ddfe36259610b25e86b69d64d1d7f32a56c0e4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/dd4eefe673b7dbf95a3d8c582ee562df986278d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/dd4eefe673b7dbf95a3d8c582ee562df986278d5"}]},"branch":"refs/heads/master"},"427656717a7739d345ab9da30a2cd94bc4827818":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-03-04 14:50:23.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/01/776201/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/01/776201/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/7"}}},"commit":{"parents":[{"commit":"6de26d851958f3ff60ee338e1ca92747362afde7","subject":"Use ansible_facts[] instead of fact variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6de26d851958f3ff60ee338e1ca92747362afde7"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 13:48:13.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-03-04 14:48:27.000000000","tz":0},"subject":"Optionally build container image as a local chroot","message":"Optionally build container image as a local chroot\n\nThis change uses dnf to build the container image for Centos-8\nusing dnf locally rather than rely on an external image that is\ndownloaded and unpacked.\n\nThe existing image prestage commands are made conditional, and\nan operating system specific command can be provided via role\nvariables to build a chroot in /var/lib/machines.\n\nDepends-On: I31880ca995735b737d33532eaa4c29be02523117\nDepends-On: I74f02669b013b8580d3469a8ffe214d88cd0f525\nChange-Id: I1ddfe36259610b25e86b69d64d1d7f32a56c0e4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/427656717a7739d345ab9da30a2cd94bc4827818"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/427656717a7739d345ab9da30a2cd94bc4827818"}]},"branch":"refs/heads/master"},"64f13d829062f7987a42438d2bec28dafdbeb739":{"kind":"REWORK","_number":8,"created":"2021-04-15 09:33:33.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/01/776201/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/01/776201/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/8"}}},"commit":{"parents":[{"commit":"6de26d851958f3ff60ee338e1ca92747362afde7","subject":"Use ansible_facts[] instead of fact variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6de26d851958f3ff60ee338e1ca92747362afde7"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 13:48:13.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-04-15 09:30:30.000000000","tz":60},"subject":"Optionally build container image as a local chroot","message":"Optionally build container image as a local chroot\n\nThis change uses dnf to build the container image for Centos-8\nusing dnf locally rather than rely on an external image that is\ndownloaded and unpacked.\n\nThe existing image prestage commands are made conditional, and\nan operating system specific command can be provided via role\nvariables to build a chroot in /var/lib/machines.\n\nDuring the transition from Centos-8 to Centos-8-Stream, the\nvars files are separated, with vars/redhat.yml covering Stream,\nand vars/redhat-8.3.yml covering legcay Centos-8.\n\nIn addition, the systemd-logind service is masked from the base\nimage. This is masked in the previously downloaded container base\nimage, so we ensure that the same is done for locally built chroots.\n\nDepends-On: I31880ca995735b737d33532eaa4c29be02523117\nDepends-On: I74f02669b013b8580d3469a8ffe214d88cd0f525\nChange-Id: I1ddfe36259610b25e86b69d64d1d7f32a56c0e4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/64f13d829062f7987a42438d2bec28dafdbeb739"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/64f13d829062f7987a42438d2bec28dafdbeb739"}]},"branch":"refs/heads/master"},"149b17e49e823457760aea44653fb0872c59ea28":{"kind":"REWORK","_number":9,"created":"2021-04-15 11:06:16.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/01/776201/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/01/776201/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/9"}}},"commit":{"parents":[{"commit":"6de26d851958f3ff60ee338e1ca92747362afde7","subject":"Use ansible_facts[] instead of fact variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6de26d851958f3ff60ee338e1ca92747362afde7"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 13:48:13.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-04-15 11:05:47.000000000","tz":60},"subject":"Optionally build container image as a local chroot","message":"Optionally build container image as a local chroot\n\nThis change uses dnf to build the container image for Centos-8\nusing dnf locally rather than rely on an external image that is\ndownloaded and unpacked.\n\nThe existing image prestage commands are made conditional, and\nan operating system specific command can be provided via role\nvariables to build a chroot in /var/lib/machines.\n\nDuring the transition from Centos-8 to Centos-8-Stream, the\nvars files are separated, with vars/redhat.yml covering Stream,\nand vars/centos-8.3.yml covering legcay Centos-8.\n\nIn addition, the systemd-logind service is masked from the base\nimage. This is masked in the previously downloaded container base\nimage, so we ensure that the same is done for locally built chroots.\n\nDepends-On: I31880ca995735b737d33532eaa4c29be02523117\nDepends-On: I74f02669b013b8580d3469a8ffe214d88cd0f525\nChange-Id: I1ddfe36259610b25e86b69d64d1d7f32a56c0e4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/149b17e49e823457760aea44653fb0872c59ea28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/149b17e49e823457760aea44653fb0872c59ea28"}]},"branch":"refs/heads/master"},"2de86a014c7386714d0442ce75b79b06f90ff0e6":{"kind":"REWORK","_number":10,"created":"2021-04-15 12:06:28.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/01/776201/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/01/776201/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/10"}}},"commit":{"parents":[{"commit":"6de26d851958f3ff60ee338e1ca92747362afde7","subject":"Use ansible_facts[] instead of fact variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6de26d851958f3ff60ee338e1ca92747362afde7"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 13:48:13.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-04-15 12:06:23.000000000","tz":0},"subject":"Optionally build container image as a local chroot","message":"Optionally build container image as a local chroot\n\nThis change uses dnf to build the container image for Centos-8\nusing dnf locally rather than rely on an external image that is\ndownloaded and unpacked.\n\nThe existing image prestage commands are made conditional, and\nan operating system specific command can be provided via role\nvariables to build a chroot in /var/lib/machines.\n\nDuring the transition from Centos-8 to Centos-8-Stream, the\nvars files are separated, with vars/redhat.yml covering Stream,\nand vars/centos-8.3.yml covering legcay Centos-8.\n\nIn addition, the systemd-logind service is masked from the base\nimage. This is masked in the previously downloaded container base\nimage, so we ensure that the same is done for locally built chroots.\n\nDepends-On: I31880ca995735b737d33532eaa4c29be02523117\nDepends-On: I74f02669b013b8580d3469a8ffe214d88cd0f525\nChange-Id: I1ddfe36259610b25e86b69d64d1d7f32a56c0e4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2de86a014c7386714d0442ce75b79b06f90ff0e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2de86a014c7386714d0442ce75b79b06f90ff0e6"}]},"branch":"refs/heads/master"},"3b4194892945f2ee2f8fe93beb81dd806cb636d6":{"kind":"REWORK","_number":11,"created":"2021-04-18 12:16:09.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/01/776201/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/01/776201/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/11"}}},"commit":{"parents":[{"commit":"6de26d851958f3ff60ee338e1ca92747362afde7","subject":"Use ansible_facts[] instead of fact variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6de26d851958f3ff60ee338e1ca92747362afde7"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 13:48:13.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-04-18 12:14:01.000000000","tz":0},"subject":"Optionally build container image as a local chroot","message":"Optionally build container image as a local chroot\n\nThis change uses dnf to build the container image for Centos-8\nusing dnf locally rather than rely on an external image that is\ndownloaded and unpacked.\n\nThe existing image prestage commands are made conditional, and\nan operating system specific command can be provided via role\nvariables to build a chroot in /var/lib/machines.\n\nDuring the transition from Centos-8 to Centos-8-Stream, the\nvars files are separated, with vars/redhat.yml covering Stream,\nand vars/centos-8.3.yml covering legcay Centos-8.\n\nIn addition, the systemd-logind service is masked from the base\nimage. This is masked in the previously downloaded container base\nimage, so we ensure that the same is done for locally built chroots.\n\nDepends-On: I31880ca995735b737d33532eaa4c29be02523117\nDepends-On: I74f02669b013b8580d3469a8ffe214d88cd0f525\nChange-Id: I1ddfe36259610b25e86b69d64d1d7f32a56c0e4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3b4194892945f2ee2f8fe93beb81dd806cb636d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3b4194892945f2ee2f8fe93beb81dd806cb636d6"}]},"branch":"refs/heads/master"},"3df27b07f59d4367369ff385b8296d296dc98beb":{"kind":"REWORK","_number":12,"created":"2021-04-18 15:53:45.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/01/776201/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/01/776201/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/12"}}},"commit":{"parents":[{"commit":"6de26d851958f3ff60ee338e1ca92747362afde7","subject":"Use ansible_facts[] instead of fact variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6de26d851958f3ff60ee338e1ca92747362afde7"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 13:48:13.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-04-18 15:53:40.000000000","tz":0},"subject":"Optionally build container image as a local chroot","message":"Optionally build container image as a local chroot\n\nThis change uses dnf to build the container image for Centos-8\nusing dnf locally rather than rely on an external image that is\ndownloaded and unpacked.\n\nThe existing image prestage commands are made conditional, and\nan operating system specific command can be provided via role\nvariables to build a chroot in /var/lib/machines.\n\nDuring the transition from Centos-8 to Centos-8-Stream, the\nvars files are separated, with vars/redhat.yml covering Stream,\nand vars/centos-8.3.yml covering legcay Centos-8.\n\nIn addition, the systemd-logind service is masked from the base\nimage. This is masked in the previously downloaded container base\nimage, so we ensure that the same is done for locally built chroots.\n\nDepends-On: I31880ca995735b737d33532eaa4c29be02523117\nDepends-On: I74f02669b013b8580d3469a8ffe214d88cd0f525\nChange-Id: I1ddfe36259610b25e86b69d64d1d7f32a56c0e4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3df27b07f59d4367369ff385b8296d296dc98beb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3df27b07f59d4367369ff385b8296d296dc98beb"}]},"branch":"refs/heads/master"},"7ecf7ba5a82771cc8becba86431e18c12cbe0949":{"kind":"REWORK","_number":13,"created":"2021-04-18 15:54:22.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/01/776201/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/01/776201/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/13"}}},"commit":{"parents":[{"commit":"6de26d851958f3ff60ee338e1ca92747362afde7","subject":"Use ansible_facts[] instead of fact variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6de26d851958f3ff60ee338e1ca92747362afde7"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 13:48:13.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-04-18 15:54:17.000000000","tz":0},"subject":"Optionally build container image as a local chroot","message":"Optionally build container image as a local chroot\n\nThis change uses dnf to build the container image for Centos-8\nusing dnf locally rather than rely on an external image that is\ndownloaded and unpacked.\n\nThe existing image prestage commands are made conditional, and\nan operating system specific command can be provided via role\nvariables to build a chroot in /var/lib/machines.\n\nDuring the transition from Centos-8 to Centos-8-Stream, the\nvars files are separated, with vars/redhat.yml covering Stream,\nand vars/centos-8.3.yml covering legcay Centos-8.\n\nIn addition, the systemd-logind service is masked from the base\nimage. This is masked in the previously downloaded container base\nimage, so we ensure that the same is done for locally built chroots.\n\nDepends-On: I31880ca995735b737d33532eaa4c29be02523117\nDepends-On: I74f02669b013b8580d3469a8ffe214d88cd0f525\nChange-Id: I1ddfe36259610b25e86b69d64d1d7f32a56c0e4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7ecf7ba5a82771cc8becba86431e18c12cbe0949"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7ecf7ba5a82771cc8becba86431e18c12cbe0949"}]},"branch":"refs/heads/master"},"f12ec395a5c78595cd3011f0a4ddb863097015bf":{"kind":"REWORK","_number":14,"created":"2021-04-18 16:58:42.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/01/776201/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/01/776201/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/14"}}},"commit":{"parents":[{"commit":"6de26d851958f3ff60ee338e1ca92747362afde7","subject":"Use ansible_facts[] instead of fact variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6de26d851958f3ff60ee338e1ca92747362afde7"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 13:48:13.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-04-18 16:54:47.000000000","tz":0},"subject":"Optionally build container image as a local chroot","message":"Optionally build container image as a local chroot\n\nThis change uses dnf to build the container image for Centos-8\nusing dnf locally rather than rely on an external image that is\ndownloaded and unpacked.\n\nThe existing image prestage commands are made conditional, and\nan operating system specific command can be provided via role\nvariables to build a chroot in /var/lib/machines.\n\nDuring the transition from Centos-8 to Centos-8-Stream, the\nvars files are separated, with vars/redhat.yml covering Stream,\nand vars/centos-8.3.yml covering legcay Centos-8.\n\nIn addition, the systemd-logind service is masked from the base\nimage. This is masked in the previously downloaded container base\nimage, so we ensure that the same is done for locally built chroots.\n\nDepends-On: I31880ca995735b737d33532eaa4c29be02523117\nDepends-On: I74f02669b013b8580d3469a8ffe214d88cd0f525\nChange-Id: I1ddfe36259610b25e86b69d64d1d7f32a56c0e4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f12ec395a5c78595cd3011f0a4ddb863097015bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f12ec395a5c78595cd3011f0a4ddb863097015bf"}]},"branch":"refs/heads/master"},"f50777f1a62b39e11d12b532d81f89f15d8a9987":{"kind":"REWORK","_number":15,"created":"2021-04-19 06:39:07.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/01/776201/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/01/776201/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/15"}}},"commit":{"parents":[{"commit":"6de26d851958f3ff60ee338e1ca92747362afde7","subject":"Use ansible_facts[] instead of fact variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6de26d851958f3ff60ee338e1ca92747362afde7"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 13:48:13.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-04-19 06:38:30.000000000","tz":0},"subject":"Optionally build container image as a local chroot","message":"Optionally build container image as a local chroot\n\nThis change uses dnf to build the container image for Centos-8\nusing dnf locally rather than rely on an external image that is\ndownloaded and unpacked.\n\nThe existing image prestage commands are made conditional, and\nan operating system specific command can be provided via role\nvariables to build a chroot in /var/lib/machines.\n\nDuring the transition from Centos-8 to Centos-8-Stream, the\nvars files are separated, with vars/redhat.yml covering Stream,\nand vars/centos-8.3.yml covering legcay Centos-8.\n\nIn addition, the systemd-logind service is masked from the base\nimage. This is masked in the previously downloaded container base\nimage, so we ensure that the same is done for locally built chroots.\n\nDepends-On: I31880ca995735b737d33532eaa4c29be02523117\nDepends-On: I74f02669b013b8580d3469a8ffe214d88cd0f525\nChange-Id: I1ddfe36259610b25e86b69d64d1d7f32a56c0e4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f50777f1a62b39e11d12b532d81f89f15d8a9987"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f50777f1a62b39e11d12b532d81f89f15d8a9987"}]},"branch":"refs/heads/master"},"a5cc9436c8fd337332db8767d1a8fda382d1d54d":{"kind":"REWORK","_number":16,"created":"2021-04-19 06:45:08.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/01/776201/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/01/776201/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/16"}}},"commit":{"parents":[{"commit":"f3bc93d1a310c4d7a3e2142e5b614955cbfbfe96","subject":"Merge \"Remove references to unsupported operating systems\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f3bc93d1a310c4d7a3e2142e5b614955cbfbfe96"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 13:48:13.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-04-19 06:44:26.000000000","tz":60},"subject":"Optionally build container image as a local chroot","message":"Optionally build container image as a local chroot\n\nThis change uses dnf to build the container image for Centos-8\nusing dnf locally rather than rely on an external image that is\ndownloaded and unpacked.\n\nThe existing image prestage commands are made conditional, and\nan operating system specific command can be provided via role\nvariables to build a chroot in /var/lib/machines.\n\nDuring the transition from Centos-8 to Centos-8-Stream, the\nvars files are separated, with vars/redhat.yml covering Stream,\nand vars/centos-8.3.yml covering legcay Centos-8.\n\nIn addition, the systemd-logind service is masked from the base\nimage. This is masked in the previously downloaded container base\nimage, so we ensure that the same is done for locally built chroots.\n\nDepends-On: I31880ca995735b737d33532eaa4c29be02523117\nDepends-On: I74f02669b013b8580d3469a8ffe214d88cd0f525\nChange-Id: I1ddfe36259610b25e86b69d64d1d7f32a56c0e4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a5cc9436c8fd337332db8767d1a8fda382d1d54d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a5cc9436c8fd337332db8767d1a8fda382d1d54d"}]},"branch":"refs/heads/master"},"ccec617bbd87fed31c2d07797c1e7d2282f20264":{"kind":"REWORK","_number":17,"created":"2021-04-19 06:49:31.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/01/776201/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/01/776201/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/17"}}},"commit":{"parents":[{"commit":"f3bc93d1a310c4d7a3e2142e5b614955cbfbfe96","subject":"Merge \"Remove references to unsupported operating systems\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f3bc93d1a310c4d7a3e2142e5b614955cbfbfe96"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 13:48:13.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-04-19 06:46:58.000000000","tz":60},"subject":"Optionally build container image as a local chroot","message":"Optionally build container image as a local chroot\n\nThis change uses dnf to build the container image for Centos-8\nusing dnf locally rather than rely on an external image that is\ndownloaded and unpacked.\n\nThe existing image prestage commands are made conditional, and\nan operating system specific command can be provided via role\nvariables to build a chroot in /var/lib/machines.\n\nDuring the transition from Centos-8 to Centos-8-Stream, the\nvars files are separated, with vars/redhat.yml covering Stream,\nand vars/centos-8.3.yml covering legcay Centos-8.\n\nIn addition, the systemd-logind service is masked from the base\nimage. This is masked in the previously downloaded container base\nimage, so we ensure that the same is done for locally built chroots.\n\nDepends-On: I31880ca995735b737d33532eaa4c29be02523117\nDepends-On: I74f02669b013b8580d3469a8ffe214d88cd0f525\nChange-Id: I1ddfe36259610b25e86b69d64d1d7f32a56c0e4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ccec617bbd87fed31c2d07797c1e7d2282f20264"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ccec617bbd87fed31c2d07797c1e7d2282f20264"}]},"branch":"refs/heads/master"},"13742f908f4dce99bded79399928891773f3ea4b":{"kind":"REWORK","_number":18,"created":"2021-04-19 07:37:34.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/01/776201/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/01/776201/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/18"}}},"commit":{"parents":[{"commit":"f3bc93d1a310c4d7a3e2142e5b614955cbfbfe96","subject":"Merge \"Remove references to unsupported operating systems\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f3bc93d1a310c4d7a3e2142e5b614955cbfbfe96"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 13:48:13.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-04-19 07:37:26.000000000","tz":0},"subject":"Optionally build container image as a local chroot","message":"Optionally build container image as a local chroot\n\nThis change uses dnf to build the container image for Centos-8\nusing dnf locally rather than rely on an external image that is\ndownloaded and unpacked.\n\nThe existing image prestage commands are made conditional, and\nan operating system specific command can be provided via role\nvariables to build a chroot in /var/lib/machines.\n\nDuring the transition from Centos-8 to Centos-8-Stream, the\nvars files are separated, with vars/redhat.yml covering Stream,\nand vars/centos-8.3.yml covering legcay Centos-8.\n\nIn addition, the systemd-logind service is masked from the base\nimage. This is masked in the previously downloaded container base\nimage, so we ensure that the same is done for locally built chroots.\n\nDepends-On: I31880ca995735b737d33532eaa4c29be02523117\nDepends-On: I74f02669b013b8580d3469a8ffe214d88cd0f525\nChange-Id: I1ddfe36259610b25e86b69d64d1d7f32a56c0e4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/13742f908f4dce99bded79399928891773f3ea4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/13742f908f4dce99bded79399928891773f3ea4b"}]},"branch":"refs/heads/master"},"791b9e813ed29c1187803c53ad46e369b24eaad4":{"kind":"REWORK","_number":19,"created":"2021-04-20 08:56:26.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/01/776201/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/01/776201/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/01/776201/19"}}},"commit":{"parents":[{"commit":"f3bc93d1a310c4d7a3e2142e5b614955cbfbfe96","subject":"Merge \"Remove references to unsupported operating systems\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f3bc93d1a310c4d7a3e2142e5b614955cbfbfe96"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 13:48:13.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-04-20 08:56:20.000000000","tz":0},"subject":"Optionally build container image as a local chroot","message":"Optionally build container image as a local chroot\n\nThis change uses dnf to build the container image for Centos-8\nusing dnf locally rather than rely on an external image that is\ndownloaded and unpacked.\n\nThe existing image prestage commands are made conditional, and\nan operating system specific command can be provided via role\nvariables to build a chroot in /var/lib/machines.\n\nDuring the transition from Centos-8 to Centos-8-Stream, the\nvars files are separated, with vars/redhat.yml covering Stream,\nand vars/centos-8.3.yml covering legcay Centos-8.\n\nIn addition, the systemd-logind service is masked from the base\nimage. This is masked in the previously downloaded container base\nimage, so we ensure that the same is done for locally built chroots.\n\nDepends-On: I31880ca995735b737d33532eaa4c29be02523117\nDepends-On: I74f02669b013b8580d3469a8ffe214d88cd0f525\nChange-Id: I1ddfe36259610b25e86b69d64d1d7f32a56c0e4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/791b9e813ed29c1187803c53ad46e369b24eaad4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/791b9e813ed29c1187803c53ad46e369b24eaad4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
