)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~776229","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I4c53307d0d4f7a72010b878953f34ba0c8784922","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"lxc-hosts-cleanup","hashtags":[],"change_id":"I4c53307d0d4f7a72010b878953f34ba0c8784922","subject":"Remove support for openSUSE","status":"MERGED","created":"2021-02-17 16:28:44.000000000","updated":"2021-02-23 10:52:39.000000000","submitted":"2021-02-23 10:50:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"776229-lxc-hosts-cleanup","meta_rev_id":"05576c5db496637f9483caa6cb432614f50a4992","_number":776229,"virtual_id_number":776229,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-23 10:50:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2021-02-22 13:11:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":2,"date":"2021-02-23 09:34:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":1,"date":"2021-02-23 09:34:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-17 16:44:10.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2021-02-17 18:51:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-22 13:11:38.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2021-02-23 09:34:07.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"d0b40bef24d25b3617964e61d40cb26375cf5a04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-17 16:28:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9189e8e268bd070f0cc126fe648ebf54978dd8b2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-17 16:44:10.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"2fbdef1d67c0ccc6b8bfe471a6a1ff7ee32b659d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-17 17:15:35.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"152f116456eee497064bb74f116f2f8de973b128","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-17 17:32:43.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3d2b930865ace84f3da66dbd7167c029e9af5294","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-17 17:32:47.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"bbc7b8e8340027dc43eed7396376bcb611914b69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 18:51:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f28225bad2594a678fa584f66b740c57 : SUCCESS in 11m 46s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/b6503e5649f74d78a5975b112e258287 : SUCCESS in 13m 11s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/0ee9678bf65a4471b6739d56e562076c : SUCCESS in 13m 38s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dbefe6de37b64f7991d0e3352a62040d : SUCCESS in 11m 03s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5210073f05d94d44869f268f9d451041 : SUCCESS in 13m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/feaa44e8a53d44e6b46054d8e22e63bf : SUCCESS in 5m 58s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/b23d9c1df2a84eb0bef415c4561d49bc : SUCCESS in 1h 12m 43s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/4eee879ac49c472096da004f473b4038 : SUCCESS in 1h 02m 22s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03cc46d21ed6498f9aca410d458a5eb3 : SUCCESS in 1h 00m 14s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/65046b60db314b8dbabee9605e6b9888 : SUCCESS in 24m 57s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/450fac9cb3c04cb4a242c43a0a474565 : SUCCESS in 16m 18s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12b4f6671ad348dead16b804a369cc32 : SUCCESS in 18m 41s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9e45ed8917874b8299866065563d2150 : SUCCESS in 17m 52s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18538b56558849f59a606c2f5c06427f : SUCCESS in 17m 56s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/73e5f71b19fd4e34aa976c5748ce038c : SUCCESS in 16m 42s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f36016448d194414a223ede93c982ed7 : SUCCESS in 17m 54s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c332a8db8e3648d2b2ce638c9e60edfa : SUCCESS in 17m 29s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cccd54d1f3e8430c8739d2b7a6ec92ed : SUCCESS in 18m 15s","accounts_in_message":[],"_revision_number":3},{"id":"e9f76605768f32990ffdb4ab275e917d4ad7e8d1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-22 08:59:02.000000000","message":"Topic set to lxc-hosts-cleanup","accounts_in_message":[],"_revision_number":3},{"id":"6a3a3740eebf11dc1e56a6721f3420bafbdade54","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2021-02-22 13:11:38.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"dfb3cd603075644bbb1411eb0eac34570dce358f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-23 09:34:07.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"7fe87e0dc08acb138418ed48fd614ceb6cc7798e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 09:34:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"17bece99d1f9ca7469dc03aa7816565059a0afdb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 10:50:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/dfc6ce1438b94668b621cb1b01c1ca9f : SUCCESS in 10m 18s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/a5e019a7f8c94c2b8d31e52d0cb2272e : SUCCESS in 9m 45s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/19bcf62f70d54ad6bfee9377f9ae7e2b : SUCCESS in 11m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/804396fa702d4f0298932ffc8b8c111f : SUCCESS in 5m 52s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/431edb1b93614c99b0e3df7ac33678cc : SUCCESS in 1h 08m 50s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/18ecb7351f794a29afd6074dca41b8c7 : SUCCESS in 55m 26s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b459570486d46f6bfb6e4e9b28a9d37 : SUCCESS in 56m 12s","accounts_in_message":[],"_revision_number":3},{"id":"268c8c59c404447b5c5fd02885d0aada61225224","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 10:50:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"05576c5db496637f9483caa6cb432614f50a4992","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 10:52:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dbed52eb2e34dbcba7023b50efb9bd0 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1e8055c5bf6a2ae9c25b6e9deee4dc56e5bfc7ef","revisions":{"533b413c52ae1604ae8aa513124117480f19d779":{"kind":"REWORK","_number":1,"created":"2021-02-17 16:28:44.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/29/776229/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/29/776229/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/776229/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/776229/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/776229/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/776229/1"}}},"commit":{"parents":[{"commit":"bd40afa00ad88c35b7916c9f9bf8f25c5fef0647","subject":"Fix lxc_hosts_container_image_url condition","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/bd40afa00ad88c35b7916c9f9bf8f25c5fef0647"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 16:28:01.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 16:28:01.000000000","tz":0},"subject":"Remove support for openSUSE","message":"Remove support for openSUSE\n\nOpenStack-Ansible no longer supports openSUSE so we can simplify this\nansible role.\n\nChange-Id: I4c53307d0d4f7a72010b878953f34ba0c8784922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/533b413c52ae1604ae8aa513124117480f19d779"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/533b413c52ae1604ae8aa513124117480f19d779"}]},"branch":"refs/heads/master"},"67d24e0fe9b033aaf8f2c6da32caf4a57cf278b0":{"kind":"REWORK","_number":2,"created":"2021-02-17 17:15:35.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/29/776229/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/29/776229/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/776229/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/776229/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/776229/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/776229/2"}}},"commit":{"parents":[{"commit":"bd40afa00ad88c35b7916c9f9bf8f25c5fef0647","subject":"Fix lxc_hosts_container_image_url condition","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/bd40afa00ad88c35b7916c9f9bf8f25c5fef0647"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 16:28:01.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 17:15:21.000000000","tz":0},"subject":"Remove support for openSUSE","message":"Remove support for openSUSE\n\nOpenStack-Ansible no longer supports openSUSE so we can simplify this\nansible role.\n\nChange-Id: I4c53307d0d4f7a72010b878953f34ba0c8784922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/67d24e0fe9b033aaf8f2c6da32caf4a57cf278b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/67d24e0fe9b033aaf8f2c6da32caf4a57cf278b0"}]},"branch":"refs/heads/master"},"1e8055c5bf6a2ae9c25b6e9deee4dc56e5bfc7ef":{"kind":"REWORK","_number":3,"created":"2021-02-17 17:32:47.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/29/776229/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/29/776229/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/776229/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/776229/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/776229/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/776229/3"}}},"commit":{"parents":[{"commit":"bd40afa00ad88c35b7916c9f9bf8f25c5fef0647","subject":"Fix lxc_hosts_container_image_url condition","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/bd40afa00ad88c35b7916c9f9bf8f25c5fef0647"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 16:28:01.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 17:31:23.000000000","tz":0},"subject":"Remove support for openSUSE","message":"Remove support for openSUSE\n\nOpenStack-Ansible no longer supports openSUSE so we can simplify this\nansible role.\n\nChange-Id: I4c53307d0d4f7a72010b878953f34ba0c8784922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/1e8055c5bf6a2ae9c25b6e9deee4dc56e5bfc7ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/1e8055c5bf6a2ae9c25b6e9deee4dc56e5bfc7ef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
