)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~776230","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~If2a63a2743714745e0f0b0eea2ee3d5b8d4c9a35","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"lxc-hosts-cleanup","hashtags":[],"change_id":"If2a63a2743714745e0f0b0eea2ee3d5b8d4c9a35","subject":"Remove support for gentoo","status":"MERGED","created":"2021-02-17 16:41:59.000000000","updated":"2021-02-23 10:56:47.000000000","submitted":"2021-02-23 10:55:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"776230-lxc-hosts-cleanup","meta_rev_id":"e6248a89dc913b32f9fae08268dedcfdfe714391","_number":776230,"virtual_id_number":776230,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-23 10:55:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2021-02-22 13:13:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":2,"date":"2021-02-23 09:36:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":1,"date":"2021-02-23 09:36:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-17 19:10:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-22 13:13:40.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2021-02-23 09:36:22.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"332a5d9ac3015edb36e893cce164f9d7c1cc6c03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-17 16:41:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b34e34723b94881fc069fcde95146226b29a5b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-17 17:54:32.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"856d1cbaf16f0eb7b8294c7287573d4d98eb0e86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-17 18:40:56.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"58ea5c7b0e8f5f282333e6a6c2818e77467e0891","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 19:10:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/97aacd2159b141d08516b1395e063103 : SUCCESS in 12m 18s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/c504d2e1263741cf89cb677d1d5288ab : FAILURE in 9m 36s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/8aedd18a16e642329345c3dc2095939b : FAILURE in 13m 04s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/85b8c697d38b450786fb65d97428caaf : FAILURE in 10m 23s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/880ca342bc0e4c6992731b415a24d4bc : FAILURE in 11m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/052e927e59fe4ab3abb779c996c31dcb : SUCCESS in 10m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73fc7908da5b4635b749c9b7bc6cf1a9 : SUCCESS in 4m 41s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/bca397cb71a041bdb229398dcf5cb42b : FAILURE in 17m 31s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/a45850051fe44b8aadea5845e3a94281 : FAILURE in 13m 58s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b718744928b540a5a77ed2556e5fc789 : FAILURE in 17m 38s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/71c7fc39086f4ddba4737b25c470b251 : FAILURE in 15m 04s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/44d155732ec748fcb72964c1c3e36e01 : FAILURE in 11m 08s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b322ec6b9994409b5024521554cf14c : FAILURE in 10m 20s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6ffaed0514a34352beee35770558e5d1 : FAILURE in 10m 48s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa3de616c41041128d75823cba7e2bda : FAILURE in 13m 38s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4dce986b80934490959299368690c963 : FAILURE in 11m 38s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/14c2996e05a24eaa8948cb2d4eea1096 : FAILURE in 11m 41s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1c6031f2eda345819a2da91bd33241a3 : FAILURE in 11m 36s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8874fc2fd542435582a291aae82d4b2a : FAILURE in 12m 42s","accounts_in_message":[],"_revision_number":3},{"id":"6daa72b332ac59cbd51d7c9a4620390341ede565","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-17 19:15:16.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2048ccc48ebf09ecd8599a511d8045d755385502","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 20:31:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/cb549d943baf41f196ad3b3a5dda5af4 : SUCCESS in 12m 39s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/88c4c814881a4763a6e1b12366377f5c : SUCCESS in 10m 42s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/eefda8f69859431eaa88f5f3f862013b : SUCCESS in 15m 12s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1038622b7f1744febbd88a6f525d4789 : SUCCESS in 12m 53s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c08f44de80ff4f4d812ff5d4cf6c03d3 : SUCCESS in 15m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d8e8219c7e04e208b254dad6b823465 : SUCCESS in 6m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/68c0bc4a06ee418fa7159e413cb14817 : SUCCESS in 4m 15s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/2769303422bc4d879e85c476128ea6f3 : SUCCESS in 1h 09m 27s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/889a1d3b881c467986a10db43d553370 : SUCCESS in 59m 34s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8109597646a94e4e818f8d947bcf9ca5 : SUCCESS in 1h 00m 23s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/d58ea142c5684a3289f0b1833b73f0a8 : SUCCESS in 20m 32s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/151758fa3fa74d7b91fef26f1a702070 : SUCCESS in 17m 20s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9614296f6f340ec8369a86810bfab1a : SUCCESS in 17m 22s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/95e1808c9f2e44f4af06690be6f76d77 : SUCCESS in 17m 40s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4adfa56b62564242a7e3ac97c41feac2 : SUCCESS in 23m 22s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/049baee9c32141138a750226dda8941d : SUCCESS in 16m 24s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d53876e769554816bca8bb456881d7d7 : SUCCESS in 20m 55s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/83d87ac4a0e742039a2eb6e526c4e909 : SUCCESS in 16m 46s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16394025663c441db0f8692f075479c2 : SUCCESS in 18m 21s","accounts_in_message":[],"_revision_number":4},{"id":"4784e9d36cfdb081faaf3c10a519899040904c95","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-22 08:59:10.000000000","message":"Topic set to lxc-hosts-cleanup","accounts_in_message":[],"_revision_number":4},{"id":"dfaafc537439be3d37e14c02c099f7f938fb02b0","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2021-02-22 13:13:40.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"9ecebf10559218cdb5ba23acf0c08e4e94adb239","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-23 09:36:22.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4259a525a2aa45ef541b08761f7338e1c2aa58c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 09:36:38.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"12c7ab05d5f2984f37846557f8e43c5344d3652e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 10:55:03.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/07e6980c0f2242faaa0ca5483e5a6cd3 : SUCCESS in 9m 43s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/8a44622af35f498584bfb873a2a2b591 : SUCCESS in 10m 52s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a7397d8e4f344bf7b3453fdd0cc22f5c : SUCCESS in 11m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3060e07bc652498192c1a864874aca83 : SUCCESS in 5m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/681a250b688146d3a0a9f38129429741 : SUCCESS in 4m 29s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/a3661c756f934c98bb980ee409a589c6 : SUCCESS in 1h 11m 34s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/88030ead08394f0d973fefca760ded00 : SUCCESS in 1h 04m 14s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/230ad94c66d64828998212b21fab5868 : SUCCESS in 58m 04s","accounts_in_message":[],"_revision_number":4},{"id":"80315bbb758160901caf40162cd1523a3289417d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 10:55:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"e6248a89dc913b32f9fae08268dedcfdfe714391","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 10:56:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c60a81dbac14367b980b30e7491af45 : SUCCESS in 1m 19s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36e76dd701df4e1fba85b46295dfe28f : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"59abc5a28882039ff3463caf451940b5a85a963a","revisions":{"d77f3b47abbafc293d595e08545d6d16a43f9aa4":{"kind":"REWORK","_number":1,"created":"2021-02-17 16:41:59.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/30/776230/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/30/776230/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/30/776230/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/30/776230/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/30/776230/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/30/776230/1"}}},"commit":{"parents":[{"commit":"533b413c52ae1604ae8aa513124117480f19d779","subject":"Remove support for openSUSE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/533b413c52ae1604ae8aa513124117480f19d779"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 16:41:14.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 16:41:14.000000000","tz":0},"subject":"Remove support for gentoo","message":"Remove support for gentoo\n\nOpenstack-Ansible does not maintain support for deploying on gentoo\nso we can simplify this ansible role\n\nChange-Id: If2a63a2743714745e0f0b0eea2ee3d5b8d4c9a35\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d77f3b47abbafc293d595e08545d6d16a43f9aa4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d77f3b47abbafc293d595e08545d6d16a43f9aa4"}]},"branch":"refs/heads/master"},"023d4c7389d4b58415b5fbf9c0fb46ae0f8a3aa5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-02-17 17:54:32.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/30/776230/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/30/776230/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/30/776230/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/30/776230/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/30/776230/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/30/776230/2"}}},"commit":{"parents":[{"commit":"1e8055c5bf6a2ae9c25b6e9deee4dc56e5bfc7ef","subject":"Remove support for openSUSE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/1e8055c5bf6a2ae9c25b6e9deee4dc56e5bfc7ef"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 16:41:14.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 17:54:32.000000000","tz":0},"subject":"Remove support for gentoo","message":"Remove support for gentoo\n\nOpenstack-Ansible does not maintain support for deploying on gentoo\nso we can simplify this ansible role\n\nChange-Id: If2a63a2743714745e0f0b0eea2ee3d5b8d4c9a35\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/023d4c7389d4b58415b5fbf9c0fb46ae0f8a3aa5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/023d4c7389d4b58415b5fbf9c0fb46ae0f8a3aa5"}]},"branch":"refs/heads/master","description":"Rebase"},"b6cd9d6f4efca555c65d24070514a091b4729b1b":{"kind":"REWORK","_number":3,"created":"2021-02-17 18:40:56.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/30/776230/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/30/776230/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/30/776230/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/30/776230/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/30/776230/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/30/776230/3"}}},"commit":{"parents":[{"commit":"1e8055c5bf6a2ae9c25b6e9deee4dc56e5bfc7ef","subject":"Remove support for openSUSE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/1e8055c5bf6a2ae9c25b6e9deee4dc56e5bfc7ef"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 16:41:14.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 18:40:45.000000000","tz":0},"subject":"Remove support for gentoo","message":"Remove support for gentoo\n\nOpenstack-Ansible does not maintain support for deploying on gentoo\nso we can simplify this ansible role\n\nChange-Id: If2a63a2743714745e0f0b0eea2ee3d5b8d4c9a35\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b6cd9d6f4efca555c65d24070514a091b4729b1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b6cd9d6f4efca555c65d24070514a091b4729b1b"}]},"branch":"refs/heads/master"},"59abc5a28882039ff3463caf451940b5a85a963a":{"kind":"REWORK","_number":4,"created":"2021-02-17 19:15:16.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/30/776230/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/30/776230/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/30/776230/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/30/776230/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/30/776230/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/30/776230/4"}}},"commit":{"parents":[{"commit":"1e8055c5bf6a2ae9c25b6e9deee4dc56e5bfc7ef","subject":"Remove support for openSUSE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/1e8055c5bf6a2ae9c25b6e9deee4dc56e5bfc7ef"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 16:41:14.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 19:14:55.000000000","tz":0},"subject":"Remove support for gentoo","message":"Remove support for gentoo\n\nOpenstack-Ansible does not maintain support for deploying on gentoo\nso we can simplify this ansible role\n\nChange-Id: If2a63a2743714745e0f0b0eea2ee3d5b8d4c9a35\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/59abc5a28882039ff3463caf451940b5a85a963a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/59abc5a28882039ff3463caf451940b5a85a963a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
