)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~776237","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I2dbd74d88ca3d04b55536fdd07f67c26b74f1b6c","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"lxc-hosts-cleanup","hashtags":[],"change_id":"I2dbd74d88ca3d04b55536fdd07f67c26b74f1b6c","subject":"Remove lxc_install task file which contains only one task","status":"MERGED","created":"2021-02-17 17:06:04.000000000","updated":"2021-02-23 19:46:25.000000000","submitted":"2021-02-23 19:29:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"776237-lxc-hosts-cleanup","meta_rev_id":"39a3ab94ee50fda329b8caeea2230e5e67a82aac","_number":776237,"virtual_id_number":776237,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-23 19:29:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2021-02-22 13:14:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":2,"date":"2021-02-23 09:36:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":1,"date":"2021-02-23 09:36:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-17 18:26:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-22 13:14:27.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2021-02-23 09:36:51.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"943e8467b1478fc438b6a1e55247b12e9117b879","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-17 17:06:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e7001ae8449ddd37bd5058063f013f36928ec456","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-17 17:54:42.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"f9776a4090285c15ead4c53bbcf66222ae41c660","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 18:26:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/48337a13069f4692b8b21b498295ba6d : SUCCESS in 8m 38s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/cfd8e6d83aad40c683e4450a1c688eaa : FAILURE in 7m 47s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/b589d13810ba43e6be662d1f2da921a5 : FAILURE in 9m 39s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/863edf141447478282215eede763e4b4 : FAILURE in 9m 05s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/430d57ebf3f1426ebb9909f85c2765d9 : FAILURE in 9m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e552e6570b441d1acefeadbbce44890 : SUCCESS in 5m 52s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/2762db67690d45729d77c5ddefe1b268 : FAILURE in 16m 43s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/ec64c0b3fb704323ba6b3384191f0576 : FAILURE in 20m 48s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3eb2ecd39d964d93bb38ca6f463c5ecb : FAILURE in 13m 45s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/fdf0e7e3b1f5477a9033b88180dbd375 : FAILURE in 11m 10s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/80dc9bf4271b460196952014b93851bd : FAILURE in 11m 22s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af256f00b9ef4a5ba747b0757c56d5bc : FAILURE in 9m 47s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d1f3c9be5bb84b2b8dc563877974dc8e : FAILURE in 10m 19s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a7ce5f9351dc475ea3ae35a093478ad1 : FAILURE in 14m 37s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/255d0079278f4a57b662dd09def5780d : FAILURE in 11m 02s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67dfd7764cbc43b2b3ed80d427e510fd : FAILURE in 10m 55s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fe521fa20e4c4d5c9307a364ff6efa23 : FAILURE in 10m 39s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/500b974b12d5465fbb6bdb89d5c564e7 : FAILURE in 19m 20s","accounts_in_message":[],"_revision_number":2},{"id":"47e9d8d3107e871871b65bd626cbde8b87f99257","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-17 18:43:47.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"6c114094b7f2bab0894456339b0a188c50e80912","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-17 19:15:16.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"c32473de041149e2e87ca356f59d254c2d86e4b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-17 19:41:00.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8a2db1d7a76b2fb44aa48704e7f2e334eb1aceb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 20:58:19.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4893c036f16242d3b8a3efbae78f01ee : SUCCESS in 10m 11s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/8a654e6bf292407e82b60ecdd72ed610 : SUCCESS in 10m 33s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/28c985f6be964a13b5ed3c784efbc5a5 : SUCCESS in 12m 48s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1370b44401bf40b796b00498778fda0a : SUCCESS in 11m 07s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/edc4be0ec2044140bb4e63718a920722 : SUCCESS in 11m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60d0e0059c2b477e9c3b0fea415d836f : SUCCESS in 5m 29s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/11fab508c4614d95a4f4423cf5d443f5 : SUCCESS in 1h 06m 57s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/d8aae375e9b04583af575e59d994a9cc : SUCCESS in 1h 00m 15s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1925409d90840559a9d52f445d3424b : SUCCESS in 57m 04s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/9cdd9161cfe84c78b384306201df5693 : SUCCESS in 19m 38s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f31ae2feb41c4e5d80a32bfdacbbb73f : SUCCESS in 16m 20s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/30300ed8db3e40ff9b6a794c94e85653 : SUCCESS in 15m 52s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cfe0b3da6afc49d49782d6644d371870 : SUCCESS in 17m 09s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a02764b2ca946d6aec8f8543d75aa6f : SUCCESS in 18m 15s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b356f9c5c3594b3081698ec7087c99a2 : SUCCESS in 17m 56s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cae1842c54314ff79b53948ca4719e07 : SUCCESS in 22m 18s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8c9b759d02084369a8108941d0f1a55f : SUCCESS in 19m 26s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1aef81de2c104712831c4bc805bdc8a0 : SUCCESS in 17m 12s","accounts_in_message":[],"_revision_number":5},{"id":"e9b5960bc2ecbb91d82ee42127ee65212e0c72b3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-22 08:59:18.000000000","message":"Topic set to lxc-hosts-cleanup","accounts_in_message":[],"_revision_number":5},{"id":"10b8d96bd256a3e40233ad90396ca9f4dd7a89f7","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2021-02-22 13:14:27.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"05a661dbc4415e15c88fb58fbd23d5f9bb915618","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-23 09:36:51.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"375f618f5a35ad738320b071a0f5689311d051c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 09:37:11.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"6c158da50324f0e997e69d5e5463ab749caf0d0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 11:00:54.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/71a0e64858a34e82b6acf42401849044 : SUCCESS in 9m 32s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/71e60a81f1004b4db84baa643fcf25a9 : SUCCESS in 10m 31s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7b4d62470b0043c2a42695528d672755 : SUCCESS in 10m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f76b1e6dbf7e46f790fc9a1585da1a85 : SUCCESS in 7m 17s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/0fcc24a580d246a68fc79ef0e154763a : FAILURE in 55m 26s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/242143a986164434839329f628263fc1 : SUCCESS in 1h 09m 55s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8d1f297398e4f6eb3cd8c6761e2d06f : SUCCESS in 1h 01m 17s","accounts_in_message":[],"_revision_number":5},{"id":"6484f699a20c177be25e353510c931fcf9e8e3ea","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-23 16:14:43.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"580a55397cb9fb57d10613b3d24c059e66be6eda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 17:52:23.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/59dd14eed66f45beb96a6f5d5f4320dd : SUCCESS in 8m 54s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/d15e6e3d9f814146bb8d5222aa4e126e : SUCCESS in 13m 45s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/78d480bb7c2d4bbcb16e820398f9a4bf : SUCCESS in 12m 43s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1ba1fe2bdc2d45d8b3e89374dba327d5 : SUCCESS in 15m 01s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c617e62ceb454147ae1bcedfa715dd07 : SUCCESS in 15m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0e01c70d9ad41a18bbcae2fb4e7e363 : SUCCESS in 5m 25s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/89aa71a18697404592d015ed04e5b8f3 : SUCCESS in 1h 16m 43s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/c25a566a0a1c4ea99fda60d0f36d085e : SUCCESS in 1h 04m 32s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4abe6f9d069e41939bf3bba653459f8b : SUCCESS in 1h 00m 37s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/4ede840b514249ae97e71ad51a4e7acc : SUCCESS in 25m 05s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/06656931d82148ba97c64b4c2609f52b : SUCCESS in 18m 51s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8f0ed9ceb404686808ea6b60f11b4fd : SUCCESS in 16m 23s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/95dd42b8bc46402b8863328bf0e0d524 : SUCCESS in 16m 39s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a906e4c4e82e40178f6b0c194a43c21b : SUCCESS in 17m 11s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/01ac382588cd440fa39ff36a4ad8e17c : SUCCESS in 15m 51s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ace72ef831249e6870cc6ab5ef61af7 : SUCCESS in 20m 45s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fb35e79872fa40b8861c0e4680a8108c : SUCCESS in 16m 03s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66bca9ad4794411c8cbf84156587e0bd : SUCCESS in 15m 58s","accounts_in_message":[],"_revision_number":5},{"id":"ee5759f34629350693b7f867a5856935bc2cd6fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 17:52:37.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"9dafc8cd4a5ad18e3c0497ba1bba61d03cfb6b25","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 19:29:09.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2f8707759a15429aacea4ef6b6c067d4 : SUCCESS in 13m 54s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/66deef4ce03e407aac3df788442bd800 : SUCCESS in 16m 44s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4da7f171a38c4ca3a15954298206e98e : SUCCESS in 12m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfe6333cd17040b6a9fd43291539d162 : SUCCESS in 5m 57s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/7aa6dc7591364fa79c3e596718b6715c : SUCCESS in 1h 22m 43s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/e572bf34f1bd4794b8b83d78774dc70e : SUCCESS in 1h 09m 14s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea10e6376e094fc2a04923c00b94aba3 : SUCCESS in 1h 08m 03s","accounts_in_message":[],"_revision_number":5},{"id":"666d7aa0fdfa159b89f6dbd79eb11f96c86f856f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 19:29:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"39a3ab94ee50fda329b8caeea2230e5e67a82aac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 19:46:25.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e7829c360044385a1171a22121e9678 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"478abec9363b696458e37268c360757e4f79bf9e","revisions":{"1b726c6d5621ff9ab28bab8614bcf42f5e84f468":{"kind":"REWORK","_number":1,"created":"2021-02-17 17:06:04.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/37/776237/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/37/776237/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/776237/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/776237/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/776237/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/776237/1"}}},"commit":{"parents":[{"commit":"d77f3b47abbafc293d595e08545d6d16a43f9aa4","subject":"Remove support for gentoo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d77f3b47abbafc293d595e08545d6d16a43f9aa4"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 16:43:46.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 16:43:46.000000000","tz":0},"subject":"Remove lxc_install task file which contains only one task","message":"Remove lxc_install task file which contains only one task\n\nChange-Id: I2dbd74d88ca3d04b55536fdd07f67c26b74f1b6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/1b726c6d5621ff9ab28bab8614bcf42f5e84f468"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/1b726c6d5621ff9ab28bab8614bcf42f5e84f468"}]},"branch":"refs/heads/master"},"c31a777bed8953db5962d27d44c80e5ee253c981":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-02-17 17:54:42.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/37/776237/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/37/776237/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/776237/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/776237/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/776237/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/776237/2"}}},"commit":{"parents":[{"commit":"023d4c7389d4b58415b5fbf9c0fb46ae0f8a3aa5","subject":"Remove support for gentoo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/023d4c7389d4b58415b5fbf9c0fb46ae0f8a3aa5"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 16:43:46.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 17:54:42.000000000","tz":0},"subject":"Remove lxc_install task file which contains only one task","message":"Remove lxc_install task file which contains only one task\n\nChange-Id: I2dbd74d88ca3d04b55536fdd07f67c26b74f1b6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c31a777bed8953db5962d27d44c80e5ee253c981"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c31a777bed8953db5962d27d44c80e5ee253c981"}]},"branch":"refs/heads/master","description":"Rebase"},"90218034d667a161006a12e9d055a18de5f6a6fe":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-02-17 18:43:47.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/37/776237/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/37/776237/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/776237/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/776237/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/776237/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/776237/3"}}},"commit":{"parents":[{"commit":"b6cd9d6f4efca555c65d24070514a091b4729b1b","subject":"Remove support for gentoo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b6cd9d6f4efca555c65d24070514a091b4729b1b"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 16:43:46.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 18:43:47.000000000","tz":0},"subject":"Remove lxc_install task file which contains only one task","message":"Remove lxc_install task file which contains only one task\n\nChange-Id: I2dbd74d88ca3d04b55536fdd07f67c26b74f1b6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/90218034d667a161006a12e9d055a18de5f6a6fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/90218034d667a161006a12e9d055a18de5f6a6fe"}]},"branch":"refs/heads/master","description":"Rebase"},"4d17f476f3862c66f6e432d36b7ce94f0cb5f9dc":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-02-17 19:15:16.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/37/776237/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/37/776237/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/776237/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/776237/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/776237/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/776237/4"}}},"commit":{"parents":[{"commit":"59abc5a28882039ff3463caf451940b5a85a963a","subject":"Remove support for gentoo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/59abc5a28882039ff3463caf451940b5a85a963a"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 16:43:46.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 19:14:59.000000000","tz":0},"subject":"Remove lxc_install task file which contains only one task","message":"Remove lxc_install task file which contains only one task\n\nChange-Id: I2dbd74d88ca3d04b55536fdd07f67c26b74f1b6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/4d17f476f3862c66f6e432d36b7ce94f0cb5f9dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/4d17f476f3862c66f6e432d36b7ce94f0cb5f9dc"}]},"branch":"refs/heads/master"},"478abec9363b696458e37268c360757e4f79bf9e":{"kind":"REWORK","_number":5,"created":"2021-02-17 19:41:00.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/37/776237/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/37/776237/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/776237/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/776237/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/776237/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/776237/5"}}},"commit":{"parents":[{"commit":"59abc5a28882039ff3463caf451940b5a85a963a","subject":"Remove support for gentoo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/59abc5a28882039ff3463caf451940b5a85a963a"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 16:43:46.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 19:40:28.000000000","tz":0},"subject":"Remove lxc_install task file which contains only one task","message":"Remove lxc_install task file which contains only one task\n\nChange-Id: I2dbd74d88ca3d04b55536fdd07f67c26b74f1b6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/478abec9363b696458e37268c360757e4f79bf9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/478abec9363b696458e37268c360757e4f79bf9e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
