)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~776239","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I59bbbcd8a66970a4fb30339aed457b50fb4dad50","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"lxc-hosts-cleanup","hashtags":[],"change_id":"I59bbbcd8a66970a4fb30339aed457b50fb4dad50","subject":"Remove per-backend container creation tasks files","status":"MERGED","created":"2021-02-17 17:06:04.000000000","updated":"2021-02-24 10:36:59.000000000","submitted":"2021-02-24 10:35:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"776239-lxc-hosts-cleanup","meta_rev_id":"d73c8b364e4ea0b1d705e553cf66ccb7544aa840","_number":776239,"virtual_id_number":776239,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-24 10:35:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2021-02-24 08:57:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2021-02-23 09:40:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2021-02-24 08:57:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-17 17:49:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-23 09:40:05.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2021-02-24 08:57:37.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"8d9444dc288a6992ccb20b8b995566ad06f68c1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-17 17:06:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a4941cd3c5e9c9964d311ed5a37ab6ef5775a5a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 17:49:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/96ab373073284287b6453fb42f639543 : SUCCESS in 9m 21s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/53aa6eab7af4418bafc3eee82baeb85e : FAILURE in 11m 38s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/b28b2060e0014ba78f4d50686bd63404 : FAILURE in 10m 55s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e9fef8e2493d4457af056964c642702f : FAILURE in 13m 15s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a9d6d373ea94afd87cf573f677d6454 : FAILURE in 10m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99c431723f434811a3dc468026a3edb1 : SUCCESS in 6m 23s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/ec9960caf393482dbcafe022f399cfcf : FAILURE in 23m 33s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/d4163351dd0246ba8b8f9f789452b652 : FAILURE in 14m 11s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d16ddd5a9e7d4f4ea3b69f928da6a33f : FAILURE in 15m 58s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/023a0b52e28c4090acec3d64ded5d21e : FAILURE in 13m 31s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f337a919f4a74ee18fbb82d56ce01c92 : FAILURE in 13m 20s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1841b7b73ee64786b350d3ddc48a41d8 : FAILURE in 14m 07s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/63d56860536843bdb48bfdbcdbf1ab34 : FAILURE in 13m 40s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a10be3744c649faac1597ace6b27ada : FAILURE in 12m 48s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d87bc3ebca344f5b892eb1992815f34c : FAILURE in 13m 27s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca1a243b343d410894ea9e20b6f5b536 : FAILURE in 12m 37s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/128d77a0a1404363bca7399e6b77dbb6 : FAILURE in 13m 58s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01fce64b91f442208b43f5217029d339 : FAILURE in 14m 13s","accounts_in_message":[],"_revision_number":1},{"id":"d7849d6a44f6e30a3b965eecf1d8dc38cbdd9452","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-17 17:55:31.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"df03212ac18017d8ecf29d67b620e948db787623","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-17 18:42:07.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"4ed8797e05cd83ab17debe2b45ecde3a8731a3bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 18:42:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c247e639acf74f94864651ce5c11c6fd : SUCCESS in 8m 20s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/895cff4eb3f14354871824d566f2b0d5 : FAILURE in 10m 28s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/6c5c7ddbd5034fd58badac8750b9cc65 : FAILURE in 10m 25s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f04195e5add14ad6b5ab835bdbbd81ea : FAILURE in 13m 46s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a17d9e4a0dea40c497fd19df2d56d2b3 : FAILURE in 15m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74da1644b4934853bfa14bc674d167c8 : SUCCESS in 6m 44s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/20849b0549284468af8c10a933467931 : FAILURE in 17m 15s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/3e4a63e6b5784e808cb601376ccd84a0 : FAILURE in 14m 57s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47bbf78f96244de9b4c83fbee0111e9b : FAILURE in 28m 41s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/4d7055ec9fd6406bb7372a2eaf72e24c : FAILURE in 17m 06s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1cd8f6f3e38f4053a161f7b42d75f239 : FAILURE in 11m 53s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8de6bf937654410b6a5b9fc43caf706 : FAILURE in 13m 19s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/83f78651b195440393137eb61c630733 : FAILURE in 12m 19s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/033c81dfd61543a18551374055dd931a : FAILURE in 13m 44s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d6d88a4fc4414a2f889e7380cd0f33ee : FAILURE in 13m 33s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8977d28c2f554a2abe7b56b5200a2f0d : FAILURE in 14m 32s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eea10d2b08b046969138f921b2e7f09a : FAILURE in 11m 22s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ffdd03212544f61a0b946baad466b7f : FAILURE in 16m 37s","accounts_in_message":[],"_revision_number":2},{"id":"baf8fef04d67973c02faa1e9007812e281f5c8b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-17 18:44:48.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"8df9488d1e2717054596ce3122c5af0e4291d431","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-17 19:15:16.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"53745914b84087c9257461e09dc7e5bab865c86a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-17 19:41:00.000000000","message":"Uploaded patch set 6: New patch set was added with same tree, parent, and commit message as Patch Set 5.","accounts_in_message":[],"_revision_number":6},{"id":"4d4cfe3aa764dc2dd632d4986d202efcc7f3adba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 20:17:30.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/39acbd9d0aeb46e2b2caa9622f17b866 : SUCCESS in 11m 29s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/6fef1ed01ede48d8b7a8fc282dc231b6 : FAILURE in 10m 11s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/a1177ff84ec34c36bc0a4af7afce6147 : FAILURE in 13m 13s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a75ba230ff6749109cb3ae220d7cbdf6 : FAILURE in 12m 32s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/476991241f7d4988a678f049847a0727 : FAILURE in 14m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9dccc2dd3a249659ff947574ac70159 : SUCCESS in 5m 46s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/fa1719a6ef1b4e108ca90a21a87e68d6 : FAILURE in 21m 14s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/68d188bb0c0f4425a80ccc3d5e779209 : FAILURE in 17m 00s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21aab286442d4919882156e23d4d2005 : FAILURE in 17m 43s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/c4caa66e7e824276a89bf6c26ab93466 : FAILURE in 14m 28s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d6c3d6c33cdf4172b6effb6990d903ca : FAILURE in 13m 36s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/477ed016b0664e36b7eb1c906a838972 : FAILURE in 14m 34s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9d08e2799dcd48409f33453b783fa12b : FAILURE in 14m 57s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc3466feaa0c4e87a92ec1c602d31c24 : FAILURE in 13m 33s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6cdf52627f454455a23975d727c1a3f9 : FAILURE in 13m 09s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec101bb87c2640c790ef9fe9569228c5 : FAILURE in 14m 24s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/71f4f24f349a40318c4b0b71ff27ac44 : FAILURE in 13m 46s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27b942b493d34e44bba0efc32f6d330d : FAILURE in 14m 21s","accounts_in_message":[],"_revision_number":6},{"id":"bd1bbcc6b9ef9e646773157237b6a3945af16a25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-17 20:41:26.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"9959daa11e7f8ce8c12395a2e41269255976a25e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 21:10:27.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/41d3a465db93485fa68ae65f941cbde5 : SUCCESS in 8m 40s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/a8b2ad2e969a4a438b4cea035f9924a8 : FAILURE in 12m 03s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/f99ee34c5bdd486f9094a81eae0f890b : FAILURE in 17m 13s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c32ca602b50c45dbabb927a74c85d3c4 : FAILURE in 11m 49s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8446cea26c642cab82b5173351b9754 : FAILURE in 11m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/991d7b6d9a9f4da9bdf6a1d3ffe4ae8f : SUCCESS in 7m 11s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/110fe8ca3c694eada1f5e90e1dfa500f : FAILURE in 19m 56s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/12f4c085be1a48f6b111209c38759461 : FAILURE in 21m 03s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46c8ed4b1d3842bfbf8f48980464b0bc : FAILURE in 16m 44s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/e9a17d17ee334fbfa62f743eda1ffdc3 : FAILURE in 16m 50s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b9f72080047240bfab2bd88fb548c01b : FAILURE in 13m 32s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f365f0782f141998893ee338d12cd5b : FAILURE in 13m 27s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/94cf8bb9175c47f5b5c4ad02f7c3145d : FAILURE in 11m 49s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3dd2ea1269544519bfb908fee93e1925 : FAILURE in 12m 25s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/955a4bc2d17d40dfa204228eb79fe0ca : FAILURE in 11m 47s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25671c64d5f645dd974fd06bdfe54fb8 : FAILURE in 13m 14s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/29520f8f02c04364b68ed79d671f6534 : FAILURE in 13m 08s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa430cf8f021494f9f404325898b689b : FAILURE in 12m 22s","accounts_in_message":[],"_revision_number":7},{"id":"4668ca0e737dedfef555ffd3fb72bfcfb096c322","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-17 22:11:34.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"bf6c199c2ae8a6a8d50b7200faf454677a570b56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 23:04:43.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4598c710a57c4d45a341135bef159e87 : SUCCESS in 10m 47s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/e6471310e5ab484a9b9008108e37a087 : FAILURE in 14m 37s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/4ee4aacfd5ad45d7bca05d2ae99cce73 : FAILURE in 15m 27s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e18c95a90be94ec49191a72a3be88bd7 : FAILURE in 11m 34s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41235003c6a54a379646c73d3dfc1d12 : FAILURE in 14m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68f0e5f06b004914b01dda413d91cb65 : SUCCESS in 7m 45s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/c9db74f4950a43c3910234c22637a65a : FAILURE in 25m 37s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/96187960d0f94c39b0dc9eea719fa792 : FAILURE in 16m 15s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0720e5f301b4c5486e69c54401216e1 : FAILURE in 22m 19s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/071ef36e3ccd4c2f87671f3e5a6fb691 : FAILURE in 17m 48s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a87e6829b0ad49e79d638a4baafddd3f : FAILURE in 12m 39s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d523e1ab6884367b153254cb5abf9ce : FAILURE in 14m 12s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/749aa91f328a4cdb9b6a7bffcbcd5bfb : FAILURE in 11m 59s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab4a1e23bc5240b885db7099997e42f9 : FAILURE in 14m 20s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a3834b3f1fa3453ca2b84fe51058e58c : FAILURE in 11m 47s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a497ad60fb9b44b8b7ddbf45d6e7a360 : FAILURE in 14m 47s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/365877949365423b8724a16c5c59ac99 : FAILURE in 12m 15s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e98b704e965d450794d065a468179eec : FAILURE in 12m 43s","accounts_in_message":[],"_revision_number":8},{"id":"30fd03650e933fa0da44151dd57fb84fd8cf1d85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-18 08:20:41.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"971119dd5d409bd0e3910fe1971e7e4426a88a0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 09:04:20.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b48a5a8b81d94a5cad982276bc3c033e : SUCCESS in 9m 45s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/f8e064ac597b41639c78b75e6ccc3285 : FAILURE in 10m 21s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/eda9f9717f504d19a469e6602f04dd1d : FAILURE in 16m 34s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e8df60027fbf42b384780a3a25da8dcb : FAILURE in 11m 40s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e4ac87d17bf24e9b87a339e81659af50 : FAILURE in 11m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7b517a88e8a490a82b845034cdf012f : SUCCESS in 5m 34s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/7b4e123c73b64ca79f7d48d126f5a75c : FAILURE in 22m 24s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/8325981d9e634ee5b9445e8d6cee1389 : FAILURE in 17m 08s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b29062e269cc46dc91518fe9cc9b2342 : FAILURE in 19m 05s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/2f1f8367b75f449a897911aa5f11f9e5 : FAILURE in 18m 24s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/af172473399b409c8535c6fc886cb94e : FAILURE in 11m 23s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/876d0bdd87e14a899f6cbc1255c30680 : FAILURE in 11m 32s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b7216a818fee461e838063b55b9d4bb7 : FAILURE in 14m 00s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1bf41001e3144f489c398965a961cc43 : FAILURE in 12m 08s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8f816406f3ea433dbaa2427edd0f7972 : SUCCESS in 18m 39s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f53c8cbebac24baea9a3ff4c0ef65c06 : SUCCESS in 16m 06s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b511256a5adc4b4c800708d07732a05d : FAILURE in 12m 01s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a86af23839cb4fafa93de819b85501e3 : FAILURE in 12m 37s","accounts_in_message":[],"_revision_number":9},{"id":"3b9603fc8711b550c23773a0ad3b5ca71527ae29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-18 09:19:51.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"1c36a18dd36a07f53f6b80a9505c5d37a6e28676","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 10:34:34.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6d920154676e4e40b9ee14cdb46e56dd : SUCCESS in 9m 37s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/b421eeeb4aad4e0983ba9bb27f67e6c0 : SUCCESS in 11m 28s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/e72fb0b342844d569fea3251a5865831 : SUCCESS in 15m 15s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6ab985346f614a4e93d80e85e208fb54 : SUCCESS in 13m 32s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c814cf01249140c8b1036d546265a7c0 : SUCCESS in 15m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f9fb4991b9249b8ae909ebe62b163ad : SUCCESS in 7m 23s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/76446b5eea2a44f3b66832eaa3b3973d : SUCCESS in 1h 10m 04s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/97aea2c31c844209855e71e8cf6901d8 : SUCCESS in 1h 00m 47s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e1c2bf309fe48339e183b64a38f3872 : SUCCESS in 57m 01s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/d51ed68d53d34d759628c1bd14f95c20 : SUCCESS in 26m 06s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a56ebb497fa542b7ba6e5032620109dd : SUCCESS in 18m 29s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/73ae47ef54da4603bc049c8108ca9350 : SUCCESS in 20m 23s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3dc1af56d99c4621bca87519e882e6e4 : SUCCESS in 19m 12s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03bf645d421647eabf46b0eba06907d8 : SUCCESS in 19m 56s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7c015291a4f54493b27a799b2c840d82 : SUCCESS in 19m 56s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/051b16939d50493a8f798923e00660f6 : SUCCESS in 17m 56s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3e6a8a2201944b089dfa09fa8f2d7e9c : SUCCESS in 16m 09s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc21cb062b19429095eeb2ce2edeb269 : SUCCESS in 17m 43s","accounts_in_message":[],"_revision_number":10},{"id":"ec2585875451df2e6a5fcf5bdc74321ae9b2e852","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-22 08:59:36.000000000","message":"Topic set to lxc-hosts-cleanup","accounts_in_message":[],"_revision_number":10},{"id":"a55acb240f0164aaa43cdb6e5338bd7fbae2f195","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-23 09:40:05.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"b3adbc4e8787e9a83488e3040063be5552b26a7e","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-02-24 08:57:37.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"5115fedfb91938757c214d7cfe200eb9ec87608a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 08:57:50.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"98756f79d92921681013f3a86d4612bcbc1ae095","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 10:35:41.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/60d27d5ba9b34d70b13001cef89ce1b2 : SUCCESS in 9m 06s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/95e08de7fca4429ab4a4b09d7cb7e3c6 : SUCCESS in 14m 30s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b815ba42ecab430a936dd237dab65fc0 : SUCCESS in 13m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c84ba10937804aabb49a2c2d49e44ebc : SUCCESS in 5m 12s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/53b50214eb804400b7f429999e2c6d50 : SUCCESS in 1h 31m 14s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/5329d490211a484ba2bc5e9a8a762e1f : SUCCESS in 58m 42s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42b4dda666c041c99b53e2480973fc8f : SUCCESS in 58m 51s","accounts_in_message":[],"_revision_number":10},{"id":"62183b8a9d4dd3d4bcc00f842aad7938cb707b5c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 10:35:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"d73c8b364e4ea0b1d705e553cf66ccb7544aa840","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 10:36:59.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e914a7b56cbc4b65a5dcd5709c1b5eb6 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"7fc5e9950ea9e8b379295837fede70f864502439","revisions":{"aee8743f0bae492aab320c9dbd4f8eec405f4aa4":{"kind":"REWORK","_number":1,"created":"2021-02-17 17:06:04.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/39/776239/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/39/776239/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/39/776239/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/39/776239/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/39/776239/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/39/776239/1"}}},"commit":{"parents":[{"commit":"fa540804f3645afea1d5924735616b99f9ad8e6b","subject":"Move kernel tuning tasks to lxc_post_install.yml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/fa540804f3645afea1d5924735616b99f9ad8e6b"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 16:58:25.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 16:58:25.000000000","tz":0},"subject":"Remove per-backend container creation tasks files","message":"Remove per-backend container creation tasks files\n\nThe same can be achieved with two ternary operators. Much less code\nand duplication.\n\nChange-Id: I59bbbcd8a66970a4fb30339aed457b50fb4dad50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/aee8743f0bae492aab320c9dbd4f8eec405f4aa4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/aee8743f0bae492aab320c9dbd4f8eec405f4aa4"}]},"branch":"refs/heads/master"},"f865348d1e1aacd36fcf2f92c53733f81f826f53":{"kind":"REWORK","_number":2,"created":"2021-02-17 17:55:31.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/39/776239/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/39/776239/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/39/776239/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/39/776239/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/39/776239/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/39/776239/2"}}},"commit":{"parents":[{"commit":"fa540804f3645afea1d5924735616b99f9ad8e6b","subject":"Move kernel tuning tasks to lxc_post_install.yml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/fa540804f3645afea1d5924735616b99f9ad8e6b"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 16:58:25.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 17:55:28.000000000","tz":0},"subject":"Remove per-backend container creation tasks files","message":"Remove per-backend container creation tasks files\n\nThe same can be achieved with two ternary operators. Much less code\nand duplication.\n\nChange-Id: I59bbbcd8a66970a4fb30339aed457b50fb4dad50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f865348d1e1aacd36fcf2f92c53733f81f826f53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f865348d1e1aacd36fcf2f92c53733f81f826f53"}]},"branch":"refs/heads/master"},"f588b27ea5e2d6f9a51487c6c388e161186b75d3":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-02-17 18:42:07.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/39/776239/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/39/776239/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/39/776239/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/39/776239/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/39/776239/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/39/776239/3"}}},"commit":{"parents":[{"commit":"ae8af72eda54ed34eb02a33f2c568c8319f23162","subject":"Move kernel tuning tasks to lxc_post_install.yml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ae8af72eda54ed34eb02a33f2c568c8319f23162"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 16:58:25.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 18:42:07.000000000","tz":0},"subject":"Remove per-backend container creation tasks files","message":"Remove per-backend container creation tasks files\n\nThe same can be achieved with two ternary operators. Much less code\nand duplication.\n\nChange-Id: I59bbbcd8a66970a4fb30339aed457b50fb4dad50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f588b27ea5e2d6f9a51487c6c388e161186b75d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f588b27ea5e2d6f9a51487c6c388e161186b75d3"}]},"branch":"refs/heads/master","description":"Rebase"},"851450f99d44b97a0cfa9f90f8cda0115340c114":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-02-17 18:44:48.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/39/776239/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/39/776239/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/39/776239/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/39/776239/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/39/776239/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/39/776239/4"}}},"commit":{"parents":[{"commit":"68179efc16f2b4fb3c7836b2f8f53a3937ec5469","subject":"Move kernel tuning tasks to lxc_post_install.yml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/68179efc16f2b4fb3c7836b2f8f53a3937ec5469"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 16:58:25.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 18:44:48.000000000","tz":0},"subject":"Remove per-backend container creation tasks files","message":"Remove per-backend container creation tasks files\n\nThe same can be achieved with two ternary operators. Much less code\nand duplication.\n\nChange-Id: I59bbbcd8a66970a4fb30339aed457b50fb4dad50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/851450f99d44b97a0cfa9f90f8cda0115340c114"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/851450f99d44b97a0cfa9f90f8cda0115340c114"}]},"branch":"refs/heads/master","description":"Rebase"},"09a265001269688a7df63565a7f14b7335eae9fc":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-02-17 19:15:16.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/39/776239/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/39/776239/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/39/776239/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/39/776239/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/39/776239/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/39/776239/5"}}},"commit":{"parents":[{"commit":"21dd54c0e76e4e0c02f3a424807d1049ea9103ed","subject":"Move kernel tuning tasks to lxc_post_install.yml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/21dd54c0e76e4e0c02f3a424807d1049ea9103ed"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 16:58:25.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 19:14:59.000000000","tz":0},"subject":"Remove per-backend container creation tasks files","message":"Remove per-backend container creation tasks files\n\nThe same can be achieved with two ternary operators. Much less code\nand duplication.\n\nChange-Id: I59bbbcd8a66970a4fb30339aed457b50fb4dad50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/09a265001269688a7df63565a7f14b7335eae9fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/09a265001269688a7df63565a7f14b7335eae9fc"}]},"branch":"refs/heads/master"},"41a1d97a4e4dc7170ed714cabe8545ce3b72248b":{"kind":"NO_CHANGE","_number":6,"created":"2021-02-17 19:41:00.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/39/776239/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/39/776239/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/39/776239/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/39/776239/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/39/776239/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/39/776239/6"}}},"commit":{"parents":[{"commit":"51ba33d7f50a8c8dd0a9417ed7e8c2b3d00f5e55","subject":"Move kernel tuning tasks to lxc_post_install.yml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/51ba33d7f50a8c8dd0a9417ed7e8c2b3d00f5e55"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 16:58:25.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 19:40:30.000000000","tz":0},"subject":"Remove per-backend container creation tasks files","message":"Remove per-backend container creation tasks files\n\nThe same can be achieved with two ternary operators. Much less code\nand duplication.\n\nChange-Id: I59bbbcd8a66970a4fb30339aed457b50fb4dad50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/41a1d97a4e4dc7170ed714cabe8545ce3b72248b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/41a1d97a4e4dc7170ed714cabe8545ce3b72248b"}]},"branch":"refs/heads/master"},"7fbdc5fe3c48c95c8555ea6e71099bdd2200d407":{"kind":"REWORK","_number":7,"created":"2021-02-17 20:41:26.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/39/776239/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/39/776239/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/39/776239/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/39/776239/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/39/776239/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/39/776239/7"}}},"commit":{"parents":[{"commit":"51ba33d7f50a8c8dd0a9417ed7e8c2b3d00f5e55","subject":"Move kernel tuning tasks to lxc_post_install.yml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/51ba33d7f50a8c8dd0a9417ed7e8c2b3d00f5e55"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 16:58:25.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 20:41:21.000000000","tz":0},"subject":"Remove per-backend container creation tasks files","message":"Remove per-backend container creation tasks files\n\nThe same can be achieved with two ternary operators. Much less code\nand duplication.\n\nChange-Id: I59bbbcd8a66970a4fb30339aed457b50fb4dad50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7fbdc5fe3c48c95c8555ea6e71099bdd2200d407"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7fbdc5fe3c48c95c8555ea6e71099bdd2200d407"}]},"branch":"refs/heads/master"},"1594a5a5a0316c1b840fc9c3880220028ccff794":{"kind":"REWORK","_number":8,"created":"2021-02-17 22:11:34.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/39/776239/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/39/776239/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/39/776239/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/39/776239/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/39/776239/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/39/776239/8"}}},"commit":{"parents":[{"commit":"51ba33d7f50a8c8dd0a9417ed7e8c2b3d00f5e55","subject":"Move kernel tuning tasks to lxc_post_install.yml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/51ba33d7f50a8c8dd0a9417ed7e8c2b3d00f5e55"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 16:58:25.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 22:11:28.000000000","tz":0},"subject":"Remove per-backend container creation tasks files","message":"Remove per-backend container creation tasks files\n\nThe same can be achieved with two ternary operators. Much less code\nand duplication.\n\nChange-Id: I59bbbcd8a66970a4fb30339aed457b50fb4dad50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/1594a5a5a0316c1b840fc9c3880220028ccff794"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/1594a5a5a0316c1b840fc9c3880220028ccff794"}]},"branch":"refs/heads/master"},"bf5e42e6e474da63b5614cb2f28006c24e4dd668":{"kind":"REWORK","_number":9,"created":"2021-02-18 08:20:41.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/39/776239/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/39/776239/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/39/776239/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/39/776239/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/39/776239/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/39/776239/9"}}},"commit":{"parents":[{"commit":"51ba33d7f50a8c8dd0a9417ed7e8c2b3d00f5e55","subject":"Move kernel tuning tasks to lxc_post_install.yml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/51ba33d7f50a8c8dd0a9417ed7e8c2b3d00f5e55"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 16:58:25.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-18 08:20:33.000000000","tz":0},"subject":"Remove per-backend container creation tasks files","message":"Remove per-backend container creation tasks files\n\nThe same can be achieved with two ternary operators. Much\nless code and duplication.\n\nDefault lxc_container_backing_store to \u0027dir\u0027 to match the\nlxc_container_create role and further simplify the code.\nChange-Id: I59bbbcd8a66970a4fb30339aed457b50fb4dad50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/bf5e42e6e474da63b5614cb2f28006c24e4dd668"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/bf5e42e6e474da63b5614cb2f28006c24e4dd668"}]},"branch":"refs/heads/master"},"7fc5e9950ea9e8b379295837fede70f864502439":{"kind":"REWORK","_number":10,"created":"2021-02-18 09:19:51.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/39/776239/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/39/776239/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/39/776239/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/39/776239/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/39/776239/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/39/776239/10"}}},"commit":{"parents":[{"commit":"51ba33d7f50a8c8dd0a9417ed7e8c2b3d00f5e55","subject":"Move kernel tuning tasks to lxc_post_install.yml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/51ba33d7f50a8c8dd0a9417ed7e8c2b3d00f5e55"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 16:58:25.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-18 09:19:47.000000000","tz":0},"subject":"Remove per-backend container creation tasks files","message":"Remove per-backend container creation tasks files\n\nThe same can be achieved with two ternary operators. Much\nless code and duplication.\n\nDefault lxc_container_backing_store to \u0027dir\u0027 to match the\nlxc_container_create role and further simplify the code.\nChange-Id: I59bbbcd8a66970a4fb30339aed457b50fb4dad50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7fc5e9950ea9e8b379295837fede70f864502439"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7fc5e9950ea9e8b379295837fede70f864502439"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
