)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~776240","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I2c23303a355c211c1d9d50238a24ef11c3f38580","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"lxc-hosts-cleanup","hashtags":[],"change_id":"I2c23303a355c211c1d9d50238a24ef11c3f38580","subject":"Remove single task from lxc_apparmor file","status":"MERGED","created":"2021-02-17 17:06:04.000000000","updated":"2021-03-01 00:45:59.000000000","submitted":"2021-03-01 00:44:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"776240-lxc-hosts-cleanup","meta_rev_id":"7ac45971fb2f87459d0c3d4e741f18bd79c3b17b","_number":776240,"virtual_id_number":776240,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-01 00:44:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2021-02-24 08:54:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2021-02-23 09:41:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2021-02-24 08:54:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-17 18:43:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-23 09:41:48.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2021-02-24 08:54:35.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"26c5d72af4c87d1c25dfb856614798cada7979fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-17 17:06:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5d877c67aa6eef48cd393ad06ff267410c6e3870","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-17 17:55:41.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"36a597e06c4a89ec2903e4115559287f83f387f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 18:43:02.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"3d1a4b0a73a88886934b5bec64d5495aa0fa49db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-17 18:43:17.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"7348022ad95a0541879904b93507a5020e2aac48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 19:16:01.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b85ccfd1deec4f66bc70025a03f2cf5c : SUCCESS in 9m 39s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/496e813f2fd44a2e8d07d060682d6c76 : FAILURE in 8m 59s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/fa7aaa62d6ed4faf84d38ecbf4be6735 : FAILURE in 13m 36s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a2af47ac417540d38f203043437b9d6d : FAILURE in 10m 19s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fab7131e098f48cbab92f13e1765bf94 : FAILURE in 12m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26df049f0fa44c50aaf2d9b5b85ba3e4 : SUCCESS in 5m 33s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/b10a2d96d0824957af9ff0fcd5a39bbb : FAILURE in 19m 11s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/aa2240432453414e8261675700ce4f9a : FAILURE in 15m 33s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c3801f60e7dd4ba4874ac9d21df5b27b : FAILURE in 17m 39s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/86c156fbe3524058a2a7a294a29da66b : FAILURE in 12m 21s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9e941754270a499e928c0adb293c47c0 : FAILURE in 11m 24s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be16628a36804971bfdcfd2b2b6e4f2b : FAILURE in 11m 14s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/452581cf8d514455a600c53296a9fb42 : FAILURE in 12m 00s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b1a08314571c445392dae58afa2859ad : FAILURE in 11m 54s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6adff0e600474a41b7f92eec9afa14dd : FAILURE in 11m 19s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/529e195005534f149ddca7ca3e823d2d : FAILURE in 11m 25s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/beb5cb80709a4fb398f5680cbb09733b : FAILURE in 12m 02s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/afe4687e0cc44dc2ae7d75e161620173 : FAILURE in 13m 19s","accounts_in_message":[],"_revision_number":3},{"id":"fd679eea25645f4094d594007d23fc80a593fa71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-17 19:41:35.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"9af2ec8af352b28ec8b88abeab9f99647bdc7e90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-17 20:41:37.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"834f7c31f4ab1bfc41d2e9fe87d01bd1d230fdbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 21:59:30.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2d98a472aa40457b9ca7cedb4258a156 : SUCCESS in 9m 38s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/a7a134ea96b54fae9817187614ed8be6 : SUCCESS in 14m 26s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/240fbd2e0ffc4d699750d22d885b59b3 : SUCCESS in 16m 26s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/06b8d77448af4ff9b577aac64b2ac21d : SUCCESS in 10m 25s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6fbca90c8170463e9a6d541a2907ba9d : SUCCESS in 15m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54f08a62f7444069a6f76b4b2fd3fcb9 : SUCCESS in 6m 46s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/821adf85264f4f4c9de8cd7f501a3614 : SUCCESS in 1h 12m 44s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/7fbd2d2c27a3484394f3fd9d37e1defe : SUCCESS in 1h 01m 44s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dbdff8d238e04df9bd18289991e59152 : SUCCESS in 1h 03m 50s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/973544382eb4466a86f1fa70b8bacacb : SUCCESS in 19m 45s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1efd8248a35943b889664328cd2dbbb2 : SUCCESS in 15m 16s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a61bf5cefbc407399287773354b0805 : SUCCESS in 17m 28s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a4505dcdc3c14499a0f9e8251e171d7b : SUCCESS in 17m 26s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb15626667cd4702a6c76d754651dc21 : SUCCESS in 16m 38s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d17b44cf6ea5454caef57927b8d52007 : SUCCESS in 17m 16s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/060a055f73604972bfb716a378698f3e : SUCCESS in 16m 41s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a5fba9666427459cbae3e6b327623d5f : SUCCESS in 16m 51s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e6098979d8bf4a05a3b69c41c9318db8 : SUCCESS in 24m 03s","accounts_in_message":[],"_revision_number":5},{"id":"363afa568693689980dd4c2ff4735b891a18e9ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-17 22:12:12.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"5239c092dbafbada2a834b7ca8b98e3ba8a6579b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 22:40:18.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/653e0d0484164672a061146534052479 : SUCCESS in 9m 14s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/bd87b49c230f496bb6cf4b9615a4206d : FAILURE in 10m 05s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/3871268125c54cd0aec5c6c265172379 : FAILURE in 13m 32s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9f07456f638a4fef9481d51c2dd72d47 : FAILURE in 10m 24s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9b5da2a684f41fabc7dd3945fa9b88c : FAILURE in 11m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b197cf8e0e854385b57ea07d252554bb : SUCCESS in 5m 05s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/ab94f91d4b534e9f8ef87b61a43c2a85 : FAILURE in 21m 05s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/385188ef4d5646b88ce77743704b6dd1 : FAILURE in 16m 00s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4ec4050304584118bb45ee3fbf5f7a3d : FAILURE in 17m 17s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/cdc576a1df65485292dcfc1ee22a4b7e : FAILURE in 15m 36s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8d774b5f9d2c4d7baa06a5f8c30af489 : FAILURE in 11m 47s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61db2ccb30354691ba9de8ca6ef799fd : FAILURE in 11m 34s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c277e76e11c2427bbb099add0d338c9f : FAILURE in 12m 24s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a8bb32f37be4f28a48e6ad1c23f115b : FAILURE in 11m 59s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a9ec1b8eaebd40d3a7d1ce6c7820990d : FAILURE in 12m 53s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8c0a91fd5fd41218d1fdd78cfbe29a8 : FAILURE in 12m 52s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3d2ec661f9e64ee684c283b04cc328f1 : FAILURE in 12m 27s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3bc76d7e551e415193c99da7f6f37140 : FAILURE in 12m 14s","accounts_in_message":[],"_revision_number":6},{"id":"c9eafb24a4830ba1287373a2225d9613d601bb6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-18 08:20:52.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"9fca35d4613ff215f429f1a656334239c130f38c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 08:54:56.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4e7d3000e97041d7939e85a3e19799d4 : SUCCESS in 11m 59s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/1ee8f9a44dd647269f73ae5d74d148a9 : FAILURE in 9m 03s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/3cc5b065cb9b41f4b6048584f75c995d : FAILURE in 16m 57s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e6423514ee4344c9b913fa67a645f2e2 : FAILURE in 10m 48s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a83b8e3c1634a04a4329aad8416f75c : FAILURE in 12m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f05a5ff07c1845b8a92ad63aa2c411de : SUCCESS in 6m 10s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/b8e1926053284e5ea5a728859dc31bd5 : FAILURE in 23m 31s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/9a2d11f6be314f13a905dfda6c1a48e6 : FAILURE in 15m 43s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8e9b1de37134864856a785046f00d6d : FAILURE in 17m 28s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/fafcd430785849eebbc6823e2e0b4737 : FAILURE in 18m 57s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f18fdd6d42384c5084da21961f0e0986 : FAILURE in 14m 09s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d58f1b9aebe41d1a35accbde5211588 : FAILURE in 14m 40s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/98e90a1da07d4f2987e990235caf3c49 : FAILURE in 12m 52s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b7e925a427e4df7b61ec4f937e493a8 : FAILURE in 14m 10s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/49adbba831b7458a95ce36ce853c3380 : SUCCESS in 16m 26s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23c0e539c594492c8b07c6df53290990 : SUCCESS in 18m 33s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a982c8154f1d4a988f12ce212ed2f9af : FAILURE in 13m 17s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c30a53b1d4404d66a3b8bb992f15da95 : FAILURE in 13m 35s","accounts_in_message":[],"_revision_number":7},{"id":"e144741f85b79c7425cdda979a12faae3e5fa198","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-18 09:20:01.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"637a47fb236ebc6a69d922113c62217ba2892e16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 10:49:54.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/fbdfebe47d2a4e6ab303ebc84daf9e89 : SUCCESS in 8m 36s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/fa292a40e56e4bf1b1b10aad1363a182 : SUCCESS in 10m 29s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/a0b581b9d92a47c4ba8730068891e4f6 : SUCCESS in 13m 51s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8a25a0c0be3346808ff3f81bb947f5cb : SUCCESS in 10m 50s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/79107735076342459d7add43b14ee43f : SUCCESS in 12m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/675f912a888243d19019cc84472ab49a : SUCCESS in 6m 07s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/4640f490731d45dcbe44563eb4cdb884 : SUCCESS in 1h 15m 50s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/37b409adce9245f69a70c51a6292c728 : SUCCESS in 1h 07m 00s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ae2b4015cb848fab956ca36a312f90e : SUCCESS in 1h 08m 30s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/6a260cbbcfa74b93976bd5fa999f6465 : SUCCESS in 24m 31s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d3f77fa931b34604aecba48f76376661 : SUCCESS in 14m 41s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4ca32459f80e4a268b729dd5bdc4d53f : SUCCESS in 17m 11s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/731b5d15be7e4b9385a8c1b41392b68b : SUCCESS in 23m 05s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/74605ddfbba74db2aad7ab737f2e10d5 : SUCCESS in 19m 13s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f13c5bfcf30c43f09fc004f75a535ea9 : SUCCESS in 16m 14s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a2af03cf8e9f4185be74472e5b8d303e : SUCCESS in 15m 47s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d99db44af3ab415288c3e9c817a02614 : SUCCESS in 21m 51s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/425a60c5c87045edaaad530eab904435 : SUCCESS in 18m 01s","accounts_in_message":[],"_revision_number":8},{"id":"586dfc3f7afd3ed7cd107c5453cf177a2d5988ca","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-22 08:59:50.000000000","message":"Topic set to lxc-hosts-cleanup","accounts_in_message":[],"_revision_number":8},{"id":"e89dc1f4583e6304281d7698745dfa4d363b832d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-23 09:41:48.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"bc74322722922cbb41e815fcfe61b83f018f0347","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-02-24 08:54:35.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"608c05a2fbc2e2f9f2d25b9fba5d0b9c74a12ad9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 08:57:53.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"c7637cb88e2cd23e49f27ea07feaaa160a703ea5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 12:08:18.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7ee89485e9464969a8718f15f636ac55 : SUCCESS in 13m 41s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/d9ddd331e99340ab95548c13a207a55d : SUCCESS in 12m 38s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f432fc2d76ed487e874c3e1c0491a410 : SUCCESS in 17m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a5446691ad5487f81dd21243c9f2fe7 : SUCCESS in 9m 20s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/f1cc76b5a568476a925f5b7a0aa054de : SUCCESS in 1h 21m 35s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/f66d4e127def434ca1444c10c0ea6b54 : FAILURE in 58m 45s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fea1a4468a9349b1ab1449c34ef60871 : TIMED_OUT in 3h 04m 41s","accounts_in_message":[],"_revision_number":8},{"id":"6e644ebffdb98a942bcbb3128db3929c2a8596da","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-28 22:02:47.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"ab44de7f1c8a52a6e2f41b4cb5f1d6de8c5b78ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-28 23:19:48.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d26f3d27306142dbacb53ce360c514f4 : SUCCESS in 14m 20s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/077f29442f1d481f97fd26803c9c9f5d : SUCCESS in 10m 43s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/9876ff7d638447589a2a892abeb7b001 : SUCCESS in 13m 24s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a22266ea57374285ad8da6a538e09b5e : SUCCESS in 11m 56s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb0d3b3e6bbb42b2bf1f08c63e0f7bfd : SUCCESS in 15m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef98006a9bf946c5bbae745af87d5201 : SUCCESS in 5m 28s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/43ac492430fa42bd8ea0a0faa28f6335 : SUCCESS in 1h 09m 43s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/0a8a76e27f354ce1b5ea601ebe3b6d8b : SUCCESS in 1h 00m 01s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df2876deb32a4a3fbe7d049f10adb57e : SUCCESS in 1h 03m 04s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/ef8825061c054a77922c82d0442617d6 : SUCCESS in 22m 10s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f874ca53e7b24706a42e76531d6958bf : SUCCESS in 16m 55s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/58b0e10dbfc74d7d95d908132b6ac65e : SUCCESS in 16m 48s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/08b89e889965432f82116b471e637759 : SUCCESS in 16m 37s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a74278ea0256413a8b65fd4b7282ba5d : SUCCESS in 17m 24s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3f46dcf04ce442aab596813a8a4236e9 : SUCCESS in 17m 00s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2bfe2c5bf0ac4963bf04f169f9391d98 : SUCCESS in 17m 41s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b6fd4617691c47319ed59ddb9d2fb025 : SUCCESS in 17m 31s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/28be2fc7e194490c8bd35d562e4790ba : SUCCESS in 17m 45s","accounts_in_message":[],"_revision_number":8},{"id":"a9c5ffb6b088bd9af1f2b391e8a9fe6eb73d4b0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-28 23:20:01.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"059118a95024ae5fd765fe9412107efc893ab2ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 00:44:49.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1d4b11bc2dda4506b7d9d2049855d516 : SUCCESS in 7m 11s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/5ada5f38ec784f4c829afef716ab0d26 : SUCCESS in 10m 34s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/af8d6f384bda4fc0b93350797bed2b2f : SUCCESS in 15m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d7240ed28a84437b518bfb44b1145c2 : SUCCESS in 5m 26s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/3bc77d854e834cf989a17a2866e0ce5d : SUCCESS in 1h 13m 48s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/d54efb61fd6148e99ecb2fb508a71106 : SUCCESS in 57m 31s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1807ea4f9535499ea3c50c6c473ee06b : SUCCESS in 1h 02m 30s","accounts_in_message":[],"_revision_number":8},{"id":"e66c4436141c57c15aa35079ffb9b374cdbc81e1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 00:44:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"7ac45971fb2f87459d0c3d4e741f18bd79c3b17b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 00:45:59.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a22840e986b4beea63461a2e34fef1e : SUCCESS in 45s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"611e278642341fd2285756044d1d9099febe1b5d","revisions":{"7aa5d0277f901792ac618cec9033e4adc81eea04":{"kind":"REWORK","_number":1,"created":"2021-02-17 17:06:04.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/40/776240/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/40/776240/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/776240/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/776240/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/776240/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/776240/1"}}},"commit":{"parents":[{"commit":"aee8743f0bae492aab320c9dbd4f8eec405f4aa4","subject":"Remove per-backend container creation tasks files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/aee8743f0bae492aab320c9dbd4f8eec405f4aa4"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 17:00:16.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 17:00:16.000000000","tz":0},"subject":"Remove single task from lxc_apparmor file","message":"Remove single task from lxc_apparmor file\n\nThis can go in the lxc_install_apt tasks file where the previous\nlxc_apparmor file was included\n\nChange-Id: I2c23303a355c211c1d9d50238a24ef11c3f38580\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7aa5d0277f901792ac618cec9033e4adc81eea04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7aa5d0277f901792ac618cec9033e4adc81eea04"}]},"branch":"refs/heads/master"},"d907d7b86bbb32ef7e90dc3aac6428320e979701":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-02-17 17:55:41.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/40/776240/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/40/776240/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/776240/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/776240/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/776240/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/776240/2"}}},"commit":{"parents":[{"commit":"f865348d1e1aacd36fcf2f92c53733f81f826f53","subject":"Remove per-backend container creation tasks files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f865348d1e1aacd36fcf2f92c53733f81f826f53"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 17:00:16.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 17:55:41.000000000","tz":0},"subject":"Remove single task from lxc_apparmor file","message":"Remove single task from lxc_apparmor file\n\nThis can go in the lxc_install_apt tasks file where the previous\nlxc_apparmor file was included\n\nChange-Id: I2c23303a355c211c1d9d50238a24ef11c3f38580\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d907d7b86bbb32ef7e90dc3aac6428320e979701"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d907d7b86bbb32ef7e90dc3aac6428320e979701"}]},"branch":"refs/heads/master","description":"Rebase"},"49193044e6c6cd5319a58040b2848a0cf9b0982d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-02-17 18:43:17.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/40/776240/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/40/776240/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/776240/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/776240/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/776240/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/776240/3"}}},"commit":{"parents":[{"commit":"b6cd9d6f4efca555c65d24070514a091b4729b1b","subject":"Remove support for gentoo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b6cd9d6f4efca555c65d24070514a091b4729b1b"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 17:00:16.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 18:43:17.000000000","tz":0},"subject":"Remove single task from lxc_apparmor file","message":"Remove single task from lxc_apparmor file\n\nThis can go in the lxc_install_apt tasks file where the previous\nlxc_apparmor file was included\n\nChange-Id: I2c23303a355c211c1d9d50238a24ef11c3f38580\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/49193044e6c6cd5319a58040b2848a0cf9b0982d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/49193044e6c6cd5319a58040b2848a0cf9b0982d"}]},"branch":"refs/heads/master","description":"Rebase"},"41236520f29541aff0e15e6879f7500c0a0e1f7f":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-02-17 19:41:35.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/40/776240/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/40/776240/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/776240/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/776240/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/776240/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/776240/4"}}},"commit":{"parents":[{"commit":"59abc5a28882039ff3463caf451940b5a85a963a","subject":"Remove support for gentoo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/59abc5a28882039ff3463caf451940b5a85a963a"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 17:00:16.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 19:41:35.000000000","tz":0},"subject":"Remove single task from lxc_apparmor file","message":"Remove single task from lxc_apparmor file\n\nThis can go in the lxc_install_apt tasks file where the previous\nlxc_apparmor file was included\n\nChange-Id: I2c23303a355c211c1d9d50238a24ef11c3f38580\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/41236520f29541aff0e15e6879f7500c0a0e1f7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/41236520f29541aff0e15e6879f7500c0a0e1f7f"}]},"branch":"refs/heads/master","description":"Rebase"},"5574801ef142627546f585e51f308364c6a97b44":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-02-17 20:41:37.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/40/776240/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/40/776240/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/776240/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/776240/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/776240/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/776240/5"}}},"commit":{"parents":[{"commit":"fb1a6e1b6421e88161b386e11166c376bd9132fb","subject":"Ensure sysctl command is present in the lxc base image for centos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/fb1a6e1b6421e88161b386e11166c376bd9132fb"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 17:00:16.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 20:41:37.000000000","tz":0},"subject":"Remove single task from lxc_apparmor file","message":"Remove single task from lxc_apparmor file\n\nThis can go in the lxc_install_apt tasks file where the previous\nlxc_apparmor file was included\n\nChange-Id: I2c23303a355c211c1d9d50238a24ef11c3f38580\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/5574801ef142627546f585e51f308364c6a97b44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/5574801ef142627546f585e51f308364c6a97b44"}]},"branch":"refs/heads/master","description":"Rebase"},"ff51612ffd95ab733219f83b73bfc1b293169efd":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-02-17 22:12:12.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/40/776240/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/40/776240/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/776240/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/776240/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/776240/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/776240/6"}}},"commit":{"parents":[{"commit":"1594a5a5a0316c1b840fc9c3880220028ccff794","subject":"Remove per-backend container creation tasks files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/1594a5a5a0316c1b840fc9c3880220028ccff794"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 17:00:16.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 22:12:12.000000000","tz":0},"subject":"Remove single task from lxc_apparmor file","message":"Remove single task from lxc_apparmor file\n\nThis can go in the lxc_install_apt tasks file where the previous\nlxc_apparmor file was included\n\nChange-Id: I2c23303a355c211c1d9d50238a24ef11c3f38580\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ff51612ffd95ab733219f83b73bfc1b293169efd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ff51612ffd95ab733219f83b73bfc1b293169efd"}]},"branch":"refs/heads/master","description":"Rebase"},"95db59c27a04fe1d2080159fd954b52c81fa6748":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-02-18 08:20:52.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/40/776240/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/40/776240/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/776240/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/776240/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/776240/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/776240/7"}}},"commit":{"parents":[{"commit":"bf5e42e6e474da63b5614cb2f28006c24e4dd668","subject":"Remove per-backend container creation tasks files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/bf5e42e6e474da63b5614cb2f28006c24e4dd668"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 17:00:16.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-18 08:20:52.000000000","tz":0},"subject":"Remove single task from lxc_apparmor file","message":"Remove single task from lxc_apparmor file\n\nThis can go in the lxc_install_apt tasks file where the previous\nlxc_apparmor file was included\n\nChange-Id: I2c23303a355c211c1d9d50238a24ef11c3f38580\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/95db59c27a04fe1d2080159fd954b52c81fa6748"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/95db59c27a04fe1d2080159fd954b52c81fa6748"}]},"branch":"refs/heads/master","description":"Rebase"},"611e278642341fd2285756044d1d9099febe1b5d":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2021-02-18 09:20:01.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/40/776240/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/40/776240/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/776240/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/776240/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/776240/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/776240/8"}}},"commit":{"parents":[{"commit":"7fc5e9950ea9e8b379295837fede70f864502439","subject":"Remove per-backend container creation tasks files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7fc5e9950ea9e8b379295837fede70f864502439"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 17:00:16.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-18 09:20:01.000000000","tz":0},"subject":"Remove single task from lxc_apparmor file","message":"Remove single task from lxc_apparmor file\n\nThis can go in the lxc_install_apt tasks file where the previous\nlxc_apparmor file was included\n\nChange-Id: I2c23303a355c211c1d9d50238a24ef11c3f38580\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/611e278642341fd2285756044d1d9099febe1b5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/611e278642341fd2285756044d1d9099febe1b5d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
