)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~776241","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I3c427bc670815ee5bac677925b56df644f460271","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","hashtags":[],"change_id":"I3c427bc670815ee5bac677925b56df644f460271","subject":"Remove workarounds for old systemd bugs","status":"MERGED","created":"2021-02-17 17:06:04.000000000","updated":"2021-03-01 14:07:10.000000000","submitted":"2021-03-01 14:05:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"776241","meta_rev_id":"f97dd6e0346960dd308288a03e9d963a633e09df","_number":776241,"virtual_id_number":776241,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-01 14:05:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":2,"date":"2021-03-01 12:51:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":2,"date":"2021-02-28 22:02:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":1,"date":"2021-03-01 12:51:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-17 18:32:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-23 09:42:40.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2021-02-28 22:02:35.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2021-03-01 12:51:25.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"}],"messages":[{"id":"484660503ab018961e64fe3419b88b0a96edeb77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-17 17:06:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3047e519ed585c75c9c17f7c7555ff47b9b01063","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-17 17:55:50.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"d90e15531aa91288ac6ae1067e66bc5cc8bae7fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 18:32:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1ee9294ba46e496ea16a6f020c8c6e0a : SUCCESS in 8m 53s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/f4307fc0ab0245d9b59e81f37f2413ec : FAILURE in 8m 59s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/a92053512d36436ca301afae285a8c68 : FAILURE in 9m 30s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/51eb973df31c41b09c5d673e9218a6e7 : FAILURE in 10m 12s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/098c1168618d40148ac8b99f3e721f89 : FAILURE in 11m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40eeee4bb8b34430a0e8b313f2a0c9b5 : SUCCESS in 5m 18s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/930f8c6fce1344d6a6644b1fc2217eb2 : FAILURE in 16m 24s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/1a7f47ff28434c8d881584c00545d2f8 : FAILURE in 14m 06s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69da6e873f7b4db4a89fa4e1b15c0ab8 : FAILURE in 18m 10s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/c5c63bd3b6c44b1e96bbf2de80ed5b9c : FAILURE in 16m 00s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e84cb69ba02e4e84b0e4aacc107a9b0c : FAILURE in 13m 18s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca9bf67ac1344a888e6f27bdb7be04b2 : FAILURE in 12m 47s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a8af890a455b480b854d2cd9badd5587 : FAILURE in 11m 29s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af23aee7b3db4843b66eebe20df2a4b2 : FAILURE in 11m 41s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1b600dbf94594959828177fa07795c88 : FAILURE in 12m 23s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/86944a9771b7401f9e505d13e0c00489 : FAILURE in 12m 41s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/661508a6920c41a5832a9d182cdb12e5 : FAILURE in 12m 13s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52ff0bddc1c94f6cadabfb99bbf45ad3 : FAILURE in 13m 19s","accounts_in_message":[],"_revision_number":2},{"id":"474dacb42ddbbc1faa6c648c65b34f3d8c344587","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-17 18:47:45.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"379da9305aa07b3dc3d354af92d727af7d03b895","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-17 19:15:16.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"8632d450d5fc2ebf1d5a103d5afbec1fa5b5b67d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-17 19:41:00.000000000","message":"Uploaded patch set 5: New patch set was added with same tree, parent, and commit message as Patch Set 4.","accounts_in_message":[],"_revision_number":5},{"id":"f2a618ccda4112e318708399b83c16a1ed2415ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 20:20:37.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8c52a4ffa97f4e9681d3e9753aa07469 : SUCCESS in 9m 16s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/a1921d16677d45dcba59caeb1692e24e : FAILURE in 11m 06s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/15a96a40778b451da56d6893b5953e40 : FAILURE in 15m 09s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/361aad1eda084cb98e0ea0d3ee4cd331 : FAILURE in 11m 54s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1fba08fe04c64c1eb91918fec0987625 : FAILURE in 11m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/731a9e6b6c8649d59332004600ffc789 : SUCCESS in 7m 30s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/2d60a0364b4046f487c20fdd014ea51b : FAILURE in 22m 45s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/4354a2e7191746838f3156313c1ddf50 : FAILURE in 17m 27s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6190daffbd0649878dd879821301fb0c : FAILURE in 19m 31s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/4debe110241647e3a30d29e2ea3049aa : FAILURE in 18m 22s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/206038614f384c16a035b905312ab459 : FAILURE in 13m 50s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39c569140be64bbc93926195fdc770b4 : FAILURE in 12m 38s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c4577ee0efcd4b7c92a9b21d6ccfabb9 : FAILURE in 12m 16s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b4b780bc8a6e47378551f4d2562f4a80 : FAILURE in 12m 41s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/83f89bd3a32245dc852a9f6fe73f378e : FAILURE in 13m 26s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3096a8537c434ee096e9023488c3ad52 : FAILURE in 15m 14s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/80065455fd8349b6be744ec4ba50775e : FAILURE in 13m 33s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5b00003c1124c34b68af10551dfe5fd : FAILURE in 15m 06s","accounts_in_message":[],"_revision_number":5},{"id":"b759e9a1fc87eaa4d6b21f9c725b1ba028efeff9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-17 20:41:47.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"0648e63173dc139cdd44b500af4474d7e544c3fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 21:24:21.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/171d4937114b4dc696b104810fa413e8 : SUCCESS in 13m 07s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/94c5f9f5db6541dea5ad55243c434e77 : FAILURE in 10m 49s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/aa6579716aa8402496ff67cb75d4b7a1 : FAILURE in 13m 53s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b57f74f3aecb4f8ca91c925466be77fe : FAILURE in 11m 39s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34066b1d66fd4d30b711daca9c08923f : FAILURE in 11m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3dc837761954101952b0ca1d88e1f89 : SUCCESS in 6m 00s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/120f970c5bab4ec09063fdb4c5a679af : FAILURE in 26m 20s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/bc004c3927e54949a9b0fa6edc65f5d2 : FAILURE in 18m 05s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/242a845b093d438ead6e7383022adaf1 : FAILURE in 20m 34s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/b86b6bccf0ae491d92c5a80de3681b0f : FAILURE in 17m 00s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e1fe5f80f6744c6cae0edbc35d47ae6f : FAILURE in 13m 43s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7bad6a697423479aa9827da91afcc5ea : FAILURE in 12m 09s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c85e347e786c4a2cb9f6254bc387fb28 : FAILURE in 13m 21s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/680d0acbd9f541da96cf0ff29918b1ce : FAILURE in 11m 43s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e6ed2ae192634eb5993977eb76178f74 : FAILURE in 13m 18s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77a463ad82bb48409d9fe2e6f9ab10a5 : FAILURE in 12m 15s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b025dc74f1314622aead48c093d4d7f4 : FAILURE in 19m 25s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e2660b04c284e10ad5b6798d2b47cfb : FAILURE in 14m 17s","accounts_in_message":[],"_revision_number":6},{"id":"6b574ac02d970f96e434205e5953b0107a248339","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-17 22:12:27.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"c3e13dc9ae748e425d5aa345178b437055141bb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 22:36:14.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/704c7668269243ccbbfc689b543fd65d : SUCCESS in 7m 20s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/449bc68dfcd042e0864b9482bde4164f : FAILURE in 10m 26s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/f2a86fa63e894e36986869d598a715ae : FAILURE in 14m 28s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2e69cc68979e4705badd514e4e5542ee : FAILURE in 11m 03s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0640849b10da40c69cb34320e8e7aa1b : FAILURE in 11m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/942eb5235e9e4462bcdc42c937a117ac : SUCCESS in 5m 35s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/f0d87b76ffe04cafa1c964cc447bb888 : FAILURE in 19m 04s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/3064ddfa00354d75a461da02d1e8d4cf : FAILURE in 15m 49s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a75015f9eee446fba3724d3c0a132f01 : FAILURE in 19m 07s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/9d387654621a4a3e9f780c02c63e7149 : FAILURE in 15m 40s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6efafe2d705344628b89211ee1597da3 : FAILURE in 13m 56s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a712deb7a5c5452b97145d75fcba7b43 : FAILURE in 11m 53s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/481befb09ed84c178dc745f11edcf335 : FAILURE in 11m 45s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/50047277f3f448ee8f0323889037ee96 : FAILURE in 12m 54s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/24de103795684931a9f2c6e936ae9572 : FAILURE in 12m 37s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00111b2ad2b94371b3c6e43f80b76b86 : FAILURE in 12m 03s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5ae1f516dc52488d9eb1439e149883f6 : FAILURE in 14m 32s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ebe09c5536f4b29876e6a067cd8f06d : FAILURE in 12m 07s","accounts_in_message":[],"_revision_number":7},{"id":"bad1338c45a6a7e42e3ac403ed5de25a391d3c31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-18 08:21:03.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"afab509c2dcb33f26a043960456bfe5c01ac7cb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 08:51:12.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2df17774bcf04a5b9cbfe4d998dc0dd5 : SUCCESS in 9m 15s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/c456bca43a7c4c088d9ffb4e20e483c4 : FAILURE in 10m 52s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/c92a0080e9ab4c718a849a778614f983 : FAILURE in 16m 33s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6f80d2d42c4244bcb1bf42c4585c0424 : FAILURE in 11m 11s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d3fedb9f7824d56830b3ee94c33c0a3 : FAILURE in 10m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f1a6ca0957e4392904291b4082e8e0e : SUCCESS in 6m 25s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/a49bc299c60b4508b3d06d7cb0dde5e5 : FAILURE in 22m 03s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/c470a1d3ad1342afa5c4e35d5f37e19a : FAILURE in 16m 41s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11536047d30d43f5b508d99f08294505 : FAILURE in 16m 50s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/ad808cb32cd3433493c04bdf256b145c : FAILURE in 17m 26s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/31b7c41414804fe98f6bef9ddc63297d : FAILURE in 12m 57s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8efa543a145047189709560d6dc74568 : FAILURE in 12m 11s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f9589f8b788548f7826e7f5837c75f52 : FAILURE in 13m 16s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3a243dbe7b54f4aaef25e115961d622 : FAILURE in 12m 32s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4f4048a25d5b41a58b23214f15a93a98 : SUCCESS in 16m 28s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9409d77b63604eaea2f6adeaab282837 : SUCCESS in 17m 21s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/91c16b37357a49c29fa38fed10fd6e27 : FAILURE in 13m 21s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b40552324934bdd8d8b62ed70b16e45 : FAILURE in 12m 18s","accounts_in_message":[],"_revision_number":8},{"id":"0917b8e66e4433d6a2dffcc99ce1c1d89532197f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-18 09:20:12.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"a701c39f7d9cde5823087083c021a587bd7fb1ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 11:00:26.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c7f078f3acd845a4a73a57758a9863a6 : SUCCESS in 8m 33s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/54140b55b0804cb9bbc4f98939fc7470 : SUCCESS in 11m 04s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/3acab25695ee40c194680b64fabbbce5 : SUCCESS in 14m 17s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f7b8475debb645fea7aa7c0ba6b8bfe2 : SUCCESS in 12m 07s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f85853f6e2d4e78a6d31c95b395fe92 : SUCCESS in 12m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec97e164c95d4eea970142c4b9a9443d : SUCCESS in 12m 16s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/c3c0ab1110454c2f9e6058b6cd7b9369 : SUCCESS in 1h 13m 38s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/6119cf8146e349f1b516cd1ac568644f : SUCCESS in 1h 08m 11s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f216320fae24c2aae31512369fdbffd : SUCCESS in 1h 00m 53s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/c59d72cc226c4b58b3332a0dccbf1666 : SUCCESS in 22m 21s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/80ab2cabae264be0bf8afa3c95bc92c9 : SUCCESS in 18m 25s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/15798e5f8e5742d39c735068cd4e021f : SUCCESS in 19m 38s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a6db2c61fb29471f84f161d395e3d169 : SUCCESS in 17m 06s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70b245ac443242d89d81effbf546748a : SUCCESS in 20m 24s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/01cba03c198641b08cf4b147211e0d2b : SUCCESS in 16m 14s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c564034916144a085cf9462e9a75fcc : SUCCESS in 19m 25s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/59b3a0c8e3614d3089dc5943df380276 : SUCCESS in 15m 51s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/63fc561efdcf4fe0bc4c60111ac7467e : POST_FAILURE in 23m 07s","accounts_in_message":[],"_revision_number":9},{"id":"e05b98757b5129a67a0d925a45a9bd2f6aff2c95","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-23 09:42:40.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"f431b541f37a7a06c4a2d07a8ae3f8b230b69e9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 11:17:40.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c269f60e3b2e4071a690b5697bccfce2 : SUCCESS in 9m 56s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/aa83cf392c254d56af482cd71195cfb7 : SUCCESS in 13m 45s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/b3919f2580384e7c97aa62fe8a185169 : SUCCESS in 13m 49s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7d307c3ce7224d36bc8fba081a18a8ae : SUCCESS in 11m 31s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7033e304dc744b6e9c9109d3755437ad : SUCCESS in 16m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/641c67c17d2342c0b92c5c844c73ac89 : SUCCESS in 6m 25s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/876b1dd4ba6e4a27ad2503786a8368cc : SUCCESS in 1h 16m 50s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/4348b5e1b7ee4f97940742cefce28aab : SUCCESS in 59m 21s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e57d21c2aee74579899396306201dfc5 : SUCCESS in 1h 10m 53s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/3db92a3e114d4830add58bfee7acb2c9 : SUCCESS in 21m 12s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/121b4f9746da4d3696948c878712f386 : SUCCESS in 17m 50s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ffe927881d464c10946f3b494b64d7eb : SUCCESS in 21m 17s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aee59798d8304290a9c79ccfb4a90997 : SUCCESS in 17m 07s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf34fbbd45b44f2897724f25b39a7d3f : SUCCESS in 17m 36s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f11418a8c3794feaa941dc40f5814817 : SUCCESS in 21m 05s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf66fd3ce2264cd88f899ddab1e9556e : SUCCESS in 16m 58s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/435f223163dd4d7cb4680821fb2c0c56 : SUCCESS in 16m 21s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6976cc5189c942db9765d650fd09c6d4 : SUCCESS in 18m 33s","accounts_in_message":[],"_revision_number":9},{"id":"7c13924edfd723c796020a3dfdff6a5a201cbdae","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-28 22:02:35.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"5fe14ad74378727d54308238753455566ae7d3be","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2021-03-01 12:51:25.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"da104bd0ace0e15908483017c82b638aef78f49f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 12:51:40.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"99f1ba610cc28fa5e4f30ebebf3cf3630aff9723","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 14:05:03.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/26c4c119410048e7a915c4542645ac43 : SUCCESS in 7m 50s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/3ba23807c002465da39075e788d190a9 : SUCCESS in 11m 03s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/78eaef241c5548df9520bda132cf6ce0 : SUCCESS in 11m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10a73ba601734d4983e3f26860f34eae : SUCCESS in 6m 04s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/b09d327d2186459ea988a3d9d49364d7 : SUCCESS in 1h 06m 43s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/1d0a573ab8384f7796563194ae32cbb0 : SUCCESS in 1h 04m 57s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/841eba40968b4c0b8b945d29e658e7b0 : SUCCESS in 1h 04m 31s","accounts_in_message":[],"_revision_number":9},{"id":"5bbfd302c683f8177a3975dde29a108b4c4b5900","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 14:05:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"f97dd6e0346960dd308288a03e9d963a633e09df","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 14:07:10.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7141b790da4446829b1c8c1c21e435fa : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"3f8a5b4ca368d468446256992c456fc37fa2cf28","revisions":{"816cb3db3b06d5787fe235a86ede7bd7353774f7":{"kind":"REWORK","_number":1,"created":"2021-02-17 17:06:04.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/41/776241/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/41/776241/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/776241/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/776241/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/776241/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/776241/1"}}},"commit":{"parents":[{"commit":"7aa5d0277f901792ac618cec9033e4adc81eea04","subject":"Remove single task from lxc_apparmor file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7aa5d0277f901792ac618cec9033e4adc81eea04"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 17:02:43.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 17:02:43.000000000","tz":0},"subject":"Remove workarounds for old systemd bugs","message":"Remove workarounds for old systemd bugs\n\nThese are now resolved:\n\nWere are now later than systemd v231\nhttps://bugs.launchpad.net/ubuntu/+source/lxc/+bug/1497420\nhttps://github.com/lxc/lxc/issues/713\n\nChange-Id: I3c427bc670815ee5bac677925b56df644f460271\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/816cb3db3b06d5787fe235a86ede7bd7353774f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/816cb3db3b06d5787fe235a86ede7bd7353774f7"}]},"branch":"refs/heads/master"},"c2adcdac7cc02e80d87f4c586381b5bc452f458e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-02-17 17:55:50.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/41/776241/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/41/776241/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/776241/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/776241/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/776241/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/776241/2"}}},"commit":{"parents":[{"commit":"d907d7b86bbb32ef7e90dc3aac6428320e979701","subject":"Remove single task from lxc_apparmor file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d907d7b86bbb32ef7e90dc3aac6428320e979701"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 17:02:43.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 17:55:50.000000000","tz":0},"subject":"Remove workarounds for old systemd bugs","message":"Remove workarounds for old systemd bugs\n\nThese are now resolved:\n\nWere are now later than systemd v231\nhttps://bugs.launchpad.net/ubuntu/+source/lxc/+bug/1497420\nhttps://github.com/lxc/lxc/issues/713\n\nChange-Id: I3c427bc670815ee5bac677925b56df644f460271\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c2adcdac7cc02e80d87f4c586381b5bc452f458e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c2adcdac7cc02e80d87f4c586381b5bc452f458e"}]},"branch":"refs/heads/master","description":"Rebase"},"e9c5a9f5e3f6e81a5be82f7da27b9c728f8bd592":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-02-17 18:47:45.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/41/776241/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/41/776241/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/776241/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/776241/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/776241/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/776241/3"}}},"commit":{"parents":[{"commit":"851450f99d44b97a0cfa9f90f8cda0115340c114","subject":"Remove per-backend container creation tasks files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/851450f99d44b97a0cfa9f90f8cda0115340c114"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 17:02:43.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 18:47:28.000000000","tz":0},"subject":"Remove workarounds for old systemd bugs","message":"Remove workarounds for old systemd bugs\n\nThese are now resolved:\n\nWere are now later than systemd v231\nhttps://bugs.launchpad.net/ubuntu/+source/lxc/+bug/1497420\nhttps://github.com/lxc/lxc/issues/713\n\nChange-Id: I3c427bc670815ee5bac677925b56df644f460271\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/e9c5a9f5e3f6e81a5be82f7da27b9c728f8bd592"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/e9c5a9f5e3f6e81a5be82f7da27b9c728f8bd592"}]},"branch":"refs/heads/master"},"97a8230766330edfcf01a710e3200b24aa7776ba":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-02-17 19:15:16.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/41/776241/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/41/776241/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/776241/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/776241/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/776241/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/776241/4"}}},"commit":{"parents":[{"commit":"09a265001269688a7df63565a7f14b7335eae9fc","subject":"Remove per-backend container creation tasks files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/09a265001269688a7df63565a7f14b7335eae9fc"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 17:02:43.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 19:14:59.000000000","tz":0},"subject":"Remove workarounds for old systemd bugs","message":"Remove workarounds for old systemd bugs\n\nThese are now resolved:\n\nWere are now later than systemd v231\nhttps://bugs.launchpad.net/ubuntu/+source/lxc/+bug/1497420\nhttps://github.com/lxc/lxc/issues/713\n\nChange-Id: I3c427bc670815ee5bac677925b56df644f460271\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/97a8230766330edfcf01a710e3200b24aa7776ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/97a8230766330edfcf01a710e3200b24aa7776ba"}]},"branch":"refs/heads/master"},"0c574cc3d149bf204553a1d0f04b9b1a96f02738":{"kind":"NO_CHANGE","_number":5,"created":"2021-02-17 19:41:00.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/41/776241/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/41/776241/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/776241/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/776241/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/776241/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/776241/5"}}},"commit":{"parents":[{"commit":"41a1d97a4e4dc7170ed714cabe8545ce3b72248b","subject":"Remove per-backend container creation tasks files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/41a1d97a4e4dc7170ed714cabe8545ce3b72248b"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 17:02:43.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 19:40:30.000000000","tz":0},"subject":"Remove workarounds for old systemd bugs","message":"Remove workarounds for old systemd bugs\n\nThese are now resolved:\n\nWere are now later than systemd v231\nhttps://bugs.launchpad.net/ubuntu/+source/lxc/+bug/1497420\nhttps://github.com/lxc/lxc/issues/713\n\nChange-Id: I3c427bc670815ee5bac677925b56df644f460271\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/0c574cc3d149bf204553a1d0f04b9b1a96f02738"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/0c574cc3d149bf204553a1d0f04b9b1a96f02738"}]},"branch":"refs/heads/master"},"67bc3c46bd0c82ecfbc60568bc4c5f72afdd17b7":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-02-17 20:41:47.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/41/776241/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/41/776241/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/776241/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/776241/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/776241/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/776241/6"}}},"commit":{"parents":[{"commit":"7fbdc5fe3c48c95c8555ea6e71099bdd2200d407","subject":"Remove per-backend container creation tasks files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7fbdc5fe3c48c95c8555ea6e71099bdd2200d407"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 17:02:43.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 20:41:47.000000000","tz":0},"subject":"Remove workarounds for old systemd bugs","message":"Remove workarounds for old systemd bugs\n\nThese are now resolved:\n\nWere are now later than systemd v231\nhttps://bugs.launchpad.net/ubuntu/+source/lxc/+bug/1497420\nhttps://github.com/lxc/lxc/issues/713\n\nChange-Id: I3c427bc670815ee5bac677925b56df644f460271\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/67bc3c46bd0c82ecfbc60568bc4c5f72afdd17b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/67bc3c46bd0c82ecfbc60568bc4c5f72afdd17b7"}]},"branch":"refs/heads/master","description":"Rebase"},"f76525f9ff82fef74c452350d074deb83b623c4c":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-02-17 22:12:27.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/41/776241/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/41/776241/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/776241/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/776241/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/776241/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/776241/7"}}},"commit":{"parents":[{"commit":"ff51612ffd95ab733219f83b73bfc1b293169efd","subject":"Remove single task from lxc_apparmor file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ff51612ffd95ab733219f83b73bfc1b293169efd"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 17:02:43.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 22:12:27.000000000","tz":0},"subject":"Remove workarounds for old systemd bugs","message":"Remove workarounds for old systemd bugs\n\nThese are now resolved:\n\nWere are now later than systemd v231\nhttps://bugs.launchpad.net/ubuntu/+source/lxc/+bug/1497420\nhttps://github.com/lxc/lxc/issues/713\n\nChange-Id: I3c427bc670815ee5bac677925b56df644f460271\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f76525f9ff82fef74c452350d074deb83b623c4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f76525f9ff82fef74c452350d074deb83b623c4c"}]},"branch":"refs/heads/master","description":"Rebase"},"804ff834fa055c999d353f9be67ced55b5d292a9":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2021-02-18 08:21:03.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/41/776241/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/41/776241/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/776241/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/776241/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/776241/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/776241/8"}}},"commit":{"parents":[{"commit":"95db59c27a04fe1d2080159fd954b52c81fa6748","subject":"Remove single task from lxc_apparmor file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/95db59c27a04fe1d2080159fd954b52c81fa6748"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 17:02:43.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-18 08:21:03.000000000","tz":0},"subject":"Remove workarounds for old systemd bugs","message":"Remove workarounds for old systemd bugs\n\nThese are now resolved:\n\nWere are now later than systemd v231\nhttps://bugs.launchpad.net/ubuntu/+source/lxc/+bug/1497420\nhttps://github.com/lxc/lxc/issues/713\n\nChange-Id: I3c427bc670815ee5bac677925b56df644f460271\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/804ff834fa055c999d353f9be67ced55b5d292a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/804ff834fa055c999d353f9be67ced55b5d292a9"}]},"branch":"refs/heads/master","description":"Rebase"},"3f8a5b4ca368d468446256992c456fc37fa2cf28":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2021-02-18 09:20:12.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/41/776241/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/41/776241/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/776241/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/776241/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/776241/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/776241/9"}}},"commit":{"parents":[{"commit":"611e278642341fd2285756044d1d9099febe1b5d","subject":"Remove single task from lxc_apparmor file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/611e278642341fd2285756044d1d9099febe1b5d"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-17 17:02:43.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-18 09:20:12.000000000","tz":0},"subject":"Remove workarounds for old systemd bugs","message":"Remove workarounds for old systemd bugs\n\nThese are now resolved:\n\nWere are now later than systemd v231\nhttps://bugs.launchpad.net/ubuntu/+source/lxc/+bug/1497420\nhttps://github.com/lxc/lxc/issues/713\n\nChange-Id: I3c427bc670815ee5bac677925b56df644f460271\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3f8a5b4ca368d468446256992c456fc37fa2cf28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3f8a5b4ca368d468446256992c456fc37fa2cf28"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
