)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~778141","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~Ibf405ab9334892705c9bf519e180e608c9d3adad","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","hashtags":[],"change_id":"Ibf405ab9334892705c9bf519e180e608c9d3adad","subject":"DNM - test log collection","status":"ABANDONED","created":"2021-03-02 06:54:05.000000000","updated":"2021-03-03 10:08:54.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0a67c2e9bf3d68f802a601e3213cffe184c51f1a","_number":778141,"virtual_id_number":778141,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-03-02 09:33:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-02 07:44:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"530d58fd948b56474580172443ce2f68c1d0d1d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-02 06:54:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d4a53700e5dc4b9eeae7040a22982eed021f95db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 07:44:37.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"91943bd5818af6dcd2fdba2ef5d38ac2e1915353","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-02 07:45:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e64b7ceaa07c9b167d88c198f52ec89439e35aba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 08:18:14.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"0124034d965bd1f44c317d4552c1ebfac6b6bb1e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-02 08:18:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e50dd6761d179941f1c7cb6b52d098cba133af80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 09:33:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/786d49d29b0240b3952776d4b49a1fc1 : SUCCESS in 11m 57s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/22c7f3ace88243f58eca2547bec04c20 : SUCCESS in 13m 55s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/59531428bfe54c329195068b4496668d : SUCCESS in 14m 34s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a6e55715d1dd4b2e86aa87658e91a706 : SUCCESS in 14m 40s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e8af3b871304affb06dcb6ff9a8c00b : SUCCESS in 11m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bd9eba4f5464e629be5a060ff76e9ce : SUCCESS in 6m 50s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/28d562e95e6e40af8529cd64719d5e00 : SUCCESS in 1h 03m 51s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/dc049f142eeb46e4bf2e714a866424bf : SUCCESS in 1h 08m 01s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61a54f4291e74480832ac162eb748638 : SUCCESS in 58m 02s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/b86d50ff175b442f86edd17af750bd5a : SUCCESS in 24m 01s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b2ae731557d0422380ddcc225716ca7b : SUCCESS in 18m 56s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66c2519bb0374dbab3f5a8ea5360e8ff : SUCCESS in 17m 29s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7174f412fbcd49cc8b84f6c81a0e603c : SUCCESS in 19m 12s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/64a4a707f6f34b908c089e1dea74947d : SUCCESS in 20m 39s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e2475578ceb54bbd9a3c361ab41737f8 : SUCCESS in 18m 53s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa6f75cfe5eb4b97af466e02cb50cbe4 : SUCCESS in 16m 41s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b2d474fc492e4b2ebaf74a5c2ceb0ed6 : SUCCESS in 20m 36s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42a59eb39c5647d1aa153a27c63a73ad : SUCCESS in 18m 54s","accounts_in_message":[],"_revision_number":1},{"id":"199df16ae253dc79c7d6eb12efa4b0b4a3eeaa94","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-02 09:43:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dc73020fa82ee718d33beb491745a62d4783ad47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 11:05:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/568054f9a4cb4974a7be143289b3b9bf : SUCCESS in 12m 08s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/60a6d7563cf84e26a94d23ef5351e9ee : SUCCESS in 12m 56s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/8c75389f1b79404c932b9dd04d05996a : SUCCESS in 16m 29s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d59f841685254c0f930b9838ab0cba28 : SUCCESS in 14m 39s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66dd43d3761e44e49687795ee923a738 : SUCCESS in 15m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e220dc7f6962496d9e33f6604abcc516 : SUCCESS in 6m 54s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/27ff81f6fa274173a6315295a3777c57 : SUCCESS in 1h 15m 04s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/f06a716184b64ac0afedc6d32f623eff : SUCCESS in 1h 04m 22s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e90e065a114444e0a69dd294f2151ac2 : SUCCESS in 1h 05m 53s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/faedd42afe1045958a9271d99bf95b78 : SUCCESS in 22m 05s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/92b18a8a4a714b63af4e2e3aa53c5dc8 : SUCCESS in 17m 32s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98e62f9739f84c8ebdad41196b48468c : SUCCESS in 19m 24s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d40d2e49bd62499bbab7bedfa0ebba7d : SUCCESS in 17m 56s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/72ed2a7f2ece49e6b721887b8b141db2 : SUCCESS in 18m 14s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d63d417aa84e4f1bb1e7602a805427fd : SUCCESS in 18m 10s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7189d2ca1f604ef3b688f6c4f23369fb : SUCCESS in 19m 53s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0f213a8f8b2144739ca786c958fb1c2b : SUCCESS in 18m 28s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9257bbb246a340fa9f9a9b55a9c386c2 : SUCCESS in 18m 24s","accounts_in_message":[],"_revision_number":1},{"id":"0a67c2e9bf3d68f802a601e3213cffe184c51f1a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-03 10:08:54.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"770e5e101d96949241e6b45a793a477c236370cf","revisions":{"770e5e101d96949241e6b45a793a477c236370cf":{"kind":"REWORK","_number":1,"created":"2021-03-02 06:54:05.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/41/778141/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/41/778141/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/778141/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/778141/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/778141/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/778141/1"}}},"commit":{"parents":[{"commit":"477ee3e0869cf95fae6e3bbe3474bcb531082b6e","subject":"Merge \"Remove duplicate lxc volume tasks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/477ee3e0869cf95fae6e3bbe3474bcb531082b6e"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-03-02 06:53:51.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-03-02 06:53:51.000000000","tz":0},"subject":"DNM - test log collection","message":"DNM - test log collection\n\nDepends-On: Ia7ccd5da55119273cf1fac26316ce37aa484d95a\nChange-Id: Ibf405ab9334892705c9bf519e180e608c9d3adad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/770e5e101d96949241e6b45a793a477c236370cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/770e5e101d96949241e6b45a793a477c236370cf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
