)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~782035","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I3366854a3f822c4bd71d95e33726695482a6b786","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I3366854a3f822c4bd71d95e33726695482a6b786","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2021-03-22 08:44:51.000000000","updated":"2021-03-22 10:54:17.000000000","submitted":"2021-03-22 10:48:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"782035-openstack/openstack-ansible-tests/sync-tests","meta_rev_id":"a774fd8c9a426fca7d766321b4e8ef5b66ee067f","_number":782035,"virtual_id_number":782035,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-22 10:48:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-22 10:16:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-22 10:16:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-22 08:56:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-22 10:16:37.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"6e3c763435d1e2c00ae571cd457ac024519b9235","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-22 08:44:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6708d6d02a125dc18b176c2b23cef38989df7aec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 08:56:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1ec61d332b4b465eb1c68d4ed324ac8b : SUCCESS in 8m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8c68f496d3b4b6ca4864f8cf7f3d710 : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":1},{"id":"07a39f04b49e577b17e2f6110fc9bcb1371626a3","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-22 10:16:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2b36701d092aeb2c51b66b3f818ff530ac303284","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 10:21:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"39b186ff0e193a1c234f0bb2f2dd6e35cb708599","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 10:48:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9b04fac43f0f4c378a7554dc2587d04d : SUCCESS in 13m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6a6a83a02b341ed95d3708d6ce0ad58 : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":1},{"id":"6952dde9a551346a34f3cb7e3bd523328d6e60ad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 10:48:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a774fd8c9a426fca7d766321b4e8ef5b66ee067f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 10:54:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e1ff2d0232f46daa6718c668e1fb4b0 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"786617c3c945ccdfea5ed09ba7ca225903ff9430","revisions":{"786617c3c945ccdfea5ed09ba7ca225903ff9430":{"kind":"REWORK","_number":1,"created":"2021-03-22 08:44:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/35/782035/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/35/782035/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/35/782035/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/35/782035/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/35/782035/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/35/782035/1"}}},"commit":{"parents":[{"commit":"1e1db042e5f10d69c006a0ecabc2a1a5810c6fc3","subject":"Merge \"Use ansible_facts[] instead of fact variables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/1e1db042e5f10d69c006a0ecabc2a1a5810c6fc3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-22 08:44:49.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-22 08:44:49.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I3366854a3f822c4bd71d95e33726695482a6b786\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/786617c3c945ccdfea5ed09ba7ca225903ff9430"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/786617c3c945ccdfea5ed09ba7ca225903ff9430"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
