)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~803630","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I7231e013a13682d872fd889e561b35159fb2a566","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"osa/lxc-c8-stream","hashtags":[],"change_id":"I7231e013a13682d872fd889e561b35159fb2a566","subject":"Add yum vars for centos-8-stream lxc containers","status":"ABANDONED","created":"2021-08-05 16:15:21.000000000","updated":"2022-05-20 18:35:08.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"874fd6a04e82878e21cea01fae279ce2c613c68a","_number":803630,"virtual_id_number":803630,"owner":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-08-09 17:52:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-05 17:44:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-09 18:23:38.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"}],"messages":[{"id":"afd6d29f7e6239f41cb392f9cf0ad56c9c0e46e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2021-08-05 16:15:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b0fc44e3b7659e1d1979f2b23bdeb96ea0c6e07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 17:44:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a23b2b5c90504e51b191f23c6a1da816 : SUCCESS in 13m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0124518d32e54defba462ca1b5b7ade1 : SUCCESS in 5m 52s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/d3b3883e2c2e4937a89fce1214c897b1 : SUCCESS in 1h 08m 30s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/30e9b21f30774dd7937bf9e26f540172 : SUCCESS in 1h 11m 49s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/74acc51b3a714adfa83bd89614aeebde : SUCCESS in 1h 07m 20s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/2ce9b4b3617648d5aab90bd60f49feb5 : SUCCESS in 1h 03m 32s (non-voting)\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b542c5ef09f44e46ab088dd4578d873b : SUCCESS in 1h 02m 15s\n- openstack-ansible-upgrade-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/243ecc44031347ca9d51708db2eba84a : SUCCESS in 1h 23m 55s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a815f6f75084f888dc91f587b30a813 : SUCCESS in 1h 18m 30s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/06575bc9d93444389c260b04b92d73a0 : SUCCESS in 19m 18s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ca85d8ac0a3146db96f40976b5a116de : SUCCESS in 17m 47s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab9b86a7b18e43c799094a52d900bc87 : SUCCESS in 17m 23s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/44d0774981754c14a511cd234851880a : SUCCESS in 20m 06s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/888c845d4db745b2bd9dfa058aa7e677 : SUCCESS in 17m 04s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/28022525eace41099ecfb769a43cadf8 : SUCCESS in 16m 44s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9926ae4627da4dcbae0db05f93e00d83 : SUCCESS in 16m 48s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d09f2e0a99d94ce58742ea2b85d375fd : SUCCESS in 17m 53s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3abc5d7549f847dda8b27df17f56e475 : SUCCESS in 16m 43s","accounts_in_message":[],"_revision_number":1},{"id":"7c683fcddb0bdb4683d5a23f1aaed8d58dd81248","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2021-08-05 18:12:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7432c6a9e92f09f336420ed2b462e5856fef471c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2021-08-05 18:12:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0a5fa2ab4fb5d43799fcbc699f8cf059de538e4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 20:25:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bfad7f64fd01448083ea91522affbb43 : SUCCESS in 13m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c80dcf19e4a453a93d5199dd363f752 : SUCCESS in 5m 38s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/c1c6fb35a44844f4924f8a8a661dabc2 : SUCCESS in 1h 17m 19s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/e72ec00d1881415f8adf5ee052d6af14 : SUCCESS in 1h 19m 31s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/e20a751505c74a2a8ff7b586ff987d47 : SUCCESS in 1h 07m 43s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/9cb8ce2ab0b8483186be7629cd0b2223 : SUCCESS in 1h 24m 41s (non-voting)\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f75dbcf6e06c4b549411fc464d167195 : SUCCESS in 1h 10m 21s\n- openstack-ansible-upgrade-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/d2524b4fddc94ff19ffa679e8f8d270e : SUCCESS in 2h 01m 16s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/270a3ec3273141ec86c93f9e3cd083f6 : SUCCESS in 1h 26m 13s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/b653d08762a742888f0a4f668ffd7561 : SUCCESS in 25m 08s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d27e75157fd2410d9a5d38a0ec59fc60 : SUCCESS in 19m 20s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b4755ab2db8e4e79a1aaaf051c498043 : SUCCESS in 18m 26s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/61b84ebff7f946be8ddfaf1d9f6e7f55 : SUCCESS in 18m 04s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11fbcbdf5ec947c8892488940575315a : SUCCESS in 19m 17s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cd0dba7067fe4af78183afccc9df41fd : SUCCESS in 19m 06s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a2cb02cccf9b448a9d3aa7f8a88a3b5b : SUCCESS in 16m 49s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d4aaaf8a40254955ba3c861fa44c76e1 : SUCCESS in 20m 21s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3255d3d779f425191adcacca0158aa6 : SUCCESS in 20m 21s","accounts_in_message":[],"_revision_number":3},{"id":"5f09013249acbf4d5128750895af979826a1f4be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2021-08-09 16:15:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3cf9209388d1132fa0adbea1cc5767a61a390cb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 17:52:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a192648aec184407afa6b50ccf59bf94 : SUCCESS in 12m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4196e6ca51348748c9761613e393341 : SUCCESS in 5m 08s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/2a4ae41228fe45908b4289084dcc83bf : SUCCESS in 1h 21m 44s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/dd37cea597eb4051902d8ada555d018a : SUCCESS in 1h 14m 20s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/be7156363e2e4569bed33020caa51901 : SUCCESS in 1h 05m 26s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/e7b6078dffdd4a12ab57e6f65de40a47 : SUCCESS in 1h 12m 26s (non-voting)\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e0d2ee33ca74e579590cf3e2a56fed6 : SUCCESS in 1h 06m 25s\n- openstack-ansible-upgrade-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/23bf61bd3097468183e8829977616a11 : SUCCESS in 1h 27m 03s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/936f9eb5f355422c81cb0ac547488428 : SUCCESS in 1h 29m 48s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/2ef979989b3e41cfb449a77542c824af : SUCCESS in 19m 52s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a9c53153a2f0405ebb8fe426b01e3ea7 : SUCCESS in 17m 38s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d23e06d45e549bebb4f4cca57b37809 : SUCCESS in 17m 53s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/191c9035cfb64da6bd568302253b6f61 : SUCCESS in 22m 40s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca7c16a0d4bd4619a7827e2d1608db87 : SUCCESS in 18m 20s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bc4ddee9a150494bb23ab39a81b3e921 : SUCCESS in 17m 57s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/79abd45c065c4626925fce05ba35b876 : SUCCESS in 19m 31s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ae8864efb36d471498a4ab1ca9258e77 : SUCCESS in 17m 29s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0192d81da6a47348bba0428c5b56420 : SUCCESS in 17m 39s","accounts_in_message":[],"_revision_number":4},{"id":"ee94e68f0f38a213a6b451f878f8694e7efbad9c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-08-09 18:23:38.000000000","message":"Patch Set 4:\n\nthis takes the whole of the /etc/yum and /etc/dnf directories - the existing code is very specific. Is it possible to just copy the vars?","accounts_in_message":[],"_revision_number":4},{"id":"1e38f1828b1c6d7b3213f9142c5240c29215b9db","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2021-08-10 15:29:21.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 4:\n\u003e \n\u003e this takes the whole of the /etc/yum and /etc/dnf directories - the existing code is very specific. Is it possible to just copy the vars?\n\nvar directory doesn\u0027t exist on chroot environment so when i tried /etc/yum/vars it failed to create var directory. may be we need to create first /etc/yum/vars directory first and then copy contents","accounts_in_message":[],"_revision_number":4},{"id":"874fd6a04e82878e21cea01fae279ce2c613c68a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-20 18:35:08.000000000","message":"Abandoned\n\nfixed slightly differently here https://review.opendev.org/c/openstack/openstack-ansible-lxc_hosts/+/842086","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a73d24abe6ddd530478dca0e8baa5120f626aea8","revisions":{"7853220f66e78340c8074524dfe23bb4bcd9eaa5":{"kind":"REWORK","_number":1,"created":"2021-08-05 16:15:21.000000000","uploader":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"ref":"refs/changes/30/803630/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/30/803630/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/30/803630/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/30/803630/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/30/803630/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/30/803630/1"}}},"commit":{"parents":[{"commit":"a640dd6f389582d7f79fad5369410487cb01632f","subject":"Add iproute to CentOS LXC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a640dd6f389582d7f79fad5369410487cb01632f"}]}],"author":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2021-08-05 16:13:00.000000000","tz":0},"committer":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2021-08-05 16:13:00.000000000","tz":0},"subject":"Add yum vars for centos-8-stream lxc containers","message":"Add yum vars for centos-8-stream lxc containers\n\ncentos-release-nfv-openvswitch repo required some vars to download\nyum mirrorlist to prepare lxc environment.\n\nChange-Id: I7231e013a13682d872fd889e561b35159fb2a566\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7853220f66e78340c8074524dfe23bb4bcd9eaa5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7853220f66e78340c8074524dfe23bb4bcd9eaa5"}]},"branch":"refs/heads/master"},"8c0fc917cbabde661dafc94dcb2f84791f9efa2e":{"kind":"REWORK","_number":2,"created":"2021-08-05 18:12:23.000000000","uploader":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"ref":"refs/changes/30/803630/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/30/803630/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/30/803630/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/30/803630/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/30/803630/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/30/803630/2"}}},"commit":{"parents":[{"commit":"a640dd6f389582d7f79fad5369410487cb01632f","subject":"Add iproute to CentOS LXC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a640dd6f389582d7f79fad5369410487cb01632f"}]}],"author":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2021-08-05 16:13:00.000000000","tz":0},"committer":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2021-08-05 18:12:16.000000000","tz":0},"subject":"Add yum vars for centos-8-stream lxc containers","message":"Add yum vars for centos-8-stream lxc containers\n\ncentos-release-nfv-openvswitch repo required some vars to download\nyum mirrorlist to prepare lxc environment.\n\nChange-Id: I7231e013a13682d872fd889e561b35159fb2a566\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/8c0fc917cbabde661dafc94dcb2f84791f9efa2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/8c0fc917cbabde661dafc94dcb2f84791f9efa2e"}]},"branch":"refs/heads/master"},"ff7ba4404448b9e418c4189d5b3053818846c5e6":{"kind":"REWORK","_number":3,"created":"2021-08-05 18:12:59.000000000","uploader":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"ref":"refs/changes/30/803630/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/30/803630/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/30/803630/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/30/803630/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/30/803630/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/30/803630/3"}}},"commit":{"parents":[{"commit":"a640dd6f389582d7f79fad5369410487cb01632f","subject":"Add iproute to CentOS LXC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a640dd6f389582d7f79fad5369410487cb01632f"}]}],"author":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2021-08-05 16:13:00.000000000","tz":0},"committer":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2021-08-05 18:12:53.000000000","tz":0},"subject":"Add yum vars for centos-8-stream lxc containers","message":"Add yum vars for centos-8-stream lxc containers\n\ncentos-release-nfv-openvswitch repo required some vars to download\nyum mirrorlist to prepare lxc environment.\n\nChange-Id: I7231e013a13682d872fd889e561b35159fb2a566\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ff7ba4404448b9e418c4189d5b3053818846c5e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ff7ba4404448b9e418c4189d5b3053818846c5e6"}]},"branch":"refs/heads/master"},"a73d24abe6ddd530478dca0e8baa5120f626aea8":{"kind":"REWORK","_number":4,"created":"2021-08-09 16:15:20.000000000","uploader":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"ref":"refs/changes/30/803630/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/30/803630/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/30/803630/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/30/803630/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/30/803630/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/30/803630/4"}}},"commit":{"parents":[{"commit":"a640dd6f389582d7f79fad5369410487cb01632f","subject":"Add iproute to CentOS LXC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a640dd6f389582d7f79fad5369410487cb01632f"}]}],"author":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2021-08-05 16:13:00.000000000","tz":0},"committer":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2021-08-09 16:15:14.000000000","tz":0},"subject":"Add yum vars for centos-8-stream lxc containers","message":"Add yum vars for centos-8-stream lxc containers\n\ncentos-release-nfv-openvswitch repo required some vars to download\nyum mirrorlist to prepare lxc environment.\n\nChange-Id: I7231e013a13682d872fd889e561b35159fb2a566\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a73d24abe6ddd530478dca0e8baa5120f626aea8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a73d24abe6ddd530478dca0e8baa5120f626aea8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
