)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~812528","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I6527d2c0c361c03cbb6fc43aa3a03896894fb8e1","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"allow-releaseinfo-change-on-apt-update","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2021-11-11 22:18:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6527d2c0c361c03cbb6fc43aa3a03896894fb8e1","subject":"allow release info change on apt update","status":"MERGED","created":"2021-10-05 16:59:17.000000000","updated":"2021-11-11 22:19:51.000000000","submitted":"2021-11-11 22:18:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"812528-allow-releaseinfo-change-on-apt-update","meta_rev_id":"b552d66693a4b4ea53d1d1ff51be900a198ff5e3","_number":812528,"virtual_id_number":812528,"owner":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-11 22:18:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2021-11-11 20:50:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-11-09 12:21:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":1,"date":"2021-11-11 20:50:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-05 18:32:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-09 12:21:25.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2021-11-11 20:50:11.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"}],"messages":[{"id":"3cb778256ad09d80cb12886b18464125d7d1b6f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-05 16:59:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5197ec2e2ea0bc3e5af12deb493110364c145c34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 18:32:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c81638fea78f4e8daf152875c60b6f2d : SUCCESS in 16m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bf0d5edbd8e49e69f67d2ab754d385f : SUCCESS in 7m 19s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/db395bbffa724a01a475a32720aa0255 : SUCCESS in 1h 20m 31s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/88caa13197c043039b5f425790e9cf4f : SUCCESS in 1h 08m 11s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/c2b5905f3f4d4b918a08944ac9464e8d : SUCCESS in 1h 03m 39s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/526a1b63676f4c549a1e573aca0ebf1b : FAILURE in 1h 04m 11s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e72f5bb52fca42be91607d3c0c9b82a6 : SUCCESS in 56m 28s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/27c8e807cf9c4432955578660d311a7a : SUCCESS in 1h 15m 56s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e23de52c00f425dad68106b1a024144 : SUCCESS in 1h 28m 51s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/8d9bcba66ef5471380811df7c51bfe2d : SUCCESS in 20m 53s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/08214c4f46bc4296b6ff00da8d4356f2 : SUCCESS in 19m 16s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f98a639a96384455960512d3f6ea033a : SUCCESS in 20m 42s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/62af2b954d9942f297740fded64ad338 : SUCCESS in 18m 03s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3867284d05b24ffc93ac9e6f1cc2e196 : SUCCESS in 21m 24s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/86041a1c5d6e459a875e4f528cf0b21f : SUCCESS in 22m 37s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab1fd49d0eff487fae67842bf77add04 : SUCCESS in 18m 54s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7d794a3e12dc47cdb2dca13c12cec0db : SUCCESS in 20m 14s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c3a6a4dfeb864dffb2a6792e79d95108 : SUCCESS in 23m 20s","accounts_in_message":[],"_revision_number":1},{"id":"2927ed0eb4141080121d58924cac04d8b4a9a273","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-05 18:47:32.000000000","message":"Patch Set 1:\n\nrecheck\n\nunrelated","accounts_in_message":[],"_revision_number":1},{"id":"a4bb9a68d7353a0d04b422938593d7a6d5f50121","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 22:09:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/451a19bf7fb5446b8f6cc781b34393df : SUCCESS in 13m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/686adb53e2f642e88e8cf915931f3933 : SUCCESS in 6m 21s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/2c54cfb74b3c4729aaa51db3524f6887 : TIMED_OUT in 3h 11m 22s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/0b5d93668640444a89ba1870f3e96f80 : SUCCESS in 1h 12m 27s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/e5029dceeb8a4ef3850ccd33aee31a0d : SUCCESS in 1h 10m 01s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/7828e0cbbade4a61aac121c578e1bb38 : SUCCESS in 1h 00m 19s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc51a00f91aa4117b683fcdf1f50f66b : SUCCESS in 1h 02m 34s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/8d9902012111419280db3a6ca08ca75a : SUCCESS in 1h 26m 39s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b4eeb5b7710412385a1acd36b701c46 : SUCCESS in 2h 08m 39s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/d8fce0ab25864f84982ee95cd2e98816 : SUCCESS in 20m 25s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d8d196082ac14e06907acdc5bb7ba7d8 : SUCCESS in 18m 11s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/862d1052450e41128a1621430be485b8 : SUCCESS in 20m 57s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e8a6e5b407a94ebb89312c77866d75d1 : SUCCESS in 18m 46s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d7602076001d47d593953e9b952bb7cd : SUCCESS in 21m 28s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/49544faab9554109945d3bb060b4ac48 : SUCCESS in 17m 50s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/80c44e836f56431f803ed7ccb78260f4 : SUCCESS in 20m 38s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ab6010a0657f4866b7657d8ac76e8e72 : SUCCESS in 17m 26s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c1e9f447393741f19328c63842c0da66 : SUCCESS in 21m 32s","accounts_in_message":[],"_revision_number":1},{"id":"1644c77d7c93a864b8a395f8b18c3fda3dd1c976","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-05 22:11:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"084b09e60d9bf84fb968502098f92bc320caf4e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 23:46:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68abdde71df947d19132daef2d5e58ae : SUCCESS in 14m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d22bb283a3924ca98cc5baaf433823bf : SUCCESS in 6m 50s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/96dd3fdc789b45d0a70ca80be266fac9 : SUCCESS in 1h 21m 48s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/08b1b158db864b6f95e2e60900ea7707 : RETRY_LIMIT in 23m 51s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/fc1b814164dc4657b388a6c9f473c7c8 : SUCCESS in 1h 11m 09s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/73f80970721f4236a6a3c342c8a8a75d : SUCCESS in 1h 08m 01s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c1cf067773b48cf9df937ac67ed0a57 : SUCCESS in 1h 16m 22s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/bf40fae72d1a4da5a038234dc7d3fb61 : FAILURE in 18m 59s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fed11d67a90248758e856b1009ec8d2b : SUCCESS in 1h 30m 58s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/a3a0f9b0be0f40279cfecafd083ecadb : SUCCESS in 21m 32s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fbb96b2e48624130be9872a090baa613 : SUCCESS in 21m 27s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e4ad2c85b5c74995b8d58a8364ca58c8 : SUCCESS in 21m 31s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/af342bd9839e4acc97274af3114bc34a : SUCCESS in 17m 09s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b321a5050dbf4290b1de05fd4eeffa18 : SUCCESS in 22m 33s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a81e4e86a168406089e32ba75002bb08 : SUCCESS in 18m 06s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d8e72ac6a7444cca930b3cd02d0e416 : SUCCESS in 26m 40s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/35f0415dc3184ff49e610d39c7f49597 : SUCCESS in 19m 05s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f98cd88d483466eb06f849208ac0c66 : SUCCESS in 21m 46s","accounts_in_message":[],"_revision_number":1},{"id":"1650aa384a0c3fe20c2b723d6da3f69aff7b7a81","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-05 23:50:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3953664d29109e1b5116d0ab2e2dcb29e3494fe7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 01:15:25.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77e6b9c99cbd4ac686a257e8551de70e : SUCCESS in 17m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9825a115549b44148b28e4083ad4af17 : SUCCESS in 6m 01s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/6f4650fcf5c54e52bcde9c64ad73486c : SUCCESS in 1h 13m 39s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/8189ee55a0df42e894bdd52df3a1d081 : RETRY_LIMIT in 18m 12s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/d81821e22ec744c3a68ab782f083b70a : SUCCESS in 1h 01m 18s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/4c9076a5f6bd4831902434328173eb06 : SUCCESS in 59m 31s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ff1906548114c9d8c491bdd73819164 : SUCCESS in 1h 19m 30s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/56e31b8072fc44958ddb0fb49d4a16c9 : FAILURE in 19m 40s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5201aceac1ea487c8e6eea113ad7e915 : SUCCESS in 1h 03m 13s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/4ea0744afbc54df5b23a1ce1235e1640 : SUCCESS in 19m 24s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/52e7caf1106d4b0898f1b5c4ffa67750 : SUCCESS in 16m 13s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b89a8f052f024449b3c093e028bda63d : SUCCESS in 17m 21s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2d5521b606114ab49209bc307cbb29e5 : SUCCESS in 17m 47s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08ffd2f7a60e4f8a9cfa1be6d19d1bdc : SUCCESS in 23m 08s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d8ba9cdedf1743e5bc73f88d1ae7a576 : SUCCESS in 15m 32s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ac4be3c51c449a9b1b39c8dccc72558 : SUCCESS in 16m 24s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8cd9073546cd4559b76cdbb82162d455 : SUCCESS in 18m 29s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9039b4913724694b4ef1545d8b3b437 : SUCCESS in 18m 17s","accounts_in_message":[],"_revision_number":1},{"id":"6b14e87498238d9709f76280cc1255b2f5112dd8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-06 01:26:00.000000000","message":"Patch Set 1:\n\nrecheck\n\ngood reproducability","accounts_in_message":[],"_revision_number":1},{"id":"f8b00ff7cf4a21381148cad22960d5847246aa76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 03:39:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc0e327ef56343e5a23fba9ed35b48c3 : SUCCESS in 13m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c760ef33683d471ea021e0e87131181c : SUCCESS in 6m 14s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/9cda4d77a0fc4be2a6a65fd02ec1f743 : SUCCESS in 1h 20m 24s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/c8d8db7172ab4f77bccab6afae7e02d0 : SUCCESS in 1h 19m 25s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/b988af33e79043b0ab51c71401b11648 : SUCCESS in 1h 10m 53s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/f6dbd8a2b3bd405eb5431c7dcf4e38bc : SUCCESS in 1h 07m 09s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1fa14ff39ba84befaabe1e1fd7f31c3c : SUCCESS in 1h 14m 30s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/306f8f6196454277bc9ae8ab045f751b : FAILURE in 18m 01s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b77d02d864741a99975e10844a7ba92 : SUCCESS in 1h 29m 28s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/8db478adafaa430bb28417487f8c1477 : SUCCESS in 22m 48s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/148557b795064f6db2534ddab57b11c2 : SUCCESS in 19m 53s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f490af6aaa24e9bad070467b0a3bc16 : SUCCESS in 20m 41s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6bab5e932f7f472e9fbdd073b5c1afac : SUCCESS in 21m 00s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f582b440b5df496ca3eded01c22de8eb : SUCCESS in 20m 37s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/42b9b9c308f748dda83c2e615c25cb1c : SUCCESS in 19m 48s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc5df396785944b3a48a85bcbd9c5ffa : SUCCESS in 20m 46s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/14589c29b13647368ce42245a4b98204 : SUCCESS in 20m 39s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b1696e5551e1456ebf7b1047e68dc7a8 : SUCCESS in 20m 40s","accounts_in_message":[],"_revision_number":1},{"id":"75050fb11600adcbc4fc5de4a32293477c2fca4b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-11-09 12:21:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e7f7c163dc2fad1504074f57404bc43b493d3a87","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2021-11-11 20:50:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"602545df27d4e06ac84a748116b1b4c16e32c1df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 20:54:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"81bd6c0e28e18400dc832b36eaa2ced8f91eea4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 22:18:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea9a5b70a677474d854ceccd58b5c87d : SUCCESS in 15m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b2e1349be414ee9845ecf2500295c3a : SUCCESS in 7m 01s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/146620a470b74acda0ae8c54afba359e : SUCCESS in 1h 19m 24s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/e2d955f533714c29862e563c561cdecb : SUCCESS in 1h 22m 59s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/b9d1da11bed14f19841627559d4abbc4 : SUCCESS in 1h 06m 55s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/4810b638a10448ec890c606af8974a13 : SUCCESS in 1h 05m 29s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60d59fca321e42718faa0a7401bd52a0 : SUCCESS in 1h 08m 12s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94b53d14cea64915a1ac156d2adf99f7 : SUCCESS in 1h 21m 50s","accounts_in_message":[],"_revision_number":1},{"id":"89e4cc972928706e7043c5eff291f844b311031b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 22:18:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b552d66693a4b4ea53d1d1ff51be900a198ff5e3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 22:19:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc472d89056a48108791473b70ccb290 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f7ea823d80e1d831e23ec14f4e9ece8c2bdc30a3","revisions":{"f7ea823d80e1d831e23ec14f4e9ece8c2bdc30a3":{"kind":"REWORK","_number":1,"created":"2021-10-05 16:59:17.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/28/812528/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/28/812528/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/28/812528/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/28/812528/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/28/812528/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/28/812528/1"}}},"commit":{"parents":[{"commit":"a640dd6f389582d7f79fad5369410487cb01632f","subject":"Add iproute to CentOS LXC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a640dd6f389582d7f79fad5369410487cb01632f"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-10-05 16:58:08.000000000","tz":-300},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-10-05 16:58:08.000000000","tz":-300},"subject":"allow release info change on apt update","message":"allow release info change on apt update\n\ndebian buster changed from stable to oldstable, without this change\napt-get update fails to run on old systems.\n\nChange-Id: I6527d2c0c361c03cbb6fc43aa3a03896894fb8e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f7ea823d80e1d831e23ec14f4e9ece8c2bdc30a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f7ea823d80e1d831e23ec14f4e9ece8c2bdc30a3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
