)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~818486","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2Fvictoria~Iee818c26a2d6b0b7c7712611be826f8d2f23ccc0","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/victoria","topic":"osa/el84-stable/victoria","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2021-11-25 13:16:22.000000000","reason":"Dmitriy Rabotyagov replied on the change"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2021-11-24 17:09:17.000000000","reason":"Jonathan Rosser replied on the change"}},"hashtags":[],"change_id":"Iee818c26a2d6b0b7c7712611be826f8d2f23ccc0","subject":"Revert \"Add CentOS 8.4 support\"","status":"MERGED","created":"2021-11-19 12:25:56.000000000","updated":"2021-11-25 16:20:16.000000000","submitted":"2021-11-25 16:16:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"revert_of":795282,"submission_id":"818486-osa/el84-stable/victoria","meta_rev_id":"c3ca438fef3f21f51bde9cec5ff6aa2ce2ff77de","_number":818486,"virtual_id_number":818486,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-25 16:16:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-11-24 17:09:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2021-11-25 09:26:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-19 12:25:56.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2021-11-19 12:25:56.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"604d292173c6f73d3054279e282c9239f7d7e54c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-11-19 12:25:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3cfafc2bf64463c2aaec84045cf0b22469f24b4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 14:04:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f47cab0221a44001b53c2a458668c448 : SUCCESS in 13m 00s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/f86fac69408f41b395a84b8afcd6bb1d : SUCCESS in 12m 36s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/f84a1189a35d4de6822c45098aee6497 : SUCCESS in 21m 57s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5143c676ca144ab89fa00a420f607e19 : SUCCESS in 14m 22s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cdd02a494fd9403ebc5eb1f4f738d136 : SUCCESS in 13m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dcfb4811481453185b9dcd0ad2d843e : SUCCESS in 6m 19s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/5d559b437bb544f3bc1b1b1dde36edec : SUCCESS in 1h 30m 50s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/d8899112e9d94ea6a6b3a1af21b93ce2 : SUCCESS in 1h 14m 22s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0fa0d7ed5066415380b436f7bbacd94b : SUCCESS in 1h 09m 10s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/d2656dc000694f91921ec3ae37d242a8 : SUCCESS in 28m 54s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/39db164aee78485b919265503b97511d : SUCCESS in 27m 25s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c5a02083fa14ec8a3bc1b48fc3ba9ee : SUCCESS in 22m 11s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/78f5fdf48715495f896dcb8d446f6c71 : SUCCESS in 24m 19s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1abcc1cf62f46968df6b64c0057b1c6 : SUCCESS in 21m 25s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5a2a39340f4846e9840d6592f949a42d : SUCCESS in 20m 05s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ddafaa4dd8fa444e8430566df4677c61 : SUCCESS in 29m 46s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5ac989b59bf24a928ed48a73c87d7eea : SUCCESS in 20m 32s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bbf5f54b860442628e2032fdd73e881b : SUCCESS in 20m 48s","accounts_in_message":[],"_revision_number":1},{"id":"7748aaed53574df514dbac8811a22e2f62368a1a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-11-24 17:09:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fcdf8b1c6842af8b385d995185461333d6f1eb3b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-11-25 09:26:11.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"52a3c1d269d8719118d9b74d9de18425b8768897","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 09:28:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dd7f0225f62f7d7ddbd851196479265a18d92240","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 12:39:13.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5ceee2341de544d187c4d6e35726bc14 : SUCCESS in 9m 41s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/f6589c433f004a2ba289f881078819d0 : SUCCESS in 14m 12s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cdd34dc13e99415b8b17b31a5bd4b29c : SUCCESS in 11m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16d47310c97944f58879fc2ee8fca5bf : SUCCESS in 5m 38s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/1d4463ffa82c42aa8f41b3a6004184c9 : SUCCESS in 1h 18m 10s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/dc39c305b37e4a2ca7f2d20eecac6320 : SUCCESS in 1h 04m 58s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a39bea7e105642b3a2f9317781672d40 : TIMED_OUT in 3h 09m 42s","accounts_in_message":[],"_revision_number":1},{"id":"905562cc377786671df70e7573689e43257135ab","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-11-25 13:16:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"96aed06ba3c3c2e05fc7eb99cff8388eecc6b786","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 14:40:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/22cfd1e163a447d5a018ee2e450271a1 : SUCCESS in 11m 00s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/4646f264d9bb4c8e983a0221075f2c66 : SUCCESS in 12m 46s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/b84292f3178e43e790c29606b5b65168 : SUCCESS in 20m 17s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8b2733ac8c9a42ebbcb8a5e1c553e55a : SUCCESS in 12m 31s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e541d7a4ca6b499bbb14b9be263e43e5 : SUCCESS in 13m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a524907e459437ab6b63392de5da224 : SUCCESS in 5m 03s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/1f22ed8605274535a0e330ea2435dd68 : SUCCESS in 1h 15m 19s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/37d1d0b2edfb463ca9e41310bebe2871 : SUCCESS in 1h 05m 01s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96ed33ab9a324d15a2e9343bf29576f9 : SUCCESS in 1h 09m 04s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/44178788a2c84b8f852fcdad7309362d : SUCCESS in 22m 14s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/541f7e06cc904e24b250fbd974da7f35 : SUCCESS in 19m 01s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62810ff1e5714bd0a9eaad1f027f249f : SUCCESS in 19m 37s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1a172c8cdc1e400ab427a7d065da49f0 : SUCCESS in 18m 20s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52619914cf884c01a5b39cd599c6c31e : SUCCESS in 20m 53s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4cd6f8cf13d6488c8d9a4b62e9608c8a : SUCCESS in 20m 07s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e9ce3ee59e1439798d6964276b8c611 : SUCCESS in 17m 41s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5d6a751cdfd24fc991df56ec382cdda3 : SUCCESS in 18m 07s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a9ab2a7f3e94648bf58c479dbcb72dd : SUCCESS in 18m 45s","accounts_in_message":[],"_revision_number":1},{"id":"42febce19e8359c250b108f57b207ecd7096f3e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 14:44:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9d8ba1cbf5a19b27f270b9bea9f28b84291e1878","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 16:16:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e026f6b326ea4e918ca7e9d9a44efc8a : SUCCESS in 11m 11s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/abaac55989944ce0926b37cf108a6dd6 : SUCCESS in 12m 28s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/040aeefd5e1744c38b03281f9e6c6da0 : SUCCESS in 14m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76e4a4ae59dd494f8801601caf180543 : SUCCESS in 5m 40s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/b14183f6cd04404bb886a36182aded6c : SUCCESS in 1h 29m 50s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/2df64e321a704f27af6c2fc717296dec : SUCCESS in 1h 09m 15s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b3f2cda5d0345cab9b244b593058bc5 : SUCCESS in 1h 09m 41s","accounts_in_message":[],"_revision_number":1},{"id":"d7fa34aa3d9a598508ec751f553b0065bac00211","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 16:16:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c3ca438fef3f21f51bde9cec5ff6aa2ce2ff77de","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 16:20:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bdeb52092fb4157a1cb842f98468420 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"51eca809d2330ca707953d5fdfd308175f738331","revisions":{"51eca809d2330ca707953d5fdfd308175f738331":{"kind":"REWORK","_number":1,"created":"2021-11-19 12:25:56.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/86/818486/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/86/818486/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/86/818486/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/86/818486/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/86/818486/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/86/818486/1"}}},"commit":{"parents":[{"commit":"fc0eedf9de8a4024b51abbe5771234f89b5bfa32","subject":"Add CentOS 8.4 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/fc0eedf9de8a4024b51abbe5771234f89b5bfa32"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-11-19 12:25:56.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-11-19 12:25:56.000000000","tz":0},"subject":"Revert \"Add CentOS 8.4 support\"","message":"Revert \"Add CentOS 8.4 support\"\n\nThis reverts commit fc0eedf9de8a4024b51abbe5771234f89b5bfa32.\n\nReason for revert: We don\u0027t have CentOS Stream support so don\u0027t need that symlink\n\nChange-Id: Iee818c26a2d6b0b7c7712611be826f8d2f23ccc0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/51eca809d2330ca707953d5fdfd308175f738331"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/51eca809d2330ca707953d5fdfd308175f738331"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
