)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~828120","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I5ac99e2b6c6e6ccd9da18ae68e1f8801f95f4f4e","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"osa/systemd_roles","attention_set":{},"removed_from_attention_set":{"13095":{"account":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"last_update":"2022-11-28 16:25:12.000000000","reason":"\u003cGERRIT_ACCOUNT_13095\u003e replied on the change","reason_account":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},"7353":{"account":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"last_update":"2022-11-28 18:02:55.000000000","reason":"Change was submitted"},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2022-11-28 18:02:55.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-11-28 18:02:55.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2022-10-14 10:01:51.000000000","reason":"Only change owner, uploader, reviewers, and cc can be in the attention set"}},"hashtags":[],"change_id":"I5ac99e2b6c6e6ccd9da18ae68e1f8801f95f4f4e","subject":"Replace ifupdown with native ip-link","status":"MERGED","created":"2022-02-07 12:30:38.000000000","updated":"2022-11-28 18:03:57.000000000","submitted":"2022-11-28 18:02:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"828120-osa/systemd_roles","meta_rev_id":"495417a6dc7d586c70b2bea8aab8f66c2fb0bc95","_number":828120,"virtual_id_number":828120,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-28 18:02:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2022-11-28 16:25:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-11-15 17:22:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":1,"date":"2022-11-28 16:25:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"CC":[{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-07 13:41:11.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"CC"},{"updated":"2022-02-07 16:20:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-17 16:56:38.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"CC"},{"updated":"2022-10-18 23:53:57.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2022-11-15 17:22:51.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2022-11-28 16:25:12.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"}],"messages":[{"id":"c3499a9753be8268fa8b90a8731acb7ac5d7195f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-02-07 12:30:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"59b8feeae1891d9bb2e841bfa8b00f139cf9960a","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-02-07 13:41:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"410722ae6899255eabeecf9105d09e04863c7dda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-02-07 14:42:16.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"c8d6fb2a4bcdebd0448f194d075d7f838caaff9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-02-07 14:42:50.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"dc2d21ae0a7883b2db7b3b068a0e37576a9499eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-02-07 14:44:34.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"6d710ba4cf80a1ff60b39f5080b771a4ad461364","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-02-07 14:44:40.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"634cc0ca1c3ce1e4af77671fca072288b7659c48","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-02-07 14:44:44.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":4},{"id":"53e50aa4bfb73762635c909124ed51cfbade45d9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-02-07 14:45:16.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3f896f66bc50c525cdab8cd7ce93d4e35df58c34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 16:20:26.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c0657264bc84746b011e42179112305 : SUCCESS in 12m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71d80538fcfc44b4980ad8ff8e303f76 : SUCCESS in 5m 03s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/017132cbfbac4759aadeedd61a2702e9 : FAILURE in 22m 02s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/30b5506b754046a399cb896b9b24d7cc : FAILURE in 20m 06s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f58820a15d334faa912d691c4144ab13 : FAILURE in 20m 30s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/54a3546cdfb147d294b4551e7cda1bb1 : FAILURE in 20m 59s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d8f852404f94b069334f0a4fb70d2a9 : SUCCESS in 1h 31m 54s\n- openstack-ansible-lxc-dir-centos-8-stream https://zuul.opendev.org/t/openstack/build/01e4b60ecb3f403094c38173dd740a59 : FAILURE in 11m 01s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/06b76d4444bf43deb868990db3854a24 : FAILURE in 10m 55s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9be88e7419ff48e8a94eac5593b3f4f7 : FAILURE in 15m 06s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62c1bc92d01b446ebaa8e5b2988a647b : FAILURE in 14m 20s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39626d0f3e2e450a8091b578a5b1bed2 : FAILURE in 10m 23s","accounts_in_message":[],"_revision_number":4},{"id":"3a1e93384c34fc78baf18fd6d3b22c8e29e91992","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-02-19 15:32:54.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"47e43a56675da67ac45342574427f56e3000a3af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-19 17:06:19.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6815b1c987e649c386792dacc00eb79c : SUCCESS in 21m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ceb05f8e7e94511a36379dd9385bb0c : SUCCESS in 5m 18s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/8c9b562f02ee49af8d4da98d7c735340 : FAILURE in 27m 39s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/e5bc709aa08947bea21fc00a7711ec87 : FAILURE in 25m 48s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42dbcb546c4e4e50ad6bcb852c60b4fe : FAILURE in 23m 51s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/2e329b9799334da3b6371aac90e09e6e : FAILURE in 27m 03s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c19e982ebf3444349be57de4ca5a5246 : SUCCESS in 1h 24m 15s\n- openstack-ansible-lxc-dir-centos-8-stream https://zuul.opendev.org/t/openstack/build/47e09c3a56ef4f24ba223da7de82fe16 : FAILURE in 18m 58s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb8bac99a57542dca7de564fba8d94a8 : FAILURE in 17m 32s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0789e0f052f846a89863e95ba1c77e20 : FAILURE in 10m 48s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d696d680bbb94c90ab2dd822f5bc15a5 : FAILURE in 16m 43s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a214e069a0941dbaf1e219a657eca86 : FAILURE in 12m 44s","accounts_in_message":[],"_revision_number":5},{"id":"ffad357ed2077c7ea3d41bcecdd53b1d635e78e9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-14 08:27:03.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ed48b4a07a6f3ff53f77ad0bd5ab86291880ac95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 10:01:51.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dfe80ba99e3a4ec1a215427520c24962\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9452b6a2c944e5d89e0ab997a30a44e : SUCCESS in 13m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41fa397965664d6a80b223cf4be0a336 : SUCCESS in 5m 33s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/79fc810ff53b4cdc83d0d816d43479c5 : FAILURE in 29m 07s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ae044916845a4fc2994bcb951e7148f7 : FAILURE in 24m 39s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/0bc2df78a4e145dcb72fcca9d44a86b2 : FAILURE in 21m 54s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3aa4de5041544d8298cd0bcf09dd9c43 : FAILURE in 21m 43s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/913bdb5042344f87b4528728bb492ab2 : FAILURE in 22m 53s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/30cade60f41543f281439c50cb8c0a83 : FAILURE in 1h 19m 55s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/82abcdbc01194d87af344a89df4e45d9 : SUCCESS in 1h 31m 41s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/020aa1521c9447bdac5ad61d7a08ac2e : FAILURE in 25m 52s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c3e38170d6d45409224c4ebad2fbdb9 : FAILURE in 26m 11s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12269ec572c74fde8723847dfdbac123 : FAILURE in 26m 52s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/94635205c9694a8997a404978a6ac338 : FAILURE in 26m 37s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/993aa637696b42bab9ee4a63a2bd767b : FAILURE in 25m 54s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-9-stream https://zuul.opendev.org/t/openstack/build/1b940fec0d304e4c8950506c0e4d287b : FAILURE in 28m 44s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5a059171ad514962a5e7f71732d3539d : FAILURE in 23m 51s","accounts_in_message":[],"_revision_number":6},{"id":"b4fb66ae9a762dc06865f4bab63c3fda278e87dc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-14 10:23:32.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"12aeccc020fcd594690b79050c428dc80e11e30d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-14 10:52:47.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"87881e830f46b52916c93228816510a03f732031","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-14 11:03:28.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"58d55f3c1babcd1c1854892071182a8b422fea20","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-14 11:16:13.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"01f0c63f8f0bcc85a9f3094089e61be78d3bce80","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-14 11:21:37.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":10},{"id":"e0e3c8ed06baf56152f60fafb4e987113efa982e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 12:56:06.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca00a90b0f674e11919263647f0cd62d\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1890e6e8a6e94a8bb6e60270d81addf8 : SUCCESS in 17m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/300bbf4c5c3f40a48bbf31d6d38892b7 : SUCCESS in 5m 16s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/7596d1e42478434e931d7ccf1091656d : SUCCESS in 1h 32m 18s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b63d3de48f3c42f1a72689c6d3961cf6 : SUCCESS in 1h 17m 40s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/87850eb61fe449b0b9421cae3c3af8f3 : SUCCESS in 1h 10m 44s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60b267973d0241068baf2149cf720639 : SUCCESS in 1h 23m 54s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ffc66ab60640443bbd290e9c074fb90a : SUCCESS in 1h 07m 55s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc5e30c740c143aaa8f90f724b22794a : FAILURE in 1h 17m 56s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e1d7bf5db2fd46be9fea7a77b4dd7df2 : SUCCESS in 1h 14m 12s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d26aa87c0f264bfd96a90b176910d184 : SUCCESS in 33m 46s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/655f056c721a4d8d82b5e8fb38434eb0 : SUCCESS in 31m 19s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7bd0cce2f2f1468f819472b143db93ad : SUCCESS in 33m 42s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/da0f6184ee2447c58fe73bd27f957e08 : SUCCESS in 29m 13s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/eba1198abd83433ebc7ef07d11955dc9 : SUCCESS in 35m 58s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-9-stream https://zuul.opendev.org/t/openstack/build/febd522647a34573bedcc5cc9c7e920a : SUCCESS in 33m 35s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d421e30f60d74056b5e3734de20e7328 : SUCCESS in 31m 58s","accounts_in_message":[],"_revision_number":10},{"id":"e0f40885d29ab72be0827ab7b22c01cd65952291","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-17 10:21:53.000000000","message":"Topic set to osa/systemd_roles","accounts_in_message":[],"_revision_number":10},{"id":"76147738affa612afc828d65daa89ff84961c0e5","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2022-10-17 16:56:38.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"673c02efcc162a76d66c72f82249fa55296a33db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-18 08:32:18.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"23cfa32df6dac616b21e9304d1919a3ddd7660da","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-18 08:32:36.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"b064a827ac295aa1cc01e9512e395be9528b737f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-18 08:33:08.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"26f07fbae6ef3965fa144dbd50e93feb05b9148d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-18 08:35:44.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"a20fbd5ec0b48d6d0534295625f10a4da2ea3af8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-18 08:37:25.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"18bd6ea91b1e19781ea42515a30cfd8b64adca89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 10:20:50.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39a0e1f11657418b94244f7848d8eb1b\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9dfdb0b11d7346119174431893cc9923 : SUCCESS in 12m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ed08311f9834db59840159aa6292678 : SUCCESS in 4m 40s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/2c0f5a93a97143fa8cc394809642aaa7 : SUCCESS in 1h 36m 32s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/281ab4b257704af6b72e66c1566ca548 : SUCCESS in 1h 26m 08s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/e3534cb1eb7d4a228cc9e93cfa7c21fd : SUCCESS in 1h 13m 57s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6056a8fc42d4294a37f58bfed71bf82 : SUCCESS in 1h 28m 15s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d97ca7769fe74725b72ec6b0f20198a7 : SUCCESS in 1h 17m 28s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/43b2fab15ba54ea59dd0dd6e5ce27286 : FAILURE in 1h 16m 47s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1381c56a8a4d4f168dca9fed0f80268f : SUCCESS in 1h 18m 24s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/335e17c4dd7e4bc4bc63baebca880202 : SUCCESS in 36m 05s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c4437dfdf38a404b994b1536afe84de5 : SUCCESS in 30m 51s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a7b050bb45e43409015ec717e458cad : SUCCESS in 33m 50s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/b3a27681cd6345ac83030ff0049d567b : SUCCESS in 30m 46s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/89d1021cfde940769c7792e983cbb3e1 : SUCCESS in 34m 36s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-9-stream https://zuul.opendev.org/t/openstack/build/d2a03aae04df46d48f7ddeef646913d8 : SUCCESS in 35m 43s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a2ce37607bf6423bbd330d4606f4920a : SUCCESS in 33m 04s","accounts_in_message":[],"_revision_number":13},{"id":"8cce6ab8f1e553ce4ba377e58e37d8f2087c99d8","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2022-10-18 10:56:26.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"0257e2d7ee5458ef47fef0c39ed46d908ccce565","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2022-10-18 23:53:57.000000000","message":"Patch Set 13: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"397af7be8d770ae014acaad732630196d2f62718","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2022-10-18 23:55:58.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"50c3a49972170ae9e48c49175e3be9a40d29a65c","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2022-10-18 23:56:26.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"b0445f9cfa54aae6cff6305f7a36ea049b3785cc","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2022-10-18 23:57:45.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"c89f024594782dd0741604b4ec902e6b65fb6c48","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-11-02 12:02:28.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"df9a5788b2f9d1877165ca9f12589db7047031f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-11-02 12:11:08.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"4a90ece89bfbf883ad4fee7e8de596e1b298794c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 13:52:37.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3da8c35d43214562b5b26fb5fd754f50\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1183c1d882bb4d8c8e10f4330ac10f5e : SUCCESS in 13m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/484cfa8150434ce0a2b077d6125af8ca : SUCCESS in 4m 17s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/0913e57c0e044d69ac1eb8742bd1c022 : SUCCESS in 1h 18m 37s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b31c27f30386446e96e37da1b98c8efa : SUCCESS in 1h 23m 35s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/32a420c1baaa46fd8ba725ecbf4ab24e : SUCCESS in 1h 08m 54s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4aa8870e3c8c4e9ba0f38b9fcb4e3d48 : SUCCESS in 1h 14m 44s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/98ff64862e904ef4b00d841a9152c5f7 : SUCCESS in 1h 10m 21s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/32d51ceeee904cc2b001646341df4d27 : FAILURE in 1h 04m 07s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/296cb922f69746c18da116198590646d : SUCCESS in 1h 17m 16s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/780abb6f6ac34f7583aaa57200a532f4 : SUCCESS in 33m 41s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c4f4124a3404ec3842a42e8c640bdbf : SUCCESS in 32m 05s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eafe478248834963846c0f7e78feb93e : SUCCESS in 30m 23s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/90d1787657b54057ba1aa8a787f8aec7 : SUCCESS in 32m 43s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/01de4867cc2f46a4b38160a0e8b598ec : SUCCESS in 32m 17s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-9-stream https://zuul.opendev.org/t/openstack/build/c7a5e2bfb73541438d3d4d88121b19d7 : SUCCESS in 31m 59s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b70f38b64cd0435e8a298d60ceb829b3 : SUCCESS in 28m 46s","accounts_in_message":[],"_revision_number":14},{"id":"1ee5917c6800192ac2c5e5bbe7b503ea54c000dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-11-11 08:58:05.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"a395a5edacaae36b31cabe3d040ca5eed1e381a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-11 10:35:46.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4413af9043584a698d43f6190966abfa\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e13960c814644e2843295eded222e46 : SUCCESS in 14m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86002cf1db7e4d798f29235c5299f901 : SUCCESS in 4m 14s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/6e9d03236b5a49098f666b8a94d6d68c : SUCCESS in 1h 33m 06s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6b90bd44465e4610ab68bf1dd462ef44 : SUCCESS in 1h 27m 58s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/914d366dc5e04cdb9855e0b7108a67e2 : SUCCESS in 1h 08m 19s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6c005dbeb534116a3aaae5601e59e75 : SUCCESS in 1h 20m 33s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2a3af1b1b214481b38351956721b071 : SUCCESS in 1h 12m 09s\n- openstack-ansible-upgrade-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/724f51fa5e0b4ddcb65356132632690f : FAILURE in 1h 06m 09s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f469c5c0206741d49206d29251c97e46 : SUCCESS in 1h 34m 23s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92935159bae4424aa19371c14bce84fc : SUCCESS in 29m 09s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c4c76d11a7c146a0a58144d7fb0b6e00 : SUCCESS in 27m 10s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e126341a2d3249079d44e5caddb6f1fd : SUCCESS in 34m 31s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/6120032852a846d9aa0ab2230cd073d2 : SUCCESS in 33m 11s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/8e9d186994fb4f3aa5a5cce6257593f8 : SUCCESS in 28m 45s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-9-stream https://zuul.opendev.org/t/openstack/build/8b3719430b3d43b9af463cd385f1a3c1 : SUCCESS in 35m 47s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bfdebb007d2848a081057f0081da04bf : SUCCESS in 27m 40s","accounts_in_message":[],"_revision_number":15},{"id":"5cbb87257c86ea72dbfcb309c48a76ecbe10ff24","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-11-15 17:22:51.000000000","message":"Patch Set 15: Code-Review+2","accounts_in_message":[],"_revision_number":15},{"id":"e2d3e89567f4f014c15b86bd72c97b655bd0121d","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-11-28 16:25:12.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":15},{"id":"eaa402da73c1f93d402148fb302efb6087edbe92","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 16:27:22.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"c2ad717e0f5c0d152a1cf0eeefdc35af3f194300","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 18:02:55.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6de6b69785214e0c9a50bdb0f3588f2d\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2625f3aee6e441e296c693f65c5b7cea : SUCCESS in 12m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da503f10ed4046b9b2f15f5b9ca5be71 : SUCCESS in 5m 31s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/3184e839174d451fb2c9135de79e8db8 : SUCCESS in 1h 26m 17s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d313d7ff29004be59eb73c2876959bb9 : SUCCESS in 1h 27m 14s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/692424ce00b34d7c8664817bcd6d2535 : SUCCESS in 1h 17m 29s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2ae0e24d34334df3853a1ece8fadb6e4 : SUCCESS in 1h 13m 44s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9063781c8a154d1e886aff6a8c25c25f : SUCCESS in 1h 24m 16s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e4a2cf8aaae46908c28ddfd37c2fedd : SUCCESS in 32m 27s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa601a03e6dd42aaad3c132df9c94e62 : SUCCESS in 35m 11s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a0cc6c6e78d405cb63757520713080e : SUCCESS in 30m 41s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/bc866638e38749438ac062ac75275f90 : SUCCESS in 32m 35s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/3a140294c8e447d491071aa869495830 : SUCCESS in 36m 21s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-9-stream https://zuul.opendev.org/t/openstack/build/955587d6a30a4d7cb623bab6893b3489 : SUCCESS in 31m 52s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a4f01225154644ebbe4dec7a736f3c0a : SUCCESS in 35m 09s","accounts_in_message":[],"_revision_number":15},{"id":"53dcd53752b637b4cdd36f15685c50d56b250fad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 18:02:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":15},{"id":"495417a6dc7d586c70b2bea8aab8f66c2fb0bc95","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 18:03:57.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebe39c9523ba403d9b0558fa461c8168\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/918906f1ce0c46baad0587b1ee111ccb : SUCCESS in 41s","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"3d8e3690ba620d1724129f8ed1a6a040c5ccdac9","revisions":{"b929e677617247ac5eea42126ee11c7f76909c0d":{"kind":"REWORK","_number":1,"created":"2022-02-07 12:30:38.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/20/828120/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/20/828120/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/1"}}},"commit":{"parents":[{"commit":"b94c5599066436489f299494f6c95d743c35ae26","subject":"Clean up bionic variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b94c5599066436489f299494f6c95d743c35ae26"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-02-07 12:30:27.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-02-07 12:30:27.000000000","tz":120},"subject":"Replace ifupdown with native ip-link","message":"Replace ifupdown with native ip-link\n\nChange-Id: I5ac99e2b6c6e6ccd9da18ae68e1f8801f95f4f4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b929e677617247ac5eea42126ee11c7f76909c0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b929e677617247ac5eea42126ee11c7f76909c0d"}]},"branch":"refs/heads/master"},"edd80cf9e11a80e9cb6d1fe6505dbb0f34006d56":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-02-07 14:42:16.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/20/828120/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/20/828120/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/2"}}},"commit":{"parents":[{"commit":"9a4004169450a0bb68ffe0fec31e6887a9c075f3","subject":"Ensure that the legacy network-scripts package is present","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/9a4004169450a0bb68ffe0fec31e6887a9c075f3"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-02-07 12:30:27.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-02-07 14:42:16.000000000","tz":0},"subject":"Replace ifupdown with native ip-link","message":"Replace ifupdown with native ip-link\n\nChange-Id: I5ac99e2b6c6e6ccd9da18ae68e1f8801f95f4f4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/edd80cf9e11a80e9cb6d1fe6505dbb0f34006d56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/edd80cf9e11a80e9cb6d1fe6505dbb0f34006d56"}]},"branch":"refs/heads/master","description":"Rebase"},"59f44ab99bc0d958031ef01f9b43f6ed2de46383":{"kind":"REWORK","_number":3,"created":"2022-02-07 14:42:50.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/20/828120/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/20/828120/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/3"}}},"commit":{"parents":[{"commit":"9a4004169450a0bb68ffe0fec31e6887a9c075f3","subject":"Ensure that the legacy network-scripts package is present","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/9a4004169450a0bb68ffe0fec31e6887a9c075f3"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-02-07 12:30:27.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-02-07 14:42:49.000000000","tz":0},"subject":"Replace ifupdown with native ip-link","message":"Replace ifupdown with native ip-link\n\nChange-Id: I5ac99e2b6c6e6ccd9da18ae68e1f8801f95f4f4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/59f44ab99bc0d958031ef01f9b43f6ed2de46383"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/59f44ab99bc0d958031ef01f9b43f6ed2de46383"}]},"branch":"refs/heads/master"},"83af9d96f35a5d65a916e0acc3b19b6a7f278f36":{"kind":"REWORK","_number":4,"created":"2022-02-07 14:44:34.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/20/828120/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/20/828120/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/4"}}},"commit":{"parents":[{"commit":"9a4004169450a0bb68ffe0fec31e6887a9c075f3","subject":"Ensure that the legacy network-scripts package is present","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/9a4004169450a0bb68ffe0fec31e6887a9c075f3"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-02-07 12:30:27.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-02-07 14:44:34.000000000","tz":0},"subject":"Replace ifupdown with native ip-link","message":"Replace ifupdown with native ip-link\n\nChange-Id: I5ac99e2b6c6e6ccd9da18ae68e1f8801f95f4f4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/83af9d96f35a5d65a916e0acc3b19b6a7f278f36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/83af9d96f35a5d65a916e0acc3b19b6a7f278f36"}]},"branch":"refs/heads/master"},"c6729553f977da0764b2bca13cfbc7f5fca425ea":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-02-19 15:32:54.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/20/828120/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/20/828120/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/5"}}},"commit":{"parents":[{"commit":"94d2856291c31ace551eb4da7da4a4c48f6ec2a5","subject":"Merge \"Simplify installtion of libpython\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/94d2856291c31ace551eb4da7da4a4c48f6ec2a5"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-02-07 12:30:27.000000000","tz":120},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-02-19 15:32:54.000000000","tz":0},"subject":"Replace ifupdown with native ip-link","message":"Replace ifupdown with native ip-link\n\nChange-Id: I5ac99e2b6c6e6ccd9da18ae68e1f8801f95f4f4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c6729553f977da0764b2bca13cfbc7f5fca425ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c6729553f977da0764b2bca13cfbc7f5fca425ea"}]},"branch":"refs/heads/master","description":"Rebase"},"9e341409b1ecf78d63fa514a91c4d8eadcf830cb":{"kind":"REWORK","_number":6,"created":"2022-10-14 08:27:03.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/20/828120/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/20/828120/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/6"}}},"commit":{"parents":[{"commit":"a09612c1fbc4f2c93047ecea77c51b2ebb326496","subject":"Use policy_rc_d attribute instead of copy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a09612c1fbc4f2c93047ecea77c51b2ebb326496"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-02-07 12:30:27.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-10-14 08:26:43.000000000","tz":120},"subject":"Replace ifupdown with native ip-link","message":"Replace ifupdown with native ip-link\n\nChange-Id: I5ac99e2b6c6e6ccd9da18ae68e1f8801f95f4f4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/9e341409b1ecf78d63fa514a91c4d8eadcf830cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/9e341409b1ecf78d63fa514a91c4d8eadcf830cb"}]},"branch":"refs/heads/master"},"9d48d917dd3a9a5d4c95242494b71d09e91804d3":{"kind":"REWORK","_number":7,"created":"2022-10-14 10:23:32.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/20/828120/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/20/828120/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/7"}}},"commit":{"parents":[{"commit":"a09612c1fbc4f2c93047ecea77c51b2ebb326496","subject":"Use policy_rc_d attribute instead of copy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a09612c1fbc4f2c93047ecea77c51b2ebb326496"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-02-07 12:30:27.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-10-14 10:21:24.000000000","tz":120},"subject":"Replace ifupdown with native ip-link","message":"Replace ifupdown with native ip-link\n\nWe also leverage systemd-networkd for managing lxc-net and replace\nusing of custom service template for lxc-dnsmasq service with our\nsystemd-service role. These changes are quite tighten together, so\nit\u0027s quite hard to split them in different patchsets.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-systemd_service/+/861350\nChange-Id: I5ac99e2b6c6e6ccd9da18ae68e1f8801f95f4f4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/9d48d917dd3a9a5d4c95242494b71d09e91804d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/9d48d917dd3a9a5d4c95242494b71d09e91804d3"}]},"branch":"refs/heads/master"},"99325ab127c064fc1c61c911f7e58a5707a80025":{"kind":"REWORK","_number":8,"created":"2022-10-14 10:52:47.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/20/828120/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/20/828120/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/8"}}},"commit":{"parents":[{"commit":"a09612c1fbc4f2c93047ecea77c51b2ebb326496","subject":"Use policy_rc_d attribute instead of copy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a09612c1fbc4f2c93047ecea77c51b2ebb326496"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-02-07 12:30:27.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-10-14 10:52:40.000000000","tz":120},"subject":"Replace ifupdown with native ip-link","message":"Replace ifupdown with native ip-link\n\nWe also leverage systemd-networkd for managing lxc-net and replace\nusing of custom service template for lxc-dnsmasq service with our\nsystemd-service role. These changes are quite tighten together, so\nit\u0027s quite hard to split them in different patchsets.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-systemd_service/+/861350\nChange-Id: I5ac99e2b6c6e6ccd9da18ae68e1f8801f95f4f4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/99325ab127c064fc1c61c911f7e58a5707a80025"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/99325ab127c064fc1c61c911f7e58a5707a80025"}]},"branch":"refs/heads/master"},"0a75340f7d4185ba970438bb8290acd18d5e968a":{"kind":"REWORK","_number":9,"created":"2022-10-14 11:03:28.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/20/828120/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/20/828120/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/9"}}},"commit":{"parents":[{"commit":"a09612c1fbc4f2c93047ecea77c51b2ebb326496","subject":"Use policy_rc_d attribute instead of copy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a09612c1fbc4f2c93047ecea77c51b2ebb326496"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-02-07 12:30:27.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-10-14 11:03:21.000000000","tz":120},"subject":"Replace ifupdown with native ip-link","message":"Replace ifupdown with native ip-link\n\nWe also leverage systemd-networkd for managing lxc-net and replace\nusing of custom service template for lxc-dnsmasq service with our\nsystemd-service role. These changes are quite tighten together, so\nit\u0027s quite hard to split them in different patchsets.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-systemd_service/+/861350\nChange-Id: I5ac99e2b6c6e6ccd9da18ae68e1f8801f95f4f4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/0a75340f7d4185ba970438bb8290acd18d5e968a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/0a75340f7d4185ba970438bb8290acd18d5e968a"}]},"branch":"refs/heads/master"},"de38595061cd846dd7dcf5c7d54293d6a3c99907":{"kind":"REWORK","_number":10,"created":"2022-10-14 11:16:13.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/20/828120/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/20/828120/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/10"}}},"commit":{"parents":[{"commit":"a09612c1fbc4f2c93047ecea77c51b2ebb326496","subject":"Use policy_rc_d attribute instead of copy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a09612c1fbc4f2c93047ecea77c51b2ebb326496"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-02-07 12:30:27.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-10-14 11:16:07.000000000","tz":120},"subject":"Replace ifupdown with native ip-link","message":"Replace ifupdown with native ip-link\n\nWe also leverage systemd-networkd for managing lxc-net and replace\nusing of custom service template for lxc-dnsmasq service with our\nsystemd-service role. These changes are quite tighten together, so\nit\u0027s quite hard to split them in different patchsets.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-systemd_service/+/861350\nChange-Id: I5ac99e2b6c6e6ccd9da18ae68e1f8801f95f4f4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/de38595061cd846dd7dcf5c7d54293d6a3c99907"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/de38595061cd846dd7dcf5c7d54293d6a3c99907"}]},"branch":"refs/heads/master"},"6134f97cd562534cd5fca28021fc893c41e33e25":{"kind":"REWORK","_number":11,"created":"2022-10-18 08:32:18.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/20/828120/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/20/828120/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/11"}}},"commit":{"parents":[{"commit":"a09612c1fbc4f2c93047ecea77c51b2ebb326496","subject":"Use policy_rc_d attribute instead of copy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a09612c1fbc4f2c93047ecea77c51b2ebb326496"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-02-07 12:30:27.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-10-18 08:32:18.000000000","tz":0},"subject":"Replace ifupdown with native ip-link","message":"Replace ifupdown with native ip-link\n\nWe also leverage systemd-networkd for managing lxc-net and replace\nusing of custom service template for lxc-dnsmasq service with our\nsystemd-service role. These changes are quite tighten together, so\nit\u0027s quite hard to split them in different patchsets.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-systemd_service/+/861350\nChange-Id: I5ac99e2b6c6e6ccd9da18ae68e1f8801f95f4f4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6134f97cd562534cd5fca28021fc893c41e33e25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6134f97cd562534cd5fca28021fc893c41e33e25"}]},"branch":"refs/heads/master"},"957c0419dcfecfbb7612fd1e7cbae34287c30dea":{"kind":"REWORK","_number":12,"created":"2022-10-18 08:35:44.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/20/828120/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/20/828120/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/12"}}},"commit":{"parents":[{"commit":"a09612c1fbc4f2c93047ecea77c51b2ebb326496","subject":"Use policy_rc_d attribute instead of copy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a09612c1fbc4f2c93047ecea77c51b2ebb326496"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-02-07 12:30:27.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-10-18 08:35:35.000000000","tz":120},"subject":"Replace ifupdown with native ip-link","message":"Replace ifupdown with native ip-link\n\nWe also leverage systemd-networkd for managing lxc-net and replace\nusing of custom service template for lxc-dnsmasq service with our\nsystemd-service role. These changes are quite tighten together, so\nit\u0027s quite hard to split them in different patchsets.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-systemd_service/+/861350\nChange-Id: I5ac99e2b6c6e6ccd9da18ae68e1f8801f95f4f4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/957c0419dcfecfbb7612fd1e7cbae34287c30dea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/957c0419dcfecfbb7612fd1e7cbae34287c30dea"}]},"branch":"refs/heads/master"},"0cc0dc613054f915dae79c8f9f90fc5c10a13a3b":{"kind":"REWORK","_number":13,"created":"2022-10-18 08:37:25.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/20/828120/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/20/828120/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/13"}}},"commit":{"parents":[{"commit":"a09612c1fbc4f2c93047ecea77c51b2ebb326496","subject":"Use policy_rc_d attribute instead of copy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a09612c1fbc4f2c93047ecea77c51b2ebb326496"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-02-07 12:30:27.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-10-18 08:37:18.000000000","tz":120},"subject":"Replace ifupdown with native ip-link","message":"Replace ifupdown with native ip-link\n\nWe also leverage systemd-networkd for managing lxc-net and replace\nusing of custom service template for lxc-dnsmasq service with our\nsystemd-service role. These changes are quite tighten together, so\nit\u0027s quite hard to split them in different patchsets.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-systemd_service/+/861350\nChange-Id: I5ac99e2b6c6e6ccd9da18ae68e1f8801f95f4f4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/0cc0dc613054f915dae79c8f9f90fc5c10a13a3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/0cc0dc613054f915dae79c8f9f90fc5c10a13a3b"}]},"branch":"refs/heads/master"},"ad3d4b0d134f8c8b83b447ea52160107c60cd9b4":{"kind":"REWORK","_number":14,"created":"2022-11-02 12:11:08.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/20/828120/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/20/828120/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/14"}}},"commit":{"parents":[{"commit":"a09612c1fbc4f2c93047ecea77c51b2ebb326496","subject":"Use policy_rc_d attribute instead of copy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a09612c1fbc4f2c93047ecea77c51b2ebb326496"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-02-07 12:30:27.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-11-02 12:11:07.000000000","tz":0},"subject":"Replace ifupdown with native ip-link","message":"Replace ifupdown with native ip-link\n\nWe also leverage systemd-networkd for managing lxc-net and replace\nusing of custom service template for lxc-dnsmasq service with our\nsystemd-service role. These changes are quite tighten together, so\nit\u0027s quite hard to split them in different patchsets.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-systemd_service/+/861350\nChange-Id: I5ac99e2b6c6e6ccd9da18ae68e1f8801f95f4f4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ad3d4b0d134f8c8b83b447ea52160107c60cd9b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ad3d4b0d134f8c8b83b447ea52160107c60cd9b4"}]},"branch":"refs/heads/master"},"3d8e3690ba620d1724129f8ed1a6a040c5ccdac9":{"kind":"REWORK","_number":15,"created":"2022-11-11 08:58:05.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/20/828120/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/20/828120/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/828120/15"}}},"commit":{"parents":[{"commit":"d1fe9f7bece5ff11a1cac9b355f759bc29e62a46","subject":"Cleanup CentOS 8 Stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d1fe9f7bece5ff11a1cac9b355f759bc29e62a46"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-02-07 12:30:27.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-11-11 08:57:56.000000000","tz":60},"subject":"Replace ifupdown with native ip-link","message":"Replace ifupdown with native ip-link\n\nWe also leverage systemd-networkd for managing lxc-net and replace\nusing of custom service template for lxc-dnsmasq service with our\nsystemd-service role. These changes are quite tighten together, so\nit\u0027s quite hard to split them in different patchsets.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-systemd_service/+/861350\nChange-Id: I5ac99e2b6c6e6ccd9da18ae68e1f8801f95f4f4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3d8e3690ba620d1724129f8ed1a6a040c5ccdac9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3d8e3690ba620d1724129f8ed1a6a040c5ccdac9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
