)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~828237","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2Fvictoria~I3adb4dd75198935c4656d0208043ad8051f29312","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2022-02-11 20:17:06.000000000","reason":"Jonathan Rosser replied on the change"}},"hashtags":[],"change_id":"I3adb4dd75198935c4656d0208043ad8051f29312","subject":"Ensure that the legacy network-scripts package is present","status":"MERGED","created":"2022-02-08 09:28:17.000000000","updated":"2022-02-14 18:21:52.000000000","submitted":"2022-02-14 18:19:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"828237","cherry_pick_of_change":828113,"cherry_pick_of_patch_set":1,"meta_rev_id":"51efd66edffa4513ef2ef8ab5313997a69fe4932","_number":828237,"virtual_id_number":828237,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-14 18:19:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2022-02-08 14:22:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-14 17:03:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-14 17:03:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-08 09:33:24.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2022-02-08 10:49:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bd0ac7510a416ab279a628df805e328933505a63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-02-08 09:28:17.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2dbbbc2049c2f406da6f4c455454e95049369720","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-02-08 09:33:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"02205f752569305de602fe2eb201a9a59bfa9ba7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 10:49:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d5e9cc9839234230b0d80ae266e4d7aa : SUCCESS in 11m 18s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/082fb246cec246458a9a0f32090e26e8 : SUCCESS in 11m 10s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f6a18215e89434aa64df66e20328033 : SUCCESS in 13m 50s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d900356800f34cfe88199c1fdc3906a4 : SUCCESS in 11m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/980e0fbca3054933b115400e7a18c890 : SUCCESS in 7m 04s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/c6aa828e36094a65944c125fab87c93c : SUCCESS in 1h 02m 57s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/757760806c4e46c3b9f67a7e2b72b1b4 : SUCCESS in 58m 49s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/6c2c888146df4995a8a0ac8cee0bd1fe : FAILURE in 14m 26s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/99252ae8d1d14b3886a19a08d4d1caf5 : SUCCESS in 19m 15s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/50272ef6fbd1474184620d08cb050ae3 : SUCCESS in 16m 55s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/58c8e5bfdea44cc8a829b36621bc4e64 : SUCCESS in 17m 56s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aaf5dbc3c0004bb785ff81dac305063c : SUCCESS in 21m 06s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/faf3b846ffec423fb2a7c00999c3dcf5 : SUCCESS in 19m 29s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c84a6579ace42fdb5c49fc79fa05b1a : SUCCESS in 17m 18s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/47d0697112064c848e19e64c783321ed : SUCCESS in 21m 06s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/760757dcf42849b9aabc8aaba2dd640e : SUCCESS in 17m 10s","accounts_in_message":[],"_revision_number":1},{"id":"23078ca45d37b5f216a200ccceb488acdb15f90d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-02-08 14:19:56.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e3fbcc4c406c423c4c88fda428cdb59389222538","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-02-08 14:22:37.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5cc434a2ad8d8f9cb0d395d3606603f0904c3b65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 16:04:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6102e561b4d543689d0c41a5c2f16f45 : SUCCESS in 14m 13s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/7d1439a2cbf3445a97cead0fc30dfb6a : SUCCESS in 17m 32s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1f6060e297774cc0a1796a92f5179f6e : SUCCESS in 17m 21s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/857f3a26a5ab43c79379c73e38f9619f : SUCCESS in 18m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8baa5aba9624a37a4943d3709248585 : SUCCESS in 6m 59s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/e7e417cce31c45cc89e4188ad48b1d51 : SUCCESS in 1h 15m 54s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36cea9a7d3e74c3db7e34f9e1576ed3f : SUCCESS in 1h 06m 13s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/e46140d2eb0143e78a99cce084176cf9 : FAILURE in 22m 22s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/285ed81f153342288c12e1071f2e8576 : SUCCESS in 26m 04s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/835a028d6bcf4631996256b8a65f42ed : SUCCESS in 23m 52s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/da61df5c83444705821a95ba4ff98c60 : SUCCESS in 24m 43s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4226808ad3d04fa8a07a695bbec66282 : SUCCESS in 25m 05s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7c698bac4b284c25815e86819f0851d5 : SUCCESS in 23m 41s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd2f7f7b1be3443a9d8182d9dc537e5b : SUCCESS in 24m 53s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d417425fd91a4c47b8f01374c663aea8 : SUCCESS in 26m 00s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc9602f45ecf4c31a2b4e510c254cb49 : SUCCESS in 23m 46s","accounts_in_message":[],"_revision_number":2},{"id":"95e98541f80f80b69b2ca3cac7050e2ba8119810","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-02-08 16:06:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2c473f80bcc8818e0f075987c4fda8239b30b708","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 18:30:32.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d52f1f16495940779cd15fd700102e64 : SUCCESS in 13m 59s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/5384fe2b03ff4ba8b9abd47ec65a223b : SUCCESS in 16m 57s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/57fd70d7da004ebaa34a691b6922cfbb : FAILURE in 12m 59s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36feb27f40ce48e5a131d037bc7722a5 : SUCCESS in 14m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6256a59509fb40ce80f77ca6de1eff6b : SUCCESS in 8m 01s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/c793a1d952294ed08cccdba7ab85aa0f : SUCCESS in 1h 39m 18s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e1568255c4db4533995dee59814b5614 : SUCCESS in 1h 19m 55s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3259e0eada1c4dacb9c2372479ca490f : SUCCESS in 22m 25s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/71cbcc2c7e4c41fbb1852921386e1947 : FAILURE in 10m 59s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c8634b01d9bd4196ba9c3d84d830fb70 : FAILURE in 12m 23s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c0582b7fdd4446c8a2b022fe5d0abd7 : SUCCESS in 19m 31s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7688128204074ac1b5668b3c85af701d : SUCCESS in 23m 46s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8acef0ad26940b68fa4f63c8a593f03 : SUCCESS in 23m 07s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/39c60c0930e34231961a0b5361aa1d57 : SUCCESS in 22m 23s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa7d65548bf04530af149b9413ae6271 : SUCCESS in 22m 33s","accounts_in_message":[],"_revision_number":2},{"id":"358af427ae472a755e672eac595a3d220e955ead","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-02-11 20:17:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8664ade70fc6d791ffdd99c1b1910e7ad62cb210","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 21:38:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c14495c7af5e483ca4c0a33191051e9e : SUCCESS in 10m 57s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/0986f26749dc4d3a9c7e3d1446f7449f : SUCCESS in 13m 53s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3749b53c13dc459ba2a9fdbe8f48d2ef : SUCCESS in 14m 06s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f904a3625ba7465eb580d0c3feb47489 : SUCCESS in 14m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56d1d6bb92934aa4930009c61685526e : SUCCESS in 5m 07s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/bc97674e5e2a4ccd93e6286bd06eb7e0 : SUCCESS in 1h 12m 07s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f4e7cd662ea44a8902f622d1746426a : SUCCESS in 1h 08m 33s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/06853c9e1aed46c7a7c45750fbd25e37 : SUCCESS in 17m 15s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d42d98684f5429297553d703c49a8fe : SUCCESS in 20m 06s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9736a547714c4139bc9fda6218924dbd : SUCCESS in 18m 55s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/15920a72cad64b9aa2208b91d45b4398 : SUCCESS in 21m 48s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2fab9d4189e74873926f83b1cd677100 : SUCCESS in 20m 13s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7942d114e2734a4ea41a1d9caf456093 : SUCCESS in 19m 16s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b632e57eb6cf4f9b989aa2c5d99edde2 : SUCCESS in 19m 05s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91e33e57a3a34179861eced8e46eee4f : SUCCESS in 21m 25s","accounts_in_message":[],"_revision_number":2},{"id":"9153b9cfa4913367e4da7dcd6769301c60297d9c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-02-14 17:03:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c29a01a417dcb1c3f9cc93a5216aec70a4ac92be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 17:05:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"66203605747a619adebd06c5722491b51f08984a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 18:19:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7c4ba3d0a69f410cb0ff407a4578acd9 : SUCCESS in 13m 59s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/3216b7c341e64e86b580b116f36aa281 : SUCCESS in 17m 15s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0c1134352e4d40bd9319b44305960694 : SUCCESS in 15m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83c4e703f553485c8f9f73e326d7ef01 : SUCCESS in 5m 34s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/a4321dd5c79940b68d8a4da58216a4d7 : SUCCESS in 1h 02m 45s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/475f668c49b4405cb6b0da64eb0c1323 : SUCCESS in 1h 13m 10s","accounts_in_message":[],"_revision_number":2},{"id":"ea8878171588d40bb50d3d16529cc076d215daf8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 18:19:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"51efd66edffa4513ef2ef8ab5313997a69fe4932","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 18:21:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47c8d0cdfab249ba8703cdf004e1f229 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1ed262ebbd4226d9af9e0e9325fe5013520b48e6","revisions":{"13030b77994e352cc5336eef2d2ba52b5adde8f3":{"kind":"REWORK","_number":1,"created":"2022-02-08 09:28:17.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/37/828237/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/37/828237/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/828237/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/828237/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/828237/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/828237/1"}}},"commit":{"parents":[{"commit":"51eca809d2330ca707953d5fdfd308175f738331","subject":"Revert \"Add CentOS 8.4 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/51eca809d2330ca707953d5fdfd308175f738331"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-02-07 11:31:52.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-02-08 09:28:17.000000000","tz":0},"subject":"Ensure that the legacy network-scripts package is present","message":"Ensure that the legacy network-scripts package is present\n\nThe lxc_hosts ansible role currently relies on the ifup / ifdown\ncommands.\n\nChange-Id: I3adb4dd75198935c4656d0208043ad8051f29312\n(cherry picked from commit 9a4004169450a0bb68ffe0fec31e6887a9c075f3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/13030b77994e352cc5336eef2d2ba52b5adde8f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/13030b77994e352cc5336eef2d2ba52b5adde8f3"}]},"branch":"refs/heads/stable/victoria"},"1ed262ebbd4226d9af9e0e9325fe5013520b48e6":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-02-08 14:19:56.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/37/828237/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/37/828237/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/828237/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/828237/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/828237/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/828237/2"}}},"commit":{"parents":[{"commit":"51eca809d2330ca707953d5fdfd308175f738331","subject":"Revert \"Add CentOS 8.4 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/51eca809d2330ca707953d5fdfd308175f738331"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-02-07 11:31:52.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-02-08 14:19:56.000000000","tz":0},"subject":"Ensure that the legacy network-scripts package is present","message":"Ensure that the legacy network-scripts package is present\n\nThe lxc_hosts ansible role currently relies on the ifup / ifdown\ncommands.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-openstack_hosts/+/828346\nChange-Id: I3adb4dd75198935c4656d0208043ad8051f29312\n(cherry picked from commit 9a4004169450a0bb68ffe0fec31e6887a9c075f3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/1ed262ebbd4226d9af9e0e9325fe5013520b48e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/1ed262ebbd4226d9af9e0e9325fe5013520b48e6"}]},"branch":"refs/heads/stable/victoria","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
