)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~833119","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I5412e9b4e85373c920b127a96dc9d505ca9fe337","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"ipaddr_filter","attention_set":{},"removed_from_attention_set":{"13095":{"account":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"last_update":"2022-03-15 16:48:43.000000000","reason":"Marc Gariépy replied on the change"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-03-15 20:33:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5412e9b4e85373c920b127a96dc9d505ca9fe337","subject":"Change location of ipaddr filter","status":"MERGED","created":"2022-03-10 16:10:25.000000000","updated":"2022-03-15 20:35:15.000000000","submitted":"2022-03-15 20:33:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"833119-ipaddr_filter","meta_rev_id":"7cf666ae279191a4e4d6e160c146dd3734aa36df","_number":833119,"virtual_id_number":833119,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-15 20:33:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2022-03-14 17:09:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-15 10:35:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":1,"date":"2022-03-15 12:19:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-15 10:35:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-10 17:42:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-14 13:49:57.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2022-03-14 17:09:28.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2022-03-15 10:35:40.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"799f7be5eeb466f9b9bb59adde617ccb6e17ab4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-03-10 16:10:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"33711bfb9d4a98011eaf0fb6ab3454a0da4ebeb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 17:42:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1bcbb2c342a464cb955978b8bf0cf20 : SUCCESS in 19m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b36fc19454434410b40f011b62fef31c : SUCCESS in 7m 19s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/2882e260807b4682ab0db0f34bceadd6 : FAILURE in 1h 07m 11s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/cb73e6cf5ef0420db2e92396da106857 : FAILURE in 56m 38s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49c5cc368f3f42b3a81617fc75f0ec03 : FAILURE in 1h 03m 02s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/bbb8d1fa28494b72bfd9541409bb1c38 : FAILURE in 1h 24m 52s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd5c7aa76f694528b19e862a1d8daefb : FAILURE in 1h 11m 09s\n- openstack-ansible-lxc-dir-centos-8-stream https://zuul.opendev.org/t/openstack/build/ab27d2734c2a42f481ebd5297b20f3b7 : FAILURE in 17m 22s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3430d233b6ea484396b8a7b61028c2e0 : FAILURE in 19m 27s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f520a895e38648cc97e196efbca8023e : FAILURE in 16m 48s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd55b451881344f49e97e2d97f0c3fe5 : FAILURE in 16m 53s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3106912fa36a4008a67f1de2928a429b : FAILURE in 18m 57s","accounts_in_message":[],"_revision_number":1},{"id":"2a819291b7aeaed9baae41ec7ba77a5e3fd6f44d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-03-10 20:32:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"917cca5be099ed60a35c41bf8ed267efc2a3e0c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 21:55:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92fd8462d9dc49a9bda86acebcb1c6dd : SUCCESS in 22m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/975f520d1f4341029583a34cd743e7d3 : SUCCESS in 9m 37s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/deb3cf629f2041c2b5a9389b620a11a2 : FAILURE in 1h 10m 27s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/fd52b2d84f274d21acec3ce36b6a0f08 : FAILURE in 52m 40s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/19ba1a732b3f475d857a94342a5d6b77 : FAILURE in 1h 03m 30s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/1497f7097d7848988c9db92ff88ef063 : FAILURE in 1h 11m 38s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/997e18200c844410a747f275114b9a32 : FAILURE in 1h 12m 51s\n- openstack-ansible-lxc-dir-centos-8-stream https://zuul.opendev.org/t/openstack/build/ec070d96b7a343df862835237078cbc7 : FAILURE in 21m 03s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe6c2411bb3d4669a6d6cdd1dfbefcbc : FAILURE in 14m 52s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2f55f058d5f4cd68725a5008d544703 : FAILURE in 19m 11s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/072259d6c22f479cbcd7f6f0aa827e02 : FAILURE in 17m 23s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2028e2eba7f64c01a83f5e954e71439b : FAILURE in 19m 35s","accounts_in_message":[],"_revision_number":1},{"id":"edd617eef7479cbafb7cb4d2aac6018d938f767f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-03-14 13:49:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f4b6a61793b53dee011eac0cf246e876803ef0d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-03-14 14:39:47.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"994e4dacc9649d29a57d8d2640ff8b8cc42d2d53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-03-14 15:10:30.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"dd4584d47e673c76e4aa0f31a0f24190229d1075","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 17:03:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11331e0b00084180bd05cf8033a31005 : SUCCESS in 22m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27dce7e16bbe41358bcb5148003229bb : SUCCESS in 7m 44s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/c5941be96ccf475aa8159b1852f6d93c : SUCCESS in 1h 35m 41s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/81e6591b3cb64a55b0404e7ffe735050 : SUCCESS in 58m 10s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/3c8d470f060045aab110f6032ae3ac0a : SUCCESS in 1h 28m 10s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8031706db1a54353b9d2fa0b2e35d348 : SUCCESS in 1h 16m 51s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/16e00dcd9f1849329c836e2bf35f5580 : SUCCESS in 1h 46m 12s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d612902919db48e8944bb6e9125d7882 : SUCCESS in 1h 37m 22s\n- openstack-ansible-lxc-dir-centos-8-stream https://zuul.opendev.org/t/openstack/build/9d55536d194e4fe78b489259d1f0773b : SUCCESS in 24m 00s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/496b5e450fb943a1ba6b10e7aff77014 : SUCCESS in 22m 13s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd3a5529f6be465db15466be084bf7b2 : SUCCESS in 23m 26s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b2da6ffdcb148589685b55a0758674a : SUCCESS in 23m 49s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/83fb1dcecc024b8cab8bf06ecfc753d0 : SUCCESS in 22m 24s","accounts_in_message":[],"_revision_number":3},{"id":"a5d40562da3611bcf6db4b7855ef49895f745dfc","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-03-14 17:09:28.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"8cfb80a332059bd2510bbe34fe9f8ddfdc20d397","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-03-15 10:35:40.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ab185ec0ffc3f27891a5c134dface8b811ac457b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-03-15 12:19:56.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3863f944fef08bac2a1ba2d20a042ff88cb090fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 12:22:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c0abb91974228dcb9ba4052f31efb15504790144","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 13:47:44.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd839a2792b843a588c403134b719019 : SUCCESS in 16m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e75e79c80834cbdaeca9214035f1759 : SUCCESS in 7m 33s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/5d0db5c2dccd4ba097bcf72f00eac304 : SUCCESS in 1h 19m 43s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/bfbb3acc74614c9e9adc9f1ed3fb9e96 : SUCCESS in 1h 02m 02s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/e7a20d5be02f4ba2afc6f42b2fba290b : SUCCESS in 1h 20m 48s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7fbae8dc149741b49f8fff04c66a95e5 : SUCCESS in 1h 10m 47s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b8478ca0bcb44a6851b68826701cdca : FAILURE in 1h 09m 48s","accounts_in_message":[],"_revision_number":3},{"id":"3609531d4754da395ec694792d61a83e94694745","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-03-15 16:48:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3026839184d362db102d69aacab6eaf8484a8c35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 18:51:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78213f1db53f4815b2f4ff319b80d4fe : SUCCESS in 17m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db40ea526a744ccca7d2010416d85a1d : SUCCESS in 7m 08s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/288fe892c6164744b5ce53794c079ecf : SUCCESS in 1h 23m 21s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/459939fb49324cd2bdafaebc51ee1862 : SUCCESS in 1h 05m 57s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/3648a9b4a469432e86227d161287614f : SUCCESS in 1h 28m 29s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7acdfc25696d4770b9ec9c95553ca981 : SUCCESS in 1h 18m 45s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/c7cce045909c478892099cdee9b7df38 : SUCCESS in 1h 51m 57s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b8266fe21b44e15909f25ba40c6d070 : SUCCESS in 1h 36m 36s\n- openstack-ansible-lxc-dir-centos-8-stream https://zuul.opendev.org/t/openstack/build/a07a111fd4314901ad8c9224efee2918 : SUCCESS in 22m 44s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b51958a133f34273b6cfe93aff9f8ba2 : SUCCESS in 21m 56s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/58d76b1bb7bd47c0ae1c5303254ba233 : SUCCESS in 22m 17s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a7e1d8feb74410eba481b84e08323b9 : SUCCESS in 22m 45s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a2636c4092b4c5b862ab8789f7a3121 : SUCCESS in 21m 28s","accounts_in_message":[],"_revision_number":3},{"id":"0ef38b321fa537d1b401cf4258001df3dd3eda5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 18:55:14.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d800af5eaa3f752669ef47d31b8fd7d5beba6f30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 20:33:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/207226cbfc59406ea26d95d7811c1cbf : SUCCESS in 19m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f434020fbda48b68874409507945fa6 : SUCCESS in 7m 59s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/7191c3a26fe24efc9be98a5db0b1bd91 : SUCCESS in 1h 25m 38s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/d1c775832af44795aee9ef52d1b1896b : SUCCESS in 1h 01m 29s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/230dc9942d664d3c944c139e0947bbb2 : SUCCESS in 1h 27m 52s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f4fda17a55f54b57b609540ccf6d76e0 : SUCCESS in 1h 02m 45s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e2ef51aa613343c598fe04135295697f : SUCCESS in 1h 37m 41s","accounts_in_message":[],"_revision_number":3},{"id":"b23f23a586fe7d98d541a1ac9d7c5cabc122991c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 20:33:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"7cf666ae279191a4e4d6e160c146dd3734aa36df","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 20:35:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a881c7351564596858a0406a3c9ac11 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ed38f3393d15fd6028e1fad76438af0dbcb6c1d3","revisions":{"fad3058ce79185fe166df6a3d525fd11b5d47439":{"kind":"REWORK","_number":1,"created":"2022-03-10 16:10:25.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/19/833119/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/19/833119/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/19/833119/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/19/833119/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/19/833119/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/19/833119/1"}}},"commit":{"parents":[{"commit":"f16e0c81125ee9dea60f01eee7d4fc5dd6dc44e2","subject":"Merge \"Clean up bionic variables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f16e0c81125ee9dea60f01eee7d4fc5dd6dc44e2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-03-10 16:10:15.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-03-10 16:10:15.000000000","tz":60},"subject":"Change location of ipaddr filter","message":"Change location of ipaddr filter\n\nipaddr filter has been moved from netcommon to utils collection [1]\n\nBased on that we must add that collection to requirements.\n\n[1] https://github.com/ansible-collections/ansible.netcommon/commit/db4920ebf6bae6476ff8829e2cf475f19f83a990\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/831525\nChange-Id: I5412e9b4e85373c920b127a96dc9d505ca9fe337\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/fad3058ce79185fe166df6a3d525fd11b5d47439"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/fad3058ce79185fe166df6a3d525fd11b5d47439"}]},"branch":"refs/heads/master"},"b38715efcc51a7bbf711fa2820531f5f086a6caf":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-03-14 14:39:47.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/19/833119/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/19/833119/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/19/833119/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/19/833119/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/19/833119/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/19/833119/2"}}},"commit":{"parents":[{"commit":"f16e0c81125ee9dea60f01eee7d4fc5dd6dc44e2","subject":"Merge \"Clean up bionic variables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f16e0c81125ee9dea60f01eee7d4fc5dd6dc44e2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-03-10 16:10:15.000000000","tz":60},"committer":{"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","date":"2022-03-14 14:39:47.000000000","tz":0},"subject":"Change location of ipaddr filter","message":"Change location of ipaddr filter\n\nipaddr filter has been moved from netcommon to utils collection [1]\n\nBased on that we must add that collection to requirements.\n\n[1] https://github.com/ansible-collections/ansible.netcommon/commit/db4920ebf6bae6476ff8829e2cf475f19f83a990\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/831525\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-systemd_networkd/+/831603\nChange-Id: I5412e9b4e85373c920b127a96dc9d505ca9fe337\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b38715efcc51a7bbf711fa2820531f5f086a6caf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b38715efcc51a7bbf711fa2820531f5f086a6caf"}]},"branch":"refs/heads/master","description":"Edit commit message"},"ed38f3393d15fd6028e1fad76438af0dbcb6c1d3":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-03-14 15:10:30.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/19/833119/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/19/833119/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/19/833119/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/19/833119/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/19/833119/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/19/833119/3"}}},"commit":{"parents":[{"commit":"f16e0c81125ee9dea60f01eee7d4fc5dd6dc44e2","subject":"Merge \"Clean up bionic variables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f16e0c81125ee9dea60f01eee7d4fc5dd6dc44e2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-03-10 16:10:15.000000000","tz":60},"committer":{"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","date":"2022-03-14 15:10:30.000000000","tz":0},"subject":"Change location of ipaddr filter","message":"Change location of ipaddr filter\n\nipaddr filter has been moved from netcommon to utils collection [1]\n\nBased on that we must add that collection to requirements.\n\n[1] https://github.com/ansible-collections/ansible.netcommon/commit/db4920ebf6bae6476ff8829e2cf475f19f83a990\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/831525\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-systemd_networkd/+/831603\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-lxc_container_create/+/833118\nChange-Id: I5412e9b4e85373c920b127a96dc9d505ca9fe337\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ed38f3393d15fd6028e1fad76438af0dbcb6c1d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ed38f3393d15fd6028e1fad76438af0dbcb6c1d3"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
