)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~834183","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I23632f9c145be334b1d19067352f8b82114a1209","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-04-29 20:56:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I23632f9c145be334b1d19067352f8b82114a1209","subject":"Replace systemd-mount template with role","status":"MERGED","created":"2022-03-17 15:24:57.000000000","updated":"2022-04-29 20:58:24.000000000","submitted":"2022-04-29 20:56:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"834183","meta_rev_id":"ebf4dde99148d2abaf19447b4b6099be55c1c94d","_number":834183,"virtual_id_number":834183,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-29 20:56:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2022-04-12 10:44:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-11 09:18:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2022-04-12 10:44:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2022-04-29 11:59:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-17 16:55:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-04 20:38:12.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2022-04-05 07:29:38.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2022-04-29 11:58:53.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"CC"},{"updated":"2022-04-29 11:59:33.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"}],"messages":[{"id":"d901318ef5c1ec6dcad699071105d85b3a93d972","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-03-17 15:24:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b75d2f354dffee603919b298796887f1e815fb43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 16:55:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6925fc301c474f3690949d10a90d395b : SUCCESS in 17m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68682ff450a148f7a0158fa5bf80d046 : SUCCESS in 7m 16s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/0e4cecc64ca44969ae9b4f359f366e46 : SUCCESS in 1h 26m 04s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/dcc95900eadd44ecb4aa99d2a0e964ce : SUCCESS in 1h 08m 07s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/45b18bbdf25946299eb188173f696fb6 : FAILURE in 1h 17m 12s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/71468b20854642638a3f6fd7199229ef : FAILURE in 1h 10m 28s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/6899dac3bac541599f9b8fdda71da53d : SUCCESS in 1h 20m 33s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b8b0f543cae4324ad2eec8a96166136 : FAILURE in 52m 39s\n- openstack-ansible-lxc-dir-centos-8-stream https://zuul.opendev.org/t/openstack/build/a48c3d0a242f4095a92d9552d2fdbeea : SUCCESS in 20m 45s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b45a23749924f4585421df45b855f40 : SUCCESS in 23m 02s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d6e60223dd647589d0e2558ad6eccea : SUCCESS in 20m 49s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/140f86ec21da4493ab9400b8957d81cc : SUCCESS in 22m 12s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6ba345714964b9e89a71eabe66323c3 : SUCCESS in 22m 25s","accounts_in_message":[],"_revision_number":1},{"id":"3f02bccfc3c18c20c17dddccf0408327b77a3421","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-03-17 17:26:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"844023ca5eaf35f021bd2ccaa4f2bc92e60107b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 19:23:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32b2d03356c64c12ad51aaea18d8977a : SUCCESS in 21m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b57524d5c28a4257a98fb075e95edd5b : SUCCESS in 9m 33s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/ecc943e314474a11ad73a4b5e22eea13 : SUCCESS in 1h 29m 01s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/25c30f8b26df4ad4825bc80e75c1bb4b : SUCCESS in 1h 17m 31s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/97bdca5709cb43f4affed272dd2911a0 : FAILURE in 1h 39m 26s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/968580c6f0c849bb9432c3dbdfe62bd4 : FAILURE in 1h 24m 52s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/9b91ef62b8764adbafbe7944f6c99bb6 : SUCCESS in 1h 51m 28s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3208244337dc4c0694e37768d5364a9d : FAILURE in 1h 13m 35s\n- openstack-ansible-lxc-dir-centos-8-stream https://zuul.opendev.org/t/openstack/build/63400daf2f2b4b6daac97f121afd5e56 : SUCCESS in 25m 36s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12b7312e5eda4fac9e671b9dba881b14 : SUCCESS in 25m 00s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0beee50b0bf149acb21025158ba589ea : SUCCESS in 25m 05s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3ad874e073e4d5cb90a336388c0ff2c : SUCCESS in 22m 29s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c698945654d74e9baa1ff8b7dee0b765 : SUCCESS in 23m 15s","accounts_in_message":[],"_revision_number":1},{"id":"2f3c50b56ca2c9219ce5514a4add60958a37bc25","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-03-18 08:14:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d8b2a2f0262460250e7ee0053759e38acdfa6e2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 10:00:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f074aaf15c64c57bb969b640097edcf : SUCCESS in 20m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3de63b64675b406c8a91acb5f5f1cfdf : SUCCESS in 8m 10s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/49db9f42ba0a41e99ba14137ba69d0e6 : SUCCESS in 1h 26m 48s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/ab147cc7d9894caa9f2faa2a94cfd623 : SUCCESS in 1h 17m 10s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/bbf85181908543bc9012559bbf3538f8 : SUCCESS in 1h 20m 59s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84fb65f4733e4a27962858fc44f5bee4 : FAILURE in 1h 15m 21s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/097300b936b244bcb0fe45d47eb3d15c : SUCCESS in 1h 42m 40s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/685766fed59943448c33934591b5e78c : SUCCESS in 1h 35m 41s\n- openstack-ansible-lxc-dir-centos-8-stream https://zuul.opendev.org/t/openstack/build/939e6792348344329f1a5789246932ec : SUCCESS in 25m 55s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6766cb64679449139baf48ab76c74425 : SUCCESS in 18m 50s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e334e82836df49558fcedb7fa283ef0a : SUCCESS in 22m 01s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e3db695c19c44bf9692dd9688f0c4f2 : SUCCESS in 18m 21s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/464943a64eef4029aab358f63cc3e0e1 : SUCCESS in 22m 59s","accounts_in_message":[],"_revision_number":1},{"id":"a5c374b3041d09656187b1b7c466ec1b5e04dffd","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-04-04 10:19:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"77ade8179fb28abf5cc07a99820ff9e3b5b7d869","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 12:03:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b278b14330745ac809de49d6132b2b6 : SUCCESS in 14m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bac3b268afb47419032b9cfe74b02dd : SUCCESS in 5m 03s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/642cb24e84224df486b13ab402e5e66e : SUCCESS in 1h 25m 11s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/c57901ff1f0f4a369651d7dffb515139 : SUCCESS in 59m 19s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/dfbd4a348e7a44d18b5cb308c8859f9e : SUCCESS in 1h 25m 31s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9fcdacf039624c3abf5eb8c42b45b55e : SUCCESS in 1h 22m 46s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/a135b31c5e4242e2a4bcf3a6566c6575 : SUCCESS in 1h 40m 23s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/852d898d41c54fc381935ad738d4a407 : SUCCESS in 1h 36m 22s\n- openstack-ansible-lxc-dir-centos-8-stream https://zuul.opendev.org/t/openstack/build/7794006e40db44f39c39345c01e36c16 : SUCCESS in 19m 25s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dea1fda54e9c4d38a74be67877b883fc : SUCCESS in 21m 37s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/549105283ace4fa793d0d380b3d85a43 : SUCCESS in 18m 51s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/759860b49993431aa97226b691ba20cd : SUCCESS in 19m 42s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/142fbd80db7f4d8abe275b8b0e5f3f02 : SUCCESS in 19m 02s","accounts_in_message":[],"_revision_number":1},{"id":"ebe580ef45e52652704fa3a99d6dbe1a85081702","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-04 20:38:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"15182193548136659ffe300a4604b9afb18d8a2e","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-04-05 07:29:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"139b3e1b25b078b31c9c8805e743014eb6a5b7a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 07:32:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"83e7080200da2fd8e991c31d73094dd8624005df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 10:43:15.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e09bbab5c7e04cc48283a57302f0ea75 : SUCCESS in 10m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc1c88d6a1b6495598aca3962cf56731 : SUCCESS in 5m 46s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/7e47c36801704d0290338ce87c9281c1 : TIMED_OUT in 3h 10m 05s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/199f20b626be4d828e8746f5648a80bb : SUCCESS in 1h 16m 53s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/1cab41cf8cf446b69b474bbe7681b76d : SUCCESS in 1h 30m 11s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32a886b086ba4872ba74d0a8369e946b : SUCCESS in 1h 13m 35s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7fdecfa22c414d3da9c988a9c4888a53 : SUCCESS in 1h 28m 08s","accounts_in_message":[],"_revision_number":1},{"id":"1c896e96be7623df43d8b9c3141905283bfa19a0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-04-05 11:20:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"74127c5bd2ef2e54ade927b38e5221b598d282d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 13:11:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7a99b5cfd754d9a8726f33b83d7f420 : SUCCESS in 16m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07ebe7daf0c642f88fd0118ad1500753 : SUCCESS in 7m 35s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/ebd210e31b014faa834f66809c24c3f4 : SUCCESS in 1h 29m 13s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/4e801e4defdd450ba3b2d52cb568f97a : SUCCESS in 1h 10m 09s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/de7c84df069d40e4aa52780d65b08007 : SUCCESS in 1h 17m 35s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55c185c9f5d2429f95e0310d1f34a6fd : SUCCESS in 1h 08m 46s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/49d8dc85ce7b47889f8a897cdbeb0485 : SUCCESS in 1h 36m 26s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41789341474647a68f07fb9683885f01 : SUCCESS in 1h 38m 36s\n- openstack-ansible-lxc-dir-centos-8-stream https://zuul.opendev.org/t/openstack/build/1659103fd602420cb6f308ef5eadc9e1 : SUCCESS in 23m 29s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cfff09fd152e43fa855a74e1d8848209 : SUCCESS in 19m 18s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8bc1c6479f444db6a8d92d8a8e9dfbc5 : SUCCESS in 20m 33s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b76fedc6db6a4e359776c050f85464c8 : SUCCESS in 20m 31s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/73957e8088e344e3b2dc165640fafbc6 : SUCCESS in 16m 35s","accounts_in_message":[],"_revision_number":1},{"id":"0813280b45adc4af9c6662d14f2addbe26fe85dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 13:14:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"539e446a6f9e1c974d864804b3b793af363fe71e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 14:56:25.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69020d1a89cd4171bf3c0a41a364b304 : SUCCESS in 13m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f06bbf0488943a5b87d239b936709c4 : SUCCESS in 5m 05s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/c0aa25afb4234395be134cb764922164 : SUCCESS in 1h 09m 38s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/7604b50863fd4a7980027093323b7c1a : SUCCESS in 1h 03m 15s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/29b7857fc00a498ba34d5e14ef20141f : SUCCESS in 1h 21m 31s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/791c50d5d807469694c0118e236940b0 : FAILURE in 1h 05m 00s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9aad663cc5e843c8a927db500e68e88c : FAILURE in 1h 04m 06s","accounts_in_message":[],"_revision_number":1},{"id":"bca02c2ba3e5421330f8f6335c20932370ea78df","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-06 08:35:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"639ce10db5c5e9b03ea31dd80c158fe5fb9dae26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 10:13:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d2dfffcd70b54cd2bc5db9f4f3c6ccf1 : SUCCESS in 17m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38ee70be755547f3882e62f78faedc33 : SUCCESS in 5m 35s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/75d03e83892748629b9808abd9b2643d : SUCCESS in 1h 25m 48s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/39735b3b5e3a4e8fafc851ce90b2c52b : SUCCESS in 1h 09m 14s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/bca64e9d06b647e4b378c6cc12093207 : SUCCESS in 1h 25m 57s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ef543af37954cf38b06fe6f174c33bc : SUCCESS in 1h 33m 28s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/4b89dff26fa842f9b1395a1fef621e0d : FAILURE in 1h 09m 43s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bec14685f9154dbbb70b6bc1037fb56e : FAILURE in 1h 14m 05s\n- openstack-ansible-lxc-dir-centos-8-stream https://zuul.opendev.org/t/openstack/build/ea4f778ed97c4a0298e175f13dfeb1f3 : SUCCESS in 22m 17s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f633a47cb6b74346b074c6741b70c167 : SUCCESS in 20m 36s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2fe8de8d9534ffd93c7de1fb28134dd : SUCCESS in 21m 03s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3cdee89c3f594b218376d9ba31479f94 : SUCCESS in 20m 25s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e71440eb96bf46c492317357ce78a029 : SUCCESS in 20m 37s","accounts_in_message":[],"_revision_number":1},{"id":"62a8b6a622306be6526db3f64b7098e3d8c58b98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-04-07 11:40:09.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a282071c187efe5b703fd86b33972262dd70caf6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 13:22:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f606075063f44fbb3b22abe65485027 : SUCCESS in 14m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9aaf53ec1194b20add06a2c05fcfbc6 : SUCCESS in 5m 18s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/54c1582313744738bbbb480b3b02caf5 : SUCCESS in 1h 12m 12s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/6c193066968049f29b7c6883e782cacc : SUCCESS in 57m 45s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/94d7f686aa594d41b1a8b18f604674e7 : SUCCESS in 1h 23m 47s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3fef1060821d4d1ca774c17a1f6afc02 : SUCCESS in 1h 18m 04s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/23ad8c3509064ba285ef54d4af0a1b0b : SUCCESS in 1h 33m 32s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e38a6dd6997849a78c6a298873907932 : SUCCESS in 1h 37m 34s\n- openstack-ansible-lxc-dir-centos-8-stream https://zuul.opendev.org/t/openstack/build/a2bd04c29de84beeba5e27b8cb194fe9 : SUCCESS in 21m 04s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9cb61da0171e44bd93c5f806b949ef1f : SUCCESS in 21m 18s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f79c3a44edb04cda9408b13635311925 : SUCCESS in 19m 50s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09f901f13a3a49cc82768863d6a410b8 : SUCCESS in 20m 44s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3fd6b7e3a18d4254a64ebe6ab706c0f2 : SUCCESS in 20m 56s","accounts_in_message":[],"_revision_number":2},{"id":"b8a6466c84a94da62de6572f55cea5d66980b4ed","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-11 09:18:36.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7b9cf542359ab9de646938ba9a810274730cadc2","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-04-12 10:44:38.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cd2aba14848dd58a8191628d5238cc8c3fb1c1e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 10:52:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b944a4780facf2671dce21e54974972903117882","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 12:26:20.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56ae44d19e124d65ab87ec444260c7f0 : SUCCESS in 20m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1563364d8f954b938d15abc552623a98 : SUCCESS in 6m 00s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/e8a1a84ce441439a965fdee4108d35c0 : FAILURE in 1h 17m 00s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/6c80761042a042ab8f3f970b68ca76da : SUCCESS in 1h 14m 30s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/db0fea827ec243d691ef583e4ab2a3ac : SUCCESS in 1h 25m 06s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a2737e216f3c4b96857696a387123afc : SUCCESS in 1h 10m 58s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8558154745e842d99b7cccaffd10fb74 : SUCCESS in 1h 26m 07s","accounts_in_message":[],"_revision_number":2},{"id":"0f092ae9f9ef6a174c892dea12e2aabd6943779a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-04-14 09:09:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e65fbd0e1416318b989c4e45d3dedd27505097b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 10:42:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5033c8605daa4f9f8d34b61c52e27d0e : SUCCESS in 14m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9048fc7e365742b99702a593a0f8cfbc : SUCCESS in 5m 04s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/65e3d4095bad4009b444175eea744573 : SUCCESS in 1h 13m 49s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/417f2ba494cd478eb095317e2263eec3 : SUCCESS in 1h 02m 07s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/a1c046bdcc71429e9552ea049702a845 : SUCCESS in 1h 25m 03s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/82f93d1553314188943422f378e6ab5f : SUCCESS in 1h 04m 13s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/7315b3bde3614c2e8ba4c2c40db8997c : SUCCESS in 1h 24m 31s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69cd5cb70a634389a5974ef4b2668e4a : FAILURE in 1h 07m 08s\n- openstack-ansible-lxc-dir-centos-8-stream https://zuul.opendev.org/t/openstack/build/b2b69b92d83147f7b34ade53bda6489b : SUCCESS in 19m 06s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/976766a6618546679a455298921b9141 : SUCCESS in 17m 26s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/153ad62d1cd54267ab15f2adf00cf03f : SUCCESS in 17m 23s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46ab022756664309bd7189a70918ddb2 : SUCCESS in 17m 01s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8cbd060740254d649e878051049fc622 : SUCCESS in 18m 46s","accounts_in_message":[],"_revision_number":2},{"id":"eea0ec118d213f2c66529abb00c7fc9f1e154471","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-04-20 08:51:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ce53daa07d1972e618ccc51ebbd7ec37b8e2b9c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 10:23:01.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b1058df48464b879159226ee5c500ce : SUCCESS in 16m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a45089790093499cbbb477d0379d50a8 : SUCCESS in 6m 12s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/0d3b88473aba4ce1a5060a9d936359d1 : SUCCESS in 1h 11m 13s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/9f02d3bc53724acabfbd997b30efe6c2 : SUCCESS in 1h 08m 53s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/22fe2c882f614ee3ac1862f1947d8b38 : SUCCESS in 1h 25m 35s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52314e27f0b945e4a9f0b40f473a11de : SUCCESS in 1h 14m 45s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/1eedfb091fdf45d08282838090a39566 : FAILURE in 12m 49s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8beab09179e943a5a268a626cc40b6a7 : FAILURE in 12m 35s\n- openstack-ansible-lxc-dir-centos-8-stream https://zuul.opendev.org/t/openstack/build/aaa50b4478a2473b8050495a46d638b6 : SUCCESS in 22m 20s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af7d8a53691843de952e913928bccfbb : SUCCESS in 20m 39s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08e9de26360e45a2915654c816f99c73 : SUCCESS in 18m 25s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bea5814610434912a090f223c2e7181f : SUCCESS in 20m 40s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f4290afb585c4ef1aec3a8f0cb27915a : SUCCESS in 17m 33s","accounts_in_message":[],"_revision_number":2},{"id":"2909fcb156e5bab75b1053f13892c90b394f18a2","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-23 11:28:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b2d469c8a1090254b729cba61fa69fc094361c78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-23 13:17:05.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34f41a924b194fc1b9c3fd70044e1b86 : SUCCESS in 15m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e083eab2b7d4684a987b31e681c30ce : SUCCESS in 5m 09s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/c7d6b41684ee48bdadbf3a8722fcee03 : SUCCESS in 1h 20m 50s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/fe6f3c5ab8f24fa483a3e34295372e6f : SUCCESS in 1h 08m 02s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/e30225008090405fbe198ad7cb28a0db : SUCCESS in 1h 23m 57s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0727820ac47b430184b0ae6ede737276 : FAILURE in 1h 06m 18s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/d7ddc367439d4627a1555b68ee636d16 : SUCCESS in 1h 40m 38s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b23162e8fe704fa2804e2a1b6681f51f : SUCCESS in 1h 34m 06s\n- openstack-ansible-lxc-dir-centos-8-stream https://zuul.opendev.org/t/openstack/build/ef3baf6d41fb4fcb9d22fa1eeb7216a4 : SUCCESS in 16m 38s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/192a01b1a8cf409b85b7089c6986834a : SUCCESS in 21m 10s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/515ce1dc3f034899a308980e27b7467c : SUCCESS in 20m 00s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6371702d8e58479d8d2023281809962b : SUCCESS in 17m 23s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70218285d38e49db9dfa59f83e2cd7d6 : SUCCESS in 21m 53s","accounts_in_message":[],"_revision_number":2},{"id":"bb43cf58b2a6ba745b2ee75cc1c3d5aeca61d530","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-27 16:01:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c01c78947856bdb5813a9300529acd6f1dbd3409","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-04-29 11:58:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9ff333ec9e660d430e3e562fadcafedca1bbb9f7","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-04-29 11:59:33.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f493b531cd90a509725f313d203397a430f16a80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 13:54:15.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d099fb6ca484900b71e8ffeec580be0 : SUCCESS in 12m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b05bc9473ef49b4acdd720c72204f83 : SUCCESS in 5m 41s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/9733fdfc90504bf98f4b08be09d46768 : SUCCESS in 1h 35m 34s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/b9a09aea2e3f4a4993b30c6767332480 : SUCCESS in 1h 02m 55s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/984f10cf15f740a3a2e14ea4128573ac : SUCCESS in 1h 28m 35s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6de6e95e1c00424bb2418fe6eccd2723 : FAILURE in 19m 31s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/aad344b89a40443f8fed63b7e2e23ab2 : SUCCESS in 1h 51m 22s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6f3983928f243d0a4a0db974110c314 : SUCCESS in 1h 35m 45s\n- openstack-ansible-lxc-dir-centos-8-stream https://zuul.opendev.org/t/openstack/build/94d4cd3005424d37b19561baf55e855f : SUCCESS in 19m 53s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0e4a2f19c9048a4a86ffbe4b671bfb4 : FAILURE in 15m 34s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/010247158f8546c197b044a3fd13b9fe : FAILURE in 20m 53s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f9e208fd5174b5a980c213cc3dffd74 : FAILURE in 16m 17s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98d0154c07614899a185e7f591f312ae : FAILURE in 7m 27s","accounts_in_message":[],"_revision_number":2},{"id":"02b963c8cd593229bf203cab23e1a60c623d41de","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-04-29 17:06:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b7e9b7dd8fa9e879c503d7f741970369729835d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 19:15:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/796ffd4648524c2e9fd7cbbe267ac385 : SUCCESS in 15m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99d132cd7e1640f286c52d22c214586d : SUCCESS in 6m 07s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/eb9e4f4b7afa41719daa2159ff2e28b2 : SUCCESS in 1h 22m 13s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/8f1207ba574f48a8902b7f39674ca0b9 : SUCCESS in 1h 15m 55s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/cbae38f1e2994f77ba9ed1048f605518 : SUCCESS in 1h 21m 33s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb70f98c8daa4aa89c35e4b4c87a3094 : SUCCESS in 1h 19m 06s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/829e278a69b74fdcb73b9508e08a1dca : SUCCESS in 1h 59m 24s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c16c6e56694d4947af343fd86b5865db : SUCCESS in 1h 33m 23s\n- openstack-ansible-lxc-dir-centos-8-stream https://zuul.opendev.org/t/openstack/build/7bb11978baec4a66ab1f1ed0de4210a9 : SUCCESS in 21m 10s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2313bbb12a44366a7a09ba845a00c4b : SUCCESS in 19m 40s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2bb593c41974592ae78d0b286a0002d : SUCCESS in 16m 17s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2c7a5cc36f74c14a6725976e2713700 : SUCCESS in 15m 14s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff32a3df0da8482496b67db00e75d495 : SUCCESS in 18m 54s","accounts_in_message":[],"_revision_number":2},{"id":"ec9c74027e2f0805ce13b68c26d88b416c439b91","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 19:18:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"025160c755b0b20554fb28dd41d7f7067094c6bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 20:56:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42bbd03fe9334e70bef667e357853213 : SUCCESS in 14m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3292feb6ebb2477db0d642f0455f3638 : SUCCESS in 4m 49s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/f549feca152a4fdb8a5e71c401a9f37a : SUCCESS in 1h 04m 23s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/c6a64fe4bdd44622979d60de45e9e590 : SUCCESS in 1h 12m 09s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/2986c508a8e64a80a2a579e1e35694be : SUCCESS in 1h 21m 41s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0133e343928a4341a9035da302c82dc0 : SUCCESS in 58m 19s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1c5a3f36e8124a81856ada94536a2506 : SUCCESS in 1h 37m 38s","accounts_in_message":[],"_revision_number":2},{"id":"86321b183e28d1e32dd7204088f274396b24da12","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 20:56:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ebf4dde99148d2abaf19447b4b6099be55c1c94d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 20:58:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb52e2dde6604393905d35cea02f500c : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"337ddf87802c7597b7c51a071eb9829a4b87defa","revisions":{"c11d4867c0acf56c314ff143626cdcba54265015":{"kind":"REWORK","_number":1,"created":"2022-03-17 15:24:57.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/83/834183/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/83/834183/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/83/834183/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/83/834183/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/83/834183/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/83/834183/1"}}},"commit":{"parents":[{"commit":"ed38f3393d15fd6028e1fad76438af0dbcb6c1d3","subject":"Change location of ipaddr filter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ed38f3393d15fd6028e1fad76438af0dbcb6c1d3"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-03-17 15:24:44.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-03-17 15:24:44.000000000","tz":60},"subject":"Replace systemd-mount template with role","message":"Replace systemd-mount template with role\n\nTo reduce role complexity we replace separatelly maintained template\nwith systemd_mount role that is widely used across OSA.\n\nChange-Id: I23632f9c145be334b1d19067352f8b82114a1209\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c11d4867c0acf56c314ff143626cdcba54265015"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c11d4867c0acf56c314ff143626cdcba54265015"}]},"branch":"refs/heads/master"},"337ddf87802c7597b7c51a071eb9829a4b87defa":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-04-07 11:40:09.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/83/834183/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/83/834183/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/83/834183/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/83/834183/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/83/834183/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/83/834183/2"}}},"commit":{"parents":[{"commit":"ed38f3393d15fd6028e1fad76438af0dbcb6c1d3","subject":"Change location of ipaddr filter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ed38f3393d15fd6028e1fad76438af0dbcb6c1d3"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-03-17 15:24:44.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-04-07 11:40:09.000000000","tz":0},"subject":"Replace systemd-mount template with role","message":"Replace systemd-mount template with role\n\nTo reduce role complexity we replace separatelly maintained template\nwith systemd_mount role that is widely used across OSA.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/836945\nChange-Id: I23632f9c145be334b1d19067352f8b82114a1209\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/337ddf87802c7597b7c51a071eb9829a4b87defa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/337ddf87802c7597b7c51a071eb9829a4b87defa"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
