)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~837865","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I6a6401debad4937eb9f6a5be31c8cee42d7035cd","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"osa-gluster","attention_set":{},"removed_from_attention_set":{"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2022-04-25 15:17:26.000000000","reason":"Jonathan Rosser replied on the change"}},"hashtags":[],"change_id":"I6a6401debad4937eb9f6a5be31c8cee42d7035cd","subject":"Ensure systemd-udev is present in centos containers","status":"MERGED","created":"2022-04-14 13:00:09.000000000","updated":"2022-04-25 18:45:42.000000000","submitted":"2022-04-25 18:44:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"837865-osa-gluster","meta_rev_id":"161d2c49cc2f177433177647ea2d6b03eb8b4361","_number":837865,"virtual_id_number":837865,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-25 18:44:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2022-04-20 13:35:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":2,"date":"2022-04-19 15:24:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2022-04-20 13:35:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-14 15:17:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-14 20:17:28.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2022-04-19 15:24:46.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2022-04-20 13:35:28.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"7aa37922cd4f32a06646e9c8bc20b33ff46dd1a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-14 13:00:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3bb9b27378555f50028b52bdcf3cbdcd93838fe9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-14 13:36:17.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"94f1f0fa0d858a63fef4bd728e68e81096cb3e3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 15:17:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/431940d9172c4fe3abe880a431d5c72b : SUCCESS in 12m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/128df85d4a4540c898d0f9090adb3a09 : SUCCESS in 5m 08s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/bc7080685f554eb9bc010bd192cbeb08 : SUCCESS in 1h 21m 16s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/fcb2203c585247efac06fc9df0791bf0 : SUCCESS in 1h 05m 12s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/b5b0f08aafd7494caaaaa0dee68cd33a : SUCCESS in 1h 13m 22s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/583f6292d69242dc8496d11bef3de773 : SUCCESS in 1h 04m 25s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/f5b745671b5d4fa2afa24db92936c0e3 : SUCCESS in 1h 33m 58s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd06095205af4308bec3e3d9d34a65e6 : SUCCESS in 1h 25m 38s\n- openstack-ansible-lxc-dir-centos-8-stream https://zuul.opendev.org/t/openstack/build/9f30cbbef2304072b4ed3bdf21db6575 : SUCCESS in 19m 45s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d53f08e17afa423f91bd1c10c23ae3b9 : SUCCESS in 17m 19s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a4c634d7d8141148f5c4473725ba147 : SUCCESS in 17m 14s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/82baafb32c99462780c175fefbbbd430 : SUCCESS in 16m 18s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4555614b10604738afa41bdf4d31e32c : SUCCESS in 17m 23s","accounts_in_message":[],"_revision_number":2},{"id":"e6dbeef91f2e619d15cc11ae3cf7a7d956ff409e","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2022-04-14 20:17:28.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"040fb4a0afd09b59f2d48c7f654a5e44ea5899a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-19 11:07:20.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"99e94d3044a8d32011de2ab2268def0d2eb040cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 13:03:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01f2d015cc564c408c0e653ee8ee44a5 : SUCCESS in 14m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3a23bd6055545d397cfb012d8b67529 : SUCCESS in 5m 01s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/7682291bc52848b28cbfe3dc5c0c3027 : SUCCESS in 1h 16m 12s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/19674938da364ce78eb9f7bbeceb63b5 : SUCCESS in 1h 04m 18s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/c332e4193dcb40508bc1d1fad18c425e : SUCCESS in 1h 13m 37s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45bd7e1d52714cea9db145dc1abf5335 : SUCCESS in 1h 07m 02s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/835b43b54e1e452a91cca9929fdd2a47 : SUCCESS in 1h 52m 57s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd53638e26324c10b792555da9ab77f4 : SUCCESS in 1h 32m 17s\n- openstack-ansible-lxc-dir-centos-8-stream https://zuul.opendev.org/t/openstack/build/0fbd674bdfd645fb9f2856793e563ebf : SUCCESS in 21m 31s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8211515b29644448f39038be1d1868d : SUCCESS in 21m 07s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5630df6609df4a53b8412a9941a9c001 : SUCCESS in 17m 25s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec5d0f918f194a6d9c4664c1d38dbce1 : SUCCESS in 20m 22s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c0819b89bdcb484f8809df29160be1b3 : SUCCESS in 21m 09s","accounts_in_message":[],"_revision_number":3},{"id":"e36aa4820ba02bcceef32ed945dcb5bab0025ecf","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-04-19 15:24:46.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"b8f5586d4bea542c47df3d6499c82d3d5ecc599c","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-04-20 13:35:28.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b76522a8635e93dd6ac02b068b24a12095907143","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 13:37:43.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b21599ceb25ef318d9a76811d2d3f8de7f69a326","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 15:11:53.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d43f2768915448019aa90fa82b7463ee : SUCCESS in 19m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66550581ed984482beb357212e8f8624 : SUCCESS in 4m 43s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/9744e3740d35414587876ee4afea2a9e : RETRY_LIMIT in 8m 46s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/bb42e2f24bb64413ae988630ee04a2db : SUCCESS in 1h 02m 26s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/911fecfef21e42e18c5122f054822f23 : SUCCESS in 1h 28m 34s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a191570ad5ac443fb8f3a7827825be3f : SUCCESS in 1h 14m 52s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66b2a69d201f4198a2217cb88c35545c : FAILURE in 11m 50s","accounts_in_message":[],"_revision_number":3},{"id":"b1f2adad00b4881fbb0d486bbe38dc280f27e6d8","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-20 21:41:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ce8e89bccf0b8f53d3ee4e99f828f82fd215bf34","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-21 12:57:06.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e5ce26eec587feba7e9c71b2a378a6c7e931ae5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 14:26:24.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1234afd0c8f44b63b9da51249aa1c105 : SUCCESS in 14m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a8eb251e8324d9b8acfe77948d1928d : SUCCESS in 5m 43s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/cd5dec77f3d64e0bb0b583c3e806ffa8 : SUCCESS in 1h 14m 55s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/18e76cda976a45d494816009bc3b34f7 : SUCCESS in 1h 05m 34s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/57539b7b6e6a4dd791c7d638bf055f15 : SUCCESS in 1h 19m 30s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/609224f82ca343fe81bb13c98be16e51 : SUCCESS in 1h 02m 47s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/1ac3ee362fb44ab1834a91a6ca4b431e : FAILURE in 12m 25s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1fe9417e11664663aef862d5e33099f4 : FAILURE in 8m 56s\n- openstack-ansible-lxc-dir-centos-8-stream https://zuul.opendev.org/t/openstack/build/9896cc45a9f24994919118cedcc42994 : SUCCESS in 19m 17s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fca85bb73a8d40b1aca4d08beb371ac6 : SUCCESS in 15m 46s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8589792393284c38b27c17a9d31bb5d3 : SUCCESS in 18m 11s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e4d0fb483544c38b49e1967f4200531 : SUCCESS in 15m 26s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/808abfc1f4df4f71abb5ec47e5b05ae3 : SUCCESS in 17m 42s","accounts_in_message":[],"_revision_number":3},{"id":"e440c929ada5d2e6282e295c493f4fb56328e626","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-25 15:17:26.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"89be114f1d979c8bcf74acf3c9fd35c5a58aa0f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 17:02:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8f00ecb64344a84abe1c63ab8ff6b84 : SUCCESS in 12m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b530b3d345c4ea389cd6a9a24ce704a : SUCCESS in 5m 09s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/11a61dcd702a4628b593e256b0002ade : SUCCESS in 1h 16m 09s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/b16e552b9cf84713a857e57ef8c6e9d2 : SUCCESS in 1h 16m 53s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/25bc6f337e83417dbca2093d5dc10a23 : SUCCESS in 1h 17m 56s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87876c7d84914ff9897fd558154f6dac : SUCCESS in 1h 09m 04s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/d63d8726cdbb48bc8da1f1565b93927d : SUCCESS in 1h 36m 27s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c0754f8c2b184f78bae8a427fd1cf7f3 : SUCCESS in 1h 25m 41s\n- openstack-ansible-lxc-dir-centos-8-stream https://zuul.opendev.org/t/openstack/build/cb3f822525c44e6babaa06bd0e49030b : SUCCESS in 18m 43s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59797197061f41068f83827d542571a5 : SUCCESS in 16m 38s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e4fa4369606437799eb41280c1bdb3f : SUCCESS in 17m 35s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7844598b9f3c40faab6c8f626a0dc049 : SUCCESS in 16m 23s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1eae0eab7e8347408bc01df4f11926de : SUCCESS in 17m 04s","accounts_in_message":[],"_revision_number":3},{"id":"ffd3bfd0a1829556e95072cf762482f1e43198b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 17:05:52.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ea486110b5b0af8f57c5391399b05f045de36113","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 18:44:07.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a79762cba074c1bae1003191df4f685 : SUCCESS in 13m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c04a14685db24450980c3e81cc5f3e09 : SUCCESS in 4m 44s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/d34f429ec3fd4f568347dd50161adef3 : SUCCESS in 1h 26m 06s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/efc6bc818e7545268ecc690fb5fc94ce : SUCCESS in 1h 23m 33s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/830abc3152bc41a0bd95866ec22966c0 : SUCCESS in 1h 17m 12s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d53d77b85614e2a9d32dcc1daeb9edc : SUCCESS in 1h 32m 40s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a0f388bbbc34dce89fa9dfd6fc33180 : SUCCESS in 1h 28m 46s","accounts_in_message":[],"_revision_number":3},{"id":"3d7b04074bacc42f03ff8b82d466b8ae5aa7bbe3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 18:44:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"161d2c49cc2f177433177647ea2d6b03eb8b4361","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 18:45:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7c7d5de37104fe2855eb214a0fe2f14 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8626a26d9acf92759c7b0e7850e40df4fdf5c8e9","revisions":{"6f019da5ef7af0611bfc3cc6d9284bad99618299":{"kind":"REWORK","_number":1,"created":"2022-04-14 13:00:09.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/65/837865/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/65/837865/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/65/837865/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/65/837865/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/65/837865/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/65/837865/1"}}},"commit":{"parents":[{"commit":"469a0e5fee23a1a8f8435304902e2ff51f7999a7","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/469a0e5fee23a1a8f8435304902e2ff51f7999a7"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-14 12:53:40.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-14 12:59:48.000000000","tz":60},"subject":"Ensure systemd-udev is present in centos containers","message":"Ensure systemd-udev is present in centos containers\n\nThis is needed to ensure that systemd-tmpfiles-setup service\nis present.\n\nChange-Id: I6a6401debad4937eb9f6a5be31c8cee42d7035cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6f019da5ef7af0611bfc3cc6d9284bad99618299"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6f019da5ef7af0611bfc3cc6d9284bad99618299"}]},"branch":"refs/heads/master"},"c873ef7a19943349ce37a9dd7e2542540714cab8":{"kind":"REWORK","_number":2,"created":"2022-04-14 13:36:17.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/65/837865/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/65/837865/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/65/837865/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/65/837865/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/65/837865/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/65/837865/2"}}},"commit":{"parents":[{"commit":"469a0e5fee23a1a8f8435304902e2ff51f7999a7","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/469a0e5fee23a1a8f8435304902e2ff51f7999a7"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-14 12:53:40.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-14 13:36:11.000000000","tz":0},"subject":"Ensure systemd-udev is present in centos containers","message":"Ensure systemd-udev is present in centos containers\n\nThis is needed to ensure that systemd-tmpfiles-setup service\nis present.\n\nChange-Id: I6a6401debad4937eb9f6a5be31c8cee42d7035cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c873ef7a19943349ce37a9dd7e2542540714cab8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c873ef7a19943349ce37a9dd7e2542540714cab8"}]},"branch":"refs/heads/master"},"8626a26d9acf92759c7b0e7850e40df4fdf5c8e9":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-04-19 11:07:20.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/65/837865/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/65/837865/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/65/837865/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/65/837865/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/65/837865/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/65/837865/3"}}},"commit":{"parents":[{"commit":"469a0e5fee23a1a8f8435304902e2ff51f7999a7","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/469a0e5fee23a1a8f8435304902e2ff51f7999a7"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-14 12:53:40.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-19 11:07:20.000000000","tz":0},"subject":"Ensure systemd-udev is present in centos containers","message":"Ensure systemd-udev is present in centos containers\n\nThis is needed to ensure that systemd-tmpfiles-setup service\nis present, which is used to create /dev/fuse in centos containers\nin other parts of the osa-gluster patch series.\n\nChange-Id: I6a6401debad4937eb9f6a5be31c8cee42d7035cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/8626a26d9acf92759c7b0e7850e40df4fdf5c8e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/8626a26d9acf92759c7b0e7850e40df4fdf5c8e9"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
