)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~838836","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I2cb97a25237a0495ea8d9001d80d06b134b8e500","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"osa-jammy","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-04-29 17:05:26.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2022-04-29 17:05:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2cb97a25237a0495ea8d9001d80d06b134b8e500","subject":"Unify debian and ubuntu cache prep scripts","status":"MERGED","created":"2022-04-21 07:46:14.000000000","updated":"2022-04-29 17:07:01.000000000","submitted":"2022-04-29 17:05:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"838836-osa-jammy","meta_rev_id":"fd8c6f76a890da1489400510c3da276d3460f65b","_number":838836,"virtual_id_number":838836,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-29 17:05:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2022-04-29 15:25:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2022-04-27 19:46:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2022-04-29 15:25:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-21 09:23:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-21 09:26:53.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2022-04-27 19:46:24.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2022-04-29 15:25:34.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"c59859fe79ac298172cca588ebc7c2530b523c2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-21 07:46:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cd36c5817750bd1a478b6734e83d0cd257b4e2f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 09:23:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f66391db1d2e4e148a32722ec9f72a8d : SUCCESS in 13m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84cded40a98040b8a325110a9e7e0231 : SUCCESS in 5m 04s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/e5da9372b221479f84e6f274c472921a : SUCCESS in 1h 15m 54s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/c5bbc18c4daa4af8a24f4e58748383af : FAILURE in 15m 13s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/cc81e0cc868e436db3aef0806a61f769 : SUCCESS in 1h 28m 39s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f1e4eba7e774d88a6912815bd794931 : SUCCESS in 1h 10m 42s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/545c8ff6e5b948b9ae9b2492aef74a1a : FAILURE in 11m 40s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e2b28b7ad37546ef90e1b9707c81a250 : FAILURE in 9m 41s\n- openstack-ansible-lxc-dir-centos-8-stream https://zuul.opendev.org/t/openstack/build/2cadd87e31d84527b4bdb4e9cb6e5bb8 : SUCCESS in 20m 58s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/943b749f34414f31ab2c87dde4ae1b0e : SUCCESS in 19m 03s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e39618c8a51467e905245f9b05cd180 : SUCCESS in 17m 16s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6958e219623451c803569aa5d00e4ed : SUCCESS in 16m 26s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6163538211924d0384ccf54f4975c252 : SUCCESS in 18m 11s","accounts_in_message":[],"_revision_number":1},{"id":"7c52f6bcad56a8b62ae4837f030da2b67247ec60","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-04-21 09:26:53.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6aea1a55b7551d4360856e8d6198dd68ea9d8115","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-21 12:55:29.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"4e41a4bf82a56609691cb15dfbbdfcbe69ad7c8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-21 12:55:43.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"e83046536712bb0266712cbec80b19507867c1ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 14:45:30.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a280e09d431b4a2bb055e714cc38773f : SUCCESS in 15m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9811d7a9b224645a64ce36f34e1c60c : SUCCESS in 5m 58s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/f57a0fa7167945be862c784c4e1d47bd : SUCCESS in 1h 26m 54s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/944a529e5daf49bd8508dcd532e7dbef : FAILURE in 14m 43s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/1fb66f707d0a456abc3feb266f11b876 : SUCCESS in 1h 22m 55s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a6d9bb009614ffbb1f2ed5bafdb7c1e : SUCCESS in 1h 23m 43s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/e4ebeadca903495896650426ee263110 : SUCCESS in 1h 44m 43s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a162512514e4f4ca2b3d09acd23b90a : SUCCESS in 1h 35m 04s\n- openstack-ansible-lxc-dir-centos-8-stream https://zuul.opendev.org/t/openstack/build/946ff4bc5b8047d4bd8203bbce2af429 : SUCCESS in 19m 59s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7dfb776f2d344ce99504fc2bfd02025 : SUCCESS in 17m 42s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3ea8f5ac45c4e158dcc28438154914b : SUCCESS in 17m 36s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f4eeff1dcdf4eba848f0e05fce51d73 : SUCCESS in 16m 31s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d4393cbf70a4e9d8eb4dd6b04273418 : SUCCESS in 16m 58s","accounts_in_message":[],"_revision_number":3},{"id":"8c66095588334b8acfaa7ecf85a221e3c6af092d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-21 14:47:42.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"90a24741a2987c684c65316cd304ac2f9ea02a5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-21 16:37:02.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"857fa86272aeedc290865c1bd6df8be4702e3f1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-21 17:27:22.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"896e5b589d2f3f51b967d9dc6c0096a2a62e7c49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-21 17:50:23.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"9135e08825a41dd5f019756ffa909b382c220266","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-21 18:13:16.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"da733ced8596ca4998155d29cbb85fcad23183bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 21:33:43.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/807188e713894daca5d598e832ab994d : SUCCESS in 14m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/801d917845f941368d160c3b85d27d2c : SUCCESS in 5m 10s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/dc74d12b8d2140068365f85f49016b66 : SUCCESS in 1h 21m 42s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/52e0b217c21b4f498d8a9e3203e1fc43 : FAILURE in 15m 30s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/9ec66ceada2b4e66b1c21d0bba4f7db0 : SUCCESS in 1h 15m 39s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/621b086fb15b495da231a1c1d1276914 : SUCCESS in 1h 06m 07s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/64919cee8c7649d29b82d2e9bdfbe41e : SUCCESS in 1h 43m 56s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db4ba6355d9f46ac905667eb7f4a9430 : SUCCESS in 1h 23m 28s\n- openstack-ansible-lxc-dir-centos-8-stream https://zuul.opendev.org/t/openstack/build/964bf28012a94c18885385500694eeb7 : SUCCESS in 16m 19s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa80f17871e747eaa6cde24ca0a10c6a : SUCCESS in 16m 08s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e3198fed6594a3ba6038501ef190d60 : SUCCESS in 16m 39s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1de92c4b7f64d89b6c1179595496357 : SUCCESS in 15m 38s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/750ba16aad344ae0b1c322cbdf2982d9 : SUCCESS in 17m 17s","accounts_in_message":[],"_revision_number":8},{"id":"1615a64416f3fab2522f7d8390e2fb628ef865ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-25 10:46:34.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"ea829596a9744acff118442278a6cc5de52d10bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 12:22:28.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9cd36d985ad8435d83c69ebbdf3a1020 : SUCCESS in 12m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3871fdf332e940d5b0ec2bc6acc19a26 : SUCCESS in 5m 39s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/abbe4380425b47ac801f2700417ca74e : SUCCESS in 1h 23m 46s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/7d81601e169b4af99a7a21c4df11d6d3 : SUCCESS in 1h 00m 59s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/1f59078d08f94bae8d5c6a4ddd787de7 : SUCCESS in 1h 25m 45s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c585092dbef44605a57ad99777016d68 : SUCCESS in 1h 16m 03s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/191f49d83b5945b0aa062c826d46ade8 : SUCCESS in 1h 32m 21s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3327a3783e641b3b69d787b10dad597 : SUCCESS in 1h 29m 06s\n- openstack-ansible-lxc-dir-centos-8-stream https://zuul.opendev.org/t/openstack/build/2af34e76cdd64dbca3b6c05b9734d354 : SUCCESS in 18m 44s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9f91768c8f94b49a5356491be4aa6be : SUCCESS in 16m 16s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f5285b599654194b378691d061dfc74 : SUCCESS in 17m 38s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/787c4f4d0cd04748869a965dee4a03f3 : SUCCESS in 16m 47s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/721bc0e44ef14331b828d7a44fcb5c5b : SUCCESS in 16m 21s","accounts_in_message":[],"_revision_number":9},{"id":"ac34a40859e165d5e26d6be1745af03a644f9ceb","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-04-27 08:28:06.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"27fb01d788f97c3ea9619d1fb7c95a7ac0c7ddff","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-27 08:34:06.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"849fb40f721353e8ea0fe32c82b66019e5f5d0c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-27 08:35:00.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"eaff6a2236e0f8011ea250742b9b9e90e142426a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-27 08:35:12.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"2dc2e8f843250963b153870dcd433002f5fc89b3","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-27 08:37:20.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"fa7d2068a6924c58598ab079c9c63f09e3c39c55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 10:16:58.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4bfd53ed7de44daa9d0b9da6a7651456 : SUCCESS in 14m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72d1c8e0a41c41fca93a08ecd9b7e626 : SUCCESS in 5m 12s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/4a4fb7ce6a794fb0a9cf881f13d6bf49 : SUCCESS in 1h 22m 19s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/83ba465dffad4de2a24f2c963c4cc35a : SUCCESS in 1h 05m 41s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/68b63fce7ce54800a051010482652bac : SUCCESS in 1h 16m 45s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56c318ee95d04cb98cfc007392c57582 : SUCCESS in 1h 06m 11s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/a88e71fe01414f07bfd7f71a9a586bc6 : SUCCESS in 1h 34m 27s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/305620a668f3465b8c68e4a4b4712bc0 : SUCCESS in 1h 25m 57s\n- openstack-ansible-lxc-dir-centos-8-stream https://zuul.opendev.org/t/openstack/build/14dcc83cd143490fbf36b1e6c90fa053 : SUCCESS in 20m 31s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a524c7049f44b5f8c59758b2d67a65c : SUCCESS in 17m 32s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84dd56a11a1a421c8a8e1457cdface7b : SUCCESS in 16m 45s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a1de4751c1d4987bdf9156bbcccb2a6 : SUCCESS in 17m 30s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8da8818438c5487faad7a0b39bc63937 : SUCCESS in 16m 35s","accounts_in_message":[],"_revision_number":10},{"id":"f83fced6b997a2d5488316f0e36dee0d5234f7ea","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-04-27 19:46:24.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"459a8784a39848450aaeed01c6ef3c0df5993393","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-04-29 14:54:28.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"1767165bbeb28b5700a16509cf08ff038f0d825d","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-04-29 15:25:34.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"9c12b3aef6951977a696ac965d02f5d939ecbe1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 15:29:30.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"d6723d8275d50cdc2532aa62343ce666bd277664","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 16:46:15.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cde7679d76c048d2a758227f4a028c11 : SUCCESS in 16m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0266a07d370c47bfb9923d3ce5e5075d : SUCCESS in 5m 53s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/1ee99d1b89db4ace980afb366d247f84 : SUCCESS in 1h 19m 57s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/5e696c7145fe4ef4af41a4823d2b2b99 : SUCCESS in 1h 13m 52s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/e2ff287b4f36470487b4e319f93e61a7 : SUCCESS in 1h 19m 44s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46c4cd376c0a474ba14f4150750b8fb5 : SUCCESS in 1h 14m 37s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/3a9d5e8739444e94ab097d4c5e576a5a : SUCCESS in 1h 36m 43s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8fae7d4637f24f7eb6c3af81021a7d77 : SUCCESS in 1h 40m 58s\n- openstack-ansible-lxc-dir-centos-8-stream https://zuul.opendev.org/t/openstack/build/d53c47bc167849799b858f0eb8e22e36 : SUCCESS in 22m 11s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fdf5d6a5a1f943168eebbe388b106b56 : SUCCESS in 17m 34s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/816cc36f20df4c3087358c9711ee9800 : SUCCESS in 17m 32s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9715c3a2d674a9292be526bea0da410 : SUCCESS in 18m 37s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/678b3d19fef74e87a492ff1e9a3f780f : SUCCESS in 19m 19s","accounts_in_message":[],"_revision_number":10},{"id":"8d7e23376c170cdcd1c66821460975d5b225aba8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 17:05:26.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/517d36fdf3cc468d88c9b6bfcd5b51fa : SUCCESS in 15m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f243fc6a354943b38e0f1a140eb8e7d9 : SUCCESS in 5m 40s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/57daae6d61e6452ebd8a1aa0c22c2685 : SUCCESS in 1h 25m 03s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/aa4c43d0cf7f4c3a993fd155d7ca0e4f : SUCCESS in 1h 05m 11s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/5a08cc40a7954351b1e391fccd0bb287 : SUCCESS in 1h 24m 57s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe991c5a34864153a52aaf1e59167618 : SUCCESS in 1h 10m 29s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46f08af515e548fa8cc1f5eb113a35e5 : SUCCESS in 1h 34m 58s","accounts_in_message":[],"_revision_number":10},{"id":"56966b4c59124e5bce19334de4277382fc953ac8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 17:05:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"fd8c6f76a890da1489400510c3da276d3460f65b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 17:07:01.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cabe16318361478ab651e43415389fec : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"d72ed7e469a908dfe7feb36c4f2d2e8a73759f28","revisions":{"2e588719f9addfb979ef30e3a047ff48da08984f":{"kind":"REWORK","_number":1,"created":"2022-04-21 07:46:14.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/36/838836/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/36/838836/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/36/838836/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/36/838836/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/36/838836/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/36/838836/1"}}},"commit":{"parents":[{"commit":"469a0e5fee23a1a8f8435304902e2ff51f7999a7","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/469a0e5fee23a1a8f8435304902e2ff51f7999a7"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-21 07:33:55.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-21 07:45:35.000000000","tz":60},"subject":"Unify debian and ubuntu cache prep scripts","message":"Unify debian and ubuntu cache prep scripts\n\nThese should be able to be common, with a dictionary added to select\nthe right libpython version to install based on the OS release.\n\nChange-Id: I2cb97a25237a0495ea8d9001d80d06b134b8e500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2e588719f9addfb979ef30e3a047ff48da08984f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2e588719f9addfb979ef30e3a047ff48da08984f"}]},"branch":"refs/heads/master"},"8347e23738d42a3d9b09ef1a39f06d97cef7b60d":{"kind":"REWORK","_number":2,"created":"2022-04-21 12:55:29.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/36/838836/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/36/838836/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/36/838836/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/36/838836/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/36/838836/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/36/838836/2"}}},"commit":{"parents":[{"commit":"469a0e5fee23a1a8f8435304902e2ff51f7999a7","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/469a0e5fee23a1a8f8435304902e2ff51f7999a7"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-21 07:33:55.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-21 12:55:29.000000000","tz":0},"subject":"Unify debian and ubuntu cache prep scripts","message":"Unify debian and ubuntu cache prep scripts\n\nThese should be able to be common, with a dictionary added to select\nthe right libpython version to install based on the OS release.\n\nChange-Id: I2cb97a25237a0495ea8d9001d80d06b134b8e500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/8347e23738d42a3d9b09ef1a39f06d97cef7b60d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/8347e23738d42a3d9b09ef1a39f06d97cef7b60d"}]},"branch":"refs/heads/master"},"f254f2e4c90f5ef8b14c3ebc26484218d57dce31":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-04-21 12:55:43.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/36/838836/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/36/838836/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/36/838836/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/36/838836/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/36/838836/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/36/838836/3"}}},"commit":{"parents":[{"commit":"469a0e5fee23a1a8f8435304902e2ff51f7999a7","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/469a0e5fee23a1a8f8435304902e2ff51f7999a7"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-21 07:33:55.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-21 12:55:43.000000000","tz":0},"subject":"Unify debian and ubuntu cache prep scripts","message":"Unify debian and ubuntu cache prep scripts\n\nThese should be able to be common, with a dictionary added to select\nthe right libpython version to install based on the OS release.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/838762\nChange-Id: I2cb97a25237a0495ea8d9001d80d06b134b8e500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f254f2e4c90f5ef8b14c3ebc26484218d57dce31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f254f2e4c90f5ef8b14c3ebc26484218d57dce31"}]},"branch":"refs/heads/master","description":"Edit commit message"},"d987dc32da38385f0ae8e8130235ae95ff6f5b47":{"kind":"REWORK","_number":4,"created":"2022-04-21 14:47:42.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/36/838836/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/36/838836/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/36/838836/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/36/838836/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/36/838836/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/36/838836/4"}}},"commit":{"parents":[{"commit":"469a0e5fee23a1a8f8435304902e2ff51f7999a7","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/469a0e5fee23a1a8f8435304902e2ff51f7999a7"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-21 07:33:55.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-21 14:47:03.000000000","tz":0},"subject":"Unify debian and ubuntu cache prep scripts","message":"Unify debian and ubuntu cache prep scripts\n\nThese should be able to be common, with a dictionary added to select\nthe right libpython version to install based on the OS release.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/838762\nChange-Id: I2cb97a25237a0495ea8d9001d80d06b134b8e500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d987dc32da38385f0ae8e8130235ae95ff6f5b47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d987dc32da38385f0ae8e8130235ae95ff6f5b47"}]},"branch":"refs/heads/master"},"99e3e2f2276e55b811b9f0e33433fd402b090fd3":{"kind":"REWORK","_number":5,"created":"2022-04-21 16:37:02.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/36/838836/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/36/838836/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/36/838836/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/36/838836/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/36/838836/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/36/838836/5"}}},"commit":{"parents":[{"commit":"469a0e5fee23a1a8f8435304902e2ff51f7999a7","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/469a0e5fee23a1a8f8435304902e2ff51f7999a7"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-21 07:33:55.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-21 16:37:02.000000000","tz":0},"subject":"Unify debian and ubuntu cache prep scripts","message":"Unify debian and ubuntu cache prep scripts\n\nThese should be able to be common, with a dictionary added to select\nthe right libpython version to install based on the OS release.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/838762\nChange-Id: I2cb97a25237a0495ea8d9001d80d06b134b8e500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/99e3e2f2276e55b811b9f0e33433fd402b090fd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/99e3e2f2276e55b811b9f0e33433fd402b090fd3"}]},"branch":"refs/heads/master"},"9761f6ee01db18fdec9cbcf3beb191a235fd6829":{"kind":"REWORK","_number":6,"created":"2022-04-21 17:27:22.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/36/838836/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/36/838836/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/36/838836/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/36/838836/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/36/838836/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/36/838836/6"}}},"commit":{"parents":[{"commit":"469a0e5fee23a1a8f8435304902e2ff51f7999a7","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/469a0e5fee23a1a8f8435304902e2ff51f7999a7"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-21 07:33:55.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-21 17:27:22.000000000","tz":0},"subject":"Unify debian and ubuntu cache prep scripts","message":"Unify debian and ubuntu cache prep scripts\n\nThese should be able to be common, with a dictionary added to select\nthe right libpython version to install based on the OS release.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/838762\nChange-Id: I2cb97a25237a0495ea8d9001d80d06b134b8e500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/9761f6ee01db18fdec9cbcf3beb191a235fd6829"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/9761f6ee01db18fdec9cbcf3beb191a235fd6829"}]},"branch":"refs/heads/master"},"f1019543b5cfa238f5fb44defab20a927973ae10":{"kind":"REWORK","_number":7,"created":"2022-04-21 17:50:23.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/36/838836/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/36/838836/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/36/838836/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/36/838836/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/36/838836/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/36/838836/7"}}},"commit":{"parents":[{"commit":"469a0e5fee23a1a8f8435304902e2ff51f7999a7","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/469a0e5fee23a1a8f8435304902e2ff51f7999a7"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-21 07:33:55.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-21 17:50:23.000000000","tz":0},"subject":"Unify debian and ubuntu cache prep scripts","message":"Unify debian and ubuntu cache prep scripts\n\nThese should be able to be common, with a dictionary added to select\nthe right libpython version to install based on the OS release.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/838762\nChange-Id: I2cb97a25237a0495ea8d9001d80d06b134b8e500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f1019543b5cfa238f5fb44defab20a927973ae10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f1019543b5cfa238f5fb44defab20a927973ae10"}]},"branch":"refs/heads/master"},"5c055e14dbb44b8a73d476d431a4d325e10225d8":{"kind":"REWORK","_number":8,"created":"2022-04-21 18:13:16.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/36/838836/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/36/838836/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/36/838836/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/36/838836/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/36/838836/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/36/838836/8"}}},"commit":{"parents":[{"commit":"469a0e5fee23a1a8f8435304902e2ff51f7999a7","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/469a0e5fee23a1a8f8435304902e2ff51f7999a7"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-21 07:33:55.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-21 18:13:15.000000000","tz":0},"subject":"Unify debian and ubuntu cache prep scripts","message":"Unify debian and ubuntu cache prep scripts\n\nThese should be able to be common, with a dictionary added to select\nthe right libpython version to install based on the OS release.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/838762\nChange-Id: I2cb97a25237a0495ea8d9001d80d06b134b8e500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/5c055e14dbb44b8a73d476d431a4d325e10225d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/5c055e14dbb44b8a73d476d431a4d325e10225d8"}]},"branch":"refs/heads/master"},"d6a2795821ccdfc13ac7d4d311ab07e65a1d60cc":{"kind":"REWORK","_number":9,"created":"2022-04-25 10:46:34.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/36/838836/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/36/838836/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/36/838836/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/36/838836/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/36/838836/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/36/838836/9"}}},"commit":{"parents":[{"commit":"469a0e5fee23a1a8f8435304902e2ff51f7999a7","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/469a0e5fee23a1a8f8435304902e2ff51f7999a7"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-21 07:33:55.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-25 10:46:34.000000000","tz":0},"subject":"Unify debian and ubuntu cache prep scripts","message":"Unify debian and ubuntu cache prep scripts\n\nThese should be able to be common, with a dictionary added to select\nthe right libpython version to install based on the OS release.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-lxc_container_create/+/839167\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/838762\nChange-Id: I2cb97a25237a0495ea8d9001d80d06b134b8e500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d6a2795821ccdfc13ac7d4d311ab07e65a1d60cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d6a2795821ccdfc13ac7d4d311ab07e65a1d60cc"}]},"branch":"refs/heads/master"},"d72ed7e469a908dfe7feb36c4f2d2e8a73759f28":{"kind":"REWORK","_number":10,"created":"2022-04-27 08:35:00.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/36/838836/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/36/838836/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/36/838836/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/36/838836/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/36/838836/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/36/838836/10"}}},"commit":{"parents":[{"commit":"469a0e5fee23a1a8f8435304902e2ff51f7999a7","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/469a0e5fee23a1a8f8435304902e2ff51f7999a7"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-21 07:33:55.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-27 08:35:00.000000000","tz":0},"subject":"Unify debian and ubuntu cache prep scripts","message":"Unify debian and ubuntu cache prep scripts\n\nThese should be able to be common, with a dictionary added to select\nthe right libpython version to install based on the OS release.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-lxc_container_create/+/839167\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/838762\nChange-Id: I2cb97a25237a0495ea8d9001d80d06b134b8e500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d72ed7e469a908dfe7feb36c4f2d2e8a73759f28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d72ed7e469a908dfe7feb36c4f2d2e8a73759f28"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
