)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~843029","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~Ic62a8fd7290318ed7c76c22620ee0b6f09075a85","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"osa/lxc-ci","attention_set":{},"removed_from_attention_set":{"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2022-05-30 10:28:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic62a8fd7290318ed7c76c22620ee0b6f09075a85","subject":"Use integrated repo \u0027hosts\u0027 jobs to test different backing stores","status":"MERGED","created":"2022-05-23 21:23:57.000000000","updated":"2022-05-30 10:31:43.000000000","submitted":"2022-05-30 10:28:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"843029-osa/lxc-ci","meta_rev_id":"583617530ba3d18123ce925aaf89bab6f0844ae0","_number":843029,"virtual_id_number":843029,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-30 10:28:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2022-05-30 08:48:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2022-05-27 21:42:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2022-05-30 08:48:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}],"CC":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-23 21:34:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-25 08:21:11.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2022-05-27 15:07:37.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"CC"},{"updated":"2022-05-30 08:48:10.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"1baa8fb1ad94637067a6dcaaf5258dafc727d1b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-23 21:23:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7c49bf83107f987eaa7abf3320a397c5471ec06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 21:34:04.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"b6bb0eb50cc877ddf7219a9b82faac826fe804b3","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-23 21:35:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"167f1ed917aaa6beb9dfd6e9299603110ce75edb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 23:24:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b28c3663d13a4a36ad04c902c6236853 : SUCCESS in 16m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a39ed12a77a0425cb3b43074f2b34177 : SUCCESS in 5m 17s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/a6740ecadb964126b817ffe910540267 : SUCCESS in 1h 28m 13s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/7f724634be7841ff83719241bf82b093 : SUCCESS in 1h 16m 40s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/a4134db8e4224c0e88e99de75174d468 : FAILURE in 34m 02s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f491daf40015430ea51e1579cd995b26 : SUCCESS in 1h 09m 18s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/2a2ab4e035fb494bbb790acd7ff2f21a : FAILURE in 1h 41m 55s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12ab54a8a5ed44698be443bd4b2fb5cf : FAILURE in 1h 12m 10s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/015c740013364286ba2628f3c7dcf4ed : FAILURE in 22m 52s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96ea0e31e7a44c288e0e80e206b0a614 : FAILURE in 23m 05s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fbc8433ddeca49c1a6daba32306c653e : SUCCESS in 32m 29s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/3d0bb08244194c34a0d1c25df85666a9 : FAILURE in 23m 59s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/2466af060c724ec68483fbe4e0d29c37 : SUCCESS in 30m 10s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-8 https://zuul.opendev.org/t/openstack/build/94f4f9160562432aa018a8ee6794679c : SUCCESS in 32m 25s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-8-stream https://zuul.opendev.org/t/openstack/build/a9d9be5d65bd4c88a1aa6579158245a9 : SUCCESS in 32m 16s","accounts_in_message":[],"_revision_number":1},{"id":"d564d0f55dad6c8e01e31429125442182d432cba","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-24 08:33:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cf9b768622809ad6bdd43bea1999554ef82191c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-24 10:17:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/679d792854df46248abe53bdfab4b963 : SUCCESS in 19m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8b6b5557c4041f3a0362ab303ec4eac : SUCCESS in 6m 23s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/d831a2497ba848c4b60b98c233e14a26 : SUCCESS in 1h 16m 30s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/555a90c8a25b484997d0ca23c37e9a6e : SUCCESS in 1h 06m 03s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/f44324c55f1a46afa48042754500dbeb : SUCCESS in 1h 31m 30s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2d5d4ab196a41bd84b2cab1f4c08474 : SUCCESS in 1h 10m 44s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/82d0610b7522486ba5ebdf2667876180 : FAILURE in 1h 38m 22s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7bc29051aa984393813256ecb135fabd : SUCCESS in 1h 30m 51s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce62dc2097cd46629d9416b55cdc0e03 : FAILURE in 29m 52s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5456939e37e146beba9b96badc43f077 : FAILURE in 29m 24s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9bc4919c66d1400ca9b8477b39fdecc2 : SUCCESS in 34m 33s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/526ef26bb69e4abcb2c94e217be901bf : FAILURE in 28m 44s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/080ada1288484341a26a5bc79e230fe0 : SUCCESS in 28m 37s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-8 https://zuul.opendev.org/t/openstack/build/0891ceb56aa54ae3930f905c4d2dc47d : SUCCESS in 32m 10s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-8-stream https://zuul.opendev.org/t/openstack/build/7ad93a7900c445408023fe0f787ff539 : SUCCESS in 37m 46s","accounts_in_message":[],"_revision_number":1},{"id":"d3ce22333881c150a683d3e89b400a331dc391fb","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-24 13:38:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"391cc76b535ed44aba091c2beb709212fa4af7da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-24 15:29:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6eb3d15a2e24439a89e92230cd6ceda4 : SUCCESS in 19m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/466e4a62cc1a4968a0e697bc7452ed5a : SUCCESS in 6m 36s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/e6134a515c0f4511942f94eb725781a6 : SUCCESS in 1h 30m 30s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/95d9ea776da2494e92ac978e8685ed99 : SUCCESS in 1h 23m 22s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/287d3379102d415fac40e0a1e4a2b035 : SUCCESS in 1h 27m 11s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84ea2b46b096448d88a05846da7a79b5 : SUCCESS in 1h 16m 57s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/062cfb8571d34963ac0b2e46fd8c0319 : FAILURE in 1h 45m 05s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31bf3479b1f04eb699ef9098173022b1 : FAILURE in 1h 20m 54s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9f80b9b13d14178a640f9767bd1b35f : SUCCESS in 36m 00s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6112f7b61334b318aeed3fc20754993 : RETRY_LIMIT in 14m 51s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65b5843da02c4c16ae273b4d2a98a31f : SUCCESS in 32m 43s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/30acda259a734d7cb73ef82dfe0e5ef8 : RETRY_LIMIT in 19m 02s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/adab8914589e4b6fa681e600888f9851 : SUCCESS in 32m 58s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-8 https://zuul.opendev.org/t/openstack/build/2246d7d33c9643d5ae1986e090eb5602 : RETRY_LIMIT in 16m 15s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-8-stream https://zuul.opendev.org/t/openstack/build/5d999c6efbe64d019f0c4c4487bc91b5 : SUCCESS in 41m 08s","accounts_in_message":[],"_revision_number":1},{"id":"4b57a9402d959b0a797c314e6761110f7e19bbec","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-24 17:17:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3968309b806d0a09ff515e26fc820764a2c87f31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-24 19:01:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ef0d60ee9054059a35217e773a4f217 : SUCCESS in 14m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8a8b11bdff343e9959ae5de76d5fe00 : SUCCESS in 5m 15s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/33c4ab9310fa40aea98f4c1452e00528 : SUCCESS in 1h 06m 42s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/297590b645904bbaada6bf3b81a70df3 : SUCCESS in 1h 01m 08s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/efab90bbf69b425581aeb63ef86075d2 : SUCCESS in 1h 31m 01s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3db8d2c6fb6f42df82d8b4a8e1958667 : SUCCESS in 59m 41s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/37ac9185f9e14fcfa5286d5536640618 : FAILURE in 1h 37m 38s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/225a733a37ee493798592d186b77f6e0 : FAILURE in 1h 20m 20s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7cd12706294b4dba8a4bef6b73fec72a : SUCCESS in 28m 06s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05b9989d23b5463e8e7fbc8d19ed3663 : SUCCESS in 30m 32s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f06a17af0004ef6943e889f34843384 : SUCCESS in 33m 29s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/b34c2da051c64a84aea9beaf429f30a8 : SUCCESS in 29m 36s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/575b17c247514887afb7e39411d526f2 : SUCCESS in 23m 40s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-8 https://zuul.opendev.org/t/openstack/build/da17da6174c64fde831775b133f2000a : SUCCESS in 34m 51s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-8-stream https://zuul.opendev.org/t/openstack/build/030fecd533814d1b8d3280f194ae7bd1 : SUCCESS in 35m 20s","accounts_in_message":[],"_revision_number":1},{"id":"ad8dd87c2aa1cafb32274736753727ec10152d92","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-24 19:10:32.000000000","message":"Topic set to osa/lxc-ci","accounts_in_message":[],"_revision_number":1},{"id":"dfa96d6d4ed5962e44c0ca474d56750d4cb6c250","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-05-25 08:21:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"572df29427367972785a776db274ac4044c6b016","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-25 08:32:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"712d9ce56b3ba2e6eff1fce9ed5465adcf1fdb28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 10:09:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/775610f3efa94ab58892a3e9759551b8 : SUCCESS in 16m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4da807a678e04b82bfa6ee2933baa244 : SUCCESS in 7m 38s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/42b0854e0f984e26b2839988986c7850 : SUCCESS in 1h 30m 15s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/74724b228beb4781982c86c72a21c6b3 : SUCCESS in 1h 21m 44s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/ba63c80f13d441f9b50ec197d8575517 : SUCCESS in 1h 27m 02s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f38b7c2e13eb4ffea055356d145ae432 : SUCCESS in 1h 11m 59s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/5610b42c9d774367b3e989feba11f663 : FAILURE in 1h 28m 56s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d70cc831328545e0aaac21ce12fe04bf : FAILURE in 1h 11m 43s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eadd889bda79466cb3d7a3b837462dcd : SUCCESS in 35m 54s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9afdf0c7b3744ff59f0754ef2946f3f4 : SUCCESS in 33m 09s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd6978e4dac845d583492dd38a061547 : SUCCESS in 26m 37s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/e55d274c177d40f9b3e6c884c4dc159f : SUCCESS in 29m 02s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/ea03b0af65cb4eeda03f0e5162430c6a : SUCCESS in 25m 51s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-8 https://zuul.opendev.org/t/openstack/build/d48dcf6dee6d43418cc80fd32ac2fd5f : RETRY_LIMIT in 15m 43s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-8-stream https://zuul.opendev.org/t/openstack/build/4f56253ae0e147378319c8f62f9f0a6c : SUCCESS in 34m 36s","accounts_in_message":[],"_revision_number":1},{"id":"8892a2674cc07ef96c4e16b1a7e9a173e55c5f7b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-25 13:05:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d8cd3310c4735492df600f314df4bddff3a8127b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 14:51:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/48e2a4ee1c544d789c4892026cc0ca3a : SUCCESS in 15m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dba73918e5d74559bf6ccc6c4e6a277b : SUCCESS in 6m 02s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/8092c83b88df4674959ff4fb994d6227 : SUCCESS in 1h 27m 32s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/21a4017e5aee4585a5e49f50f4ac9445 : SUCCESS in 1h 13m 08s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/5880e3228e3a4d7e8e030f3375309f93 : SUCCESS in 1h 21m 49s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9253db23f0f34d2cbeaa041612e148fd : SUCCESS in 1h 17m 57s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/686b694c06d5423cb913e55472681c6a : FAILURE in 1h 40m 27s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b04ff93e9f64fd29545e4dfe884fcb0 : FAILURE in 1h 23m 53s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67c681266c1c4c7b9f8eef06105c6f60 : SUCCESS in 34m 52s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/50b074a2d4e44f5fb2a580486c94df6f : SUCCESS in 31m 22s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1efb738d85654cd0a7f3f8bb2fbf7cc1 : SUCCESS in 36m 13s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/3497550606f842a6a3ce408d2ec48ab7 : SUCCESS in 31m 12s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/e27c2544acc24b25b25a2301bb2ec15d : SUCCESS in 31m 30s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-8 https://zuul.opendev.org/t/openstack/build/067d4036cc5849cdb83e059b73185e1f : SUCCESS in 34m 19s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-8-stream https://zuul.opendev.org/t/openstack/build/614a8d684c804135915c658239ed76b4 : SUCCESS in 40m 50s","accounts_in_message":[],"_revision_number":1},{"id":"405cbdabecf73375ea51850976dfb032dcb63d68","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-25 18:41:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1466a388218857f431a361507d24f1992e95d41e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 20:16:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef3b219f3317425ea1b4ca45d854c4c8 : SUCCESS in 16m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fabb40201ed47e6a8055afecf06e77b : SUCCESS in 6m 09s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/ef2e81d8c65b445c933d1adc118e0f99 : SUCCESS in 1h 23m 21s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/73879aa50ca3447facc336888c8705d0 : SUCCESS in 1h 14m 57s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/26aeee93add7447ba77c20cbb7d4c4ae : FAILURE in 1h 06m 22s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22af4acc7d964b648e4953d3cd6c18b3 : SUCCESS in 1h 21m 07s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/31f586064fe546faa874a91dea7a0014 : FAILURE in 1h 21m 11s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fbc967a912514006b37300508b1f1149 : SUCCESS in 1h 30m 47s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd6d1c75783d470d8509f6c578e2af4d : SUCCESS in 35m 50s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cdf17ff4575f46cfa6130585381dc459 : SUCCESS in 34m 57s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f4e48b8a27024a06b37496f66108671d : SUCCESS in 35m 56s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/8898da4c028d491a8018f5d8053208c8 : SUCCESS in 36m 09s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/a8602ce412cc4548a8fe80a68094f188 : SUCCESS in 28m 37s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-8 https://zuul.opendev.org/t/openstack/build/42bff9c6886547a4b6e4cfefff43956a : SUCCESS in 31m 20s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-8-stream https://zuul.opendev.org/t/openstack/build/03c4b6d953d542c1b758fc4102a13fc6 : SUCCESS in 35m 59s","accounts_in_message":[],"_revision_number":1},{"id":"a1f4fd7757d907c969888f60473d8c149bbbe7ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-26 09:01:50.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"39c261c158ba41471dd44f6cfa8cf0e616453a8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-26 10:33:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f28f3fd2df24783866a3d386e8a9351 : SUCCESS in 14m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76b53703e4fb4290867013f2af6cc5d7 : SUCCESS in 5m 18s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/98b6b4572e3f4f97b8d3a10d5ae57a02 : SUCCESS in 1h 07m 33s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/5cfd5c47f3d64d8780d52bcad28f8614 : FAILURE in 56m 01s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/3cbdc659b948467c995347319712042a : SUCCESS in 1h 24m 09s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7af6a8f6098d4dd7a920fd81e32a8272 : SUCCESS in 57m 56s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/20eef4613cae470ea441f973ade51c0b : FAILURE in 1h 12m 05s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d60f00aa6a4748a4b900c803792ebd5f : SUCCESS in 1h 25m 59s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0233df5c64d9410185316c44a54db381 : SUCCESS in 25m 10s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3445240796cd43a69c51203854a59188 : SUCCESS in 29m 58s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c477f89a6109440b817ed266487312b1 : SUCCESS in 28m 37s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/e450de6b104e48ddbe8e1074810509da : SUCCESS in 25m 45s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/866c56658dd0404fa7d519ef7b7dcfba : SUCCESS in 31m 11s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-8 https://zuul.opendev.org/t/openstack/build/77453b31b2dd4120be4f197d3999e724 : SUCCESS in 28m 44s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-8-stream https://zuul.opendev.org/t/openstack/build/c80bbb1b59a146709615214ad244dd59 : SUCCESS in 31m 47s","accounts_in_message":[],"_revision_number":2},{"id":"035f71ec30c7beaf985866996dd1ad21b5ce87be","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-26 15:58:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1e5dba8202510370d57bb9ff54aaa2b9e7147448","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-26 17:42:14.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7dcb5da4d3df42db9f1aac0711073e68 : SUCCESS in 16m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/732dd22ea811498b9c0b944428f89cf8 : SUCCESS in 6m 13s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/a317431cd0b84f53a7d90d9cf43145d8 : SUCCESS in 1h 24m 03s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/2fa9fb7b4fb9466caec10082cc118b0e : SUCCESS in 1h 16m 39s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/27394d052a6942e191c8ea7dff5b8aff : SUCCESS in 1h 26m 47s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8704e4d1cbc492d8abbf028b16c5737 : SUCCESS in 1h 17m 56s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/4d912e3cd76547a886d0e446635deaa1 : FAILURE in 1h 20m 52s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/37de5aba74f0458a8564cd22998d8f11 : FAILURE in 1h 25m 45s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed02980cf94c4b5a9c46be6d03da368b : SUCCESS in 37m 17s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4b1fc497b8c4ae5a466f7ab63d408c7 : SUCCESS in 39m 51s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92eb19de383f435f8ba975a34d91cf3f : SUCCESS in 35m 15s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/8524c22867974ab6b9cf14b4a0545950 : SUCCESS in 38m 30s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/7f5f61feb7db4913aaf6fc775cec7407 : SUCCESS in 41m 22s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-8 https://zuul.opendev.org/t/openstack/build/37f3d6661ac44b19a17263fa63a30d0e : SUCCESS in 37m 57s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-8-stream https://zuul.opendev.org/t/openstack/build/a50a15fa5fad45f688c9b784b37bf74d : SUCCESS in 37m 35s","accounts_in_message":[],"_revision_number":2},{"id":"08f6a0ea32ecc0dc4dc7e06bc8a2a86633eb8328","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-26 21:29:46.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"a43094a332725a70ba87878e67d5c8971b65d111","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-26 23:01:04.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad8ffd40b152474e8cb66e8d90670827 : SUCCESS in 15m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19fbae4776c14a0c96a5b61672644851 : SUCCESS in 5m 34s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/4d127cf61b074e668d1ea19734a143bd : SUCCESS in 1h 13m 41s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/1773341633c949f0ae4b3a20bb7ab26b : SUCCESS in 1h 01m 57s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/8d146e1c4a444ae4adddc70e107584f9 : SUCCESS in 1h 26m 28s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6934bc5775f49258ccd9eb1d7e9b00d : SUCCESS in 1h 12m 17s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/edb0836d100c40e8a446f6afee232707 : FAILURE in 1h 14m 14s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fdf72789100f47c2b463426b02578b36 : FAILURE in 1h 05m 35s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6fb92d008e8d4ac0b1fb7905211c5498 : SUCCESS in 29m 29s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9905809b69b040509c437e2ba4d49c4c : SUCCESS in 25m 16s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c0a8d15e463e48b1a8b3243f56168684 : SUCCESS in 25m 32s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/6ba5209e4d1b4d489d9781a6c9413581 : SUCCESS in 28m 43s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/828fc14d61c84f29ad52b9e188c3a008 : SUCCESS in 30m 09s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-8 https://zuul.opendev.org/t/openstack/build/5c6e05880beb47e2a74d390c9c653388 : SUCCESS in 31m 18s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-8-stream https://zuul.opendev.org/t/openstack/build/d8939b73e51e4f20b4252daf5d340656 : SUCCESS in 30m 49s","accounts_in_message":[],"_revision_number":3},{"id":"536651e9d8d1786f3e7fd195b89af98de327b703","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-05-27 15:07:37.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9dad4b56d63f97b9b9277158ac17b752c7d1a041","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 16:52:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c0c4f1eab3445b495f3b2044452e94f : SUCCESS in 14m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f1ce92c6aa94a5fa7fc36c7dee7532b : SUCCESS in 5m 35s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/ef0a22ec27944471b509772f634045cb : SUCCESS in 1h 17m 32s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/1582c20f1729408da9ede6dfeb276aa2 : SUCCESS in 1h 10m 18s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/ba8cdb828bff4b0db5a0fe3de086aa40 : SUCCESS in 1h 35m 55s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93fc04bd9697409ab6a419790c68ca71 : SUCCESS in 1h 07m 19s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/1e87fdcd9eba491384a8a637f02feb98 : FAILURE in 1h 26m 26s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d8604f82f0d4506bfae202763f46f26 : SUCCESS in 1h 30m 09s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c6b2075058e43d1813afaa2165ee6e4 : SUCCESS in 33m 11s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/287b43fcd9d748c884c756d7d8ef4ccf : SUCCESS in 30m 38s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95b31f78121e4bb0b17908a15d763492 : SUCCESS in 27m 56s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/c0aff068250b45b78098bf99ff69d788 : SUCCESS in 28m 30s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/5725a0fb02db4e3c8fb3ea201cdda890 : SUCCESS in 33m 05s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-8 https://zuul.opendev.org/t/openstack/build/e100013bbfb14068a7b22a4f5a043c14 : SUCCESS in 33m 12s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-8-stream https://zuul.opendev.org/t/openstack/build/ce2cba1ff8cf4a8e9f2cb63195f50f2c : SUCCESS in 32m 28s","accounts_in_message":[],"_revision_number":3},{"id":"e481b3e5b895d008b23d87d6e87b213365e72804","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-05-27 21:42:32.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"047cd1d5fe3ada358bc90b35582001354fe3b579","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-05-30 08:48:10.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"1f6dc68718e57b4bd5d778af85abfe47f94a0cb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 08:51:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b3bb44c98ad7bdc956e28f43c6b1cdc3dc2a66f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 10:28:42.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de8dc4cf95e0498fbe8f26b19415709c : SUCCESS in 12m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6667f71a20cb4c9fa7d0d92087cfe49d : SUCCESS in 5m 30s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/d9d25c92a5be4a3abe4998be598c277e : SUCCESS in 1h 28m 03s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/8368665f0d964061870e7c4dba687535 : SUCCESS in 1h 14m 04s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/b2f76802365f47a39fce6bb633341dd7 : SUCCESS in 1h 31m 08s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b2d4d641d6149ba86ed1ac0676e147a : SUCCESS in 1h 18m 42s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a7b734d52894e39ba518744acad2a0c : SUCCESS in 1h 35m 44s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a21fa2450c544bf28b7bdbb9f32c496b : SUCCESS in 36m 12s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1ba8d8f961d4053a7d0d9a74869a3ee : SUCCESS in 35m 04s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95c4e1d8f7b148c2b47d675d594ca404 : SUCCESS in 25m 24s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/44c37d23851d4a2bbf3b2e595f49189c : SUCCESS in 29m 35s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/d500426d773243918973faa2323bb402 : SUCCESS in 32m 27s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-8 https://zuul.opendev.org/t/openstack/build/9ba2c12cfb3141bca9cc79776d7b2f3c : SUCCESS in 33m 50s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-8-stream https://zuul.opendev.org/t/openstack/build/6d85631af70542b58daf3be5c8fe88f5 : SUCCESS in 33m 35s","accounts_in_message":[],"_revision_number":3},{"id":"ceaa9253b8347f25f3eb4135cbf31545a0eb6d54","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 10:28:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"583617530ba3d18123ce925aaf89bab6f0844ae0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 10:31:43.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ffa0c7790e949e3a0cfb4ea34cea5ce : SUCCESS in 56s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"667321df9c72366c53fa97592d5f132cdf6d642a","revisions":{"56d07c229698212cb9521b4769dc5835c89a18e4":{"kind":"REWORK","_number":1,"created":"2022-05-23 21:23:57.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/29/843029/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/29/843029/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/843029/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/843029/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/843029/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/843029/1"}}},"commit":{"parents":[{"commit":"3928a0b9146532b3d37149953228c10309a3e41e","subject":"Clean up NFV OpenvSwitch repo for container build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3928a0b9146532b3d37149953228c10309a3e41e"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-23 21:21:04.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-23 21:22:12.000000000","tz":60},"subject":"Use integrated repo \u0027hosts\u0027 jobs to test different backing stores","message":"Use integrated repo \u0027hosts\u0027 jobs to test different backing stores\n\nThis uses the main openstack-ansible repo for testing these different\nbacking store scenarios rather than openstack-ansible-tests. This is\nmore maintainable and ensures that we test using the same code as is\nused for a real deployment.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/843027\nChange-Id: Ic62a8fd7290318ed7c76c22620ee0b6f09075a85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/56d07c229698212cb9521b4769dc5835c89a18e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/56d07c229698212cb9521b4769dc5835c89a18e4"}]},"branch":"refs/heads/master"},"991da893e206a421176ae613a5b0439fc4049436":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-05-26 09:01:50.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/29/843029/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/29/843029/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/843029/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/843029/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/843029/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/843029/2"}}},"commit":{"parents":[{"commit":"3928a0b9146532b3d37149953228c10309a3e41e","subject":"Clean up NFV OpenvSwitch repo for container build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3928a0b9146532b3d37149953228c10309a3e41e"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-23 21:21:04.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-26 09:01:50.000000000","tz":0},"subject":"Use integrated repo \u0027hosts\u0027 jobs to test different backing stores","message":"Use integrated repo \u0027hosts\u0027 jobs to test different backing stores\n\nThis uses the main openstack-ansible repo for testing these different\nbacking store scenarios rather than openstack-ansible-tests. This is\nmore maintainable and ensures that we test using the same code as is\nused for a real deployment.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/843027\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-repo_server/+/843418\nChange-Id: Ic62a8fd7290318ed7c76c22620ee0b6f09075a85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/991da893e206a421176ae613a5b0439fc4049436"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/991da893e206a421176ae613a5b0439fc4049436"}]},"branch":"refs/heads/master","description":"Edit commit message"},"667321df9c72366c53fa97592d5f132cdf6d642a":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-05-26 21:29:46.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/29/843029/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/29/843029/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/843029/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/843029/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/843029/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/843029/3"}}},"commit":{"parents":[{"commit":"3928a0b9146532b3d37149953228c10309a3e41e","subject":"Clean up NFV OpenvSwitch repo for container build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3928a0b9146532b3d37149953228c10309a3e41e"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-23 21:21:04.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-26 21:29:46.000000000","tz":0},"subject":"Use integrated repo \u0027hosts\u0027 jobs to test different backing stores","message":"Use integrated repo \u0027hosts\u0027 jobs to test different backing stores\n\nThis uses the main openstack-ansible repo for testing these different\nbacking store scenarios rather than openstack-ansible-tests. This is\nmore maintainable and ensures that we test using the same code as is\nused for a real deployment.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/843027\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-repo_server/+/843418\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-galera_server/+/843547\nChange-Id: Ic62a8fd7290318ed7c76c22620ee0b6f09075a85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/667321df9c72366c53fa97592d5f132cdf6d642a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/667321df9c72366c53fa97592d5f132cdf6d642a"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
