)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~843672","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I60f1b04c2378c91905440e8de223c29dd6da4d17","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"843672","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-06-06 00:52:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I60f1b04c2378c91905440e8de223c29dd6da4d17","subject":"Do not install COPR repo for CentOS LXC","status":"MERGED","created":"2022-05-27 18:19:44.000000000","updated":"2025-06-06 00:53:47.000000000","submitted":"2025-06-06 00:52:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"843672-843672","meta_rev_id":"b03cabee3e7b311c1a7707057f8676c8f31c92e2","_number":843672,"virtual_id_number":843672,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-06 00:52:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":2,"date":"2025-06-04 14:45:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-03 15:17:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"date":"2025-06-05 10:09:19.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":1,"date":"2025-06-05 14:16:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-27 21:57:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-31 10:38:31.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2025-06-03 15:17:15.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2025-06-04 14:45:57.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"}],"messages":[{"id":"af458b5a3589d21ecb1fcb4d53017674e7d20571","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-05-27 18:19:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ffd0e44879f7ad18f04db8b5aecb303ff8e9383","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-05-27 18:29:13.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"587f31ae375d64c2b0b2f133c1c89e742c6c4baa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-05-27 18:30:05.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"d2bc859edaf7362ea8ee10aec2af3e589ece2972","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-05-27 18:58:02.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"8c79f6a3238235f34674b1f7c8385f91456e69e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-05-27 20:12:38.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"c7cc8853491ac3459798bfba262526756a1cdb69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 21:57:40.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf65fd56f2b8408ea5b3f6e5e379dfe0 : SUCCESS in 11m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e67982f73d56471c81c29c233214271e : SUCCESS in 4m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b413060ce01d4bbfa804dc6567fe5d99 : SUCCESS in 2m 57s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/a9adcd83c771478ab3a70528138b3c7d : SUCCESS in 1h 16m 39s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/3c5b4da1e4cf446bad5c9d8bd215aa62 : SUCCESS in 1h 04m 53s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/76dd03b4d6604b4fb57f08baafda955a : SUCCESS in 1h 34m 34s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1cc31494caa04ab5a4c8a5b2d84e0866 : SUCCESS in 1h 03m 14s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/2021d1ff6c654acb9e73f657675968f2 : FAILURE in 1h 21m 19s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c83acd3086948dc984c037533cecc5c : SUCCESS in 1h 23m 27s\n- openstack-ansible-lxc-dir-centos-8-stream https://zuul.opendev.org/t/openstack/build/6f68097ac0a64cf4b206385163a9787d : FAILURE in 6m 30s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4cf018e3b25547a58faf8abcbb2f1dcf : SUCCESS in 16m 19s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc279aefa8e342f8903e8901972d36ef : SUCCESS in 15m 47s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/125a8a1c1caa4ee593a16d5892f9ad8a : SUCCESS in 14m 58s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9906f3e9c65d40a7a453e60e2ed3bf77 : SUCCESS in 15m 22s","accounts_in_message":[],"_revision_number":5},{"id":"b4e0d1605bc492b806f736e8414ac05b83e2dc71","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-31 10:38:31.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"871881792f47d75ecb1aade46746cb07a81a4871","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 12:18:51.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00401db60efd401d9bd8deb83c0be538 : SUCCESS in 13m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/087953f092224a4d8bc7d99cd7553a52 : SUCCESS in 4m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5e07fd70c134a148e207de89d13c363 : SUCCESS in 3m 19s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/f625b37dda4849efbeaf0ba8852a2d02 : SUCCESS in 1h 22m 01s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/3ba54e498b7247729fac5216b4aac303 : FAILURE in 1h 03m 30s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/486eca2c3e294198b672f8162d498bfe : SUCCESS in 1h 02m 14s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/8cb19966e39645fa95e7cdcff66f758f : SUCCESS in 1h 32m 19s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2a06fa710c744268269600ed3e2e7b6 : SUCCESS in 1h 17m 02s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/33858a88e33f4df0be15a925bcf97b58 : FAILURE in 1h 23m 49s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6812e0eba9ca44a18736ba29c56259d0 : SUCCESS in 1h 28m 42s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f2e21efb6024f83b9ee31832e94e889 : SUCCESS in 29m 25s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/edd1e17140e045cb8aa72933c24a4550 : SUCCESS in 30m 00s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3bb376696494862997528c81e976734 : SUCCESS in 27m 19s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/a9f517179d7f48bcb8546b7efaebdd36 : SUCCESS in 28m 48s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/73bbd7edd3864ed98756f64c63728e92 : SUCCESS in 25m 07s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-8 https://zuul.opendev.org/t/openstack/build/ccc92b382dc64c11b04cbe3291a52415 : SUCCESS in 32m 39s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-8-stream https://zuul.opendev.org/t/openstack/build/881cacac3e34467197c019dc0d229c4f : SUCCESS in 29m 54s","accounts_in_message":[],"_revision_number":5},{"id":"9f5c7f36d6c3a710b5d25247938696c82ee9a011","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-06-14 07:37:03.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"39fd1364f41d1e2e9b78af1d40131839758390eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-14 09:16:21.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46527829f5994615a03d24319780290a : SUCCESS in 13m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e2287c590394e0ca13eedd3d36eb895 : SUCCESS in 5m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99849db876984b3db6e192019557f22c : SUCCESS in 3m 32s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/4c8f75216a9949bab5896402d50c673d : SUCCESS in 1h 21m 11s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/b4abef150add41708449f35a55f068e1 : FAILURE in 28m 42s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/3718245ac99945d78504c8e17a6db452 : SUCCESS in 1h 08m 15s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/84aef17f067b48e8bd5e5cecf6148a33 : SUCCESS in 1h 18m 22s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1071077ab8dd4c10ae843af13a27d394 : SUCCESS in 1h 09m 09s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e1fca48aeca4dde9544f562523d84f1 : SUCCESS in 1h 08m 25s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/b7cb32780ef8499597eae4b58aa72dca : FAILURE in 1h 32m 33s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7fb45cf64ac24bcb895332644fabd390 : SUCCESS in 1h 31m 50s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb1bb1ad8c0243e3a087eabda2d0b78b : SUCCESS in 31m 16s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59e801d3fd4142d8a393a2cb881bf8c7 : SUCCESS in 26m 20s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd0715982d8d40189a9809999d6792c7 : SUCCESS in 27m 30s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/e3e1897c379e4ca98b98a9385f00302c : SUCCESS in 25m 55s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/87580d03a7ba43cba339508317f35230 : SUCCESS in 25m 56s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-8 https://zuul.opendev.org/t/openstack/build/b8aacbb813b94c3d97a02277e0a2a670 : SUCCESS in 29m 47s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-8-stream https://zuul.opendev.org/t/openstack/build/8fa2a77f58364a248e403f916f5ae6a9 : SUCCESS in 31m 37s","accounts_in_message":[],"_revision_number":6},{"id":"e85ddf411b1a93c994fb52aa0d34c208171bc71a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-06-17 20:40:44.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"fea549e4f8f9a9d287afc1ebd73970abeddfe9ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 22:39:46.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6ad225cac9c4b62a81b3168be4e0edc : SUCCESS in 14m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac624be030df4000bd1b98b7ccf1f7a7 : SUCCESS in 4m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a16fd2a5c6e49e3b193cfbd05726f4c : SUCCESS in 3m 24s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/01b447a26fe2409481328aad99ea9aff : SUCCESS in 1h 25m 20s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/d10fc453ba68402287e946cffd40e3ce : FAILURE in 25m 07s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/2dad7b48550e44c6b26027bf6ad97050 : SUCCESS in 1h 10m 30s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/a6c61585bf444673b435bd9d74f088ee : SUCCESS in 1h 33m 57s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a1a94f1af8c4b73954ffbcdcbea8618 : SUCCESS in 1h 18m 25s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ca435d1fce64a45982efc2835d1cf55 : SUCCESS in 1h 27m 41s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/e071b5c338864244a4663b1806c26074 : SUCCESS in 1h 52m 36s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2382eeb65ff9409f91724ec994be88f3 : SUCCESS in 1h 51m 33s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa4bc64ebd2741d5bfc7f3b335dfc96a : SUCCESS in 37m 48s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11da1c2310574921bb4ec982e75161f4 : SUCCESS in 26m 16s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32142d1dfd8d461082b817606407ee2c : SUCCESS in 32m 24s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/02101580f0194d98a6719c4cb7ee02f7 : SUCCESS in 27m 24s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/f87e81b934154844bdc1c2d80ba5af8c : SUCCESS in 30m 04s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-8 https://zuul.opendev.org/t/openstack/build/19568c278cd341c7a8d18f54126929e4 : SUCCESS in 29m 34s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-8-stream https://zuul.opendev.org/t/openstack/build/099ce7c3c2e548c8aa41249df4bff283 : SUCCESS in 34m 19s","accounts_in_message":[],"_revision_number":7},{"id":"d3f8ad99df0d41ee7e991e3a25593bca24d3a060","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-06-20 10:34:08.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"182a66de0ed4c763f61aa9ea9484739d82999b82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-06-20 10:35:39.000000000","message":"Patch Set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"8d648c936fc9675c11e34d2e6e1f0166d901c21d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 12:37:58.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93334c62386842c4945aa02c41de4e41 : SUCCESS in 18m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7283d9acdc38424ebd9ccdae12c4023f : SUCCESS in 6m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50386a7f37ec4ed1add1b67298cbf3db : SUCCESS in 4m 23s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/a76b6d4b2df14de3b5db3b863d738206 : SUCCESS in 1h 22m 30s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/55f7518ad5f7463b8e250092397d2a64 : FAILURE in 20m 30s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/f86f678f5a6548ee97a2bcbc37fd1a0a : SUCCESS in 1h 40m 09s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/2ae6619c30424ad0aeef5a2b57b48bac : SUCCESS in 1h 27m 06s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc373f8f37094656a62d80edb7373aa1 : SUCCESS in 1h 07m 37s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/704fe01421ac4b2a88e44cb11a1fef41 : SUCCESS in 1h 18m 54s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/5b8ece226ee243ba84f5c6eccfd63bef : FAILURE in 1h 53m 56s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b260db79594241459dd24f89bf7dd364 : SUCCESS in 1h 30m 48s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/609042bb89274f6da0e1c00f5035b12b : SUCCESS in 30m 39s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67dc3ca1a42b4c7c9ecf5c8d80401af4 : SUCCESS in 32m 02s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5de29d98a2d4f9aa81186082c4e50d8 : SUCCESS in 25m 39s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/3ee63327bcc2414f8514addf5b34e108 : SUCCESS in 27m 34s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/86884725eca6409b9445288cc00d1fae : SUCCESS in 29m 12s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-8 https://zuul.opendev.org/t/openstack/build/84c0633c70254c76a5806face34f14e5 : SUCCESS in 29m 00s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-8-stream https://zuul.opendev.org/t/openstack/build/4d846de366da48638520870310117f21 : SUCCESS in 31m 17s","accounts_in_message":[],"_revision_number":9},{"id":"f54cfd52b50130ee0803a56696810e50aeb9e5a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-07-28 08:43:45.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"1a5d644d3c88b707c510d5fd8f9739b2f0794057","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 10:25:50.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c305aa63c3b4ec88d809cc6ad5f167c\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f1caacda6644ca383ad8ce9b6598811 : SUCCESS in 17m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c44a4591590d4fa695009f7e11e936af : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa6918dcc3e9420c9da1fb039ff0e2fa : SUCCESS in 3m 38s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/4da5c0d308e54c0097f9d8512d414241 : FAILURE in 27m 42s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/7de5bcc86adb40ed9a0fb048093edc42 : SUCCESS in 55m 04s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32baa50385bd4891b7658506376fb73b : SUCCESS in 54m 12s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/828c47c8ac01478e937a43900974a02c : SUCCESS in 1h 17m 53s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/cdf5f7826a8647319794db0a7237bb95 : SUCCESS in 1h 38m 40s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2f6024934c2416a86317606123cd170 : SUCCESS in 1h 36m 42s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2adb5f0093c849aab4eafb607a0b87c5 : RETRY_LIMIT in 17m 16s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/76b45481f332446db298ca6ea128ac23 : SUCCESS in 30m 43s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/590acf29903b4070b051c11d148a8828 : SUCCESS in 24m 18s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/0ecb625c5e6e45a7b879a0216c785681 : SUCCESS in 49m 19s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/088d4a5183e940499ec6ecbb5e7a9da2 : SUCCESS in 26m 01s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-9-stream https://zuul.opendev.org/t/openstack/build/ed9ea7869e50428bb747e9fa698ff0eb : FAILURE in 27m 33s","accounts_in_message":[],"_revision_number":10},{"id":"bbbed1e5a820332859a990e2fb54dac7a23825c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-14 08:10:30.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"f056c12c17c075a18ac2895b667239e6494a7872","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 09:35:56.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a28ebcb196884cb99df416bce4dfca3b\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b346a8b31994de3a0647f4634929b12 : SUCCESS in 13m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24a6deb29ad34c45a7a71b0af96283d8 : SUCCESS in 5m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f1120eeb66e41b3981633f8d4f0caf6 : SUCCESS in 3m 25s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/82cf5c34da9a492e96e1f46fa491eed1 : FAILURE in 27m 06s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0a3adb8ddec24cb192555d8018dc0699 : FAILURE in 22m 00s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/6796a2a24bfd4c56beec62e631ff84f7 : SUCCESS in 1h 06m 00s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/afd814973929413abe8b61867c14f609 : SUCCESS in 1h 15m 50s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fd4a8df4e43d4b50867172b29c967000 : FAILURE in 1h 09m 33s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/32b904438790426f9d0c30a3a95505f6 : FAILURE in 1h 17m 44s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac5d3dc5b407497d80ba75150a8382f8 : FAILURE in 1h 15m 28s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18f1934b62224e78a82efc9865cfcfbb : RETRY_LIMIT in 10m 38s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/652a0cb5a1284145a5d9055f315c86f6 : SUCCESS in 29m 57s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0b539baebfb4f8e92c299a6e428b803 : SUCCESS in 32m 49s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/4677f9e6d4ae44ee974e1b626d52869b : SUCCESS in 32m 23s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/2aa47e49e42949d98eeb50be960d8d67 : SUCCESS in 33m 24s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-9-stream https://zuul.opendev.org/t/openstack/build/9287b182471648eba4fdcc266c7142ce : FAILURE in 24m 59s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d7f7380e117849b881d29cbfc6da6a6e : FAILURE in 24m 36s","accounts_in_message":[],"_revision_number":11},{"id":"c5e17137c3a205161d85c837f24bd80ff96d4d9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-28 15:45:59.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"99e0d362b1a49ef3e49ad07ce832787d38ab56ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-28 15:46:35.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"252679338df8731a42143d3a006d92b00b01bc0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-28 15:50:04.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"013dd561f1a6992cbc63bb82a84ec437e43e3a28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-28 16:43:45.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"6158144f46bb5f6b72472d5c4b91c7a34fd47f2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-28 17:17:40.000000000","message":"Patch Set 16: Patch Set 15 was rebased","accounts_in_message":[],"_revision_number":16},{"id":"1c9f1bd95c3d71482b8dc1df87ae00f7865a75c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-28 18:23:06.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"579a41fa3db388105b2b69286449d8c9179c6338","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-28 18:23:46.000000000","message":"Patch Set 18: Patch Set 17 was rebased","accounts_in_message":[],"_revision_number":18},{"id":"b160e1f07baac155c2b4a8e4aa4f1816921eb6ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 19:54:36.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca15d1b4d5344162abcff67be45b852f\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a65b4945f1c3456cb7bb936b4a5b200c : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e2380a473864b83adab6689e9ff99d3 : SUCCESS in 5m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a6e1f43854347c3b9addfaf8284eb1b : SUCCESS in 2m 32s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/73cf4740b01e454a84f26642c82f8b10 : SUCCESS in 1h 13m 41s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b0a4ec51726c488ca0fe0b66eb5808dc : SUCCESS in 1h 17m 56s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8228a736f3c0421295fbbc4e9681945f : SUCCESS in 1h 21m 43s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8761abbc87af49489972a38098f9c413 : SUCCESS in 1h 22m 28s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/810586aa8135437aa0c7b437d067f8a3 : SUCCESS in 1h 23m 48s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a3cf586affa540d0889a6f342005b2b7 : SUCCESS in 19m 20s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/481b1c4112864843bcb6e355eea4ac54 : SUCCESS in 20m 26s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/82a8bb3e174445989e75aa65503a5535 : SUCCESS in 19m 38s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/79c0035f50a84e3f95886efc212ec487 : SUCCESS in 25m 47s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/d2d506e03261470e874d806660bca853 : SUCCESS in 20m 01s","accounts_in_message":[],"_revision_number":18},{"id":"29b66b87db8d7e132372f6ac95e3d6d789fdbc11","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-06-03 15:17:15.000000000","message":"Patch Set 18: Code-Review+2","accounts_in_message":[],"_revision_number":18},{"id":"3c306d4aa050e4a033fa7d1b75c0645a79c50f9f","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2025-06-04 14:45:57.000000000","message":"Patch Set 18: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":18},{"id":"d878cb4ff1cda6d798319875d7d83af9ca728e31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 14:48:03.000000000","message":"Patch Set 18: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":18},{"id":"aea6a068be61f97dd7038adc3748af7e5e1296b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 16:28:05.000000000","message":"Patch Set 18: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/951177 is needed.","accounts_in_message":[],"_revision_number":18},{"id":"4ad05cb1a86c83f0f127d51efff3375c94d8711f","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2025-06-05 10:09:19.000000000","message":"Patch Set 18: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"5a1a7678280140b91cd279cd3d6d085ed0f84a62","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-05 14:16:03.000000000","message":"Patch Set 18: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"cf1a8a48d9c5da827d8f5bb6f8784491a5bcbb0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 15:46:03.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3fc17373279549ae846737e048c1f836\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d71d0f9ed7b94b6f87643796f29fb1f9 : SUCCESS in 7m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/180ba8d151ca4d7582e9d3c88debb8e0 : SUCCESS in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45bdd4e543eb4ef5b41344ef5fec2a07 : SUCCESS in 2m 09s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/099c42b9a17b4497af3f12baaf64ab8e : SUCCESS in 47m 28s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9ab2b4c257e4a6186b56d4f51858a3e : SUCCESS in 40m 41s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2330f2e257a94014bd6de3f8359d53e7 : SUCCESS in 1h 19m 09s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6de70e0a1624d2da17f04c6fe3c2d31 : SUCCESS in 1h 22m 46s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/462fdef7dd4642dcb8bebae0f291bb17 : FAILURE in 47m 55s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/90df27417df042e39159a7bfc3ec210f : SUCCESS in 21m 57s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d39d8274e6c7472dbc84e47211bdfb94 : SUCCESS in 24m 53s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa91e9aa538c468795b1935535c178bb : SUCCESS in 21m 05s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/77ab38a81459463fb1a9604d02a5866f : SUCCESS in 16m 53s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/cc5078a8a24845b98fa4a43b72e93bff : SUCCESS in 18m 37s","accounts_in_message":[],"_revision_number":18},{"id":"90d13ea310f5e0caf8607cdc252f1751bb9c7374","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-05 16:04:45.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"d7ecbf9ec30decb65367c94ec22d11268e890922","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 17:32:18.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdcea9e6da1b4c43b9e9230e2291ba69\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de59c00f2dac4794909ac8e947654c2a : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/601aefd2e1a145ea8890b612b34d5669 : SUCCESS in 3m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b99ea5cea9dc46ee959a99eddbfb37cc : SUCCESS in 2m 30s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/79d579f4a60d4558967ac79e1f01f073 : SUCCESS in 49m 28s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/92b162343b4748caa9187c07eebbf469 : SUCCESS in 1h 23m 08s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f11a712658ad45a2a3b47b16a7ebf366 : SUCCESS in 53m 09s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/80fd5d0129f843c88422110745058ae3 : SUCCESS in 1h 24m 15s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/13fc3c3343fd4e3aaa9622deddd3a107 : SUCCESS in 1h 04m 59s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad5b758e27c84083a1dea38937272fd8 : SUCCESS in 26m 39s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/21b3146de5ab44baaa4e319ef6bcadb5 : SUCCESS in 16m 14s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b8d7267863bd48c7b07d52e1fde7c941 : SUCCESS in 26m 09s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/ea05c23ca449451ab8a10dca4ab3dcba : SUCCESS in 26m 15s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/35c59d3b0f064c5f814206372f7c97bd : SUCCESS in 23m 42s","accounts_in_message":[],"_revision_number":18},{"id":"f7e471f437287eb52e4a5718d1e39e764fe53ab7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 17:34:10.000000000","message":"Patch Set 18: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":18},{"id":"b7ed039fabe2b1a624ec811a21de94d769236182","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 19:27:54.000000000","message":"Patch Set 18: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/283c7d5fec9d4a98b970ca5c852a977b\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3efa664a63064facb0ebbba4b507885f : SUCCESS in 8m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd5adbd9ed2a4c2ea4efd78f79118a5e : SUCCESS in 6m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de008887f64e46b5928d3ed8848213b1 : SUCCESS in 3m 28s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/218205164f2a4aecb260c85744d73fe9 : SUCCESS in 1h 19m 24s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c2c6c044669c4aec9105b2a463c88102 : SUCCESS in 1h 22m 58s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af315a8c8f94444aab759d948d976beb : SUCCESS in 1h 23m 24s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/23d77c136f3b4ba6acdf5b2989861ae8 : SUCCESS in 1h 21m 11s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a982fe9f195742d6957faf016dbe3971 : SUCCESS in 1h 50m 48s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b9ae5ed22c240609241f9c05b5a6f72 : SUCCESS in 23m 57s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9edb14d4ea274f979b1d2885fecfa5ed : SUCCESS in 26m 42s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/39fbac3fdead4f5090aa8325225adfd3 : FAILURE in 21m 44s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/9d4ddf84b6884e1a89f3b253943a9ecb : SUCCESS in 21m 32s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/ba22085d1ed54d319ae36f34c1480466 : SUCCESS in 15m 10s","accounts_in_message":[],"_revision_number":18},{"id":"0a9ff735a451a753e7fd648b94218fc6cb962576","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-06-05 21:38:44.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"c605297c182a94f4a7700e36740209f83f2e294a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 23:26:16.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68156b93e2a64f749b1a898489c138cc\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/48245b8d747542778ce2205569b31f8e : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ec55e970e004304a3cd846973b6dcfb : SUCCESS in 5m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/728953b8f5df4473895c3486b21656a6 : SUCCESS in 1m 30s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/fe7274881f514646b5801a107a085571 : SUCCESS in 1h 06m 05s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00aba0b0f95d43caa7547c1874101f72 : SUCCESS in 1h 31m 37s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/426fc11fcbf54ad1ab9a9ae0a9f45743 : SUCCESS in 1h 19m 57s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4e4532a2376c4b1ba04ded12db459a23 : SUCCESS in 1h 20m 10s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a293bd739ef46289ceb7ccef1def4c6 : SUCCESS in 1h 42m 51s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa590c4d28614dfd9c07cc3011800d72 : SUCCESS in 19m 47s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/09c5d65485d549a89d22f3b551be1c7e : SUCCESS in 24m 19s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d10bbf645b1b4e0fb1b64f141c0fcdfa : SUCCESS in 21m 23s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/21ac6b93d8a24a49826961fd8b1b3c2e : SUCCESS in 22m 50s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/7091a38422c8497a8c17ad5800df24b6 : SUCCESS in 20m 38s","accounts_in_message":[],"_revision_number":18},{"id":"86339b9deed9af810998d6229fb84410012f5cf2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 23:28:19.000000000","message":"Patch Set 18: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":18},{"id":"79070cc4ef1e8d4558d877fe4179f1dc865bba20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-06 00:52:45.000000000","message":"Patch Set 18: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1db53ea03d6e407bb3194be25f4a3feb\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/91e2b017e5b04c68a58e13a22877b072 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eefc5ec390114b1687d081434414c24c : SUCCESS in 3m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/501e8752fc7b4535a28044f40a3a37c8 : SUCCESS in 1m 28s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/735affa6a9b04102b26f94e1c93aba12 : SUCCESS in 41m 15s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5eb86ab0ed4744968d833c8a9d531dc5 : SUCCESS in 39m 54s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a4349c75037494cb2d1c6315ef56700 : SUCCESS in 1h 20m 09s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/34cee74b747846fd930f7fe6241505d9 : SUCCESS in 40m 49s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/77524cc32e754500bca5b77266052409 : SUCCESS in 1h 19m 25s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/26f465a382804ea0ac520594ffb8a61b : SUCCESS in 23m 17s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa09f2cd97534e12be7e1c909c7a5de5 : SUCCESS in 14m 17s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad3cd069007848d5ab61b0fd0ac4f641 : SUCCESS in 18m 34s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/c2b4cf0b27724976a94597bf044315c2 : SUCCESS in 22m 03s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/50315f0bd524452691d807641d2ba9ff : SUCCESS in 20m 46s","accounts_in_message":[],"_revision_number":18},{"id":"fc7c38143a5fc23590ca0cdb7bba6ac884f1b8db","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-06 00:52:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":18},{"id":"b03cabee3e7b311c1a7707057f8676c8f31c92e2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-06 00:53:47.000000000","message":"Patch Set 18:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fec6f218c389493fbee4ffb9cdec3c3c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a618cef8db64a149fbe2f80e804946d : SUCCESS in 52s","accounts_in_message":[],"_revision_number":18}],"current_revision_number":18,"current_revision":"c84b23b25851a8fca67a5a6e70d8bc687d5832d5","revisions":{"af17464a5708bd48cdfe2987602ef8d1854b8718":{"kind":"REWORK","_number":1,"created":"2022-05-27 18:19:44.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/72/843672/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/72/843672/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/1"}}},"commit":{"parents":[{"commit":"fd23eeedfc2a28556278abc03f722ae17dd77990","subject":"Add centos-9 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/fd23eeedfc2a28556278abc03f722ae17dd77990"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-05-27 18:19:35.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-05-27 18:19:35.000000000","tz":120},"subject":"Do not install COPR repo for CentOS 8 LXC","message":"Do not install COPR repo for CentOS 8 LXC\n\nEPEL does contain LXC package for CentOS 8 Stream, so we don\u0027t need to\ninstall COPR repo anymore.\n\nWith that we also add control of what packages are going to be installed\nfrom EPEL.\n\nChange-Id: I60f1b04c2378c91905440e8de223c29dd6da4d17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/af17464a5708bd48cdfe2987602ef8d1854b8718"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/af17464a5708bd48cdfe2987602ef8d1854b8718"}]},"branch":"refs/heads/master"},"c07f38ddd4048303e15a38a79aa40371e9277b81":{"kind":"REWORK","_number":2,"created":"2022-05-27 18:29:13.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/72/843672/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/72/843672/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/2"}}},"commit":{"parents":[{"commit":"fd23eeedfc2a28556278abc03f722ae17dd77990","subject":"Add centos-9 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/fd23eeedfc2a28556278abc03f722ae17dd77990"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-05-27 18:19:35.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-05-27 18:29:13.000000000","tz":0},"subject":"Do not install COPR repo for CentOS 8 LXC","message":"Do not install COPR repo for CentOS 8 LXC\n\nEPEL does contain LXC package for CentOS 8 Stream, so we don\u0027t need to\ninstall COPR repo anymore.\n\nWith that we also add control of what packages are going to be installed\nfrom EPEL.\n\nChange-Id: I60f1b04c2378c91905440e8de223c29dd6da4d17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c07f38ddd4048303e15a38a79aa40371e9277b81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c07f38ddd4048303e15a38a79aa40371e9277b81"}]},"branch":"refs/heads/master"},"2117ef8aa1a68201d2db3d2a56eabba793f554bf":{"kind":"REWORK","_number":3,"created":"2022-05-27 18:30:05.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/72/843672/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/72/843672/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/3"}}},"commit":{"parents":[{"commit":"fd23eeedfc2a28556278abc03f722ae17dd77990","subject":"Add centos-9 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/fd23eeedfc2a28556278abc03f722ae17dd77990"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-05-27 18:19:35.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-05-27 18:30:05.000000000","tz":0},"subject":"Do not install COPR repo for CentOS 8 LXC","message":"Do not install COPR repo for CentOS 8 LXC\n\nEPEL does contain LXC package for CentOS 8 Stream, so we don\u0027t need to\ninstall COPR repo anymore.\n\nWith that we also add control of what packages are going to be installed\nfrom EPEL.\n\nChange-Id: I60f1b04c2378c91905440e8de223c29dd6da4d17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2117ef8aa1a68201d2db3d2a56eabba793f554bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2117ef8aa1a68201d2db3d2a56eabba793f554bf"}]},"branch":"refs/heads/master"},"c995544fc64b2826de8512fab65a1b62c3f00854":{"kind":"REWORK","_number":4,"created":"2022-05-27 18:58:02.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/72/843672/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/72/843672/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/4"}}},"commit":{"parents":[{"commit":"fd23eeedfc2a28556278abc03f722ae17dd77990","subject":"Add centos-9 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/fd23eeedfc2a28556278abc03f722ae17dd77990"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-05-27 18:19:35.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-05-27 18:58:02.000000000","tz":0},"subject":"Do not install COPR repo for CentOS 8 LXC","message":"Do not install COPR repo for CentOS 8 LXC\n\nEPEL does contain LXC package for CentOS 8 Stream, so we don\u0027t need to\ninstall COPR repo anymore.\n\nWith that we also add control of what packages are going to be installed\nfrom EPEL.\n\nChange-Id: I60f1b04c2378c91905440e8de223c29dd6da4d17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c995544fc64b2826de8512fab65a1b62c3f00854"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c995544fc64b2826de8512fab65a1b62c3f00854"}]},"branch":"refs/heads/master"},"b66403a0484e312efa1a77d684ee0e0a70b8d8a1":{"kind":"REWORK","_number":5,"created":"2022-05-27 20:12:38.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/72/843672/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/72/843672/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/5"}}},"commit":{"parents":[{"commit":"fd23eeedfc2a28556278abc03f722ae17dd77990","subject":"Add centos-9 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/fd23eeedfc2a28556278abc03f722ae17dd77990"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-05-27 18:19:35.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-05-27 20:12:38.000000000","tz":0},"subject":"Do not install COPR repo for CentOS 8 LXC","message":"Do not install COPR repo for CentOS 8 LXC\n\nEPEL does contain LXC package for CentOS 8 Stream, so we don\u0027t need to\ninstall COPR repo anymore.\n\nWith that we also add control of what packages are going to be installed\nfrom EPEL.\n\nChange-Id: I60f1b04c2378c91905440e8de223c29dd6da4d17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b66403a0484e312efa1a77d684ee0e0a70b8d8a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b66403a0484e312efa1a77d684ee0e0a70b8d8a1"}]},"branch":"refs/heads/master"},"09805f1424b2691dd5ba3c399aa3932a44fd05de":{"kind":"REWORK","_number":6,"created":"2022-06-14 07:37:03.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/72/843672/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/72/843672/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/6"}}},"commit":{"parents":[{"commit":"844ebcdab63125f239733c2ef1023fade5543e68","subject":"Merge \"Switch sphinx language to en\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/844ebcdab63125f239733c2ef1023fade5543e68"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-05-27 18:19:35.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-06-14 07:36:25.000000000","tz":120},"subject":"Do not install COPR repo for CentOS LXC","message":"Do not install COPR repo for CentOS LXC\n\nEPEL does contain LXC package for CentOS 8 and 9 Stream, so we don\u0027t need\nto install COPR repo anymore.\n\nWith that we also add control of what packages are going to be installed\nfrom EPEL.\n\nChange-Id: I60f1b04c2378c91905440e8de223c29dd6da4d17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/09805f1424b2691dd5ba3c399aa3932a44fd05de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/09805f1424b2691dd5ba3c399aa3932a44fd05de"}]},"branch":"refs/heads/master"},"758ac41e3eced6198eac35d6336c43b620a2b46b":{"kind":"REWORK","_number":7,"created":"2022-06-17 20:40:44.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/72/843672/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/72/843672/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/7"}}},"commit":{"parents":[{"commit":"844ebcdab63125f239733c2ef1023fade5543e68","subject":"Merge \"Switch sphinx language to en\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/844ebcdab63125f239733c2ef1023fade5543e68"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-05-27 18:19:35.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-06-17 20:40:29.000000000","tz":120},"subject":"Do not install COPR repo for CentOS LXC","message":"Do not install COPR repo for CentOS LXC\n\nEPEL does contain LXC package for CentOS 8 and 9 Stream, so we don\u0027t need\nto install COPR repo anymore.\n\nWith that we also add control of what packages are going to be installed\nfrom EPEL.\n\nChange-Id: I60f1b04c2378c91905440e8de223c29dd6da4d17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/758ac41e3eced6198eac35d6336c43b620a2b46b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/758ac41e3eced6198eac35d6336c43b620a2b46b"}]},"branch":"refs/heads/master"},"0d03382f9000c2bb7e4aede6bfa12ac63408b1fd":{"kind":"REWORK","_number":8,"created":"2022-06-20 10:34:08.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/72/843672/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/72/843672/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/8"}}},"commit":{"parents":[{"commit":"844ebcdab63125f239733c2ef1023fade5543e68","subject":"Merge \"Switch sphinx language to en\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/844ebcdab63125f239733c2ef1023fade5543e68"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-05-27 18:19:35.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-06-20 10:34:08.000000000","tz":0},"subject":"Do not install COPR repo for CentOS LXC","message":"Do not install COPR repo for CentOS LXC\n\nEPEL does contain LXC package for CentOS 8 and 9 Stream, so we don\u0027t need\nto install COPR repo anymore.\n\nWith that we also add control of what packages are going to be installed\nfrom EPEL.\n\nChange-Id: I60f1b04c2378c91905440e8de223c29dd6da4d17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/0d03382f9000c2bb7e4aede6bfa12ac63408b1fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/0d03382f9000c2bb7e4aede6bfa12ac63408b1fd"}]},"branch":"refs/heads/master"},"9110e8492ecbfa487a12fc0dcc4a4c02d2bf2d92":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2022-06-20 10:35:39.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/72/843672/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/72/843672/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/9"}}},"commit":{"parents":[{"commit":"844ebcdab63125f239733c2ef1023fade5543e68","subject":"Merge \"Switch sphinx language to en\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/844ebcdab63125f239733c2ef1023fade5543e68"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-05-27 18:19:35.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-06-20 10:35:38.000000000","tz":0},"subject":"Do not install COPR repo for CentOS LXC","message":"Do not install COPR repo for CentOS LXC\n\nEPEL does contain LXC package for CentOS 8 and 9 Stream, so we can\ninstall LXC from there. For CentOS 9 COPR still needed for lxc-templates\n\nWith that we also add control of what packages are going to be installed\nfrom EPEL.\n\nChange-Id: I60f1b04c2378c91905440e8de223c29dd6da4d17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/9110e8492ecbfa487a12fc0dcc4a4c02d2bf2d92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/9110e8492ecbfa487a12fc0dcc4a4c02d2bf2d92"}]},"branch":"refs/heads/master","description":"Edit commit message"},"2272f75545c2000e8405d22c1d37f9f5d7f51eb0":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2022-07-28 08:43:45.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/72/843672/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/72/843672/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/10"}}},"commit":{"parents":[{"commit":"783076a508d64918eb474ea5069b1261efdbaf79","subject":"Take account of lxc_apt_mirror in new debootstrap command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/783076a508d64918eb474ea5069b1261efdbaf79"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-05-27 18:19:35.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-07-28 08:43:45.000000000","tz":0},"subject":"Do not install COPR repo for CentOS LXC","message":"Do not install COPR repo for CentOS LXC\n\nEPEL does contain LXC package for CentOS 8 and 9 Stream, so we can\ninstall LXC from there. For CentOS 9 COPR still needed for lxc-templates\n\nWith that we also add control of what packages are going to be installed\nfrom EPEL.\n\nChange-Id: I60f1b04c2378c91905440e8de223c29dd6da4d17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2272f75545c2000e8405d22c1d37f9f5d7f51eb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2272f75545c2000e8405d22c1d37f9f5d7f51eb0"}]},"branch":"refs/heads/master","description":"Rebase"},"936014eef4cc61faea4088262425a528d7b850e6":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2022-10-14 08:10:30.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/72/843672/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/72/843672/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/11"}}},"commit":{"parents":[{"commit":"a09612c1fbc4f2c93047ecea77c51b2ebb326496","subject":"Use policy_rc_d attribute instead of copy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a09612c1fbc4f2c93047ecea77c51b2ebb326496"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-05-27 18:19:35.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-10-14 08:10:30.000000000","tz":0},"subject":"Do not install COPR repo for CentOS LXC","message":"Do not install COPR repo for CentOS LXC\n\nEPEL does contain LXC package for CentOS 8 and 9 Stream, so we can\ninstall LXC from there. For CentOS 9 COPR still needed for lxc-templates\n\nWith that we also add control of what packages are going to be installed\nfrom EPEL.\n\nChange-Id: I60f1b04c2378c91905440e8de223c29dd6da4d17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/936014eef4cc61faea4088262425a528d7b850e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/936014eef4cc61faea4088262425a528d7b850e6"}]},"branch":"refs/heads/master","description":"Rebase"},"bda509c8ff40a9e179751ec72d8cfb8b95ec030a":{"kind":"REWORK","_number":12,"created":"2025-05-28 15:45:59.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/72/843672/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/72/843672/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/12"}}},"commit":{"parents":[{"commit":"f059f97ce2a04f1618b53f68d78c9f37f3c699f1","subject":"Patch the usr.bin.lxc-copy apparmor profile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f059f97ce2a04f1618b53f68d78c9f37f3c699f1"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-05-27 18:19:35.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-05-28 15:45:51.000000000","tz":120},"subject":"Do not install COPR repo for CentOS LXC","message":"Do not install COPR repo for CentOS LXC\n\nEPEL does contain LXC package for CentOS 8 and 9 Stream, so we can\ninstall LXC from there. For CentOS 9 COPR still needed for lxc-templates\n\nWith that we also add control of what packages are going to be installed\nfrom EPEL.\n\nChange-Id: I60f1b04c2378c91905440e8de223c29dd6da4d17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/bda509c8ff40a9e179751ec72d8cfb8b95ec030a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/bda509c8ff40a9e179751ec72d8cfb8b95ec030a"}]},"branch":"refs/heads/master"},"c9ed3b2b8485d27befd5ed064d0c61ea2b16a3b0":{"kind":"REWORK","_number":13,"created":"2025-05-28 15:46:35.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/72/843672/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/72/843672/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/13"}}},"commit":{"parents":[{"commit":"f059f97ce2a04f1618b53f68d78c9f37f3c699f1","subject":"Patch the usr.bin.lxc-copy apparmor profile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f059f97ce2a04f1618b53f68d78c9f37f3c699f1"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-05-27 18:19:35.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-05-28 15:46:35.000000000","tz":0},"subject":"Do not install COPR repo for CentOS LXC","message":"Do not install COPR repo for CentOS LXC\n\nEPEL does contain LXC package for CentOS 8 and 9 Stream, so we can\ninstall LXC from there. For CentOS 9 COPR still needed for lxc-templates\n\nWith that we also add control of what packages are going to be installed\nfrom EPEL.\n\nChange-Id: I60f1b04c2378c91905440e8de223c29dd6da4d17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c9ed3b2b8485d27befd5ed064d0c61ea2b16a3b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c9ed3b2b8485d27befd5ed064d0c61ea2b16a3b0"}]},"branch":"refs/heads/master"},"5ac1cc403f3da60f3055349d47b1e56bc7e79ac3":{"kind":"REWORK","_number":14,"created":"2025-05-28 15:50:04.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/72/843672/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/72/843672/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/14"}}},"commit":{"parents":[{"commit":"f059f97ce2a04f1618b53f68d78c9f37f3c699f1","subject":"Patch the usr.bin.lxc-copy apparmor profile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f059f97ce2a04f1618b53f68d78c9f37f3c699f1"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-05-27 18:19:35.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-05-28 15:49:57.000000000","tz":120},"subject":"Do not install COPR repo for CentOS LXC","message":"Do not install COPR repo for CentOS LXC\n\nEPEL does contain LXC package for CentOS 8 and 9 Stream, so we can\ninstall LXC from there. For CentOS 9 COPR still needed for lxc-templates\n\nWith that we also add control of what packages are going to be installed\nfrom EPEL.\n\nChange-Id: I60f1b04c2378c91905440e8de223c29dd6da4d17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/5ac1cc403f3da60f3055349d47b1e56bc7e79ac3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/5ac1cc403f3da60f3055349d47b1e56bc7e79ac3"}]},"branch":"refs/heads/master"},"7d3cc78f98e0109a94f5f91a6044f15bd508a7cd":{"kind":"REWORK","_number":15,"created":"2025-05-28 16:43:45.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/72/843672/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/72/843672/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/15"}}},"commit":{"parents":[{"commit":"6ba2079b57a387077281ab0127d246bc7d773212","subject":"Do not use distro templates by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6ba2079b57a387077281ab0127d246bc7d773212"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-05-27 18:19:35.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-05-28 16:43:36.000000000","tz":120},"subject":"Do not install COPR repo for CentOS LXC","message":"Do not install COPR repo for CentOS LXC\n\nEPEL does contain LXC package for CentOS 8 and 9 Stream, so we can\ninstall LXC from there. For CentOS 9 COPR still needed for lxc-templates\n\nWith that we also add control of what packages are going to be installed\nfrom EPEL.\n\nChange-Id: I60f1b04c2378c91905440e8de223c29dd6da4d17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7d3cc78f98e0109a94f5f91a6044f15bd508a7cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7d3cc78f98e0109a94f5f91a6044f15bd508a7cd"}]},"branch":"refs/heads/master"},"eeb608b28bb3f71da9d6868d802ceec117504ca6":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2025-05-28 17:17:40.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/72/843672/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/72/843672/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/16"}}},"commit":{"parents":[{"commit":"73c9581299aaf0a4a4b542511ba8f88913b310fa","subject":"Do not use distro templates by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/73c9581299aaf0a4a4b542511ba8f88913b310fa"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-05-27 18:19:35.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-05-28 17:17:40.000000000","tz":0},"subject":"Do not install COPR repo for CentOS LXC","message":"Do not install COPR repo for CentOS LXC\n\nEPEL does contain LXC package for CentOS 8 and 9 Stream, so we can\ninstall LXC from there. For CentOS 9 COPR still needed for lxc-templates\n\nWith that we also add control of what packages are going to be installed\nfrom EPEL.\n\nChange-Id: I60f1b04c2378c91905440e8de223c29dd6da4d17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/eeb608b28bb3f71da9d6868d802ceec117504ca6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/eeb608b28bb3f71da9d6868d802ceec117504ca6"}]},"branch":"refs/heads/master","description":"Rebase"},"d13f1b4b2cc83da8c8c3e0a51ac09d5a0e7780d6":{"kind":"REWORK","_number":17,"created":"2025-05-28 18:23:06.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/72/843672/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/72/843672/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/17"}}},"commit":{"parents":[{"commit":"6ba2079b57a387077281ab0127d246bc7d773212","subject":"Do not use distro templates by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6ba2079b57a387077281ab0127d246bc7d773212"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-05-27 18:19:35.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-05-28 18:17:15.000000000","tz":120},"subject":"Do not install COPR repo for CentOS LXC","message":"Do not install COPR repo for CentOS LXC\n\nEPEL does contain LXC package for CentOS 8 and 9 Stream, so we can\ninstall LXC from there. For CentOS 9 COPR still needed for lxc-templates\n\nWith that we also add control of what packages are going to be installed\nfrom EPEL.\n\nChange-Id: I60f1b04c2378c91905440e8de223c29dd6da4d17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d13f1b4b2cc83da8c8c3e0a51ac09d5a0e7780d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d13f1b4b2cc83da8c8c3e0a51ac09d5a0e7780d6"}]},"branch":"refs/heads/master"},"c84b23b25851a8fca67a5a6e70d8bc687d5832d5":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2025-05-28 18:23:46.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/72/843672/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/72/843672/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/843672/18"}}},"commit":{"parents":[{"commit":"73c9581299aaf0a4a4b542511ba8f88913b310fa","subject":"Do not use distro templates by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/73c9581299aaf0a4a4b542511ba8f88913b310fa"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-05-27 18:19:35.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-05-28 18:23:46.000000000","tz":0},"subject":"Do not install COPR repo for CentOS LXC","message":"Do not install COPR repo for CentOS LXC\n\nEPEL does contain LXC package for CentOS 8 and 9 Stream, so we can\ninstall LXC from there. For CentOS 9 COPR still needed for lxc-templates\n\nWith that we also add control of what packages are going to be installed\nfrom EPEL.\n\nChange-Id: I60f1b04c2378c91905440e8de223c29dd6da4d17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c84b23b25851a8fca67a5a6e70d8bc687d5832d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c84b23b25851a8fca67a5a6e70d8bc687d5832d5"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
