)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~851203","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I3e761c181c1bf3b736fff3bf9ac441e266bc4e2c","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2022-07-29 09:35:13.000000000","reason":"\u003cGERRIT_ACCOUNT_31542\u003e replied on the change","reason_account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}}},"hashtags":[],"change_id":"I3e761c181c1bf3b736fff3bf9ac441e266bc4e2c","subject":"Take account of lxc_apt_mirror in new debootstrap command","status":"MERGED","created":"2022-07-27 10:07:35.000000000","updated":"2022-07-29 14:01:10.000000000","submitted":"2022-07-29 13:59:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851203","meta_rev_id":"98da709ccf28507649c9a237c439aaf3ff8e0c83","_number":851203,"virtual_id_number":851203,"owner":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-29 13:59:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2022-07-28 08:41:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-27 20:28:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2022-07-28 08:41:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2022-07-28 08:41:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-27 10:27:37.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2022-07-27 11:48:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-27 20:28:55.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"3881af759fe585a553f5cd3f8512e16f424c8293","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-07-27 10:07:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"17fe2887f8090020cc6c76bf10f58bf29fdbf589","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-07-27 10:27:37.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"e08bc36263aeef620093bf85acf77cfad31f0672","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 11:48:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0d8e1991ec9142b4aef7fe13b1cc42d2\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/013963daddd14eb0997287e7a73d9aa6 : SUCCESS in 19m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dfd5aad19084df19415ec6183a084c7 : SUCCESS in 5m 37s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/90981e308e344882b9a2a5c129a96686 : SUCCESS in 1h 23m 21s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/66c74c80bbff416a8e20feef27f5d852 : FAILURE in 15m 31s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f5bba0b7af74a0e97eea573045f747c : FAILURE in 16m 03s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f26a5fb9ef149d3b30cf6f0fc6301de : FAILURE in 16m 20s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/fad8a5973ecc4cd7b144906e2250289f : SUCCESS in 1h 37m 29s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1fcc434c48434a1c97b818ec92e0f0f0 : SUCCESS in 1h 33m 34s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c73e18ee569d420b8fd577060d78cc17 : FAILURE in 25m 22s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c53d8b7f71845fbb9baf57ea702f551 : FAILURE in 25m 42s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/338fca58e5d14deea93dc1b65e2c10c6 : FAILURE in 23m 57s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/e1101eb48dc240ff9e26c52d80cc06af : FAILURE in 43m 47s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/730c4490448e40b19657665fa741b83c : FAILURE in 44m 33s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-9-stream https://zuul.opendev.org/t/openstack/build/453a1347959442459c845b7ffb643e90 : FAILURE in 46m 15s","accounts_in_message":[],"_revision_number":1},{"id":"c8a0b7881ab793930eb3777dd85257db66015702","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-07-27 12:00:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7e5c5528e75223ec5a7980bcc5fad8a0e1688c02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 13:52:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/04f8794eeaaf40b89893345c88a7d2db\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95a203c1c8924bef98764a67f572b858 : SUCCESS in 18m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b85d920a8cf7484684fe69c53997adeb : SUCCESS in 5m 14s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/9e697ca7e214490fbd13a4cc2f790281 : SUCCESS in 1h 33m 25s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/dfbd3a2acd354aaf89c594befa447bee : SUCCESS in 1h 02m 02s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1f37a06ce08425cbf628dc91e253c4c : SUCCESS in 1h 21m 52s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/86baee65769b493da4f2aba4e3b63979 : SUCCESS in 1h 18m 49s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/27fa626cb581484e824caa7277cb2338 : SUCCESS in 1h 49m 34s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa62fd14ed51458193abc621cfabc029 : SUCCESS in 1h 39m 47s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7fc80da7d314a4ca602675c8d918f1a : SUCCESS in 36m 36s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d2caae1ff714cf2a860a22fc566afc7 : SUCCESS in 38m 11s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee4a3f8f9da342c1a7506bdbd7105de4 : SUCCESS in 36m 10s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/035c4f4064ac4eb58634970dfc9d9387 : SUCCESS in 28m 41s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/a4d301be43834aa89b6a4ac5b9bf4b69 : SUCCESS in 27m 23s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-9-stream https://zuul.opendev.org/t/openstack/build/27af6c2695b04f1c98b8e4fb46e78b8c : FAILURE in 28m 07s","accounts_in_message":[],"_revision_number":2},{"id":"5fc215f9205faade0ffdd87028653346f714f689","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-07-27 14:37:05.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"ea1327c23adf9d1e6a7f177f97b98f49b493d5c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 16:36:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69f30c32e8a243c882afbe2c8dc82300\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78c5452f188f406783fcc89de27c52c4 : SUCCESS in 19m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b4ac7589e07450b8d4a82dc128eff33 : SUCCESS in 12m 06s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/fd3806c495064f2083d7807355e8f67a : SUCCESS in 1h 12m 32s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/48bda0bebfb54d55956940fbca6432d9 : SUCCESS in 1h 10m 24s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70fc86862a444cb89755a22c891a8596 : SUCCESS in 1h 01m 06s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a3396b58930d4f8dbaffe1a0ab72abd2 : SUCCESS in 1h 16m 15s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/a6f75d69d2d047b89a4649a9768869e3 : SUCCESS in 1h 51m 46s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c268829598bc49a79a04e1e021b85373 : SUCCESS in 1h 31m 02s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2bf3ced21ce54fb2865de149f1df08ce : SUCCESS in 50m 40s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/29c8d0796fca4264a4d8c2076f939353 : SUCCESS in 29m 40s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c1d0fb4127d43d0b39d2241d275ed58 : SUCCESS in 24m 49s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/191fb0bdf44d4447ba152b159741c396 : SUCCESS in 29m 30s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/5718f4cf2a2447b49308ffdc26656af4 : SUCCESS in 26m 43s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-9-stream https://zuul.opendev.org/t/openstack/build/fc3a90ebfdaf450897d1f53553e5633f : SUCCESS in 42m 15s","accounts_in_message":[],"_revision_number":3},{"id":"f2ec32804cc1978055c22d1dad6e0cbff9334699","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-07-27 20:28:55.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"f2de338c16f194072b5e8f1f397ae5ea1bdc0ee2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-07-28 08:41:01.000000000","message":"Patch Set 3: Backport-Candidate+1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c30ddd21169bf5c9026dad2b3de16b24564d5035","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 08:43:21.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"30d1c7ae077f00226554be6a04fd239f7756f9eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 10:26:34.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d0d0406f199749f7aaf925719b263b72\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a44f8a33c234d2984988a8f9c5ca447 : SUCCESS in 19m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8ae8be5c1c14325a00288b4d2940a0b : SUCCESS in 8m 39s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/a5344b083266454299c5229f4e81193a : FAILURE in 1h 42m 08s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/2bac14fa96694d69be8c8644ef76718d : SUCCESS in 1h 12m 33s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c7fd57ff0594517bb9acde2ee2c384b : SUCCESS in 1h 15m 29s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8665678c190a451dbc3f09bb3abba198 : SUCCESS in 1h 42m 04s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6fd49f6f6074441eb26f0b2c18cf6add : SUCCESS in 50m 58s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ee4cdff1d9740a8ba1de63c4d501893 : SUCCESS in 28m 33s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d2a03995c82483293539924f2279a03 : SUCCESS in 30m 15s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/3fef02250e354d9387c898614995582a : SUCCESS in 23m 49s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/878fac40331e44df96c0fbda924745b9 : SUCCESS in 31m 54s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-9-stream https://zuul.opendev.org/t/openstack/build/30a45951c3c94f689b19b39a01962a7f : SUCCESS in 37m 54s","accounts_in_message":[],"_revision_number":3},{"id":"79742c6b5f4319a166f1c01e1283528f960213bb","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-07-29 07:47:54.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2538eb69d2f921fe6a84a19dac18474094bdceaa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 09:33:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f67f3bef3c24a8382d2b1b820891dbe\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5cf649ffe4534ce29d924e215d0af98f : SUCCESS in 10m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c52db96e8e641da9b00ca1c364add92 : SUCCESS in 4m 15s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/45dc73cc693f4aa3ba3ac77bdceb3632 : SUCCESS in 1h 24m 49s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/a7daba1ab92f41b093c9a6e793bbdb8e : SUCCESS in 1h 09m 41s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ff4ab9861e4400f8ff07ba9df011ce4 : SUCCESS in 1h 04m 04s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0515f7894b14149b22cd457e7afd4e5 : SUCCESS in 1h 10m 22s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/5ecbb4a617454438a08fc0f90d8768e6 : SUCCESS in 1h 39m 59s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab21f5c07c174de69621d21831976f63 : SUCCESS in 1h 26m 59s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d6f348593554050a1da4bc80fbe04f9 : RETRY_LIMIT in 19m 52s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fcde832bddb84f73b885d6d89497a1db : SUCCESS in 29m 16s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ae278d0b42e46e0b1fd55dd1651b03a : SUCCESS in 29m 07s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/c43b1405492b4626ad2a7f3fae13a26d : SUCCESS in 23m 00s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/63bee75880b34aeaa79bd9996d432e33 : SUCCESS in 21m 58s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-9-stream https://zuul.opendev.org/t/openstack/build/489da0622f4c4de78845bfa124631428 : SUCCESS in 29m 05s","accounts_in_message":[],"_revision_number":3},{"id":"3999fa82593ae084e87f955fbc4e62e9b2ac5593","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-07-29 09:35:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ff856b4ea13c7813b9450c04d3d7034dabf56e50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 12:01:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75fa12104ebe48e5860a1587c58058f8\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9e2c08a10914f5cada68db83a40d581 : SUCCESS in 18m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8aaa4f579954b4e9699a76b545615db : SUCCESS in 4m 46s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/55e67ba2fce0404c84a3f5b2b632d5ce : SUCCESS in 1h 13m 01s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/7b3950b8b1274fab8f76dae2cfdf2a40 : SUCCESS in 1h 00m 39s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dddfbadbd4e342528e366208b5bb035d : SUCCESS in 1h 00m 16s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/29e44b39dd4448bb9422926ed4c1b21e : SUCCESS in 1h 19m 16s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/5c39e931f424462a83fc63f4ab551f81 : SUCCESS in 2h 06m 18s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/324f980829cb4d9695121e0c081132b2 : SUCCESS in 1h 48m 00s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70cba47b1d9247c6b68ac9ee956133ef : SUCCESS in 50m 18s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/810536e8d5124111b0629a5138bcb53f : SUCCESS in 48m 27s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23cd3148f73a452384298efbba5bdadc : SUCCESS in 50m 47s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/03a1ca14ee2a4c39a2d980c66f19724a : SUCCESS in 51m 41s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/7170e3fe9a0145fb8ba5630c75f3b076 : SUCCESS in 48m 50s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-9-stream https://zuul.opendev.org/t/openstack/build/ccfe21ee38b044b09059db0d9c4b7ec1 : SUCCESS in 56m 48s","accounts_in_message":[],"_revision_number":3},{"id":"073c46da2cb6cd8ae99f196cafd5defea9719fe6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 12:03:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"652643abebcff462d2b205ac38b398c3d7cb984e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 13:59:43.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6839593bccf456c9a1880cc2447e732\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c312a4fc23244ff4887e013469b75099 : SUCCESS in 36m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c83701c5c734d6db0002cf5b153fda3 : SUCCESS in 8m 21s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/41f50f8f4f18433e82f240946c17879f : SUCCESS in 1h 43m 25s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/0c6cd5c35c4a47d1b9b285db2fb217aa : SUCCESS in 1h 19m 16s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/737dd81fe31b427486c2479588f26c05 : SUCCESS in 1h 20m 03s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42c4cf681ac64a3daaa86bd895f37c0f : SUCCESS in 1h 16m 57s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e79e86f2a0fe45a5a4b6db9f3277b328 : SUCCESS in 28m 47s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/557376b92fb04761beb8d46e4e0c92e5 : SUCCESS in 28m 53s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c89ed32821504388b673bebe4c33477b : SUCCESS in 29m 26s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/ad2a4f1f460e4ee48ac6181d6c4eb49d : SUCCESS in 29m 24s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/cc461e2b9db6427db3e9514f18945c72 : SUCCESS in 27m 20s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-9-stream https://zuul.opendev.org/t/openstack/build/72154bea130f4aaf98780f80ab810306 : SUCCESS in 36m 12s","accounts_in_message":[],"_revision_number":3},{"id":"3a83803d2dbedadef0cd841b89483ac5c0c44edd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 13:59:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"98da709ccf28507649c9a237c439aaf3ff8e0c83","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 14:01:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e11a0f3810749c78867421bcda15223\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9e99de2086042d2b4427fc829122db3 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"783076a508d64918eb474ea5069b1261efdbaf79","revisions":{"8ddf8ff34b711d57b8cc4459833c203253df7cef":{"kind":"REWORK","_number":1,"created":"2022-07-27 10:07:35.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/03/851203/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/03/851203/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/03/851203/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/03/851203/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/03/851203/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/03/851203/1"}}},"commit":{"parents":[{"commit":"844ebcdab63125f239733c2ef1023fade5543e68","subject":"Merge \"Switch sphinx language to en\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/844ebcdab63125f239733c2ef1023fade5543e68"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2022-07-27 10:04:29.000000000","tz":60},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2022-07-27 10:07:08.000000000","tz":60},"subject":"Take account of lxc_apt_mirror in new debootstrap command","message":"Take account of lxc_apt_mirror in new debootstrap command\n\nWithout this change the mirror variable is ignored which can\ncause issues on systems running behind a proxy.\n\nChange-Id: I3e761c181c1bf3b736fff3bf9ac441e266bc4e2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/8ddf8ff34b711d57b8cc4459833c203253df7cef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/8ddf8ff34b711d57b8cc4459833c203253df7cef"}]},"branch":"refs/heads/master"},"45878f7dcde77d447bb47385a3fafdad2743f1e8":{"kind":"REWORK","_number":2,"created":"2022-07-27 12:00:13.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/03/851203/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/03/851203/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/03/851203/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/03/851203/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/03/851203/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/03/851203/2"}}},"commit":{"parents":[{"commit":"844ebcdab63125f239733c2ef1023fade5543e68","subject":"Merge \"Switch sphinx language to en\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/844ebcdab63125f239733c2ef1023fade5543e68"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2022-07-27 10:04:29.000000000","tz":60},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2022-07-27 11:59:51.000000000","tz":60},"subject":"Take account of lxc_apt_mirror in new debootstrap command","message":"Take account of lxc_apt_mirror in new debootstrap command\n\nWithout this change the mirror variable is ignored which can\ncause issues on systems running behind a proxy.\n\nChange-Id: I3e761c181c1bf3b736fff3bf9ac441e266bc4e2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/45878f7dcde77d447bb47385a3fafdad2743f1e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/45878f7dcde77d447bb47385a3fafdad2743f1e8"}]},"branch":"refs/heads/master"},"783076a508d64918eb474ea5069b1261efdbaf79":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-07-27 14:37:05.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/03/851203/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/03/851203/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/03/851203/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/03/851203/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/03/851203/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/03/851203/3"}}},"commit":{"parents":[{"commit":"844ebcdab63125f239733c2ef1023fade5543e68","subject":"Merge \"Switch sphinx language to en\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/844ebcdab63125f239733c2ef1023fade5543e68"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2022-07-27 10:04:29.000000000","tz":60},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2022-07-27 14:37:05.000000000","tz":0},"subject":"Take account of lxc_apt_mirror in new debootstrap command","message":"Take account of lxc_apt_mirror in new debootstrap command\n\nWithout this change the mirror variable is ignored which can\ncause issues on systems running behind a proxy.\n\nChange-Id: I3e761c181c1bf3b736fff3bf9ac441e266bc4e2c\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/851233\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/783076a508d64918eb474ea5069b1261efdbaf79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/783076a508d64918eb474ea5069b1261efdbaf79"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[]}
