)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~851387","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2Fyoga~I3e761c181c1bf3b736fff3bf9ac441e266bc4e2c","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2022-08-08 10:07:15.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-08-08 08:34:23.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I3e761c181c1bf3b736fff3bf9ac441e266bc4e2c","subject":"Take account of lxc_apt_mirror in new debootstrap command","status":"MERGED","created":"2022-07-29 14:01:43.000000000","updated":"2022-08-08 10:08:29.000000000","submitted":"2022-08-08 10:07:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851387","cherry_pick_of_change":851203,"cherry_pick_of_patch_set":3,"meta_rev_id":"e3e93197447d7db38a95578c84bc9c83099fe866","_number":851387,"virtual_id_number":851387,"owner":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-08 10:07:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2022-08-08 08:34:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-04 22:03:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2022-08-08 08:34:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-29 15:45:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-04 22:03:00.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2022-08-08 08:34:23.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"983b217af6508ab1f223ea6ac3cbab06ba07d5ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-07-29 14:01:43.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"589ffc8c7cbf78d5f4e242cf029dd65be189450f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 15:45:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b05d7faddda54be39c3d75eae0a96815\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/97ee5919588b4e91a28ea95f1e5f6589 : SUCCESS in 36m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ac592c428ac4be684414169de493606 : SUCCESS in 8m 19s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/00f703ba606e42a3a23ebc23bc84c00e : RETRY_LIMIT in 12m 03s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/2fcb5027f21d4e4d9df793802ac290cf : FAILURE in 1h 34m 06s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/53f9efbee3c54a3fa2975762ad9991f6 : FAILURE in 1h 15m 27s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/0067eeae33f448008cbfaedaf62e3d93 : NODE_FAILURE in 0s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7806fb9f323f4d17bfdf45a5be0c75be : FAILURE in 1h 19m 19s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/00de515b69d940189ef25a7536465ac4 : FAILURE in 1h 20m 10s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/785e3e9ac2f54ba0b56c45bbd80e511a : FAILURE in 13m 14s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d2bc384427b4686a7362bea4407f059 : FAILURE in 1h 00m 38s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab61397f9d714ac28d65335aae14d5c2 : SUCCESS in 36m 39s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/546848a47aa34d61888b700380f89999 : SUCCESS in 35m 02s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/919857e3566044b7a845d45121c788f2 : SUCCESS in 35m 18s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/dfe4ebd1c236438598c7337096a95ea5 : SUCCESS in 31m 27s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/e0bdeaff4e854f799e0283c9b31ff180 : SUCCESS in 35m 42s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-8 https://zuul.opendev.org/t/openstack/build/07ae6d8eadee4ed5a428bf40d4a60e65 : NODE_FAILURE in 0s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-8-stream https://zuul.opendev.org/t/openstack/build/8134b35354684bb89a108b71bddfe8a1 : RETRY_LIMIT in 12m 34s","accounts_in_message":[],"_revision_number":1},{"id":"7340780fbd7e74f4e6d27b344e371662f77c5a42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-08-04 10:17:49.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"d99ef70044a6454782720155fe7fe7d25c5a29f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 12:20:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78969d8aacaa47a480e56c882bb34568\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/921b25e3b55848b1a78feda1972c77c4 : SUCCESS in 19m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f37f12ee234456f978609a48bf4b238 : SUCCESS in 5m 48s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/2aa16edbb07c44f5bf5a7672ed2321f9 : SUCCESS in 1h 19m 59s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/0fbc2db808f845f79b4f3c26751dd869 : SUCCESS in 1h 22m 21s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/05577367a6514b738590aefa49463b5b : SUCCESS in 1h 01m 35s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/e31965bfb363426a8fcc4026b798ffca : SUCCESS in 1h 29m 48s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9be1939cc69f481bb639bd0f2e133960 : SUCCESS in 1h 20m 51s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dfbbbe30db7a4430870e114b7934979b : SUCCESS in 1h 18m 00s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/403316b02cb34950966e3a07595d7657 : SUCCESS in 1h 50m 35s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8b883bbb7204779a72179fc0dd7f22c : SUCCESS in 1h 41m 47s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b98901733dd94190b743c124761012b3 : SUCCESS in 50m 08s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/858ba96a01a34bf3b3caf89ab8ef643e : SUCCESS in 49m 18s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/72da0671ce7b4a0b99e4e9a852688fdf : SUCCESS in 49m 45s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/659384ca9c884b22b0f5807cf87bb02c : SUCCESS in 48m 34s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/cc36509f72634acb85d0e59a2fd3bf78 : SUCCESS in 50m 50s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-8 https://zuul.opendev.org/t/openstack/build/1e558ffc7eef4874b16517840f024623 : SUCCESS in 41m 09s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-8-stream https://zuul.opendev.org/t/openstack/build/fa7cfb9228ae4feabf492f66a9a682cf : SUCCESS in 54m 20s","accounts_in_message":[],"_revision_number":2},{"id":"dceae7fa571af186ed3cbe7362339c437370499e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-08-04 22:03:00.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b7b6845447bb35321205ddf95c9e4149d7ddd274","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-08-08 08:34:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6b6a91b0199c2db4b084b80e67b8866dfcdacc88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 08:36:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b1d744f069c846746d06c2dbe9f27685e0cd4b53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 10:07:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71ad084743864a8da1e5f2d67f00ee70\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f137791b2aef467ca527b2d558c26dbe : SUCCESS in 19m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5503921dab3c4f6ca0a579638aa8e51f : SUCCESS in 5m 41s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/bcf1f7710fa84e85bb3cdf9b32b641e0 : SUCCESS in 1h 21m 39s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/49ce2eaebf08491eb510789c5521e089 : SUCCESS in 1h 16m 21s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/eb07b98a1fe44d59bdfa6123e871661c : SUCCESS in 1h 05m 26s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/fd52aca868264e6296d423b68902c75a : SUCCESS in 1h 24m 10s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/daa46f5d4ab24508ac706a7e743c4f3c : SUCCESS in 1h 01m 22s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fac2f5c8630d42b2a0b269e53f729878 : SUCCESS in 1h 28m 45s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ccb633ef546c464f80a668b5ee949548 : SUCCESS in 32m 10s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb7ef83fb58f499482042c5aa0e3b9df : SUCCESS in 28m 53s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77e1fa4bf70943a39bec9f744f5bac4c : SUCCESS in 31m 50s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/4702b69cfe8842d4bb83586f4fc07937 : SUCCESS in 30m 31s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/fff5288f743c49f79e102052ecb248a3 : SUCCESS in 30m 33s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-8 https://zuul.opendev.org/t/openstack/build/16c27ccc2cca475da4041d8f93dc22ec : SUCCESS in 37m 52s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-8-stream https://zuul.opendev.org/t/openstack/build/2cb07f9f66504a3691c2e641a1c176c2 : SUCCESS in 36m 29s","accounts_in_message":[],"_revision_number":2},{"id":"767b3ad5b50c8f61d9853be2bcfb4d7e34dd3502","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 10:07:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e3e93197447d7db38a95578c84bc9c83099fe866","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 10:08:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6563d3a1b61e48e6b8d2291d8d6130be\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5348bc89b334fb28f40acd136303da0 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6a4b8f37789e1e3ec6ad37498124308c57f93303","revisions":{"9cdb3c7603580dd913c53509106e7e97771a0a28":{"kind":"REWORK","_number":1,"created":"2022-07-29 14:01:43.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/87/851387/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/87/851387/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/87/851387/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/87/851387/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/87/851387/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/87/851387/1"}}},"commit":{"parents":[{"commit":"c59a5d672bd404cb2e26e0e67fa18819dcf278c6","subject":"Update TOX_CONSTRAINTS_FILE for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c59a5d672bd404cb2e26e0e67fa18819dcf278c6"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2022-07-27 10:04:29.000000000","tz":60},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2022-07-29 14:01:43.000000000","tz":0},"subject":"Take account of lxc_apt_mirror in new debootstrap command","message":"Take account of lxc_apt_mirror in new debootstrap command\n\nWithout this change the mirror variable is ignored which can\ncause issues on systems running behind a proxy.\n\nChange-Id: I3e761c181c1bf3b736fff3bf9ac441e266bc4e2c\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/851233\n(cherry picked from commit 783076a508d64918eb474ea5069b1261efdbaf79)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/9cdb3c7603580dd913c53509106e7e97771a0a28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/9cdb3c7603580dd913c53509106e7e97771a0a28"}]},"branch":"refs/heads/stable/yoga"},"6a4b8f37789e1e3ec6ad37498124308c57f93303":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-08-04 10:17:49.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/87/851387/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/87/851387/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/87/851387/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/87/851387/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/87/851387/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/87/851387/2"}}},"commit":{"parents":[{"commit":"c59a5d672bd404cb2e26e0e67fa18819dcf278c6","subject":"Update TOX_CONSTRAINTS_FILE for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c59a5d672bd404cb2e26e0e67fa18819dcf278c6"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2022-07-27 10:04:29.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-08-04 10:17:48.000000000","tz":0},"subject":"Take account of lxc_apt_mirror in new debootstrap command","message":"Take account of lxc_apt_mirror in new debootstrap command\n\nWithout this change the mirror variable is ignored which can\ncause issues on systems running behind a proxy.\n\nChange-Id: I3e761c181c1bf3b736fff3bf9ac441e266bc4e2c\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_rally/+/849227\n(cherry picked from commit 783076a508d64918eb474ea5069b1261efdbaf79)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6a4b8f37789e1e3ec6ad37498124308c57f93303"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6a4b8f37789e1e3ec6ad37498124308c57f93303"}]},"branch":"refs/heads/stable/yoga","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
