)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~852497","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2Fyoga~I9140b7ab9f9266fcf4fe800e4610497f2324df4e","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-08-12 05:09:06.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I9140b7ab9f9266fcf4fe800e4610497f2324df4e","subject":"Prevent lxc.service from being restarted on package update","status":"MERGED","created":"2022-08-09 12:08:02.000000000","updated":"2022-08-12 08:44:27.000000000","submitted":"2022-08-12 08:43:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852497","cherry_pick_of_change":851071,"cherry_pick_of_patch_set":2,"meta_rev_id":"a3b623eaf1f71bcd320c3a1e99de579b6ab5828d","_number":852497,"virtual_id_number":852497,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-12 08:43:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2022-08-10 17:22:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":2,"date":"2022-08-10 19:26:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":1,"date":"2022-08-09 12:11:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":1,"date":"2022-08-10 19:26:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-09 12:11:16.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2022-08-09 13:51:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-10 17:22:22.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"}],"messages":[{"id":"cbf5e3a6ea6a067c848a01bedf4b3e2b44cddf2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-08-09 12:08:02.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"fe15f5f350ebcffa338f115e7174e7b34ad0886f","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2022-08-09 12:11:16.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dd803d7a9528f143f60a071bf7dde179dbcf23ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 13:51:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfc257a98f884bb982418ebe27904041\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59088b3c7e294cffb707f10ea38884cc : SUCCESS in 19m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a31c7225589a40c3847fb07a3a0c1479 : SUCCESS in 5m 49s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/dae332b1b7e44a77a29863848666461f : SUCCESS in 1h 28m 18s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/79740e13edb44ffc8ef8cf3d17775f71 : SUCCESS in 1h 18m 32s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/812cf6eabcf44aeb864fcb32a0305c0c : SUCCESS in 1h 06m 16s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/873672f6cec3444a97824bb915e99ba3 : SUCCESS in 1h 30m 40s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/622e4dce22c94685b979954ac0607466 : SUCCESS in 1h 03m 44s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca364bcd755e49c9af167b1d919b6e0c : SUCCESS in 59m 24s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/8753d79d95de4e8894e938eba42b67a0 : SUCCESS in 1h 40m 16s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/482601a4cce546b2835fd39ec16dc7f4 : SUCCESS in 1h 31m 49s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/acef93f597d64a2d883b96745a55ea6d : SUCCESS in 34m 53s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c8282389dc84e21b6f7415db2b26a42 : SUCCESS in 35m 00s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/26962b0133e14bfc9257c76b3743d3e0 : SUCCESS in 32m 52s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/22219f8db58d4c43912baec067fcbd46 : SUCCESS in 30m 50s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/ce5c1d4e45424ad980776764af8ffde5 : SUCCESS in 32m 59s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-8 https://zuul.opendev.org/t/openstack/build/ee2a3e07d9ee405985a72775808e3881 : SUCCESS in 36m 34s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-8-stream https://zuul.opendev.org/t/openstack/build/42d78e4f15af4df88171355ec68c826d : SUCCESS in 45m 07s","accounts_in_message":[],"_revision_number":1},{"id":"a543502cb5f34a77617975eea24089b97f130b19","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-08-10 17:22:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a802aae67ad64089f7e0fb958330070efeca57b0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-08-10 19:26:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4aa206be0e941d4d9b7a122dcb0d45ac3a2fbe07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 19:28:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ca6e17ed6385d64a681f6d3c8878a68f9d4ca78d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 21:08:28.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7bd59cc031c34b77bd218ad701314b92\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ac7d859f9fa41669370105a787f2f43 : SUCCESS in 17m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c60f3edc0c8846b7a07b807edfbb4722 : SUCCESS in 5m 11s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/1524d4ca876c42138beb17e02eaa4c92 : SUCCESS in 1h 18m 32s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/4a9bb1b4ba9f43dbaadba23222ed2a64 : RETRY_LIMIT in 21m 46s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/69b28b349bc24edaab99943b1974f9dc : SUCCESS in 1h 03m 02s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/8134859231864c278e9f1a4167f48a29 : SUCCESS in 1h 37m 26s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0554f1266b342c38089637bb18f2386 : SUCCESS in 1h 03m 04s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f8ec9d045064fdc9c659e3004a747de : SUCCESS in 1h 31m 39s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b4258b9e350149daa839bbc34dab4960 : SUCCESS in 33m 17s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df0de25cfa3b4ad38d045967f9bd9aa7 : SUCCESS in 30m 55s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b08581386d664ee0abac85b1bb321031 : SUCCESS in 31m 00s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/e883dd3af9214367b83c878412c56aaa : SUCCESS in 30m 15s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/899c3e77a8e74039953b51bc5d256dde : SUCCESS in 30m 08s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-8 https://zuul.opendev.org/t/openstack/build/e6dcd1399aa34e24969e14f23e609782 : SUCCESS in 37m 06s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-8-stream https://zuul.opendev.org/t/openstack/build/a16cc8c319ab4cf6aa7e4c08745e7a7b : SUCCESS in 37m 02s","accounts_in_message":[],"_revision_number":1},{"id":"2ffe9a1078192d3ea5ae757013019165abaae7f1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-08-12 05:09:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8d62b025615f9448a81ddfc89430366a91a651de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 07:10:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1deb9f74f724fa5b95ff886dad372cb\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d1bf65ff277489d8391c43db4c7f6d9 : SUCCESS in 20m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f60d85c733746ab84fa45ff1e8387e0 : SUCCESS in 6m 32s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/d3cffb7e61dc4354b11efca78decc5eb : SUCCESS in 1h 24m 12s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/f43ea70f8e0d439fa11ebd1363495fc1 : SUCCESS in 1h 33m 23s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/750199fcafbe47508696b7fe360b4dc6 : SUCCESS in 1h 25m 41s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/04d7da5ba6a04d3499af08f3db371c8b : SUCCESS in 1h 27m 45s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39a9e8c5556f45d087e208550a035666 : SUCCESS in 1h 14m 14s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/23119f56d2844a5c9449641e790c62e7 : SUCCESS in 1h 12m 30s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/57774a8edcce4e2e9f0b6d8ff7822a93 : SUCCESS in 1h 57m 09s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7bc1a28d4c524638bb5cf0106624adfc : SUCCESS in 1h 45m 03s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9545dc79084f46e5bcc98838af590dc9 : SUCCESS in 36m 31s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4728df9499eb4070a1752d2212927c01 : SUCCESS in 34m 50s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf2477eab0a7465d877ede911093642a : SUCCESS in 35m 05s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/0542512c3cd74e258438f4f09b989f3f : SUCCESS in 32m 55s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/dc39723a2c45464893f4cb29191d561e : SUCCESS in 32m 16s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-8 https://zuul.opendev.org/t/openstack/build/63141551583a46b08692097dbdeeeb72 : SUCCESS in 40m 10s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-8-stream https://zuul.opendev.org/t/openstack/build/c230294d7b354560a8772496c7159424 : SUCCESS in 40m 23s","accounts_in_message":[],"_revision_number":1},{"id":"d4cc16275822e8f868b2b64cb2d449bda83fb687","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 07:12:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"caa5068922d06352487e50f0d6653b9383aa7114","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 08:43:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30b24f3a41404bb4b26bab7972d14c69\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/211bdd08674a434a81e000cb15ebd1f9 : SUCCESS in 38m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8a46ed44fe94aca8e855f08ec50617c : SUCCESS in 5m 28s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/44175ff6448a4686aac408a550d8d068 : SUCCESS in 1h 22m 43s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/79e5c2c6d7c842308b827ac0cab7baf6 : SUCCESS in 1h 23m 31s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/484fb448b7e04b7c90b0ee7960f3426b : SUCCESS in 59m 49s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/f357f29b6c13453085b022d96deda4c2 : SUCCESS in 1h 22m 25s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d2a1bea35864cd29f69fad8e2a17ba3 : SUCCESS in 1h 08m 43s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f5f020de63f4e65ae19f1f230d3fbd2 : SUCCESS in 1h 29m 41s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2dbad1ec1f35428889168d5fbf712b1f : SUCCESS in 31m 47s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/04703df292bb4083980c937d7ce3598d : SUCCESS in 32m 27s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eecd5fb22fbb463ea48ddc9c812d56a6 : SUCCESS in 32m 19s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/f0a9d9d063fb4138914d05a85084109f : SUCCESS in 31m 17s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/0c5c8279647e422e902a21dcb456bf69 : SUCCESS in 28m 24s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-8 https://zuul.opendev.org/t/openstack/build/da5220f48f7248e58329d4fc1a8cb132 : SUCCESS in 38m 23s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-8-stream https://zuul.opendev.org/t/openstack/build/187dff3bae9f4fd08b1a9a31f1d2497e : SUCCESS in 38m 01s","accounts_in_message":[],"_revision_number":1},{"id":"865a4c198fbc5041164b5ee107e0221ac2d57ef2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 08:43:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a3b623eaf1f71bcd320c3a1e99de579b6ab5828d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 08:44:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5105c27bc95b4a0bb7f50efa3762cc48\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4387364f89b24ef7a0743983927656ad : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f2d17607735eeccfa2f06dc0adc9f2f27f495d67","revisions":{"f2d17607735eeccfa2f06dc0adc9f2f27f495d67":{"kind":"REWORK","_number":1,"created":"2022-08-09 12:08:02.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/97/852497/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/97/852497/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/97/852497/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/97/852497/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/97/852497/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/97/852497/1"}}},"commit":{"parents":[{"commit":"6a4b8f37789e1e3ec6ad37498124308c57f93303","subject":"Take account of lxc_apt_mirror in new debootstrap command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6a4b8f37789e1e3ec6ad37498124308c57f93303"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-07-26 16:29:24.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-08-09 12:08:02.000000000","tz":0},"subject":"Prevent lxc.service from being restarted on package update","message":"Prevent lxc.service from being restarted on package update\n\nAs of today, each lxc-utils update would lead to restart of all\ncontainers. At the same time this might be unwanted behaviour, as\nif it\u0027s run without limit, all cluster members inside containers can\ngo down at the same time.\n\nIn order to prevent that, we place policy-rc.d file that will simply\nquit with 101 code `action forbidden by policy` on service restart\nattempt.\n\nChange-Id: I9140b7ab9f9266fcf4fe800e4610497f2324df4e\n(cherry picked from commit f8594d335f0ec70768d7bccd9ba13d84a6f65151)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f2d17607735eeccfa2f06dc0adc9f2f27f495d67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f2d17607735eeccfa2f06dc0adc9f2f27f495d67"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
