)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~852498","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2Fxena~I9140b7ab9f9266fcf4fe800e4610497f2324df4e","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-08-10 19:26:44.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I9140b7ab9f9266fcf4fe800e4610497f2324df4e","subject":"Prevent lxc.service from being restarted on package update","status":"MERGED","created":"2022-08-09 12:08:13.000000000","updated":"2022-08-10 22:20:27.000000000","submitted":"2022-08-10 22:19:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852498","cherry_pick_of_change":851071,"cherry_pick_of_patch_set":2,"meta_rev_id":"e49b065de2999c6a93a28ed6b50fd7445a1b6b1a","_number":852498,"virtual_id_number":852498,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-10 22:19:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2022-08-10 17:23:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":2,"date":"2022-08-10 19:26:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":1,"date":"2022-08-09 12:12:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":1,"date":"2022-08-10 19:26:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-09 12:12:23.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2022-08-09 14:06:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-10 17:23:24.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"}],"messages":[{"id":"9a3cc8b8f8027ef65d1663bd0ddfeacbd83ce25f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-08-09 12:08:13.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c53021571dc9f338f0fcbf074d47b986ffb856cd","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2022-08-09 12:12:23.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"06d4d9e4c8f177ed7969eda3eb9f57722bda0cd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 14:06:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f577ce3205b6479a9a154194c2085700\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea58e22189c34befb3162e077992c363 : SUCCESS in 21m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03069e4ca2f34e889fdbe64b0457cda0 : SUCCESS in 9m 04s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/9d883eddb362441b84f78bd27f00b3a4 : SUCCESS in 1h 25m 44s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/602d2b48ca6d428b8adcd01eac4ce455 : SUCCESS in 1h 31m 06s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/3734860328b44c3ca7b0e6a7569933a6 : SUCCESS in 1h 06m 19s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9fcb4902020e4ecc9156b4e73e99af8a : SUCCESS in 1h 13m 04s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/6d204efc67de459a90079989dc82a6a7 : SUCCESS in 1h 55m 16s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31b3f40e7cc94ebb953fee184584fd5c : SUCCESS in 1h 31m 27s\n- openstack-ansible-lxc-dir-centos-8-stream https://zuul.opendev.org/t/openstack/build/6330d1af46634138aee7566f1dbd413d : SUCCESS in 27m 57s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b06beb88dd92440181d994436df8e5d4 : SUCCESS in 18m 57s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ee4e85cb6194bb1aa5090b52f5cb403 : SUCCESS in 23m 00s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/df58ec35376e4df4b83e91560625f74f : SUCCESS in 21m 19s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/63d276ea8e7943dd99fcb874ef2edb44 : SUCCESS in 15m 44s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/db1113b209624113bdc3c92f74adbe4a : SUCCESS in 21m 55s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/83c2f467f9144e1dab7c0d4fe5b9ac04 : SUCCESS in 19m 43s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/36307235fd974798b933c419fdd074a6 : SUCCESS in 19m 31s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/723c70cb64384cf688550d9164283fca : SUCCESS in 25m 27s","accounts_in_message":[],"_revision_number":1},{"id":"9d153dabddf0321e850e5f350411a9db00e23b2e","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-08-10 17:23:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"439e61f3b4be679c981848a07af39878a512baad","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-08-10 19:26:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d510789f9b1979063a865d9ab5a4c655b4c88ec1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 19:28:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2df6f54df30705a93d4c20ae4833866a24ac55bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 22:19:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f4c6c79043a4858b17fd112d57beb66\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a6d8ed27bce438e840af370282bb62a : SUCCESS in 17m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b73724ae7804094b957c22d7d91f673 : SUCCESS in 5m 52s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/67801c8f08cb44d890234d854378abd9 : SUCCESS in 1h 21m 44s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/c58238ef99f14758b02319d6decafcc6 : SUCCESS in 1h 12m 54s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/cc9bfd13d8ff4667b3e51b815c1f0bd4 : SUCCESS in 1h 07m 47s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0487fba42d8d432eab1572ea4cf67d03 : SUCCESS in 1h 14m 46s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/81358507d2a64a6080b5430646d9dd72 : SUCCESS in 1h 37m 30s","accounts_in_message":[],"_revision_number":1},{"id":"b4d09cccdc83a2672f7bd9c0f509f9d232b096ab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 22:19:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e49b065de2999c6a93a28ed6b50fd7445a1b6b1a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 22:20:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a2c670a7cf74bc9933535e7f12bfb64\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab899ab0962e4b09a893fcd12e8697fe : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fea49e30bcb34b7b229d6f586ba2c207b6f3f79b","revisions":{"fea49e30bcb34b7b229d6f586ba2c207b6f3f79b":{"kind":"REWORK","_number":1,"created":"2022-08-09 12:08:13.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/98/852498/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/98/852498/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/98/852498/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/98/852498/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/98/852498/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/98/852498/1"}}},"commit":{"parents":[{"commit":"fc0810cc9761335a4d2920d05a926daff5c93139","subject":"Do not mess up with distro-provided machines mount","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/fc0810cc9761335a4d2920d05a926daff5c93139"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-07-26 16:29:24.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-08-09 12:08:13.000000000","tz":0},"subject":"Prevent lxc.service from being restarted on package update","message":"Prevent lxc.service from being restarted on package update\n\nAs of today, each lxc-utils update would lead to restart of all\ncontainers. At the same time this might be unwanted behaviour, as\nif it\u0027s run without limit, all cluster members inside containers can\ngo down at the same time.\n\nIn order to prevent that, we place policy-rc.d file that will simply\nquit with 101 code `action forbidden by policy` on service restart\nattempt.\n\nChange-Id: I9140b7ab9f9266fcf4fe800e4610497f2324df4e\n(cherry picked from commit f8594d335f0ec70768d7bccd9ba13d84a6f65151)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/fea49e30bcb34b7b229d6f586ba2c207b6f3f79b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/fea49e30bcb34b7b229d6f586ba2c207b6f3f79b"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
