)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~861676","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~Icdf4a1f5ff98dc1b86c6a87ea4e606b7c74e1aac","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","attention_set":{},"removed_from_attention_set":{"7353":{"account":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"last_update":"2022-11-30 11:25:09.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-11-30 11:25:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icdf4a1f5ff98dc1b86c6a87ea4e606b7c74e1aac","subject":"Add option to disable lxc interface management","status":"MERGED","created":"2022-10-18 00:09:17.000000000","updated":"2022-11-30 11:26:11.000000000","submitted":"2022-11-30 11:25:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861676","meta_rev_id":"4f3a1e4a0b633c7a5d861e9a923c6f4db3db2df6","_number":861676,"virtual_id_number":861676,"owner":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-30 11:25:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"recommended":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"all":[{"value":1,"date":"2022-11-29 16:35:23.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":2,"date":"2022-11-29 16:36:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-11-30 09:35:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-11-30 09:35:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-18 01:49:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-29 16:36:28.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2022-11-30 09:35:59.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"2cb230e6a5da92d6dbef96c19774e7c6e8029905","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2022-10-18 00:09:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aed6430b6b56f826dca4344ae7d10972f00dfdca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 01:49:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0e077bd36e58403193faad2a78f102e9\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e29028b9c76447aa9a081c0a65749ce6 : SUCCESS in 12m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e269be5f2dc249258edfde4d810d85ce : SUCCESS in 4m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bea20e8e747545b3b056f267700df338 : SUCCESS in 3m 09s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/98a986e752f24fe595aa8c2c0c0056dc : SUCCESS in 1h 37m 45s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fa721dbf1e754bccb336f0a9b2c31c9c : SUCCESS in 1h 22m 28s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/c1487129d7c74a9e820dec73d406b3f4 : FAILURE in 1h 15m 39s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e4842d47e70d4394810666a1f44d3e66 : SUCCESS in 1h 20m 25s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67a5431360e145a48ffe63ad702fffaa : SUCCESS in 1h 19m 48s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/aa62a4335088456f85af5f3408bfb40f : FAILURE in 1h 16m 30s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f334bbd0dff42c1a789eb40bdbc85ee : SUCCESS in 1h 18m 18s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb1a5ea0102242b7998d1eaef8f85e6b : SUCCESS in 34m 34s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2bd7e7dae90a40a7a3867943308264bd : SUCCESS in 30m 12s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ad5f16b243147b6a29c0d42b853e94e : SUCCESS in 34m 09s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/ff470c60e8494d41be87264554a351db : SUCCESS in 46m 33s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/e5f0c18fe1e64439a2d5c9fe2e4b244a : SUCCESS in 32m 38s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-9-stream https://zuul.opendev.org/t/openstack/build/25ec754072614c048daf3f6a063d9f01 : SUCCESS in 36m 43s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/22163485003746fe9dd23b439227c8f1 : SUCCESS in 33m 41s","accounts_in_message":[],"_revision_number":1},{"id":"895367d1986044db56261cbfeabb87b555c158cd","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2022-10-18 23:51:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0bcd12b96712345358e87d14e369e5a8b7d5d1f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 01:45:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/35889bc0f300487aaecc9e08d205255b\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9290553636f74a43ad04508d0bbae17f : SUCCESS in 12m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fc73eaff4204e408820139b6aef91ee : SUCCESS in 5m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd44c2ba5ed14118926a87eb1cf32d88 : SUCCESS in 2m 36s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/2942854f9fa14dacb02f4c587a0607f7 : SUCCESS in 1h 23m 13s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/411f9b76500e467497f5a54caca9d0d7 : SUCCESS in 1h 17m 32s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/5097873e76bf4007a97adb241f473ddf : FAILURE in 1h 18m 24s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e1a586c442044f50b5b59543bd4eb13f : SUCCESS in 1h 14m 04s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6317fad92d254aabb156247205c7bf61 : SUCCESS in 1h 17m 04s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/5e5da873d58e4e628e011ae71e975e06 : FAILURE in 1h 18m 06s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/696ab837c9184938b2448bc5af06902b : SUCCESS in 1h 29m 43s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0430c7e3e34e4be3a5ac1eca13b34b28 : SUCCESS in 29m 08s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ee190dc8ef54082ae7bff307c3958e1 : SUCCESS in 31m 29s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c11bcbb121a4d03af63c1af5b623b91 : SUCCESS in 32m 26s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/d2ef1765ae23465b881be526a486d7d6 : SUCCESS in 32m 43s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/9eb8009f50a44fa09c2dcbd610f80e81 : SUCCESS in 30m 34s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-9-stream https://zuul.opendev.org/t/openstack/build/2583bad7953b432794f4c9478ff9a1bc : SUCCESS in 38m 02s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/50bae311958347728130735118e79193 : SUCCESS in 29m 38s","accounts_in_message":[],"_revision_number":1},{"id":"fb10d9db9378fa585f23b34dc9792db8cac8e321","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-11-29 16:27:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"94469a9c0132a32d20a00d9b09baab54565bd2ea","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2022-11-29 16:35:23.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c46fc8fe761133974cca391e2057f67d402f5736","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-11-29 16:36:28.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2ab8214cf9d64721eef81c4cdf4cb184d7ce386d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 18:18:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7075e5c425f4225b1b236aab432ac46\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/148160c860b145678f0a0476966c6786 : SUCCESS in 16m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ae647c7fabf478e9b148440872235cc : SUCCESS in 5m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b53018cc0ee4dc5884e96d08bbb693b : SUCCESS in 3m 18s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/fdebc19aa5074ac0a92af0d07f871aba : SUCCESS in 1h 40m 21s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/06f63e78431948c58ce646d319528fea : SUCCESS in 1h 32m 50s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/8aa539ecc71a489eafbfc89be6caa34a : SUCCESS in 1h 18m 44s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92868510a2bd433282888caa5ea77485 : SUCCESS in 1h 24m 17s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97b3d2448aec408994f9b1b0eb82042c : SUCCESS in 1h 18m 54s\n- openstack-ansible-upgrade-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/3b8bb18aea7e4e22bf430b78729e9398 : SUCCESS in 1h 48m 45s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9574be78f25e4d20977fbccf74e5165b : SUCCESS in 1h 20m 11s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/63185bcb79d5489d8c1699f414fb0c55 : SUCCESS in 34m 15s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5fe85383fa1844a99bee92f75adc840b : SUCCESS in 32m 45s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0a0e62e0a9e4814ac54721bcec8cedc : SUCCESS in 34m 22s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/8555b8bc32c54e8fbe5e0f199032812c : SUCCESS in 32m 32s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/0b8ca4a1ef714f8486ef5ed3a4ea69e3 : SUCCESS in 33m 23s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-9-stream https://zuul.opendev.org/t/openstack/build/355cf4a6406e4c62b0dc07ace9a9499e : SUCCESS in 41m 01s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/21e718aee27642928ea00319aa49a582 : SUCCESS in 34m 13s","accounts_in_message":[],"_revision_number":2},{"id":"465271e570ae0383ff0adfae9ac35198899b3703","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-11-30 09:35:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"63c664795d0676867fe29e8f140e30a5ff52a987","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 09:38:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1e6a2f21c80b1a64b04b2d5e0d463d75f7bd10b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 11:25:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8beabdbec6f24835bab97f8a8e445bdb\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/63a3ddd5c6204c59a98892c73aa7bdd5 : SUCCESS in 13m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edafa2718b0e40de8560e9f5e955d078 : SUCCESS in 4m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/724579242a0742258f26467ae6afd1e1 : SUCCESS in 3m 22s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/fffdeed81f334ee79dfd18b81f888282 : SUCCESS in 1h 37m 50s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/297c693f5dba4ea880d016a9839868a8 : SUCCESS in 1h 19m 18s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/65d7fe8f9b4c4ab0a5049da126f0bc82 : SUCCESS in 1h 15m 03s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c3e5f6fcfbe489797708d810037626b : SUCCESS in 1h 15m 35s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42291026fbee4831a6785cb34260fd95 : SUCCESS in 1h 19m 52s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/06aff56c97c141579c365ecd80e6f6e3 : SUCCESS in 30m 57s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c19a7734f5ed40d18675657a96d49629 : SUCCESS in 30m 27s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0dd776f73e6f4fb6af8b203157b9a351 : SUCCESS in 29m 24s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/74b97f358967417183fd5af215688940 : SUCCESS in 33m 15s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/1032a90fae3e4db0aa57f209af1f9e66 : SUCCESS in 31m 28s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-9-stream https://zuul.opendev.org/t/openstack/build/bd86a130351e422db8dfb0d5b5d21ea6 : SUCCESS in 34m 43s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6de446c7dc9b453fbcf20c28afd0afb6 : SUCCESS in 30m 22s","accounts_in_message":[],"_revision_number":2},{"id":"6d160059b642e91899811b4304c0ffe53ab23620","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 11:25:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4f3a1e4a0b633c7a5d861e9a923c6f4db3db2df6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 11:26:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29f637c58878463db2401eb582b066df\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c4bd5a966f146e6bb8fd709e94aa1db : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9385ec8011b8af129eb3354f990695e67897943a","revisions":{"c9d2f147836f1c4baed5e131157a53b84f09fb06":{"kind":"REWORK","_number":1,"created":"2022-10-18 00:09:17.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/76/861676/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/76/861676/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/76/861676/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/76/861676/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/76/861676/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/76/861676/1"}}},"commit":{"parents":[{"commit":"3d25c4f72b724c497d44281ec9dc81b6de120258","subject":"Cleanup unused templates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3d25c4f72b724c497d44281ec9dc81b6de120258"}]}],"author":{"name":"Kevin Carter","email":"kevin@cloudnull.com","date":"2022-10-18 00:02:55.000000000","tz":-300},"committer":{"name":"Kevin Carter","email":"kevin@cloudnull.com","date":"2022-10-18 00:09:15.000000000","tz":-300},"subject":"Add option to disable lxc interface management","message":"Add option to disable lxc interface management\n\nThis change adds a new role default option which will allow operators\nto omit the deployment of specific lxc bridge network config. This\nchange is being implemented because, as an operator, I have a host\nsetup specifically built for OpenStack which includes an interface\nconfig covering the lxc deployment. Currently when running a deployment\nthe role will attempt to deploy a new interface file which at best\nconflicts with the host setup and at worst fails to run due to the\ninterface being in a state unknown to OSA.\n\nThe new config option `lxc_net_managed` is default **true** keeping\nthe existing expectations, but when set to **false** the role will\nno longer deploy an interface file or attempt to bring up the interface\nusing the distro tools.\n\nSigned-off-by: Kevin Carter \u003ckevin@cloudnull.com\u003e\nChange-Id: Icdf4a1f5ff98dc1b86c6a87ea4e606b7c74e1aac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c9d2f147836f1c4baed5e131157a53b84f09fb06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c9d2f147836f1c4baed5e131157a53b84f09fb06"}]},"branch":"refs/heads/master"},"9385ec8011b8af129eb3354f990695e67897943a":{"kind":"REWORK","_number":2,"created":"2022-11-29 16:27:13.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/76/861676/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/76/861676/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/76/861676/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/76/861676/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/76/861676/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/76/861676/2"}}},"commit":{"parents":[{"commit":"3d8e3690ba620d1724129f8ed1a6a040c5ccdac9","subject":"Replace ifupdown with native ip-link","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3d8e3690ba620d1724129f8ed1a6a040c5ccdac9"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-11-29 16:27:05.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-11-29 16:27:05.000000000","tz":60},"subject":"Add option to disable lxc interface management","message":"Add option to disable lxc interface management\n\nThis change adds a new role default option which will allow operators\nto omit the deployment of specific lxc bridge network config. This\nchange is being implemented because, as an operator, I have a host\nsetup specifically built for OpenStack which includes an interface\nconfig covering the lxc deployment. Currently when running a deployment\nthe role will attempt to deploy a new interface file which at best\nconflicts with the host setup and at worst fails to run due to the\ninterface being in a state unknown to OSA.\n\nThe new config option `lxc_net_managed` is default **true** keeping\nthe existing expectations, but when set to **false** the role will\nno longer deploy an interface file or attempt to bring up the interface\nusing the distro tools.\n\nSigned-off-by: Kevin Carter \u003ckevin@cloudnull.com\u003e\nChange-Id: Icdf4a1f5ff98dc1b86c6a87ea4e606b7c74e1aac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/9385ec8011b8af129eb3354f990695e67897943a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/9385ec8011b8af129eb3354f990695e67897943a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
