)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~861744","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2Fussuri~If2c33590258c90ec8d522bad5545b6f626c9772a","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-10-19 16:36:32.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"If2c33590258c90ec8d522bad5545b6f626c9772a","subject":"Use legacy image retrieval for CentOS 7","status":"MERGED","created":"2022-10-18 12:00:31.000000000","updated":"2022-10-19 17:17:27.000000000","submitted":"2022-10-19 17:15:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861744","meta_rev_id":"245514bbe794738f4e5d624794dfd4f5fd391dc1","_number":861744,"virtual_id_number":861744,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-19 17:15:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2022-10-19 16:36:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2022-10-19 16:36:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-18 12:13:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-19 16:06:13.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"38e39458afcd2f4367f598de49409c176e9abf8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-18 12:00:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"42a8fe4443276dcb55161c64432f8cb4bac34f35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 12:13:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a9a2ed9b9314782a13b01e5e100cda6\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4f101e82b45049ccb2ec760e7af699f1 : FAILURE in 5m 20s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/86da9aa0ed424c2d8b240a002e7a862c : FAILURE in 5m 23s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/bfeb1292aa7546a7bbe13a1e4beb4ed6 : FAILURE in 6m 45s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/45da6829e5274fb19db079ac22f71da8 : FAILURE in 6m 34s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2361000e87f46aeafc59ea448c5e21d : FAILURE in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7a7c38f96bc423bb85fe4b18053bc85 : SUCCESS in 4m 30s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/8af9eedf12f44ba7902737ed79f6b5a7 : FAILURE in 7m 14s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3222b7e08fd744d4a73f67788a50238a : FAILURE in 5m 43s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd192ca8cfb04d15b4f24504645b26ab : FAILURE in 6m 07s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8976b6aab0804814baa2e49ff62b9f1a : FAILURE in 5m 30s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8bdb67b32ecb4e8291e819e9a6856761 : FAILURE in 5m 53s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2aac015be330452a9b421457e92c7766 : FAILURE in 5m 14s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/addbee060e094660940da927e61241cf : FAILURE in 5m 53s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7a0ed827bc1a46fab1d55d6d673a1fdf : FAILURE in 4m 52s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e496286c957d410a80197994ea4b3031 : FAILURE in 6m 12s","accounts_in_message":[],"_revision_number":1},{"id":"3d1ce6a2694ef42007d958a7448ac7bad6148e0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-18 12:14:45.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"19ba89666968f2727692c91408b75e48d0a31dd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 12:20:50.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/861742 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"fb7b06c4980a3e618f32ec33af1ef0f5b8eecafe","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-18 19:05:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0de3d4ce6c54b02a814e40dc5c4cf5ffc9d66770","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 19:34:28.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c1512a78e5b456fb08b9e1cff91779f\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9ba8862877ef4fceb83d28e6c52832f3 : SUCCESS in 6m 39s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/1906face689b46ce8993c13515983f81 : SUCCESS in 11m 40s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/ae6133d3f4994063a27c715f487f6289 : FAILURE in 8m 06s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/279513726bda472fb5c07bd5f7f1401d : SUCCESS in 9m 38s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/48d6b90713ff4d378831902eab31ac28 : SUCCESS in 18m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46633d97b6554c76b6e77e84e4d696e6 : SUCCESS in 5m 20s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/9676aebdb3434adab5a308e7443ee674 : FAILURE in 8m 59s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/983aa77568d84e3fa66cefe139ad02dd : SUCCESS in 16m 09s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dfff1d16c46b480ca56f9f4147aad684 : SUCCESS in 26m 21s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fc9e4832943d43f487a47f3a97ea571f : SUCCESS in 15m 15s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f90e59451c23444ebffba09858b3b810 : SUCCESS in 14m 24s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1b09d8c47cc64ef9a8e71a28284b3af2 : SUCCESS in 17m 02s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a719331c63424dd3bce88eb8872158e8 : SUCCESS in 12m 57s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f5d48ae9a35e45f9b03e06718f82e45f : SUCCESS in 18m 42s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/adfd53404b2a49c99607e88c3cc95a17 : SUCCESS in 12m 49s","accounts_in_message":[],"_revision_number":2},{"id":"e33a9f8b0e6da2c42f481992d6c1220536472a08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-18 19:36:27.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"beb58bcbba615fc9c3539d851d4cfc9e2fe0461f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 19:58:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20e68975939c40fba45670bba91b9eba\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8ec49f04180c4bdeb18f95e45e3cde27 : SUCCESS in 5m 47s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/c8794c13564c4a70aa7fb716f3b45bbd : SUCCESS in 8m 38s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/bcaf158a7fda4ed2b37c23ee9a299f41 : FAILURE in 14m 58s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5cc7fd40ed78462db5bb069a71e049f9 : SUCCESS in 15m 28s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bed2d1a49731480aa6a1f595a1d46c6d : SUCCESS in 13m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2436c12795a8424597bc9d4db355380d : SUCCESS in 4m 35s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/826486fefb304cc1a0b17c9a0e08118e : FAILURE in 8m 07s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9221c98ca7dd41db8d6d3f12523984be : SUCCESS in 12m 28s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/395905a15a224eec90ad72187112138f : SUCCESS in 19m 25s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/64f23d4e98be4a62b7638932cadc4f38 : SUCCESS in 16m 00s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a200fc9b0f0a4280865d12fa52cff42f : SUCCESS in 19m 59s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6384df3a6a3347a0b02d63656d9e4f3f : SUCCESS in 15m 54s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/730815dbf4a34c958ba0e186d5ca2372 : SUCCESS in 15m 39s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/028d83787892473eb01a4efb59b088cc : SUCCESS in 11m 45s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02b871cee42b4fb78182c949c26c920d : SUCCESS in 18m 00s","accounts_in_message":[],"_revision_number":3},{"id":"27e6292e018a144ac83b4c97cfcf50f13815ff43","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-10-19 16:06:13.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9432110e1833b3229d9eeb651ebc0661d4b17ec5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 16:07:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a55dc9d0d7c14521a6feb6d42a2b9f55e2f7e667","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 16:20:19.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0155c6225a7040e0b5d8f65cf1062364\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a61df84450884847ab284b9d6ef72a2a : SUCCESS in 6m 39s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/855c540bbd764359b20e55f44130da98 : SUCCESS in 9m 55s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/7ff11cf2cc434605b972554a2f84a437 : FAILURE in 7m 52s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8e82a4b735ec4756ba260ce637c30c4b : SUCCESS in 10m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52d72ba076a54e2fa1b5e67bfe8db3ce : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":3},{"id":"30d72fcc21778c383db18c0ff5ec2f7a4a388455","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-19 16:36:10.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"141e54c91fb24bc75f666d01c5a0bb510a437681","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-19 16:36:32.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fd90aac6d41531b9722354842bbd18fb73871004","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 16:57:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a90bb05a234f4baea32149caffaa3e40\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9206efaff81849ef902cfbec8e9dde07 : SUCCESS in 7m 52s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/943945fab13e4d08b52004de0159c8e5 : SUCCESS in 12m 34s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/8a6cda64966e47319ca0060103a5ff67 : FAILURE in 7m 04s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c53e75b8e478418ca41c1532e62823a2 : SUCCESS in 9m 59s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bfb258d8791542a58a86f27ae8c48469 : SUCCESS in 11m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce5754db22d248c7accb567ee98a758f : SUCCESS in 5m 59s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/1c283ee1c5be4f3ab688a1567c1c419d : FAILURE in 12m 30s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e1ae6fb22c584008bd2c79d1043a00e4 : SUCCESS in 15m 19s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb76884bc275405ca236fdcd72d7c9ca : SUCCESS in 14m 15s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/955bf07d7e1845cd98ca64e55077d325 : SUCCESS in 12m 52s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51018a361642471d83b3ba6225085ff0 : SUCCESS in 15m 18s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/44b8a23e4cb54fa7a559c0e5cd3739a0 : SUCCESS in 18m 14s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad019de360a847829af9b5a7e2fb50f9 : SUCCESS in 15m 32s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e46d77da8f32440abdcf6024e03d27c4 : SUCCESS in 17m 03s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09fdbdc6e6ac460786f2007bd3b26d4b : SUCCESS in 18m 08s","accounts_in_message":[],"_revision_number":4},{"id":"5a02f07ed1391051bcbf430de924bce99fe45848","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 16:58:38.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"8c63b0800c0be527dc48fbb8a44aa208083cfd09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 17:15:42.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c87b7a413b894763a5aee804c0ea1847\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e6a623c14a5d446785c97bc54975cebf : SUCCESS in 6m 17s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/9102354e35b64978a67d7956b2c6d20f : SUCCESS in 8m 30s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/290154b912a949b38dfaaf3e12200ef4 : FAILURE in 7m 39s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f635ac0d17404517a98b7059e1b3e4d7 : SUCCESS in 9m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b459274f26b249288b68b75fc8d3ce9a : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":4},{"id":"4d37cb20587571dc6b28990dbc12ad78582224bf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 17:15:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"245514bbe794738f4e5d624794dfd4f5fd391dc1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 17:17:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7684e3d9727f4ae1afa9eca08d67b2ed\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ba486b1ff374ee6ac2e56c393412d5c : SUCCESS in 37s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a9a84714f55331d36d2328ea5bdc8babec07a2a1","revisions":{"7a4288150598513cd94272be14f012b0aa9d9ac7":{"kind":"REWORK","_number":1,"created":"2022-10-18 12:00:31.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/44/861744/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/44/861744/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/44/861744/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/44/861744/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/44/861744/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/44/861744/1"}}},"commit":{"parents":[{"commit":"2f1e9dc13fe498ab15a55b27d1778ff0a66ab1ab","subject":"Revert \"Add CentOS 8.4 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2f1e9dc13fe498ab15a55b27d1778ff0a66ab1ab"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-10-18 12:00:22.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-10-18 12:00:22.000000000","tz":120},"subject":"Use legacy image retrieval for CentOS 7","message":"Use legacy image retrieval for CentOS 7\n\nCentOS has dropped cloud image we relied on for LXC contaners.\n\nLet\u0027s fetch image from linuxcontainers.org instead to workaround this.\n\nChange-Id: If2c33590258c90ec8d522bad5545b6f626c9772a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7a4288150598513cd94272be14f012b0aa9d9ac7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7a4288150598513cd94272be14f012b0aa9d9ac7"}]},"branch":"refs/heads/stable/ussuri"},"d8741484539fa8dfd47f5640e539411772733ded":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-10-18 12:14:45.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/44/861744/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/44/861744/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/44/861744/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/44/861744/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/44/861744/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/44/861744/2"}}},"commit":{"parents":[{"commit":"2f1e9dc13fe498ab15a55b27d1778ff0a66ab1ab","subject":"Revert \"Add CentOS 8.4 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2f1e9dc13fe498ab15a55b27d1778ff0a66ab1ab"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-10-18 12:00:22.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-10-18 12:14:45.000000000","tz":0},"subject":"Use legacy image retrieval for CentOS 7","message":"Use legacy image retrieval for CentOS 7\n\nCentOS has dropped cloud image we relied on for LXC contaners.\n\nLet\u0027s fetch image from linuxcontainers.org instead to workaround this.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-tests/+/861742\nChange-Id: If2c33590258c90ec8d522bad5545b6f626c9772a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d8741484539fa8dfd47f5640e539411772733ded"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d8741484539fa8dfd47f5640e539411772733ded"}]},"branch":"refs/heads/stable/ussuri","description":"Edit commit message"},"5e7e4bd536c0eb4d5a5e78bfc986a84ff66a1bc6":{"kind":"REWORK","_number":3,"created":"2022-10-18 19:36:27.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/44/861744/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/44/861744/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/44/861744/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/44/861744/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/44/861744/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/44/861744/3"}}},"commit":{"parents":[{"commit":"2f1e9dc13fe498ab15a55b27d1778ff0a66ab1ab","subject":"Revert \"Add CentOS 8.4 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2f1e9dc13fe498ab15a55b27d1778ff0a66ab1ab"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-10-18 12:00:22.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-10-18 19:36:27.000000000","tz":0},"subject":"Use legacy image retrieval for CentOS 7","message":"Use legacy image retrieval for CentOS 7\n\nCentOS has dropped cloud image we relied on for LXC contaners.\n\nLet\u0027s fetch image from linuxcontainers.org instead to workaround this.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-tests/+/861742\nChange-Id: If2c33590258c90ec8d522bad5545b6f626c9772a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/5e7e4bd536c0eb4d5a5e78bfc986a84ff66a1bc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/5e7e4bd536c0eb4d5a5e78bfc986a84ff66a1bc6"}]},"branch":"refs/heads/stable/ussuri"},"a9a84714f55331d36d2328ea5bdc8babec07a2a1":{"kind":"REWORK","_number":4,"created":"2022-10-19 16:36:10.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/44/861744/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/44/861744/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/44/861744/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/44/861744/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/44/861744/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/44/861744/4"}}},"commit":{"parents":[{"commit":"2f1e9dc13fe498ab15a55b27d1778ff0a66ab1ab","subject":"Revert \"Add CentOS 8.4 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2f1e9dc13fe498ab15a55b27d1778ff0a66ab1ab"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-10-18 12:00:22.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-10-19 16:36:10.000000000","tz":0},"subject":"Use legacy image retrieval for CentOS 7","message":"Use legacy image retrieval for CentOS 7\n\nCentOS has dropped cloud image we relied on for LXC contaners.\n\nLet\u0027s fetch image from linuxcontainers.org instead to workaround this.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-tests/+/861742\nChange-Id: If2c33590258c90ec8d522bad5545b6f626c9772a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a9a84714f55331d36d2328ea5bdc8babec07a2a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a9a84714f55331d36d2328ea5bdc8babec07a2a1"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
