)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~861789","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2Fussuri~Iba8e1b95b7ea4326873c872f8fc6cc5bdefecc97","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-10-21 07:17:43.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"Iba8e1b95b7ea4326873c872f8fc6cc5bdefecc97","subject":"Return CentOS 7 jobs to voting","status":"MERGED","created":"2022-10-18 19:39:22.000000000","updated":"2022-10-21 07:40:13.000000000","submitted":"2022-10-21 07:37:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861789","meta_rev_id":"63f33ffb16dc24d72537e1e318266d12deb91711","_number":861789,"virtual_id_number":861789,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-21 07:37:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2022-10-21 07:17:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-10-20 17:54:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2022-10-21 07:17:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-18 20:03:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-20 17:54:58.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"8cc4ac19acc56b140bae4101b561e98b2fc2f623","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-18 19:39:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3fe43e67ed9bec1fc5753d599f171046fbe9e886","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 20:03:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c8b6ca242ed048868f7b74fc0ecba4de\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/360bb7f026c4424196a357a6b012c6c7 : SUCCESS in 6m 48s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/adc99f2e7efb41429b7a7b3b41265fc7 : SUCCESS in 13m 27s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/bbc57533a5e749018cd04a516551df9e : FAILURE in 9m 31s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/88eeaf1f8b1f43309bb4373e1bf82096 : SUCCESS in 11m 52s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/127ec833e8a74fe0a82d471f48e7eda9 : SUCCESS in 9m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97cb7f954fc14db6a41d07d3a987153c : SUCCESS in 6m 59s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/3762678c889a45f0839368cb571fa0cc : FAILURE in 12m 16s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6b67792949f24c998483bfb270a4951e : SUCCESS in 13m 48s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/917a6bbab7644cda9b5da97070768bde : SUCCESS in 16m 38s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/193f9b9435be4c7fa8966048f75bbf15 : SUCCESS in 15m 32s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9bbd25de91bf4f3aba475f5ec5aff010 : SUCCESS in 16m 48s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/81ad61ae70c14aee87375d6961aad1df : SUCCESS in 16m 36s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f99acec24684621a0672079e069a13f : SUCCESS in 17m 33s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0533c95d7d73469fb6df439a8262431d : SUCCESS in 19m 58s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39bc542c045a4ab0a64fbd57d696a907 : SUCCESS in 16m 33s","accounts_in_message":[],"_revision_number":1},{"id":"597cc7baf3f520afd7d46ce6cdfe65c788a43fbd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-20 14:57:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"81f2d8a8de218f4539ccecbff969e0991bf270bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 15:20:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a09dabb906374e4d8722119af0520546\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1d7e8f47956945fc9fd67946c97077d7 : SUCCESS in 7m 23s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/cd1c97e1b1724ed2807be08c83550fb5 : SUCCESS in 10m 27s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/72e23c0b97dd4e67a119a802f259ba51 : SUCCESS in 13m 05s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b7b67ec024cb4aeaaac9a3bb724fe9f3 : SUCCESS in 10m 31s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c686a9168d7a4691b636ece8ee51848e : SUCCESS in 10m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dca52c908504ea88fb8342d3c1474c5 : SUCCESS in 6m 02s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/3973034d42874fe78eb4eff717408542 : SUCCESS in 18m 34s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b60a5e4b6155496db9741922c252417f : SUCCESS in 16m 39s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/375bee4a1ed64ea4a4a3b487e3087fd4 : SUCCESS in 15m 18s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0a83b6c1bcf04998b31a307c295360eb : SUCCESS in 15m 17s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46e9bc5bfe4a406e9f816e4506e1867f : SUCCESS in 16m 38s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2084cbe62fc4480d8f5d4d9b53a81620 : SUCCESS in 18m 23s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2fbd86e1968641ae853ed2db54e70aa1 : SUCCESS in 15m 30s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/44f40f02781c432ba74ac5b9f5f801c4 : SUCCESS in 15m 32s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/780dc13f9bd642b4b9508d074bc77509 : SUCCESS in 18m 41s","accounts_in_message":[],"_revision_number":2},{"id":"87eba4f63bd4b5a5081e286b2aacabdc5a1197ac","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-10-20 17:54:58.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"317e660876e6eaec86738799bb8df269d3a548d6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-21 07:17:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"faa8460c457f14fb8a7229d73eb994347aa2c797","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 07:19:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b63c1d091ca2251d258c85229db8d58be9645556","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 07:37:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6d0e61fcc534515a382e005972fcc70\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ba3ae06a044540dab0d4bbf256c2feb4 : SUCCESS in 6m 11s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/489010ebdccb4b9e87783150f15eb207 : SUCCESS in 7m 49s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/49c7503fba50465295c20897a6eb7419 : SUCCESS in 12m 38s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cd04c31fb1224727bd314c212e8fcb14 : SUCCESS in 12m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77e5f530e80241d3906591ddf523588d : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":2},{"id":"fd03ef3684dd71a8f487f636dcaea23beb446854","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 07:37:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"63f33ffb16dc24d72537e1e318266d12deb91711","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 07:40:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fcc367724b5744a68a0549128027dcad\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0faf4f527db4279b71792beae604304 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cfa0bd92831003bfb3d9aad64b764b3e1d35987d","revisions":{"56a782535d3133fdc2f76e3d2209d13cf58002df":{"kind":"REWORK","_number":1,"created":"2022-10-18 19:39:22.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/89/861789/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/89/861789/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/89/861789/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/89/861789/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/89/861789/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/89/861789/1"}}},"commit":{"parents":[{"commit":"5e7e4bd536c0eb4d5a5e78bfc986a84ff66a1bc6","subject":"Use legacy image retrieval for CentOS 7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/5e7e4bd536c0eb4d5a5e78bfc986a84ff66a1bc6"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-10-18 19:39:09.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-10-18 19:39:09.000000000","tz":120},"subject":"Return CentOS 7 jobs to voting","message":"Return CentOS 7 jobs to voting\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/853029\nChange-Id: Iba8e1b95b7ea4326873c872f8fc6cc5bdefecc97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/56a782535d3133fdc2f76e3d2209d13cf58002df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/56a782535d3133fdc2f76e3d2209d13cf58002df"}]},"branch":"refs/heads/stable/ussuri"},"cfa0bd92831003bfb3d9aad64b764b3e1d35987d":{"kind":"REWORK","_number":2,"created":"2022-10-20 14:57:29.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/89/861789/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/89/861789/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/89/861789/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/89/861789/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/89/861789/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/89/861789/2"}}},"commit":{"parents":[{"commit":"3d19b09595f428427f209bba2efd8b95b34b74ba","subject":"Do not user urljoin for deprecated script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3d19b09595f428427f209bba2efd8b95b34b74ba"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-10-18 19:39:09.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-10-20 14:56:26.000000000","tz":120},"subject":"Return CentOS 7 jobs to voting","message":"Return CentOS 7 jobs to voting\n\nChange-Id: Iba8e1b95b7ea4326873c872f8fc6cc5bdefecc97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/cfa0bd92831003bfb3d9aad64b764b3e1d35987d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/cfa0bd92831003bfb3d9aad64b764b3e1d35987d"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
