)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~862143","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2Fussuri~I11c83628b0c33146a36e58c9de85e4a1a99bd9ca","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-10-21 07:16:21.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I11c83628b0c33146a36e58c9de85e4a1a99bd9ca","subject":"Do not user urljoin for deprecated script","status":"MERGED","created":"2022-10-20 13:25:02.000000000","updated":"2022-10-21 07:39:22.000000000","submitted":"2022-10-21 07:37:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"862143","meta_rev_id":"f4a12153526aea3d72119bad0c87c0059bbcf5ba","_number":862143,"virtual_id_number":862143,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-21 07:37:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2022-10-21 07:16:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-10-20 17:54:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2022-10-21 07:16:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-20 13:49:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-20 17:54:54.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"da87342fd7fc319839f37b3289eb6fe6b4c1c462","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-20 13:25:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"da98f084c8b0b1c66047884db2e372764cb97ceb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 13:49:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bff8e60ca9554781901f7faf7873e90b\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/535582856f9741e3b911b3fff2ffbf74 : SUCCESS in 6m 55s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/f828cc99b473415d986cf7415e7fc658 : SUCCESS in 9m 49s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/ef8022440a7c40609fafb83db7ed7b65 : SUCCESS in 10m 58s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ce7751f09a124f04a5b7d4a4a820fb4f : SUCCESS in 9m 45s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/40802da274644815a30410645f714102 : SUCCESS in 8m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d57a7170567a449eaea9cfa8ec993355 : SUCCESS in 4m 27s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/901b6a607fd240acbea920554ac1c7f1 : SUCCESS in 17m 48s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c517adfbdb7941da9b3fa3003d6853a5 : SUCCESS in 15m 20s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f70c980ee314e04a5d9fd1be98de134 : SUCCESS in 16m 34s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ef69f9cb433a4eb5b2e76c2157113f19 : SUCCESS in 15m 19s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e9fa9e6bb744ccea244ad87dfc8b483 : SUCCESS in 12m 45s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2c0044f1b55a4b11bc205cd989d48661 : SUCCESS in 14m 17s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/30133b69cfca4468a7205254da583639 : SUCCESS in 13m 38s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/abe61a033918465aa1545130772de7f6 : SUCCESS in 15m 47s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa4724b243824e56b774d21cd738b25a : SUCCESS in 14m 44s","accounts_in_message":[],"_revision_number":1},{"id":"277317e96340933268e851f033d92d8fe0fa3919","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-10-20 17:54:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0b0ec51b5c52f5413862a4971592915781d1a159","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-21 07:16:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0899a4df16aafe99cc92dcc55ab1338eaa661f74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 07:17:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"92a6f0b884e163df80834e3ac25e74d8524076e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 07:37:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bb13cca5d0f4fa78e0588999de39c01\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/12f677e1ec284f368e5719ced21911c6 : SUCCESS in 7m 09s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/a7bfa10f925546a6ad27cfe18dbd9614 : SUCCESS in 9m 49s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/32110c712f914741a34fd4e56ea18e9e : SUCCESS in 10m 20s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fba71e199e4048bab6ba4d443ad6e6e1 : SUCCESS in 9m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdb3d94ac54a4787a70c6141f0846290 : SUCCESS in 6m 48s","accounts_in_message":[],"_revision_number":1},{"id":"488827b084be335366614218ecfe6710b4ef382b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 07:37:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f4a12153526aea3d72119bad0c87c0059bbcf5ba","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 07:39:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea2bbf9154a049ac98947c196abb5f43\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8d4a15aa26d46d29164edaaeb3033aa : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3d19b09595f428427f209bba2efd8b95b34b74ba","revisions":{"3d19b09595f428427f209bba2efd8b95b34b74ba":{"kind":"REWORK","_number":1,"created":"2022-10-20 13:25:02.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/43/862143/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/43/862143/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/43/862143/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/43/862143/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/43/862143/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/43/862143/1"}}},"commit":{"parents":[{"commit":"a9a84714f55331d36d2328ea5bdc8babec07a2a1","subject":"Use legacy image retrieval for CentOS 7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a9a84714f55331d36d2328ea5bdc8babec07a2a1"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-10-20 13:24:43.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-10-20 13:24:43.000000000","tz":120},"subject":"Do not user urljoin for deprecated script","message":"Do not user urljoin for deprecated script\n\nThe problem is that urljoin does drop path after FQDN which results in\nwrong result URL. os.path.join on contrary will just drop first argument\nSo we use stupid join of the list through `/` which can add extra one\neven if it\u0027s not needed. But by far it\u0027s the best thing we can do.\n\nChange-Id: I11c83628b0c33146a36e58c9de85e4a1a99bd9ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3d19b09595f428427f209bba2efd8b95b34b74ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3d19b09595f428427f209bba2efd8b95b34b74ba"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
