)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~867203","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~Ia51329110ffa2c634799544ac6c7b7f2016369a5","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","attention_set":{},"removed_from_attention_set":{"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2022-12-12 15:37:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia51329110ffa2c634799544ac6c7b7f2016369a5","subject":"Add git and libxml2 to container base image","status":"MERGED","created":"2022-12-12 09:50:28.000000000","updated":"2022-12-12 15:38:29.000000000","submitted":"2022-12-12 15:37:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"867203","meta_rev_id":"fd970d3ab206d5b223c2c94ab84ecf4f1270bda9","_number":867203,"virtual_id_number":867203,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-12 15:37:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2022-12-12 14:11:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2022-12-12 13:58:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2022-12-12 14:11:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-12 11:38:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-12 13:58:04.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2022-12-12 14:11:43.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"4dd2e6628c96a6b4c73e92d8d0536243332b82a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-12-12 09:50:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f19e5efa435fa54d2e62b5a4797e0e6f95670a2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 11:38:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30ef9a940b664bf088edbdabc1b0e7e5\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f703006a581d4a7d81acef577b014ff2 : SUCCESS in 14m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc9b6c6429d443aca97a0111351cb399 : SUCCESS in 5m 24s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/867b90b6f2d2451aa398bb3226403bd2 : SUCCESS in 1h 21m 46s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2102e38e4035408081fdc0d11f3d5697 : SUCCESS in 1h 23m 07s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/520271985e28413d84508a78f4f9d07c : SUCCESS in 1h 10m 12s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9817dc89e8574435827e5ec3a5838609 : SUCCESS in 1h 05m 49s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30d9bcfd66c14a3fb934029af9c89a73 : SUCCESS in 1h 10m 20s\n- openstack-ansible-upgrade-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/a2db08bb29ae40bf8417605500db3d02 : SUCCESS in 1h 44m 45s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66c3cb30d2c746ec9d508f86444d2a5d : SUCCESS in 1h 26m 25s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7df53ba45ea54cc5a2af5b100b384e7f : SUCCESS in 27m 36s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8230a29b0c7f47ce9c21bc17d9709a0c : SUCCESS in 30m 11s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a8afb528a19487ca7191f74d2bd148a : SUCCESS in 29m 00s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/a8bce4e8b6214847be9a257f0d79ce66 : SUCCESS in 27m 12s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/c3d7029a8c9c41f8ba5a64482a3ff93b : SUCCESS in 29m 16s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-9-stream https://zuul.opendev.org/t/openstack/build/149e7498e24b4799b3055ff55f82dc7d : SUCCESS in 32m 18s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3cbbe2cbe9464958903f1b6122935859 : SUCCESS in 31m 56s","accounts_in_message":[],"_revision_number":1},{"id":"71b1a3c14a45ae93d00fb416dfc4b58740b140ec","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-12-12 13:58:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cf19fb40442c27d277a8f3e262a419671226bdd1","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-12-12 14:11:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"70183524b714148660478154b3ded95714815b20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 14:13:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"81d10f8f1f003b20e4df03895b98d3b2aff77630","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 15:37:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa973e69c3d04a9abcef402d87337288\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/20c19a71142e4638b065eb1bcdd44b71 : SUCCESS in 9m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fe25e4b44c54ea18017d1ce5da296f8 : SUCCESS in 5m 18s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/06e6f3753fb6497d86167a1794a86000 : SUCCESS in 1h 19m 29s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4e9558ee457b4435a102959a449bac7d : SUCCESS in 1h 17m 27s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/e618796a78c14312836b06b6806909ea : SUCCESS in 1h 07m 15s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/619138f0c35a4c30b1e352426edd6569 : SUCCESS in 57m 28s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/657d971778c2481ebf5f1247ee0ea031 : SUCCESS in 1h 22m 24s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aac232accc864adb930297a617a6cb17 : SUCCESS in 27m 41s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/284a8c04bc2c42dbacf72094315f639d : SUCCESS in 22m 10s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ad5cca1833446d8a56d06ca3bba1c43 : SUCCESS in 27m 04s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/287287542937424cb0d19c552dffbce3 : SUCCESS in 32m 52s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/c3cc8815f6bb47afab0dcb8472983c9c : SUCCESS in 29m 12s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-9-stream https://zuul.opendev.org/t/openstack/build/585c050a158e426e9586e62a71376364 : SUCCESS in 30m 10s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/afd1c3c7a5e84336a23f27f9c1b60573 : SUCCESS in 30m 52s","accounts_in_message":[],"_revision_number":1},{"id":"3c9b5aa66bf9d69f4cb9a5824bba094fbb111ff0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 15:37:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fd970d3ab206d5b223c2c94ab84ecf4f1270bda9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 15:38:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49336906a99040379a1f20a3e9e677b9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95a4e5a963544240a93da1f7224630a0 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ca12ef136a1574931b783f15a6a6bc9d31081b49","revisions":{"ca12ef136a1574931b783f15a6a6bc9d31081b49":{"kind":"REWORK","_number":1,"created":"2022-12-12 09:50:28.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/03/867203/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/03/867203/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/03/867203/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/03/867203/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/03/867203/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/03/867203/1"}}},"commit":{"parents":[{"commit":"ef721dbf13db8f3578b45791ac67c4815bd8c59a","subject":"Use correct apt repo when ubuntu architecture is not x86_64","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ef721dbf13db8f3578b45791ac67c4815bd8c59a"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-12-12 09:49:44.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-12-12 09:49:44.000000000","tz":0},"subject":"Add git and libxml2 to container base image","message":"Add git and libxml2 to container base image\n\nThese are needed universally in the service container images so\ninstall them into the base image to save build time later.\n\nChange-Id: Ia51329110ffa2c634799544ac6c7b7f2016369a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ca12ef136a1574931b783f15a6a6bc9d31081b49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ca12ef136a1574931b783f15a6a6bc9d31081b49"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
