)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~868490","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I905609600451d579f3a3e7c732759d9e38b4c63e","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-12-26 15:33:03.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I905609600451d579f3a3e7c732759d9e38b4c63e","subject":"Allow to customize networking configuration","status":"ABANDONED","created":"2022-12-23 14:43:46.000000000","updated":"2022-12-26 15:33:03.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f1c7115b211ff29c4c632981499305d612406852","_number":868490,"virtual_id_number":868490,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-12-24 11:52:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-23 16:17:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"82b13d57b1e781d76fc74198833041c0c651d998","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-12-23 14:43:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6dd752c734fe8d258406654bbf84e37af4544396","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-23 16:17:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7e0a51f462514ba6bcee74bfa302e882\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56ed43a6ebb94ab794fef42cc0ce608b : SUCCESS in 9m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b44d399d07eb47b4a5561e48043760e3 : FAILURE in 4m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a14a4afd5bd547c598701999c241df9f : SUCCESS in 5m 40s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/f096862ac98a4ad482ce05944b4886d5 : RETRY_LIMIT in 9m 10s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ce8b08b116f6469dbce8f8fd868bbf66 : SUCCESS in 1h 13m 37s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/e68394bb29864d5099d3a9f2b3c73355 : SUCCESS in 1h 06m 10s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a7091f73fdc7450dae5e56e941ca0368 : SUCCESS in 56m 29s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78ba8284ce4a49c9b9f8e2fc71f6277b : SUCCESS in 1h 01m 30s\n- openstack-ansible-upgrade-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/c63f438189a5496e9f8c636f79a10bf2 : FAILURE in 10m 18s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6f334d7260547eaa3edbc88303b749c : SUCCESS in 1h 31m 27s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/226602c0d9d34afc92840f9d74d9deb3 : SUCCESS in 19m 48s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90f5f466d2654fadac9268ef940102f6 : SUCCESS in 23m 33s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3212911ee6bb4487bf0e9cce86f1d7e5 : SUCCESS in 31m 26s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/d8027cdcae2e48e2afc86c8b796358a9 : SUCCESS in 25m 47s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/9dbe185e6a254c40ac0fe50b6d69e7de : SUCCESS in 24m 20s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-9-stream https://zuul.opendev.org/t/openstack/build/7285f20489694428bb65c72032c90575 : RETRY_LIMIT in 13m 32s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5af3f7c78406437a9befb319a1a4943a : SUCCESS in 29m 45s","accounts_in_message":[],"_revision_number":1},{"id":"ac0c27aaafafec866ad07c05b81522a56ffb7a91","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-12-24 10:22:06.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"636dcb6fbea4595f27f62b311625317b2ed1ffb5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-12-24 10:31:51.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"84f94fb03f8205a76ffcebb86f815a406a7655c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-24 11:52:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4e625c0c626941a0ae6a57446b9dfdce\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0bb8281e66e64c39b4ae7784e17f3d50 : SUCCESS in 9m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/844ffc2880e04c4aa045f1483148fed6 : FAILURE in 4m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22c5318f5ec346d8a224803a978f1307 : SUCCESS in 4m 29s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc76c761ef094bfcbbae5259d01005e0 : RETRY_LIMIT in 9m 08s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4c220d39ea6b407db44d82d740cbb02e : SUCCESS in 1h 08m 10s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/c6f74a2b43ae4d4780879467bec75928 : SUCCESS in 58m 09s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c48bf05ff31246c390f844a7860a6f81 : SUCCESS in 1h 03m 24s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5860b50c8c7f48d3b3809df5be2d9af7 : SUCCESS in 1h 03m 34s\n- openstack-ansible-upgrade-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/3d9245b617554baaaf78cfdd4576f840 : FAILURE in 8m 45s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd5d8e9eb18d4b1f9ddc24c8e88e508a : SUCCESS in 1h 08m 51s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bdce64a82fc3458cba9b645ae0d2674e : SUCCESS in 23m 09s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/74a20952a73b4001ada8cf7423d0c5aa : SUCCESS in 22m 39s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/50563a0cfe8c4bdcbd6bff0db2d69d98 : SUCCESS in 23m 50s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/16cccb7da2a64adba4fd55d10097c704 : SUCCESS in 20m 47s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/0548973b49074c12a7e19d36ab309fb7 : SUCCESS in 24m 51s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-9-stream https://zuul.opendev.org/t/openstack/build/bbd5976a5ebb4ce99dfbf2c233b5257c : RETRY_LIMIT in 9m 21s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d9b324545b6240ae84da8ad250e4c4d2 : SUCCESS in 27m 07s","accounts_in_message":[],"_revision_number":2},{"id":"b7bd5ad8f068cc90c0153376a828b74df5e54f8b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-12-25 12:01:05.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2},{"id":"986e9747a63ecc365fe02904ae4fefb29f737624","tag":"autogenerated:gerrit:restore","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-12-26 11:23:06.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"86f06b729a6fbcea83d0ae52199434e3b549e6fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-26 12:47:25.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bbff890622ad426dae408a5a13fce9dd\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4bd6525e0f2f4a04aa7ded29300e76c1 : SUCCESS in 8m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc3a5b03ca3f4787bf862c2a93bba0b7 : FAILURE in 4m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b88c1af92ead4d589b937259e2246097 : SUCCESS in 2m 42s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/be67fe1b682543f0b5811ae61a9d1df1 : RETRY_LIMIT in 8m 42s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6699e1ad95ff4118a5472bcdd5aee4ad : SUCCESS in 1h 09m 08s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/9743e6beb4b94d9399dc486dec28bff3 : SUCCESS in 1h 07m 28s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb212f87dcda49228eceae1f2f90bbdf : SUCCESS in 52m 38s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cf39d74d7a1f485d88314ef9443d0551 : SUCCESS in 1h 04m 24s\n- openstack-ansible-upgrade-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/87d94ce97e9e441f8b972b6fc1b0ea84 : FAILURE in 8m 25s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/982ce002cf6b41a2906c1b619c3a0a4c : SUCCESS in 1h 10m 09s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/63b7c3e8238a40498de987906cdbba85 : FAILURE in 19m 40s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1599bf661af34ddba83b5e04c93c8d75 : SUCCESS in 26m 18s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a8a8a66696943e384b2d056e10fe87d : SUCCESS in 22m 25s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/05e8b97607ee47888fe485f658d6b413 : SUCCESS in 19m 25s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/13b337e6847d4977a98852b8ead0f5ff : SUCCESS in 24m 20s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-centos-9-stream https://zuul.opendev.org/t/openstack/build/74ad5870138c469a93f2e0b57487c7bd : RETRY_LIMIT in 8m 59s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/32dd12cd464c4cefbbc2da8505f3cc28 : SUCCESS in 23m 51s","accounts_in_message":[],"_revision_number":2},{"id":"f1c7115b211ff29c4c632981499305d612406852","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-12-26 15:33:03.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"25c4965ac347816ba7d206b30bbf67406ed497a1","revisions":{"ba811bd6526cd60ebb41e7d15243190455663d85":{"kind":"REWORK","_number":1,"created":"2022-12-23 14:43:46.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/90/868490/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/90/868490/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/90/868490/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/90/868490/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/90/868490/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/90/868490/1"}}},"commit":{"parents":[{"commit":"8975a4df068416033fd89ba1019cef172293cf93","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/8975a4df068416033fd89ba1019cef172293cf93"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-12-23 14:43:38.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-12-23 14:43:38.000000000","tz":60},"subject":"Allow to customize networking configuration","message":"Allow to customize networking configuration\n\nFor some scenarios it might be useful to customize bridges that are\ncreated for LXC.\nAt the same time we introduce variables lxc_net_extra_devices and\nlxc_net_extra_networks that allow to define more networking devices\nThey can be leveraged even to create a br-mgmt or br-stor interfaces\nusing systemd-networkd role.\n\nChange-Id: I905609600451d579f3a3e7c732759d9e38b4c63e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ba811bd6526cd60ebb41e7d15243190455663d85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ba811bd6526cd60ebb41e7d15243190455663d85"}]},"branch":"refs/heads/master"},"25c4965ac347816ba7d206b30bbf67406ed497a1":{"kind":"REWORK","_number":2,"created":"2022-12-24 10:31:51.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/90/868490/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/90/868490/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/90/868490/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/90/868490/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/90/868490/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/90/868490/2"}}},"commit":{"parents":[{"commit":"8975a4df068416033fd89ba1019cef172293cf93","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/8975a4df068416033fd89ba1019cef172293cf93"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-12-23 14:43:38.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-12-24 10:31:50.000000000","tz":0},"subject":"Allow to customize networking configuration","message":"Allow to customize networking configuration\n\nFor some scenarios it might be useful to customize bridges that are\ncreated for LXC.\nAt the same time we introduce variables lxc_net_extra_devices and\nlxc_net_extra_networks that allow to define more networking devices\nThey can be leveraged even to create a br-mgmt or br-stor interfaces\nusing systemd-networkd role.\n\nChange-Id: I905609600451d579f3a3e7c732759d9e38b4c63e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/25c4965ac347816ba7d206b30bbf67406ed497a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/25c4965ac347816ba7d206b30bbf67406ed497a1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
