)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~868883","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2Fxena~If040699f64a34adf54fdc0ba1dba7e60d98c577b","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/xena","topic":"osa/tox4","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-01-03 11:00:39.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2023-01-03 10:18:16.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"If040699f64a34adf54fdc0ba1dba7e60d98c577b","subject":"Update tox.ini to work with 4.0","status":"MERGED","created":"2022-12-29 16:28:49.000000000","updated":"2023-01-03 11:01:36.000000000","submitted":"2023-01-03 11:00:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868883-osa/tox4","meta_rev_id":"ccb6b9358ad64fecaa4f9144ee6ea0f66506f518","_number":868883,"virtual_id_number":868883,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-03 11:00:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2023-01-03 07:53:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-12-31 12:26:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2023-01-03 07:53:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-29 17:02:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-31 11:20:18.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2022-12-31 12:26:49.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"e09b496a911db0ba089a8f68fd91b9c17ef0fc7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-12-29 16:28:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"01ca6bd3f0782ff5a21d8f68107ac503a008cd9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-29 17:02:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9e285d8f79694a58be1aa6835d6594b7\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4218d8528bd24688a48fc9ee1bbe2a13 : SUCCESS in 10m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/282da28240c44686bac8b96ede17539c : SUCCESS in 4m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b794f8f1e18c4d2c974aef30119d8405 : SUCCESS in 2m 35s\n- openstack-ansible-lxc-dir-centos-8-stream https://zuul.opendev.org/t/openstack/build/1ac0d9d7ad5440f2bfca9df31301ec40 : SUCCESS in 25m 42s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8268b818a8d04ea791cf4f6fd097d640 : SUCCESS in 13m 19s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc70c383d13740909f388f7eb40d512f : FAILURE in 4m 22s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0bad49f18a1c4f1e9d2dd5a4baa00e2e : SUCCESS in 14m 41s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5de76d7851e4d8e8bf81c56fd2c138b : FAILURE in 3m 58s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aec93a66f44441eb90cfd2153a37596f : SUCCESS in 14m 55s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7444699317904fed8471e72da0b01202 : FAILURE in 4m 41s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7ea111afdb524f5b84a516cdee13a95b : SUCCESS in 16m 09s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f60d42510b1466ab7d7a1b3c130a08c : FAILURE in 4m 48s","accounts_in_message":[],"_revision_number":1},{"id":"c77337d758fdde0abcc28e9c2bcec5948da61798","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-12-31 11:20:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b91f0f871125465aa3db57a907639cb981fd7e07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-31 11:50:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d01c6ee3bbc407e9833ce62889003fb\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/985c38228b42497ab275ff5d4bf6671c : SUCCESS in 9m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea71d27dbb20437da6388762024bd700 : SUCCESS in 4m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef7edfef9b234a28acd75344408c1dbf : SUCCESS in 2m 45s\n- openstack-ansible-lxc-dir-centos-8-stream https://zuul.opendev.org/t/openstack/build/4dbed0530eaf492c9a0d79869dcf8c5f : SUCCESS in 18m 40s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f0d5f8dbcf3a41cf87cda8c3f0852739 : SUCCESS in 12m 54s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3879b04f65d248dbad012b4799d34d97 : FAILURE in 4m 12s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5a76ff1aa3214c8d960f82ac68ee6836 : SUCCESS in 15m 06s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce4a31a8de2042c5b801053e95ae0735 : FAILURE in 4m 31s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9018b4f5507142aab11e98c68bc762f0 : SUCCESS in 15m 30s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f48b6f8861e4dc8b1dcba3fc5698e40 : FAILURE in 3m 55s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/225d0b95ce234ef49485fe24b75624c4 : SUCCESS in 15m 10s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/73a70d1f596c4919aa3bd64cc40a797e : FAILURE in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"8c4768feb8e84e130d099d60fb4b5261530371f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-12-31 12:00:58.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"06441f0f65d4886390ad08b80c3c0d21b2666679","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-31 12:24:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66edf2a1c55a485abe1a55ffb2a4461b\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/63f6428e714b4f6eb34d66c821ea6ba5 : SUCCESS in 8m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f496f467a7284f0d9f4c37407aa36a2f : SUCCESS in 3m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab589173bddc47579d5901fd9a7fe5c2 : SUCCESS in 2m 51s\n- openstack-ansible-lxc-dir-centos-8-stream https://zuul.opendev.org/t/openstack/build/b0aeabe9678d480e8485d48cea80d6bd : SUCCESS in 19m 31s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1437e6366b4d447594107b37604bef6b : SUCCESS in 14m 10s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/112ccc1995a14dce8f066ebe3700f96e : SUCCESS in 15m 12s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/06d4ad3b505b486995b978594e3795b5 : SUCCESS in 15m 15s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9efc438b86d64c0291e779be12ea9107 : SUCCESS in 13m 11s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/17f6dfbc06784d3a8286d3ecfd07854d : SUCCESS in 14m 18s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/07a6e6d83ba14464bf64a399ff95481c : SUCCESS in 13m 27s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9101d46ad85f4cb9af727e8cab5e66b0 : SUCCESS in 13m 46s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3586dcdf21341139c8320d09bbe37b1 : SUCCESS in 15m 17s","accounts_in_message":[],"_revision_number":2},{"id":"bf2c5a4756ff4eceb55f1e2d9a90839e8f14abff","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-12-31 12:26:49.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"01a416dee6785dbdc4e21f6daae885a2693f0668","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-01-03 07:53:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"13083a6820f5eb76d8632edd75ee6d9553b2cf0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 07:54:04.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/32c16ba09fae4d769fd781ca29ae1cb4\n\n\nWarning:\n  Change 868880 in project openstack/openstack-ansible-lxc_container_create does not share a change queue with 868883 in project openstack/openstack-ansible-lxc_hosts","accounts_in_message":[],"_revision_number":2},{"id":"3c6acb9c7756482af4f32ed804155a0dba22ae86","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-01-03 10:18:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"79b9731eb87722b4cd1aad641f0f1c3e983f1906","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 10:43:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53d93a84eea94385ba8f2f321ed3e2e1\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/74e719ff7441459882f4296f5314738f : SUCCESS in 11m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/233da5f0f7474cb6b6d749f94e6919ee : SUCCESS in 4m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd587342e24b44c4a73ddbcbb93527fc : SUCCESS in 3m 08s\n- openstack-ansible-lxc-dir-centos-8-stream https://zuul.opendev.org/t/openstack/build/706a46c0994f4c538ea070373f46568c : SUCCESS in 21m 47s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/525d6bbdae164eacb3627f42b6fa1f59 : SUCCESS in 18m 09s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb5de871f3cd465992e856699057932f : SUCCESS in 15m 56s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dd6aa7b055744379aa6d910b1cd3b1bd : SUCCESS in 14m 47s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/662ee230fb6a49d4a246d94524aaa611 : SUCCESS in 14m 54s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7fb3290fb6a446f096ca8a774c18968b : SUCCESS in 14m 26s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f19aa73d3554c8f9ee8e121bfd18177 : SUCCESS in 15m 40s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/72fa0b6fd8ea4ef8bc6ef974b7bcb2e3 : SUCCESS in 15m 18s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6acb66adaa214a2d9b49baf38caafaf1 : SUCCESS in 14m 26s","accounts_in_message":[],"_revision_number":2},{"id":"8dde705b37bbf712aafd3aa9b01aa6eae978cefd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 10:45:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f4fa896aa4c1b03ef7d8dd65f0dc979bbf278dbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 11:00:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c9ca58a99df40e3a8f673aaecd04db4\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/688cbed5149c4e19bbac3db10d4fb519 : SUCCESS in 10m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db681d5191cc4cae8cb6c92273d0bc29 : SUCCESS in 4m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e5a4b5461364c879026fb83daeb5061 : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":2},{"id":"ca7f57b01024a82eabc82ad43d7e756f562f67e1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 11:00:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ccb6b9358ad64fecaa4f9144ee6ea0f66506f518","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 11:01:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b48bb2df91814521b77517ec3c41638a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7da84eab8821481aba10791b19335c21 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"16bbf43c88e25a89f9a20520fb562ab9b39aea16","revisions":{"8f10de6566f631cf6f8a26e819093440e6f3b788":{"kind":"REWORK","_number":1,"created":"2022-12-29 16:28:49.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/83/868883/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/83/868883/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/83/868883/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/83/868883/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/83/868883/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/83/868883/1"}}},"commit":{"parents":[{"commit":"fea49e30bcb34b7b229d6f586ba2c207b6f3f79b","subject":"Prevent lxc.service from being restarted on package update","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/fea49e30bcb34b7b229d6f586ba2c207b6f3f79b"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-12-29 16:28:47.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-12-29 16:28:47.000000000","tz":60},"subject":"Update tox.ini to work with 4.0","message":"Update tox.ini to work with 4.0\n\nWith tox release of 4.0, some parameters were deprecated and are ignored now\nwhich causes tox failures. One of the most spread issues we have is using\n`whitelist_externals` isntead of `allowlist_externals`\n\n\nChange-Id: If040699f64a34adf54fdc0ba1dba7e60d98c577b","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/8f10de6566f631cf6f8a26e819093440e6f3b788"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/8f10de6566f631cf6f8a26e819093440e6f3b788"}]},"branch":"refs/heads/stable/xena"},"16bbf43c88e25a89f9a20520fb562ab9b39aea16":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-12-31 12:00:58.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/83/868883/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/83/868883/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/83/868883/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/83/868883/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/83/868883/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/83/868883/2"}}},"commit":{"parents":[{"commit":"fea49e30bcb34b7b229d6f586ba2c207b6f3f79b","subject":"Prevent lxc.service from being restarted on package update","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/fea49e30bcb34b7b229d6f586ba2c207b6f3f79b"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-12-29 16:28:47.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-12-31 12:00:58.000000000","tz":0},"subject":"Update tox.ini to work with 4.0","message":"Update tox.ini to work with 4.0\n\nWith tox release of 4.0, some parameters were deprecated and are ignored now\nwhich causes tox failures. One of the most spread issues we have is using\n`whitelist_externals` isntead of `allowlist_externals`\n\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-lxc_container_create/+/868880\nChange-Id: If040699f64a34adf54fdc0ba1dba7e60d98c577b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/16bbf43c88e25a89f9a20520fb562ab9b39aea16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/16bbf43c88e25a89f9a20520fb562ab9b39aea16"}]},"branch":"refs/heads/stable/xena","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
