)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~888750","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I5546ee71cce4f4b40fbd1d38d5d49586606bbbda","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-08-18 08:34:10.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2023-08-18 06:18:11.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"I5546ee71cce4f4b40fbd1d38d5d49586606bbbda","subject":"Add retries to LXC base build command","status":"MERGED","created":"2023-07-18 06:41:44.000000000","updated":"2023-08-18 08:35:17.000000000","submitted":"2023-08-18 08:34:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"888750","meta_rev_id":"dfec0a9c87ddd823b947f852ffa11ba78f622cef","_number":888750,"virtual_id_number":888750,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-18 08:34:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":2,"date":"2023-07-20 07:27:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-18 06:18:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-18 06:18:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-18 08:23:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-18 09:49:57.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2023-07-19 06:29:21.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"20d146da2afddf33f693412784d785c5f3e75554","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-07-18 06:41:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5facba37eea7c35fd13c833fb150eaabaebd1820","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-07-18 06:41:53.000000000","message":"Topic osa/core-2.15 removed","accounts_in_message":[],"_revision_number":1},{"id":"5ce17d5ab0655d62d018fbe37883312261f357c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-18 08:23:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8952a40bbfaf4ec989a697ee0c4c6539\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70bab9f339ef4822b01c3d3a6896dbd7 : SUCCESS in 9m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/525fa8ebb3ba4634929c32b146df70d8 : SUCCESS in 4m 32s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/3e56f6d8362b4a079eeb5da6611b1bb6 : FAILURE in 35m 26s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/10871c593472402aa92d648247a48346 : SUCCESS in 1h 33m 52s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/00e879e971a6449eb6f66b0cae9a73bf : SUCCESS in 1h 08m 24s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/a72598bc52104afaaecc2600b9045bc7 : SUCCESS in 1h 13m 21s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eccc3f7c59cf403eabeaeb82d09de385 : SUCCESS in 56m 16s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7fa9a6d0c8224ae1a8ca70a4e5ecb526 : SUCCESS in 1h 01m 52s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d564468ed4842379ece518a6570a0b8 : SUCCESS in 1h 07m 37s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1e63cb2bc9654714916703adb9fe2538 : SUCCESS in 1h 03m 24s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2ea5221cad8c44059199d162726e0044 : SUCCESS in 1h 29m 28s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/725a9e2d04b94ef5882e071bf948c6a2 : SUCCESS in 1h 23m 41s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c46dd1fe9fc469d91ed0a016ab66390 : SUCCESS in 29m 28s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe0ee88c597e480397e538294b60419c : SUCCESS in 25m 26s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef36fe8cb48042df86f3b5f6c4c12fd9 : SUCCESS in 29m 39s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/c3ce80b0cc634e0391c4080bef6b9eb1 : SUCCESS in 23m 12s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/9eae0fd3425444049f570f52d1af817c : SUCCESS in 26m 44s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2b16b6b0a0654ecf80d32e78393223f7 : SUCCESS in 31m 10s","accounts_in_message":[],"_revision_number":1},{"id":"f9a3ca422b4d2379cc409c2be5b5882c92b9e1e6","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-07-18 09:49:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"886564f50ab648539c22f61477ffff18a7ad205b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-07-19 06:29:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e2fe4e6bd637f778cc9114aec068b97ba80072d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-07-20 07:27:33.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"dd056af8aad571532dac1b1d84dcac4d32fbf334","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 10:07:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2ba10b1d590941bf92f6fb35521d2307\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5f35b055066f417cb4b6641ce8f663e9 : SUCCESS in 10m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7674fbe4827742d58bc579d087a66f6a : SUCCESS in 4m 57s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/2505535039d945fab8bacc3f8f24626e : FAILURE in 39m 50s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e9cf399c7100448fae587d9bbf0583cc : SUCCESS in 1h 36m 15s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/658cb728445f4bc2bafa5cb9ac4296f8 : SUCCESS in 1h 43m 39s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/94e5d46d8e4e43c2bcc0430eac5277a9 : SUCCESS in 1h 19m 49s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d99badedbbc746aa98e0af8c22ee8722 : SUCCESS in 1h 25m 57s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d83b84cfe03412c83f3799a5b060df5 : SUCCESS in 1h 19m 31s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4f8bbe560aa84d4cb9da1c1a0af26fae : SUCCESS in 1h 22m 42s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2a5703a0ec754e409eb3b17ab582b3f0 : SUCCESS in 2h 25m 28s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf41616fc78d48669e53fe6a8321609c : FAILURE in 1h 32m 42s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8876759ef4c34edc904216b8d97f316d : SUCCESS in 32m 42s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1462e0e181b84c99817467f205a20a7d : SUCCESS in 27m 25s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7b9fc1cb844412489c4c93f464ef08c : SUCCESS in 33m 52s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/64fb4b0ee8c24f049331772e66cab9f4 : SUCCESS in 24m 46s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/8ef2389ea7d94160a95307593bda932c : SUCCESS in 30m 22s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/44fc16e1c38d44c9a6a91bb08a9a4046 : SUCCESS in 37m 13s","accounts_in_message":[],"_revision_number":2},{"id":"178684632b8c921046f46b4b7057a6446eb2f1fa","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-17 14:52:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6c8424e2e899891648dc8a4134bae9ac70b6a66c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 20:14:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7ee0435326c49dcb236a6c02863a074\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/539ddfd7dc1541fd874ee68f43e1e7b3 : SUCCESS in 12m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc366f648ac44be4ba2209822d9e9b1a : SUCCESS in 4m 36s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/1d515cf4bfe84392b10a6b8fdef60fbf : SUCCESS in 1h 29m 03s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1e7ac023e8f6462b9cbda577beffe760 : SUCCESS in 1h 28m 03s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/865c1eb152a548a19de58b441f2de2cf : SUCCESS in 1h 33m 49s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/a88552bd699f40d38990430714a7ba59 : SUCCESS in 1h 32m 02s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fabd65e5d5984ec690fe4d9947de9621 : SUCCESS in 1h 13m 55s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7f27faffbad4467949c520110dbe79e : SUCCESS in 1h 23m 00s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8204831324de46cb8fff2b06408854de : SUCCESS in 1h 21m 09s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9f22289579ea41d0a881e7b8fa4e7f67 : SUCCESS in 2h 17m 18s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d919e64c3bcd4508ab09f8ea7e746629 : SUCCESS in 1h 45m 39s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a46077eaf9804604b062c4468c1be58f : SUCCESS in 34m 29s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dffa9ea945f44edf9f19b91763fd3e23 : SUCCESS in 28m 57s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/35cec22433b14e4cbdbf7a5d64126b86 : SUCCESS in 25m 13s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/48a55dc140b746a3bac4417336ef9b4c : SUCCESS in 28m 15s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/285f279164514b6ba276111c26bce9b2 : SUCCESS in 32m 27s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ec007b3b10414ec6a9dbf7be3d90fe67 : SUCCESS in 36m 11s","accounts_in_message":[],"_revision_number":2},{"id":"a739e90becbcafaac5e0f0668b61e063700ad570","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-08-18 06:18:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"745e46a0523675d388a4d97e9a620207a0519119","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 06:20:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"78b0be144920ea239c13f55fbfcd96745ce2bde5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 08:34:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cb44441db2140cfa2c9a4fc415dec41\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b61b53f12fce442a89f98c6a37171474 : SUCCESS in 11m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92c0fe46d34741b583f86107b355502a : SUCCESS in 5m 17s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/20171c1ce1704823bb6850ce076ac9ef : SUCCESS in 1h 35m 38s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2adb854b08714229824f89ece2e8a098 : SUCCESS in 1h 25m 11s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/e72309f2d76c48b38e780c346990806b : SUCCESS in 1h 01m 34s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/031af58c59314e4aae2ecfbdd7bb0587 : SUCCESS in 1h 14m 54s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/947d602addfb40a39ea28156ad5b3a9a : SUCCESS in 1h 06m 50s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9cc34a5ada984d62bec44ef016fdd67d : SUCCESS in 1h 21m 20s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/eaa48644dc9b4ca09b6878456335a763 : SUCCESS in 2h 09m 35s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c3fae28b3eb6476894597d758924de6e : SUCCESS in 1h 31m 54s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dcffc4ffe16b4b4d878f38adf115c4c2 : SUCCESS in 24m 36s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f0afba201bed48949ce656b74c1fd962 : SUCCESS in 29m 14s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0d74dfb8130d4596b5700a4378abd780 : SUCCESS in 27m 55s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/7f71053b51864451830f9f9a7a06ee02 : SUCCESS in 24m 48s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/7832129e40214b32a5b56527090f9133 : SUCCESS in 30m 01s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/49bf5eb5763740ce933a930a2ddb947e : SUCCESS in 29m 10s","accounts_in_message":[],"_revision_number":2},{"id":"50baa266480f7a05f139595c5f6a5c72353dcdff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 08:34:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"dfec0a9c87ddd823b947f852ffa11ba78f622cef","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 08:35:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6112c18b7f634ff5a98395ca3f61edd4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0ff590c1d924caa94bc78b50200602c : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4686fac8623298bbbf7b3fc8254f308dacda1e38","revisions":{"aa9fdadfb8d37fe100d4eefc92b890f64b043bb3":{"kind":"REWORK","_number":1,"created":"2023-07-18 06:41:44.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/50/888750/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/50/888750/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/50/888750/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/50/888750/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/50/888750/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/50/888750/1"}}},"commit":{"parents":[{"commit":"4ecd67ac87b7f65310d43671a48f40400fefca87","subject":"Fix linters issue and metadata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/4ecd67ac87b7f65310d43671a48f40400fefca87"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-18 06:41:30.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-18 06:41:30.000000000","tz":120},"subject":"Add retries to LXC base build command","message":"Add retries to LXC base build command\n\nSometimes there could be intermittent issues on some of the\nmirrors that would be picked while building the base image.\n\nIn order to increases chances of image to build, we add a retries\nto increase chances to pick properly synced mirror.\n\nChange-Id: I5546ee71cce4f4b40fbd1d38d5d49586606bbbda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/aa9fdadfb8d37fe100d4eefc92b890f64b043bb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/aa9fdadfb8d37fe100d4eefc92b890f64b043bb3"}]},"branch":"refs/heads/master"},"4686fac8623298bbbf7b3fc8254f308dacda1e38":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-07-20 07:27:33.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/50/888750/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/50/888750/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/50/888750/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/50/888750/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/50/888750/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/50/888750/2"}}},"commit":{"parents":[{"commit":"a5589beb5f64e01fecffefbcd398a34f6566e101","subject":"Fix linters issue and metadata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a5589beb5f64e01fecffefbcd398a34f6566e101"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-18 06:41:30.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-07-20 07:27:33.000000000","tz":0},"subject":"Add retries to LXC base build command","message":"Add retries to LXC base build command\n\nSometimes there could be intermittent issues on some of the\nmirrors that would be picked while building the base image.\n\nIn order to increases chances of image to build, we add a retries\nto increase chances to pick properly synced mirror.\n\nChange-Id: I5546ee71cce4f4b40fbd1d38d5d49586606bbbda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/4686fac8623298bbbf7b3fc8254f308dacda1e38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/4686fac8623298bbbf7b3fc8254f308dacda1e38"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
