)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~889945","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I25729462fa6be7953e8ef0687ec4580509c21aaf","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-10-13 09:51:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I25729462fa6be7953e8ef0687ec4580509c21aaf","subject":"Stop installing openssh and rsync to containers","status":"MERGED","created":"2023-07-28 11:06:19.000000000","updated":"2023-10-13 09:52:05.000000000","submitted":"2023-10-13 09:51:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"889945","meta_rev_id":"36966be3583a235a1a540d888950f05287fd37c2","_number":889945,"virtual_id_number":889945,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-13 09:51:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2023-10-13 07:11:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-12 19:13:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2023-10-13 07:11:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-28 12:43:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-21 11:08:52.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2023-10-12 19:13:26.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2023-10-13 07:11:42.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"9b1d1a9405fa6b4a205920b666e06c3dd7af1e28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-07-28 11:06:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"872734b21045b3c8d062f87827127d6bfe004008","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-28 12:43:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/49a1de71b8de4ba1aef2fd75a808076d\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66c7669e3f5f4122a940b6e6538fa65e : SUCCESS in 11m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73cf098fb4c34b8cbcc68b283f62311b : SUCCESS in 4m 19s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/ceb8081b2fcd4d7b9c6b5a4b64488619 : FAILURE in 25m 15s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3b0db0a832cc46f79d8e5922938001ab : FAILURE in 28m 00s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e419e198433d4b77a98b910814239f93 : FAILURE in 29m 09s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/51a98b2d5bc141c98abd23be96ddb810 : FAILURE in 15m 51s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d4850b286a847dfa4ea366cf52af30b : FAILURE in 19m 22s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c1f14b867e2148f68bd655b0457129f3 : FAILURE in 18m 54s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc8c363b8e794b5f8811188d9b3d9081 : FAILURE in 16m 32s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/70ae6f209dc04d019a1d3e5f6a8aaa5a : FAILURE in 1h 20m 42s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/70c280f6e3c34708bc335f1ae220cfdb : FAILURE in 1h 15m 59s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e364d5aebd7440fa7fadbafe1558787 : FAILURE in 19m 53s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d75ce00fe7a4abbb945cea01756d260 : FAILURE in 20m 13s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f609cee9d52b4f39b65de05b15e9846a : FAILURE in 20m 26s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/af6f84221e81422b9d6bb89ce11b2d91 : FAILURE in 20m 42s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/4b6e54cfd42e4920b87489a8f09efa1d : FAILURE in 19m 36s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ef606ecb06ad4bb3a1ee772c488812fc : FAILURE in 22m 23s","accounts_in_message":[],"_revision_number":1},{"id":"f7ca0606aedb9dd09b069240a31b72a12b2e7752","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-15 14:14:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f4796aecd9cc97c49480182da8707aee3d870972","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-15 16:33:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cf17e7e9e43e4b56afaeac59023e43d5\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ecd694fb21e64b9babd66e62d658f734 : SUCCESS in 18m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfb57abd5f9346a5a2dc27dee9a7ea28 : SUCCESS in 4m 06s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/75fc452e276a45a4aa27e6296587e8da : FAILURE in 25m 55s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a9191b79776249f194a2fe9804a281b7 : FAILURE in 27m 29s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0a91476c37544fde884b321aa2962448 : FAILURE in 27m 33s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/d3724b1009a84868b34d5c554afbbe7d : FAILURE in 19m 37s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d8ae01e7c3304a8f945e3de4b4340efd : FAILURE in 19m 18s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bbfee2751ed44bcb91f982e8f5db8f8b : FAILURE in 18m 22s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6be55c2d8e904a91894d0770757eb44a : FAILURE in 19m 35s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0fccfda31ddd45cfbbebad081ed90c8a : SUCCESS in 2h 17m 09s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/94b30016342b43928057a47a0ba42677 : SUCCESS in 1h 33m 38s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4315f278b6004fa78bdfceb30986655f : FAILURE in 20m 26s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/45179b87addb416283be8b233b3fd079 : FAILURE in 25m 44s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b58cf5f71dcc478cb2496a5dd6c39d6d : FAILURE in 21m 52s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/f23557f9cfe748c7abd88f5094d6d89f : FAILURE in 26m 42s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/0fe212d028cb40a1a01e18a331b9a002 : FAILURE in 21m 42s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b4e0720cfe5445e29314addd815f5775 : FAILURE in 24m 22s","accounts_in_message":[],"_revision_number":1},{"id":"d832b82a13a31691f9e2a80917562e880fdb5492","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-17 10:30:14.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a3b98caddaf3d163f6e8408a05f7adafab03a5a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-17 10:30:46.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"a29c82adb1f6c1299614508194a6694a97dd01fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 12:35:15.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9a19c55fb43544e684daf18e8fae1f3d\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a8abe39c81d54e8b9c797e40989ffa40 : SUCCESS in 12m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcd36c7a0baa4e749b5d4891305affc3 : SUCCESS in 4m 30s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/9328a86b86af4c10b7b590f2047c1157 : SUCCESS in 1h 26m 30s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f827aab2388f49bca5c544073b051d77 : SUCCESS in 1h 28m 36s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e2c7cec4d7ed4b48bbfcb028038f2e2e : SUCCESS in 1h 31m 34s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/b7a5194b6b2749e3b7854c18c4d0c18e : FAILURE in 19m 20s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/47e15da0fed84d948d3faf09ac395334 : SUCCESS in 1h 18m 26s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c1be5f95fde413781b9f6294b2ce7da : SUCCESS in 1h 22m 06s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a5deab6d911144bd8f646a0d02de91ad : SUCCESS in 1h 23m 21s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/64f0c252385b470db1b00b530f88968e : SUCCESS in 1h 58m 54s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f8ffca9db3604d37abeeed7eaaf77f77 : SUCCESS in 1h 34m 28s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b00a3d5ff8c46858a3b6d37ea8cc7ba : SUCCESS in 29m 09s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ddcb805649744c3aa427ecb9764f661f : SUCCESS in 28m 25s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/71ac1b0c73af4e1cb9a15e32d80b9af4 : SUCCESS in 27m 45s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/243674aa836840188a9e1406356099db : FAILURE in 22m 52s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/ec81c72a063748e3b66706a9250d1e24 : FAILURE in 24m 49s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/34afbc01929540c1a1f40c0c4e1231a6 : SUCCESS in 28m 54s","accounts_in_message":[],"_revision_number":3},{"id":"72f8cc099f067b41617e96c24224a83a4a6bfd52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-18 07:44:21.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"39e4fe3a7876b2d366a6ff7a4035e07d19a19a51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 09:49:50.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d881d74a11d9409d91fa53e47ce85759\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/175de22f22bb4fb89aa9e8e069718b92 : SUCCESS in 12m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/246749be4ca54eaf9ddef682abed9b49 : SUCCESS in 4m 21s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/5b60c9fef7e34c0fba462bdd8538d7ab : SUCCESS in 1h 25m 08s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bad62bbe0ec4426eb143d3982a2c77eb : SUCCESS in 1h 32m 25s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/79f0f4fccc3e4e62bc0e938687b28f8d : SUCCESS in 1h 22m 35s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/8cd358dcc18742c19945b7a652562c28 : FAILURE in 19m 02s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6afc258b1d3b4701867c392c27480c23 : SUCCESS in 1h 17m 47s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/710bda93dc7342f89bfdbc525c321b91 : SUCCESS in 1h 14m 11s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a28cab720f044a86a0332f28e173714a : SUCCESS in 1h 19m 22s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e108fdd8badd4402abc70f2a36dd0f1c : SUCCESS in 1h 59m 12s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b532a9ebfd1740d9aedc26a1b0932b05 : SUCCESS in 1h 37m 24s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d6b2eda1e30e443eaf2c1feb4005c6dd : SUCCESS in 25m 32s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c1edc921733497790c5e8c6f6a3cabe : SUCCESS in 26m 23s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/766ba5e84c744eb78bf78e64cbf165f6 : SUCCESS in 31m 22s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/61d6c02a36be4a6b905c7b6ed937cd66 : FAILURE in 20m 05s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/521365b2c4aa4f7caea0010503720781 : FAILURE in 20m 18s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/63082d4268264480b887d29e91c05e6f : SUCCESS in 29m 37s","accounts_in_message":[],"_revision_number":4},{"id":"054708256866f0e7cba3d4ae729b0b70b3c78aca","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-21 14:09:33.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b20735c14b8a2e17081fc6016f73168adb599643","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 16:02:41.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/de3725acea6d4af396dacd2443e19bdd\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9521d60fc57047d6be1f9200dcbfbbf9 : SUCCESS in 12m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d5bc5a1c9f447c291ccc6da78cbe493 : SUCCESS in 4m 47s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/af083649542c44f0b5222b9be9e52188 : SUCCESS in 1h 34m 04s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d30b42d15b624e4bad26d21c6f1c4fd5 : SUCCESS in 1h 40m 03s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/98c5bdeb762d49c09e418049b8d7b9f3 : SUCCESS in 1h 21m 13s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/4dcf39ac89484cec883f5fba7c4be28e : FAILURE in 21m 22s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/586e645e866a42d5b034604ba0c3527b : FAILURE in 49m 29s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a2dbd4bf982456d84ee86d4ecd20f40 : SUCCESS in 1h 21m 08s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7f5068a58dd4b8daa9e4f5784274abf : SUCCESS in 1h 18m 19s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1ef2ffa8234449de83e493c26526db35 : SUCCESS in 1h 49m 19s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5493d20a07254597b0816817755224b1 : SUCCESS in 1h 41m 12s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e08d9ea6687e41f9bfa39cffc115cd5f : SUCCESS in 25m 13s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea5419dc7ee1434f96067d5ef44c65db : SUCCESS in 31m 36s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a5d8b4061eb4a91916cfa40eaaf1ed8 : SUCCESS in 27m 09s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/ecdec973753340dfbbeef4aae247d2bb : FAILURE in 22m 06s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/2b3d650c0e214d6f8d7d5ca83ed2655a : FAILURE in 22m 56s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8078574dfac14816b60f9b3d975c6174 : SUCCESS in 30m 31s","accounts_in_message":[],"_revision_number":4},{"id":"3b8abb1035b96a8321dd8d3bdc86cf922ebec608","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-09-05 06:29:54.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"702ebb18265197d708bb024eac61a2308e306586","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 07:59:29.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b121bd2ddd75435fb18bfaf408cd5492\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/17be7efd798142b085adb78121caf04b : SUCCESS in 10m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/589d1e6633484bd1aa3fd050a0cb6e7a : SUCCESS in 4m 18s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/8f8b2ecd83fd45cf85583bb78085c773 : SUCCESS in 1h 17m 41s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0178923de13648e8b0e58c1ee3705a18 : FAILURE in 22m 41s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9fff2d71e63647418bc2c413eca197ac : FAILURE in 22m 10s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/5596cd18641e46b0add9a6bfd44a8f52 : FAILURE in 19m 20s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a04f09fe3c0477ea4d0f876f07f0f9e : FAILURE in 25m 24s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b5b3bc8e1a547c49d8a4860edd7479a : FAILURE in 30m 18s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1ac4214ec3f24bdda7b26623de9ec455 : SUCCESS in 1h 19m 34s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/39999f5977104ad1bddf316bc6a5000d : FAILURE in 10m 47s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f5b810a7aac247d980841106cb751130 : FAILURE in 25m 49s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7715df3c632d4db7b4931a4dae9fd771 : SUCCESS in 25m 12s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/853c25710621405b9071a605291fede7 : SUCCESS in 26m 35s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb7f0c6e3329466fb65b5b21f85470ca : SUCCESS in 24m 29s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/53d2376e595a47589cde4a60a2f9a29c : FAILURE in 21m 00s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/13ce723e6545482b83caeac548004c15 : FAILURE in 20m 36s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cb8c759529964e009f6fab4b051051c2 : SUCCESS in 29m 20s","accounts_in_message":[],"_revision_number":5},{"id":"013a7e093b98142de349a3988711bac0f2d51c55","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-09-21 11:08:52.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2d84475850785993ab6110da172c742d8feffd40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 13:04:19.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9731065ef1ad4f2eb3f4263d0f29450f\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/91d7f1503d45481181dcbb9476d6436a : SUCCESS in 12m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57a4b1827eff450fadb9fb2c1e505cdf : SUCCESS in 5m 19s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/8cdb4620ac68455da4c12a92989d4874 : SUCCESS in 1h 35m 23s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f2e695bae48d42579cc0b4ccbd33b9ab : SUCCESS in 1h 35m 56s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c7ed102a5bfb490eb7dc0397cb774fba : SUCCESS in 1h 36m 35s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/f5a12c0dece14cb5ad844c18b2f05458 : FAILURE in 21m 50s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b5851a0d93b4ab2b957ae3e44e2843c : SUCCESS in 1h 19m 37s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f0071a2ae2646dea82ed0c546b94bcc : SUCCESS in 1h 33m 11s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a354b72186e6497bbc613a551ef5cfa7 : SUCCESS in 1h 30m 49s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/73db85177ddc4452b8b1f3bb4452fba1 : SUCCESS in 1h 40m 23s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cc8e5dbb79d443ce808d940f8e32adc5 : SUCCESS in 1h 47m 56s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b51a9b6da0ef41c3ab4fee79ccfa7dab : SUCCESS in 26m 41s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/09f03e947398487d9ac8849b3cf45c54 : SUCCESS in 27m 15s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/173966b92f064df9be347cae9f67c35a : SUCCESS in 27m 44s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/8e412d9d8da94df694d824945b50029b : FAILURE in 25m 56s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/39b8a4faeac5438b8b6e4aeeb9e8153a : FAILURE in 24m 31s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6e1afc33faa5400e99dd42e4b3904b6d : SUCCESS in 33m 52s","accounts_in_message":[],"_revision_number":5},{"id":"76acc5d9cf3514292594bde370c3c3b13425d610","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-09-28 06:49:43.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f2faebec3b9883288d23b910920eb3f7b9dca4f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 09:14:41.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3cdfbb523e8247029383c4ee7035ac85\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/029a4d16da964120b4d338d9e35d0187 : SUCCESS in 11m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adab514359eb4ef1a171ffe8c6c6acc4 : SUCCESS in 4m 38s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/c21b3cb4843d48dfa4a046cfb80a400a : SUCCESS in 1h 19m 25s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/feaa861fb29d47eaa51cda7f762cfdbf : SUCCESS in 1h 38m 39s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e074f99748a740debcc46b6b3020803d : SUCCESS in 1h 32m 32s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/125353c6cff04de48769953513acac56 : FAILURE in 18m 03s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ba59c4cf09204816896de15f1615bfa9 : SUCCESS in 1h 13m 12s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/57aceb95a29e43d0b9a30140e081b95b : SUCCESS in 1h 06m 09s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d6d240a06bd044b1a2666be4478fe746 : SUCCESS in 1h 26m 59s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1622fa0ede0a46148b74bcdd105f7a24 : SUCCESS in 2h 18m 37s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b3409514a9c84aef9d42be60e94ff5dd : SUCCESS in 1h 28m 03s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b98b2677c61649848dc1261362b28f47 : SUCCESS in 28m 11s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7164cdfee2ce4b63a934f78fa57eab1d : SUCCESS in 28m 06s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5dcbb4d0597a49a39190cbfe9fa68c93 : SUCCESS in 24m 45s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/3880c3deda09452f88e84dea8ff7cd14 : FAILURE in 23m 12s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/aa2d0412f6aa4d5c9dab8ef7208d0f97 : FAILURE in 21m 30s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0a7d1c1f17744c7681aee03e8224852b : SUCCESS in 28m 53s","accounts_in_message":[],"_revision_number":5},{"id":"64366d1c715dcd0d66c3134f30c718a1199ee2c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-06 10:55:27.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"c835448a779f74ee91abcc2ed8b59cbd8fab3e2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 12:45:14.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/70220700574e43829478d074ae56479d\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6dcf80f480be4f6cb5a823f5ba36104c : SUCCESS in 16m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f490b9b415ff444c9b7fae9d2992e0d5 : SUCCESS in 7m 55s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/8afdc774a1db437f94b7258e9a2ec29a : SUCCESS in 1h 37m 22s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9e0107350425418795c6798360f5c458 : SUCCESS in 1h 38m 01s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/35cf875166864e6196782e022566e581 : SUCCESS in 1h 38m 02s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/16236dd198e441df8d395a8dd2a4d362 : FAILURE in 23m 33s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/41cee06f8aa241e88877cb6146897bec : SUCCESS in 1h 23m 59s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a7c46d2e8bd4ff3b9fff3055adc9863 : SUCCESS in 1h 23m 22s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6184fc1a2b1a4de29dac3c01d6b01746 : SUCCESS in 1h 16m 46s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/56786d6e1d5d4d0f9722b6fc093d25e5 : FAILURE in 11m 07s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e4c6287343a14f458f3858c991e0f44c : SUCCESS in 1h 45m 58s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0fca9e932f7f4cf5a8bd1a779a2d495b : SUCCESS in 27m 55s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5eb255f0232a49189a4a4c2b69ddd649 : SUCCESS in 27m 12s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/15ce0baa45c2481da937ca722de988e3 : SUCCESS in 26m 14s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/9310713f8ed74f17bb12601b57353317 : FAILURE in 21m 23s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/4dc7fffe4415431a9ab8be6e80207fa4 : FAILURE in 26m 31s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/12555dbbb9a34a24a5f1c2b386db189f : SUCCESS in 32m 42s","accounts_in_message":[],"_revision_number":6},{"id":"daf902cb625a2d924c5a73b5058c22accdc616e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-10 16:11:08.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"5b123585e4e35b0452ecafcdac75baeb6a217010","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 18:34:37.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/97e083cd3688422d99ea41dadafe3fe1\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c7e7ae27db214f04b45485b61e3637f2 : SUCCESS in 11m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08f77a5c4d974a0d9c1f30247b1cc911 : SUCCESS in 4m 10s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/ad56d06bc72e41a8ad5ed4eb2984920f : SUCCESS in 1h 23m 45s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1a425de3253349a4a0c7b4d5863ed8b5 : SUCCESS in 1h 26m 13s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b6c4c5412ec14c198138f96cf2d1d04b : SUCCESS in 1h 34m 50s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/39d8e7e52d304d85a5a953d0ed94828b : FAILURE in 25m 42s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/695ec6a7d14a4b6584e5251144c965c2 : SUCCESS in 1h 15m 23s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/686471bb8d8b4852aed710dee0b842c0 : SUCCESS in 1h 30m 30s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8667eade54ee4782957805e873e66c58 : SUCCESS in 1h 28m 24s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1bb6075112fa40698787014ee044cf1f : SUCCESS in 2h 13m 18s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97e794386136474bb76b8400291c3802 : FAILURE in 1h 12m 48s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0b57b835fd5349de89fff61a75f95597 : SUCCESS in 28m 50s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7d9ed15c4c304db38082225ce32ffd5e : SUCCESS in 27m 07s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c673f08995b54ac48807f81d3d35723e : SUCCESS in 26m 22s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/508933aa51854b968c2f5dca904d2dea : FAILURE in 23m 09s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/9bfe3ae754624a568770f273dd10f6b3 : FAILURE in 30m 34s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7d7e9f6913294b479e7a6f6a7040939d : SUCCESS in 32m 03s","accounts_in_message":[],"_revision_number":7},{"id":"009aa243efed9eab1ee185748024cd5d60ab29fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-11 14:16:13.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"8766af40f6d59e862175f06498f84207a2f713b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-11 14:26:42.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"1bde98368fa1da9432f7d78fce936c18f732f8d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 16:51:10.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/164ba8b276c64f518d3046277786a808\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/21d653db4bf34f66af5b03bb4c93f6b6 : SUCCESS in 17m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e34e463e17746228244da61a0ca576e : SUCCESS in 4m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c1d96d1b68544dda547b8fb90b23af4 : SUCCESS in 3m 20s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/28decfe44dd14a1fa02a151195c67997 : SUCCESS in 1h 39m 31s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b7a358e163a54e02b363bbe608abfb55 : FAILURE in 42m 31s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8a6c627017f14c00bb2fa3233ab52f8d : SUCCESS in 1h 34m 33s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/0d1e99b9759a41e7ac5c33e4088b6467 : FAILURE in 26m 29s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/355b46a73b0547d7b57367b89a222b7b : SUCCESS in 1h 28m 24s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c102b184d8ee4ad890500dcfeff0f37d : SUCCESS in 1h 25m 31s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fb5f4eb3974a4a9b9f1b8e1c76bf7310 : SUCCESS in 1h 20m 10s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0a90c97f90ad465892720fd5fa3a890b : SUCCESS in 2h 17m 47s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/492a5c4812af487ea2ea2e00595be617 : SUCCESS in 1h 46m 21s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/36a4cd2369874770abcea3cd0b8866bc : SUCCESS in 34m 30s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/43052228d8804fbfbe686ae58c6997f6 : SUCCESS in 26m 05s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bfff9294f06742af87d31f0332cbad8e : SUCCESS in 28m 20s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/300cf24abf824cceaeb5b6e55069edb9 : FAILURE in 22m 18s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/ec944ee9c33448448375416dc95bff4d : FAILURE in 23m 44s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5c403f679c7a43ae887a5753f20e72d6 : SUCCESS in 35m 30s","accounts_in_message":[],"_revision_number":9},{"id":"7fd6a339c2facff88dfdd09be8bedb1f05ec196e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-12 08:01:57.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"5d04694a252ca5cba169f63dd58fff62d3c4d512","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 09:46:32.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f14ce10d80a4db082593ecb5f2b874a\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/06c612beb4ac482e86216e33de201fde : SUCCESS in 13m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69f53202140d488d8e8cbc298073a2bf : SUCCESS in 6m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5367d74f53f4da68f1b55005f261a98 : SUCCESS in 4m 21s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/27b84dd45c45489e96e31c7ab6295965 : SUCCESS in 1h 24m 36s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/75430c040747405687fc4704de6420c2 : SUCCESS in 1h 35m 43s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/43bbb78a3fa54711bf54f60fe6282435 : SUCCESS in 1h 36m 39s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/5b8a9a03a0304506b9c8849ab57919fc : SUCCESS in 1h 07m 25s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/75a2d3de3c9846cca62bffd5d17c4c09 : SUCCESS in 1h 15m 26s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee4cfa90861b44dfb6422fdae6ff8af8 : SUCCESS in 1h 21m 45s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d4ffebae4e734c66a68fa6efbddf2dc4 : SUCCESS in 1h 17m 30s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/566d23466d56482b819b0e88d9e46c0a : FAILURE in 36m 01s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a86b4decfa84caaba36f8950b96ae1d : SUCCESS in 1h 35m 17s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55cb3e5631fe45a48ca90b80babfdcf0 : SUCCESS in 26m 37s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c2294dffb96f4de88c022b7d67af376f : SUCCESS in 28m 34s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e05a2553c7db49e384e48ae90646a41d : SUCCESS in 27m 13s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/15053d45427e4bdcadf1aaf59cdf4cf5 : SUCCESS in 27m 33s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/96881f6c071a4a059d478346d91cfdd2 : SUCCESS in 25m 36s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f7f0729a86414360943001cc53941a26 : SUCCESS in 30m 55s","accounts_in_message":[],"_revision_number":10},{"id":"7b9b0366ec63667a56ddf80514d469f2b63ccb4d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-12 10:05:05.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"722acf706c98aa26e24ef1c9777786f2f1000463","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 12:16:52.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69cc66e1b6e14668845474c61625f32b\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/08bf35dcd7a24334ad20759b8c6b356c : SUCCESS in 13m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6df4e0d98ef141889863cc7c081ae8ce : SUCCESS in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ac0e4e5fb7c4a98973fb48cd19002fc : SUCCESS in 3m 32s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/1c1c18f58211465bb89c4b0ec2ae245f : SUCCESS in 1h 49m 05s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1c6b1aef95144dffb4661d52e466c944 : SUCCESS in 1h 29m 52s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/01b3133c686e489082ff820f43b7ff12 : SUCCESS in 1h 35m 10s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/6fc2624afcf94f66925fbbee6ef26f30 : SUCCESS in 1h 22m 08s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/47001b6a411047eea9936eb063acae5c : SUCCESS in 1h 13m 57s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2dc7a1989a16438a8078aa7d4de44c5a : SUCCESS in 1h 19m 26s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a857c084d584509b2a9f70cf989a5c3 : SUCCESS in 1h 24m 44s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/feb1fb378f804c01a630b6e006a532f0 : SUCCESS in 2h 05m 21s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/304b3a6d6f7e45a2991594cfb195a1e9 : SUCCESS in 1h 47m 10s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/29d49c84cf1548b8bb3a046bc434123d : SUCCESS in 30m 33s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b412ec5b8da44fbbb96e4d5e84cb3a21 : SUCCESS in 26m 43s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/10a21599b37e44bf909acc8ea5a0e235 : SUCCESS in 26m 41s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/69fb8c4fc1f846aa8180cc58fbfd3662 : SUCCESS in 30m 53s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/2bdb71a5169049f890f285d511905493 : SUCCESS in 27m 46s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/dd829ed403fd4e99b99edebc8da35de4 : SUCCESS in 31m 42s","accounts_in_message":[],"_revision_number":10},{"id":"5ad98fb8509b5a1949371a80ab5c1e80b1568041","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-10-12 19:13:26.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"1680050684936a4201b9eaf3fdb099da16dfc974","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2023-10-13 07:11:42.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"2dc5fcc2e52b3669d9a3eabd0a49cb9659c9a83c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-13 07:13:48.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"1e533b31d93c91361167acb67e46ac3364568ffa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-13 09:51:09.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e4fb3b257a94987a16cbd2285123cb1\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f6935924e69e4fc5bf2395f91c2323a2 : SUCCESS in 13m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b1aaa708129481dbdf864d4537cd93f : SUCCESS in 4m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e5ddabce5d9496b9b51779825ec2c2d : SUCCESS in 3m 16s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b9e0386a44ca499f99f28c8b6c6c66d2 : SUCCESS in 1h 37m 41s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4dd3380d7994492ea8297ab812b28bfd : SUCCESS in 1h 34m 19s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/275ef1d7ee99441e83dac96f49906d7b : SUCCESS in 59m 12s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/01b67215f6f8418db4e801e5673078ce : SUCCESS in 1h 33m 34s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b6f036c0e8d46589428a0aac429aad0 : SUCCESS in 1h 30m 59s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b70bcb779cf34639aab3d664da42c81b : SUCCESS in 1h 24m 43s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0c67be8539c84871840f8f8530c2ffe5 : SUCCESS in 2h 31m 18s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d65ada889e084f6bae072d4bc157d72e : SUCCESS in 1h 54m 05s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e4d422f46a8a457aa8c72abf8f6e9e07 : SUCCESS in 32m 56s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/46c177015efe4a52871388ebc6838b55 : SUCCESS in 32m 39s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9483a2e792be4461bf0bd60d4503037e : SUCCESS in 32m 26s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/4c4fe963dd9c496aa3602b94bd08acd6 : SUCCESS in 32m 33s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/d362900bd0a04640af7f32c7b2a6900b : SUCCESS in 31m 39s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/65e0b2a7ac33432fb0a24fffc565b329 : SUCCESS in 35m 27s","accounts_in_message":[],"_revision_number":10},{"id":"bb518e519c1f91d1f3e6772cd1a4a37ddffad960","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-13 09:51:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"36966be3583a235a1a540d888950f05287fd37c2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-13 09:52:05.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2181714d979242ff8fb29ae07f8486cb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c575eceff4d48c7b144bf6d8e17e0dc : SUCCESS in 37s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"d4d8114a11df9e38d7c77d565d85c9c3db2f11bc","revisions":{"57cbc2bb9c14cd87a977414a6407ec483ef2f7cf":{"kind":"REWORK","_number":1,"created":"2023-07-28 11:06:19.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/45/889945/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/45/889945/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/45/889945/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/45/889945/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/45/889945/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/45/889945/1"}}},"commit":{"parents":[{"commit":"caebffe51e1e7cbd5373c711d4847da4286e47f7","subject":"Cleanup old OS support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/caebffe51e1e7cbd5373c711d4847da4286e47f7"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-28 11:06:06.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-28 11:06:06.000000000","tz":120},"subject":"Stop installing openssh and rsync to containers","message":"Stop installing openssh and rsync to containers\n\nWith fixing keystone role there should be no need in explicitly\ninstalling rsync and openssh everywhere.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_keystone/+/889934\nChange-Id: I25729462fa6be7953e8ef0687ec4580509c21aaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/57cbc2bb9c14cd87a977414a6407ec483ef2f7cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/57cbc2bb9c14cd87a977414a6407ec483ef2f7cf"}]},"branch":"refs/heads/master"},"f1990a88872710da8725b227414ffef535f920b7":{"kind":"REWORK","_number":2,"created":"2023-08-17 10:30:14.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/45/889945/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/45/889945/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/45/889945/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/45/889945/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/45/889945/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/45/889945/2"}}},"commit":{"parents":[{"commit":"caebffe51e1e7cbd5373c711d4847da4286e47f7","subject":"Cleanup old OS support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/caebffe51e1e7cbd5373c711d4847da4286e47f7"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-28 11:06:06.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-08-17 10:30:14.000000000","tz":0},"subject":"Stop installing openssh and rsync to containers","message":"Stop installing openssh and rsync to containers\n\nWith fixing keystone role there should be no need in explicitly\ninstalling rsync and openssh everywhere.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_keystone/+/889934\nChange-Id: I25729462fa6be7953e8ef0687ec4580509c21aaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f1990a88872710da8725b227414ffef535f920b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f1990a88872710da8725b227414ffef535f920b7"}]},"branch":"refs/heads/master"},"f4082be3f1406b70d88ead6e783f1402256e90da":{"kind":"REWORK","_number":3,"created":"2023-08-17 10:30:46.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/45/889945/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/45/889945/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/45/889945/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/45/889945/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/45/889945/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/45/889945/3"}}},"commit":{"parents":[{"commit":"caebffe51e1e7cbd5373c711d4847da4286e47f7","subject":"Cleanup old OS support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/caebffe51e1e7cbd5373c711d4847da4286e47f7"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-28 11:06:06.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-08-17 10:30:46.000000000","tz":0},"subject":"Stop installing openssh and rsync to containers","message":"Stop installing openssh and rsync to containers\n\nWith fixing keystone role there should be no need in explicitly\ninstalling rsync and openssh everywhere.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_keystone/+/889934\nChange-Id: I25729462fa6be7953e8ef0687ec4580509c21aaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f4082be3f1406b70d88ead6e783f1402256e90da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f4082be3f1406b70d88ead6e783f1402256e90da"}]},"branch":"refs/heads/master"},"b72385e382a171c04a0286b341d21d86aad5f4b7":{"kind":"REWORK","_number":4,"created":"2023-08-18 07:44:21.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/45/889945/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/45/889945/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/45/889945/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/45/889945/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/45/889945/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/45/889945/4"}}},"commit":{"parents":[{"commit":"a5589beb5f64e01fecffefbcd398a34f6566e101","subject":"Fix linters issue and metadata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a5589beb5f64e01fecffefbcd398a34f6566e101"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-28 11:06:06.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-08-18 07:44:12.000000000","tz":120},"subject":"Stop installing openssh and rsync to containers","message":"Stop installing openssh and rsync to containers\n\nWith fixing keystone role there should be no need in explicitly\ninstalling rsync and openssh everywhere.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_keystone/+/889934\nChange-Id: I25729462fa6be7953e8ef0687ec4580509c21aaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b72385e382a171c04a0286b341d21d86aad5f4b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b72385e382a171c04a0286b341d21d86aad5f4b7"}]},"branch":"refs/heads/master"},"819503f7afaaff7dda67a319d94243e521ada0e5":{"kind":"REWORK","_number":5,"created":"2023-09-05 06:29:54.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/45/889945/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/45/889945/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/45/889945/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/45/889945/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/45/889945/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/45/889945/5"}}},"commit":{"parents":[{"commit":"2272de8f0c5977819ed38582c30cff68d6da7d5b","subject":"Fix linters issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2272de8f0c5977819ed38582c30cff68d6da7d5b"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-28 11:06:06.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-09-05 06:29:22.000000000","tz":120},"subject":"Stop installing openssh and rsync to containers","message":"Stop installing openssh and rsync to containers\n\nWith fixing keystone role there should be no need in explicitly\ninstalling rsync and openssh everywhere.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_keystone/+/889934\nChange-Id: I25729462fa6be7953e8ef0687ec4580509c21aaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/819503f7afaaff7dda67a319d94243e521ada0e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/819503f7afaaff7dda67a319d94243e521ada0e5"}]},"branch":"refs/heads/master"},"0f0beed2b18bd9b8ad562537eada2ebbce29af14":{"kind":"REWORK","_number":6,"created":"2023-10-06 10:55:27.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/45/889945/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/45/889945/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/45/889945/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/45/889945/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/45/889945/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/45/889945/6"}}},"commit":{"parents":[{"commit":"2272de8f0c5977819ed38582c30cff68d6da7d5b","subject":"Fix linters issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2272de8f0c5977819ed38582c30cff68d6da7d5b"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-28 11:06:06.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-10-06 10:55:27.000000000","tz":0},"subject":"Stop installing openssh and rsync to containers","message":"Stop installing openssh and rsync to containers\n\nWith fixing keystone role there should be no need in explicitly\ninstalling rsync and openssh everywhere.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_keystone/+/889934\nChange-Id: I25729462fa6be7953e8ef0687ec4580509c21aaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/0f0beed2b18bd9b8ad562537eada2ebbce29af14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/0f0beed2b18bd9b8ad562537eada2ebbce29af14"}]},"branch":"refs/heads/master"},"671315be53afb0d8affad2f18521c3525ed6e737":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-10-10 16:11:08.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/45/889945/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/45/889945/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/45/889945/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/45/889945/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/45/889945/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/45/889945/7"}}},"commit":{"parents":[{"commit":"39542e2c6594d3181108e383e59a423186d16c39","subject":"Switch to native systemd-resolved from resolv.conf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/39542e2c6594d3181108e383e59a423186d16c39"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-28 11:06:06.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-10-10 16:11:08.000000000","tz":0},"subject":"Stop installing openssh and rsync to containers","message":"Stop installing openssh and rsync to containers\n\nWith fixing keystone role there should be no need in explicitly\ninstalling rsync and openssh everywhere.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_keystone/+/889934\nChange-Id: I25729462fa6be7953e8ef0687ec4580509c21aaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/671315be53afb0d8affad2f18521c3525ed6e737"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/671315be53afb0d8affad2f18521c3525ed6e737"}]},"branch":"refs/heads/master","description":"Rebase"},"cab76d189d8816e17fb4004463caf74582b56fde":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2023-10-11 14:16:13.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/45/889945/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/45/889945/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/45/889945/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/45/889945/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/45/889945/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/45/889945/8"}}},"commit":{"parents":[{"commit":"a22ec2150f6b9300cbe0dd9f1acb84f034056bf4","subject":"Remove old tasks and vars from image download process","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a22ec2150f6b9300cbe0dd9f1acb84f034056bf4"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-28 11:06:06.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-10-11 14:16:13.000000000","tz":0},"subject":"Stop installing openssh and rsync to containers","message":"Stop installing openssh and rsync to containers\n\nWith fixing keystone role there should be no need in explicitly\ninstalling rsync and openssh everywhere.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_keystone/+/889934\nChange-Id: I25729462fa6be7953e8ef0687ec4580509c21aaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/cab76d189d8816e17fb4004463caf74582b56fde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/cab76d189d8816e17fb4004463caf74582b56fde"}]},"branch":"refs/heads/master","description":"Rebase"},"2456659539ad17b9fe58ed2ead3b429b1aa1ebc4":{"kind":"REWORK","_number":9,"created":"2023-10-11 14:26:42.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/45/889945/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/45/889945/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/45/889945/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/45/889945/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/45/889945/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/45/889945/9"}}},"commit":{"parents":[{"commit":"a22ec2150f6b9300cbe0dd9f1acb84f034056bf4","subject":"Remove old tasks and vars from image download process","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a22ec2150f6b9300cbe0dd9f1acb84f034056bf4"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-28 11:06:06.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-10-11 14:26:40.000000000","tz":120},"subject":"Stop installing openssh and rsync to containers","message":"Stop installing openssh and rsync to containers\n\nWith fixing keystone role there should be no need in explicitly\ninstalling rsync and openssh everywhere.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_keystone/+/889934\nChange-Id: I25729462fa6be7953e8ef0687ec4580509c21aaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2456659539ad17b9fe58ed2ead3b429b1aa1ebc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2456659539ad17b9fe58ed2ead3b429b1aa1ebc4"}]},"branch":"refs/heads/master"},"d4d8114a11df9e38d7c77d565d85c9c3db2f11bc":{"kind":"REWORK","_number":10,"created":"2023-10-12 08:01:57.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/45/889945/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/45/889945/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/45/889945/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/45/889945/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/45/889945/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/45/889945/10"}}},"commit":{"parents":[{"commit":"a22ec2150f6b9300cbe0dd9f1acb84f034056bf4","subject":"Remove old tasks and vars from image download process","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a22ec2150f6b9300cbe0dd9f1acb84f034056bf4"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-28 11:06:06.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-10-12 08:01:57.000000000","tz":0},"subject":"Stop installing openssh and rsync to containers","message":"Stop installing openssh and rsync to containers\n\nWith fixing keystone role there should be no need in explicitly\ninstalling rsync and openssh everywhere.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_keystone/+/889934\nChange-Id: I25729462fa6be7953e8ef0687ec4580509c21aaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d4d8114a11df9e38d7c77d565d85c9c3db2f11bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d4d8114a11df9e38d7c77d565d85c9c3db2f11bc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
