)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~892295","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I2564e3dcb2efad8f6a2ed21bec61668c1b6f6209","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"osa/core-2.15","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2023-09-04 14:34:25.000000000","reason":"\u003cGERRIT_ACCOUNT_31542\u003e replied on the change","reason_account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-09-04 16:46:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2564e3dcb2efad8f6a2ed21bec61668c1b6f6209","subject":"Fix linters issues","status":"MERGED","created":"2023-08-22 11:24:54.000000000","updated":"2023-09-04 16:47:39.000000000","submitted":"2023-09-04 16:46:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892295-osa/core-2.15","meta_rev_id":"9b08beebfd3d3215c4c5c9bc07092e46c9b6fae5","_number":892295,"virtual_id_number":892295,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-04 16:46:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2023-09-04 14:34:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-31 11:43:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2023-09-04 14:34:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-22 13:34:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-31 11:43:05.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2023-09-04 14:24:00.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"50a636cf96d92707bda724c20b6be7680bc21778","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-22 11:24:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"32677f55c81697d59dc3f90dbce24ae988a58221","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 13:34:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1030d489243c4b31a23ee15ebdac9c98\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4c723221aef4dfcad2e04c36e03fdcd : SUCCESS in 13m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4d1e80df8d0475e853e5d76e00d839c : SUCCESS in 4m 40s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/3b5a55a3d74647a19340deea8a85f797 : SUCCESS in 1h 26m 42s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c8d86d5df79447a3a5d5a57b3f7d733c : SUCCESS in 1h 36m 59s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fd9d9ba0049b439b80d14a946fdfe7fc : SUCCESS in 1h 35m 26s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/4cd8c4b8475846abbaff77a08a810070 : SUCCESS in 1h 08m 39s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ad277ee6b5844b7abcc146a1a97dd67 : SUCCESS in 1h 20m 23s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/56c0cf56035e4b9ca848948c55c1583c : SUCCESS in 1h 22m 02s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0e2a1fcbc1645078e58d6a3ae83b21f : SUCCESS in 1h 14m 44s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/56e09e7b6d5b4732aaf1de2636151d92 : SUCCESS in 2h 02m 54s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff870493f20e459baa072dd406cdefcc : SUCCESS in 1h 33m 31s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/178846503ec74901877518fe44085c54 : SUCCESS in 28m 30s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/710a44b888654af29e40464b1d92f6e1 : SUCCESS in 28m 40s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb15ef09ab614cc991b603df446fc028 : SUCCESS in 25m 16s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/c25952b27bf74c6d90963043074c8cdb : SUCCESS in 26m 00s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/870175f2f27e4b68a8e4e0e242dc0f19 : SUCCESS in 26m 46s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6c2f375b87c14dc790b9f70dc006edc6 : SUCCESS in 27m 14s","accounts_in_message":[],"_revision_number":1},{"id":"347f38ef5eafe16ee0c99583d9b4617e0f01af3c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-08-31 11:43:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6e8ce251ea4a706c3e755c1a736b9854a2e49f8a","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2023-09-04 14:34:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c9ed7ad2f80e11c6ea1b65b26c2d6d796bf2d489","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 14:36:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"500a713db0ff471509448a9d306a00385e8bf4eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 16:46:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c39de31780dc4e1a9a04c939a819bf84\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c654caf981734d79b98c95a4ee2be87b : SUCCESS in 17m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c504b45a9ff7495bad3cd4b69fc76625 : SUCCESS in 4m 54s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8411cd0462174a06b86c2bbb4f544b0d : SUCCESS in 1h 29m 03s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bf4f7f6d717148f9bf517e7e3710d4ec : SUCCESS in 1h 32m 40s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/ed52364fb38049168d4560d893595072 : SUCCESS in 1h 22m 10s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8ad338588ee745e78fb7415cfce0a4a8 : SUCCESS in 1h 25m 37s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9522b2166454fe3a3c700c122aebe1c : SUCCESS in 1h 15m 30s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/817857503d0a449492be6daf59549440 : SUCCESS in 1h 18m 42s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6bd817134da548168f916a82c4c3cf93 : SUCCESS in 2h 05m 40s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f87a2be48c84438a973059ff17232ba1 : SUCCESS in 1h 30m 45s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/58585bc72e774672b6cbed8b6dacc276 : SUCCESS in 26m 54s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fbe06a1f18bd42778fdb668e2a494ae0 : SUCCESS in 29m 42s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be7aa299a5d9464b8da550a41158b59e : SUCCESS in 23m 28s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/63db7e4578fb4da5a50dde72b2d67eec : SUCCESS in 26m 15s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/999bc8de6b1c4b23a37699ba596e9ba7 : SUCCESS in 28m 40s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d359b241eb124944a429a27f5e0e4882 : SUCCESS in 28m 51s","accounts_in_message":[],"_revision_number":1},{"id":"0dd3afc564470ec6ad23d0e11f4fc630942eba39","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 16:46:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9b08beebfd3d3215c4c5c9bc07092e46c9b6fae5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 16:47:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08817365096e401a859337eee487ed90\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b9794e6845e4e41a897d65a66f9c235 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2272de8f0c5977819ed38582c30cff68d6da7d5b","revisions":{"2272de8f0c5977819ed38582c30cff68d6da7d5b":{"kind":"REWORK","_number":1,"created":"2023-08-22 11:24:54.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/95/892295/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/95/892295/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/95/892295/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/95/892295/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/95/892295/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/95/892295/1"}}},"commit":{"parents":[{"commit":"4686fac8623298bbbf7b3fc8254f308dacda1e38","subject":"Add retries to LXC base build command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/4686fac8623298bbbf7b3fc8254f308dacda1e38"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-08-22 11:24:46.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-08-22 11:24:46.000000000","tz":120},"subject":"Fix linters issues","message":"Fix linters issues\n\nWith update of ansible-lint to version \u003e\u003d6.0.0 a lot of new\nlinters were added, that enabled by default. In order to comply\nwith linter rules we\u0027re applying changes to the role.\n\nThis is a follow-up change to [1].\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible-lxc_hosts/+/888180\n\nChange-Id: I2564e3dcb2efad8f6a2ed21bec61668c1b6f6209\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2272de8f0c5977819ed38582c30cff68d6da7d5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2272de8f0c5977819ed38582c30cff68d6da7d5b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
