)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~892312","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2Fzed~I42f5b43f09d3c530813dd7fd334eafce7a5eaf39","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2023-08-29 21:02:39.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-08-29 21:02:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I42f5b43f09d3c530813dd7fd334eafce7a5eaf39","subject":"Refactor LXC image expiration","status":"MERGED","created":"2023-08-22 13:30:46.000000000","updated":"2023-08-29 21:03:42.000000000","submitted":"2023-08-29 21:02:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892312","cherry_pick_of_change":888278,"cherry_pick_of_patch_set":3,"meta_rev_id":"61ca66cc0bd8a14b95e79eafa8cf662f7eb5a9a3","_number":892312,"virtual_id_number":892312,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-29 21:02:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2023-08-29 18:57:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":2,"date":"2023-08-28 11:01:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":1,"date":"2023-08-29 18:57:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-22 15:21:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-22 16:01:21.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"CC"},{"updated":"2023-08-22 16:01:25.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2023-08-29 18:57:56.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"}],"messages":[{"id":"eeacc80ba433e18286a3841de319691c4a0cba6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-22 13:30:46.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"cd3684c080999a5fd021df153afdc3e35a6ae30b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 15:21:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/469839bdf37b4ece9caee8257c312708\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42c8342a00a146f091b738a9076e1ccd : SUCCESS in 11m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4663de6b351f4980b28624f31956c32d : SUCCESS in 4m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/313208d7f0884dfa9e796df36df5bdc8 : SUCCESS in 3m 19s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/16dc5168f8a04517a26242fbf5c4b729 : FAILURE in 41m 27s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8737ce8bb2b94f75b048ab72c9ddec43 : SUCCESS in 1h 26m 36s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/6f00e7769c5f43308abbbc8b4d797e18 : SUCCESS in 1h 06m 04s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/88ecbfb548e54203b13438a6449e6428 : SUCCESS in 59m 41s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2945d9b43e15473384dd2962524804c1 : FAILURE in 1h 06m 43s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8cf8d754957b4fd1bd59b222ded4d075 : SUCCESS in 1h 42m 26s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/106651c442ae4e3fad8f32873899a0d8 : SUCCESS in 1h 38m 38s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0ec7c43481d4da9966ac97b67fc6daa : SUCCESS in 25m 18s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57b17140d57b412db233db37af1115bb : SUCCESS in 28m 26s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa25b6e90e334b4abccc165e2e0c7c02 : SUCCESS in 26m 46s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/8a7b85d97d8746d9838a84248ed19182 : SUCCESS in 30m 46s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/a43c6e091ddc45f39ebb48f4de3c1a02 : SUCCESS in 28m 09s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4864b9bb6f8643cab8b65624107a524a : SUCCESS in 28m 00s","accounts_in_message":[],"_revision_number":1},{"id":"bdaa2c9cf172a101b7e807f48864e6449f110ad2","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-08-22 16:01:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"17edbd4ad5f1142565cc4728540eca15e2b5d0d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 17:59:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fc2aee1a26a1497898eb11d14d5e8256\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6f3b6cc2e2f4746be7aa7acab252ef3 : SUCCESS in 11m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dce4bd6cc5764853b8c0fc63706ef355 : SUCCESS in 4m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cfcd6a85b203494f8c8f8956932587cc : SUCCESS in 3m 17s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/fd8a641e65d142d7b8daafef4abb6efe : FAILURE in 36m 51s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/33befc784b0e4f2c943aa46f827e2fec : SUCCESS in 1h 19m 05s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/166bb15674454eb99557972b2cc953dc : SUCCESS in 1h 04m 59s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da71eaccb1784b678dbe85543fcd00f0 : SUCCESS in 1h 06m 57s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aeb29572c84c4c9fa0b76db366765cf9 : SUCCESS in 1h 00m 37s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f3ea945be0894870a58b9a9eedeb1b6b : SUCCESS in 1h 55m 46s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b00877109704e0d9c44e22dc0b8267e : SUCCESS in 1h 25m 02s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d4d8ee568978454693597de95c1c7c37 : SUCCESS in 23m 08s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a7821d234d148719f7fe05e8d958a15 : SUCCESS in 23m 42s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/847ce65fbb1d4e3c8dccee1d77c6970b : FAILURE in 25m 23s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/53cd3ae1e9ea456d8d94de94ae758b12 : SUCCESS in 27m 31s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/2871b384fa3a4787aba4406c2dff95d1 : SUCCESS in 28m 42s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e12b9bafa86a4dbea71adbb6a8cf98fb : SUCCESS in 30m 37s","accounts_in_message":[],"_revision_number":1},{"id":"93ab496b203a3a8af614d35e24beee4c5b9bacd5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-22 18:44:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"884c34065c1d68ca45abb5d61443bfaabd65e072","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 20:42:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9aa6524876b4f47a09bdb07c5931133\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f70662f155e94313a41ad8345287d30d : SUCCESS in 9m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ea8d76bff134e3f856aba8dc014f8c6 : SUCCESS in 5m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e75c5d0f6c854c2bb8b2c61c6fd6e2a3 : SUCCESS in 3m 13s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/8baaacd26fff422f97b8bcc9ae13d1af : FAILURE in 37m 17s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c715cfa06b5943328b001c7bc98dce92 : SUCCESS in 1h 22m 54s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/35ab0f9c3bc94fcc94a4d9aa815da8fe : SUCCESS in 1h 04m 34s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e45c50e2d2f143888840be08241c9260 : SUCCESS in 1h 06m 29s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8ff9c7c2997a4ba2bbe7d18ba6b39a30 : SUCCESS in 1h 05m 19s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/992d14a840b74ec38e41bcca76add2c6 : SUCCESS in 1h 52m 50s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ff2359c5164465bb1538099a7b61d8c : SUCCESS in 1h 30m 35s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0718c8a5d12f4847923e6aba71b108f6 : SUCCESS in 24m 45s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab7b20d498fc4ddf9473f633bc80ed73 : SUCCESS in 24m 11s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46f1871065c84d22ba4277eb11a4ebb7 : SUCCESS in 32m 38s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/b1afd526478a40c5937170b4f1a4d9cc : SUCCESS in 28m 37s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/47e629cbf75c496bb75ca95a5b7f0c0e : SUCCESS in 25m 52s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5dad9a9f57a540b3ad597e27995e278c : SUCCESS in 26m 24s","accounts_in_message":[],"_revision_number":1},{"id":"69e8d791b0f5c206b97cde92f39561cc2f92b196","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-28 11:01:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"51218a79410c18c96d5ca0156e3d82afb95b47b8","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2023-08-29 18:57:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"30b768f5746fabc14b8caa3b117ea2f29e62c01c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 18:59:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"12a4b0001e6d2d2b796b3f9ccb6c4c65137c883d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 21:02:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c84362710c2484899e456b5820b6a50\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bfcb3642d2c34e98a5403cfc21120dad : SUCCESS in 8m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29bcc0ed77784302825af73b508b787f : SUCCESS in 4m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9cbe494367be400ba715318b87a98bb8 : SUCCESS in 3m 52s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e211ae903cb749e7a5e32721b176e202 : SUCCESS in 1h 17m 02s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/39235366b29a4b5190f2e2ea9487e0c3 : SUCCESS in 54m 30s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4da0371669049a0946007862370a0ff : SUCCESS in 1h 09m 40s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cec52f6ee61e4db1a1a3f5170900db8d : SUCCESS in 1h 56m 22s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f988bbddc964e59a6525f7110a3ec2f : SUCCESS in 1h 31m 41s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c97f82c035fe4820b3f2e965d915a596 : SUCCESS in 24m 34s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4fe7a9b2c12c40609ab53753351dbe05 : SUCCESS in 23m 30s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c94154e1fc141ada02ea93e471bce48 : SUCCESS in 31m 37s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/476a10ba2ceb46e79e1f374451ef8a1e : SUCCESS in 27m 30s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/fa8ac5ecd5904b0bb74f09bd01869972 : SUCCESS in 25m 45s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0161dab76e6a43e7b6423cae47480f62 : SUCCESS in 34m 46s","accounts_in_message":[],"_revision_number":1},{"id":"c1cd8d8044432cf9d12679c0528ab61e1cad8ba1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 21:02:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"61ca66cc0bd8a14b95e79eafa8cf662f7eb5a9a3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 21:03:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56b82d2d35734429835134e2ac304dc7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bfb6cd4d3ac457bbeab17e07fdd39af : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7e75a083d29e0cbf7033c0a8e4c4a6c24f0a08f5","revisions":{"7e75a083d29e0cbf7033c0a8e4c4a6c24f0a08f5":{"kind":"REWORK","_number":1,"created":"2023-08-22 13:30:46.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/12/892312/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/12/892312/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/12/892312/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/12/892312/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/12/892312/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/12/892312/1"}}},"commit":{"parents":[{"commit":"5b209f9db12c4e4294a5ef8966a3c441bac93ef4","subject":"Merge \"Ensure tar is installed on LXC host\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/5b209f9db12c4e4294a5ef8966a3c441bac93ef4"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-12 13:43:20.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-08-22 13:30:46.000000000","tz":0},"subject":"Refactor LXC image expiration","message":"Refactor LXC image expiration\n\nRight now we write output of `date -d @{{ timestamp }} to\nthe expiry file, and then attempt to comapre with timestamp.\nHowever, output of  `date -d` is datetime and not timestamp,\nso these 2 things can not be properly compared. So image cache\nwas valid forever.\n\nChange-Id: I42f5b43f09d3c530813dd7fd334eafce7a5eaf39\n(cherry picked from commit a51291f87d44a126f432bafbfd68f6fe3cd1ba1d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7e75a083d29e0cbf7033c0a8e4c4a6c24f0a08f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7e75a083d29e0cbf7033c0a8e4c4a6c24f0a08f5"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
