)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~894555","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~Ie7200d5c01948c885c3dd4e8103c8f0a65e26108","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"osa-bookworm","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-09-27 16:35:39.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2023-10-07 08:13:34.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"Ie7200d5c01948c885c3dd4e8103c8f0a65e26108","subject":"Ensure systemd-resolved is present in debian container images","status":"MERGED","created":"2023-09-11 16:14:19.000000000","updated":"2023-10-07 12:35:05.000000000","submitted":"2023-10-07 12:34:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"894555-osa-bookworm","meta_rev_id":"2ace0789f0c41bb31e1fa4eea245025214f4119c","_number":894555,"virtual_id_number":894555,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-07 12:34:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2023-10-02 14:27:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2023-09-29 08:10:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2023-10-03 12:50:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-11 18:22:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-27 07:24:48.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2023-10-02 14:27:52.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"e39d92f2d1f052a5084d9b614a67bcb3925eef03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-09-11 16:14:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"368f61eb911a6ab2237794285518c633ff9af3c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 18:22:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e65519f3b51e4389a690ce90020343aa\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f64a0ed7d52d4e7e97776a0abf5c8c06 : SUCCESS in 13m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cd488914f3647eea405f9d14906ed84 : SUCCESS in 6m 13s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/4f2e6b00b84a48f9a5684869c7cf464c : SUCCESS in 1h 33m 47s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c3174a2bd1da433fa9680b835397f8fd : SUCCESS in 1h 21m 53s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/84e8e5ec197e42d6a1dc3cf7c4bd8f5b : SUCCESS in 1h 42m 01s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/bd5d96026d3f4829ada0bfe384b06549 : FAILURE in 15m 13s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67b201ce337d4ad8bd4bb0e79f2c7b66 : FAILURE in 16m 38s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/43f482443ddd413f87711703ae843a8b : FAILURE in 20m 05s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f28d1942087b4ee79075c2a843b2c1a7 : FAILURE in 16m 03s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5b518c235fbc450292ae8578c210ba03 : SUCCESS in 1h 45m 34s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cfecdd3ed54f4c0bb38337e4e773078c : SUCCESS in 1h 31m 29s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c0435d5d3961427bb79229f9ef33df8a : FAILURE in 20m 12s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cf5775c96bc942528fbf369046fdefa7 : FAILURE in 20m 29s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/93be168169e24d689f3a399ed275953c : FAILURE in 19m 14s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/2dd4875a1621416fa1aee467fac91dd3 : FAILURE in 20m 29s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/1eb5bf7b179348baa7f5aefc59dc5053 : FAILURE in 25m 55s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ded29051783f442f9dae826b1fba1d87 : SUCCESS in 27m 24s","accounts_in_message":[],"_revision_number":1},{"id":"f1cc2eb6d112c5ab6f99d737401cbbf3202e088b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-09-11 18:30:02.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b7dbd5c06281cd1cd36b35cc0a7a8811bda64c8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 20:12:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02936b47fd524b1abc6a46a7eda8c18a\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0cb3a913d67b4d42bb0fe6d4791b48b6 : SUCCESS in 12m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1d2b4b2948145db90b4e0f3090b0c57 : SUCCESS in 4m 53s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/5ec4165f2dbb46a1b2fd0c2f8e471658 : SUCCESS in 1h 38m 10s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/56844746e4f644d6b3ca2fe01f384847 : SUCCESS in 1h 35m 36s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/846962f28b81418c9dda9d2ca08d880d : SUCCESS in 1h 32m 59s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/958f14e7923146eaa5652209ef960843 : SUCCESS in 1h 17m 25s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d8f4368fab84b39aa86fc92e3f298f9 : SUCCESS in 1h 19m 53s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4887232124994bc4bdc121390a7230af : SUCCESS in 1h 26m 17s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7c77eef70c64cbb8a97b3c3fdcaada4 : SUCCESS in 1h 24m 47s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9d8f26c9673548cd94cb18a790180b46 : SUCCESS in 1h 35m 05s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a0bb0019cac40ab870094588549743d : SUCCESS in 1h 34m 45s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1a9003d184d042e1bc94d33a441a4ff1 : SUCCESS in 30m 14s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a52d50de092b4030b5e32f5503e256af : SUCCESS in 26m 08s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/29dc9d13423e40b9985aaf00cb0012e4 : SUCCESS in 29m 47s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/dedb353da94045209b4ea1b62e84d30a : SUCCESS in 25m 39s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/0242bc42d4e245f4bd89fdde93ad0a54 : SUCCESS in 28m 02s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/de5913bb9f7e4cfe9c7fd61f85559109 : SUCCESS in 28m 44s","accounts_in_message":[],"_revision_number":2},{"id":"a48e9ab71501ffa927fc9ec465586703b375901a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-09-27 07:24:48.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"8a53297f54270a7022e52fa0a4382073f6e9166d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-09-27 16:26:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0016755bb3ed837c120747a3c89884f240b95479","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-09-27 16:28:41.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"800b251d4f47900c88ffe1b115fbcbce6dfe5cd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-09-27 16:29:20.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":4},{"id":"7185fd01e07a9f294510f879968bf808c29ac934","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-09-27 16:35:39.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8c20e27632bc5d9ace4d8c9d9659135c7c415263","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 18:30:13.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0583772a68ea4d7cbb3f41d2ede4f7cb\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/03cd263847d842e7ad5991d8206ef638 : SUCCESS in 21m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00d3d88e6a744e36a5f9bdb43925900e : SUCCESS in 5m 55s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/ecc6faae3f614c6090b5be381a74d02b : SUCCESS in 1h 32m 02s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/234f5464a2ce4430a64a8ce6584da0b9 : SUCCESS in 1h 35m 42s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2fded45e20d04c1882309264b67008bd : SUCCESS in 1h 37m 00s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/cb74791f9ab0428f827d65f5bf8dc575 : SUCCESS in 1h 27m 09s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ab1e48fdef34fb6894b7dcdd7ab93d9 : SUCCESS in 1h 19m 31s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a99ef10fabb4fe5b57c4f34613f84d6 : FAILURE in 35m 41s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f9d0ce50de743f4a20cd1df29530f15 : SUCCESS in 1h 08m 32s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c4ed5684371646ed85f8c8ac62de94f1 : SUCCESS in 1h 39m 59s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bba67124e1a34b9aaeb3a23d610b8bcf : SUCCESS in 1h 53m 25s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/614635f409854a0188fb1cb63ac8ffa2 : SUCCESS in 28m 56s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7de19c2046fe420e92d35f46058d2ccb : SUCCESS in 27m 46s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b00396229ff48d180a5c99570acd9cb : SUCCESS in 27m 16s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/9eb5bac611c944a0bc8ba513634be33a : SUCCESS in 36m 34s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/ec3bc4b38c1643c0aa3ba43ef9a67606 : SUCCESS in 29m 36s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/548f45d138964a30ba0869e8ebf9c3eb : SUCCESS in 33m 45s","accounts_in_message":[],"_revision_number":4},{"id":"7ded8eef0e49e9340f163ea1dd3853bd41c59c1f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-09-28 09:18:06.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8df85cce2ba13cebb40c8ce029090ba8f36d98f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-09-28 09:19:18.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"b81231a76ab4df635af5d8a24c6bff01a53a2dbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 11:40:59.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed6a8c99c22f47da8025809a3183092c\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee630288d7fa41db9001c3ab485c90ff : SUCCESS in 13m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/502e98c480284e2384c263ad75320c08 : SUCCESS in 5m 02s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/852343e224ff456693ddec15ffd2bfe5 : SUCCESS in 1h 45m 44s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1e7fb672282948d1ba80c21d3146fc85 : SUCCESS in 1h 43m 42s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/366418bb650344b3b5db6788c7e9e5db : SUCCESS in 1h 34m 40s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/5f3eeacf66654abc925d890a76436a6a : SUCCESS in 1h 10m 26s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/88409da337054d49886a145b2a38bc71 : SUCCESS in 1h 24m 10s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db74c79df9194869a69e2a225d2c4efd : SUCCESS in 1h 25m 44s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/62f3d58a862c4c52b30c5a942f0faf6b : SUCCESS in 1h 19m 20s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f80a73aed3984bc6aa6349952740713d : SUCCESS in 2h 18m 07s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cae4f269636648e8a805c7ee1ef3ae51 : SUCCESS in 1h 37m 52s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b51302b70514bc5a91d361f323c7599 : SUCCESS in 32m 29s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07d5e00ebe9d42519ccd224635743424 : SUCCESS in 30m 16s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0673f55d7dd242ed8a886a41533e82f2 : SUCCESS in 29m 38s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/a98521d06cc54f2690f3a6ca8f0e6abf : SUCCESS in 33m 52s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/6541607f0dd14f989ce1a7ecc3f757e8 : SUCCESS in 26m 11s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/60ade79828764afda4ccee2bd9c237d7 : SUCCESS in 34m 37s","accounts_in_message":[],"_revision_number":5},{"id":"9b9b117d62ac9ad36e839913b5f2ce7dc5c14b64","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-09-29 08:10:36.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"b4cb0020ee33d7c477f7d80a49bd84f236909994","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2023-10-02 14:27:52.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"8c0b93c1becbdde343be87a74a6f25682d5d656f","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2023-10-03 12:50:16.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"9c3ea713f576ea5d7ac3a1c94e2db61da3e23432","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 12:52:22.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"c24277977b49060c545e9b9ea81f38dd7ce1379a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 15:03:29.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/76d029e40a9f4a088d59586779000563\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d44b28f629e744a082662904496a700b : SUCCESS in 13m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eed5b696740a4d96b80ddac6714a3e55 : SUCCESS in 5m 12s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a4dceecb26cd452e965ce1ab65b01e69 : SUCCESS in 1h 35m 19s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e56549434a1440f49533aed387919eed : SUCCESS in 1h 41m 46s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/6a006a365a904a9eb05185ea5c12d1f6 : SUCCESS in 1h 31m 58s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f490a7d179a847848ee642daab69bc29 : SUCCESS in 1h 17m 58s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9fda4961cbf4cc9bef13536700760b9 : SUCCESS in 1h 30m 26s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2684b9262da84bd087b0543c60c10c92 : SUCCESS in 1h 21m 54s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d9a0d7f4ee4f4f2192f277e7c6e1b179 : FAILURE in 2h 10m 57s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1054ecada3114338991b1b3d3d07c9e7 : SUCCESS in 1h 39m 55s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3835a76d439941699c0c9612f74778f1 : SUCCESS in 30m 03s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e578dfddf9a4d4d95f04491ab21e74c : SUCCESS in 25m 38s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/35e67affc0b34ce5991c48463c606284 : FAILURE in 34m 13s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/58f6df9b5d7a4db4bbc2a0dca10d1040 : SUCCESS in 29m 06s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/3acb6485acfd4e898dde7dee634eec84 : SUCCESS in 25m 07s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e72ed7d431154810a34a226773a3e15c : FAILURE in 35m 41s","accounts_in_message":[],"_revision_number":5},{"id":"9834093b296bbba9d316d04793110ba98bcb3e5f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-04 16:22:54.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b15c962f5bebd2562468d94e07cc07163f2289e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 18:21:01.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d3454258ddb4684bbb7264eb6561888\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eb28286d81e746b3b04f3f9817fc726e : SUCCESS in 13m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e1e8db3abbd43e791c044bea6ba2348 : SUCCESS in 5m 48s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/641f1b500d7d4333bc7855e5e1b93ade : SUCCESS in 1h 40m 41s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/22c3cad8320c46b18ad638f30d2f3868 : SUCCESS in 1h 44m 51s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d58625ae79a248d3b209769c46a49033 : SUCCESS in 1h 47m 05s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/9780e20552a04fb6a5a085a5942880ea : SUCCESS in 1h 11m 31s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f0128850e82e44bf9b3257d1374a42ba : SUCCESS in 1h 20m 28s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/207a6befe4de477e89ecdab536444f94 : SUCCESS in 1h 31m 35s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e2a5c11c3af499b84baf9a3006fcea5 : SUCCESS in 1h 13m 23s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/993cd008a0ad43138af3c7c627262e57 : FAILURE in 30m 53s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd5e449c42594b1b849e270d40197f1d : SUCCESS in 1h 35m 16s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/384f9dc40ea8496f875a3740808b900f : SUCCESS in 32m 20s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c89a8b13beab4456af2d83644556b963 : SUCCESS in 37m 04s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55c28f59a8914e4783e5ce4514e2324b : SUCCESS in 30m 22s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/daca26d41207400380ed5e111dbe8397 : SUCCESS in 30m 17s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/7f8c4168049143beab7f81931cf245f5 : SUCCESS in 31m 01s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3a6a0060f8b742f08520f1de95f2b610 : SUCCESS in 35m 06s","accounts_in_message":[],"_revision_number":5},{"id":"c93f346cc00278c5838582bfdaba1f4c3d97b1a1","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-10-07 08:13:34.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a6c3d6179ac18332626873a4a87dcbce637d41de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-07 10:27:25.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9765fb71a0d47b792c8f09379a66660\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fde23ffe3f164260a082659407c00649 : SUCCESS in 11m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8d006da34be4074a36b7e1eb64fbfe3 : SUCCESS in 4m 11s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/db514a808f214134bb778d7fd8434fc4 : SUCCESS in 1h 28m 10s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6f17fc0838564932aed699a4e1673a7a : SUCCESS in 1h 37m 20s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7ba0db27231843309d6b1e8c580959fa : SUCCESS in 1h 49m 05s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/4fbc6c264fe447d388f78e873d029dd8 : SUCCESS in 1h 00m 33s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/679b24d6071945088ae446de0123c607 : SUCCESS in 1h 19m 36s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dcb9ddf234344105b91c05ab6ba9d6d3 : SUCCESS in 1h 25m 02s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7fb85dd29c4d4d4ebf996570454ebad9 : SUCCESS in 1h 15m 26s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f2220160d2e94426a84a69e6bf1d61de : SUCCESS in 2h 06m 47s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/832f81f770a74f84a01c6439963dd4fb : SUCCESS in 1h 54m 07s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f2f075bf99b4b27bf6fb860724a3282 : SUCCESS in 28m 04s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8b08cafc6b064a12acdf4c75d74cd0b6 : SUCCESS in 27m 07s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4172656665bd4c05ba8400d85ca9569d : SUCCESS in 25m 39s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/a07b675383b64a76b9882ffb6bce4ac9 : SUCCESS in 27m 06s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/80fb7b0e796d41e2b295120058f4600d : SUCCESS in 24m 19s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7c5cade8f31c4e689cc720d52e6db168 : SUCCESS in 28m 03s","accounts_in_message":[],"_revision_number":5},{"id":"b2ad73d8ac37af5fa1ed83193fefefcacab792c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-07 10:29:47.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ec963a7c12d539c8d5821b8c74b8813aa4be7d3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-07 12:34:04.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5037adad9ff141738d4716c272187846\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eaf5b688644a4246871d64951f5119a4 : SUCCESS in 10m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f99c06f15fe543808e8c5dc5e10ebd42 : SUCCESS in 4m 11s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/37805f7a7b3240c49f57037130354b83 : SUCCESS in 1h 33m 28s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2ad0fb5ac3824e18b667be7d2ff49795 : SUCCESS in 1h 33m 35s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/10ff3a7da48e479d83f0c212b40d51fb : SUCCESS in 1h 19m 37s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/76c16cc9ceb145d082f196dabd9fbf4d : SUCCESS in 1h 17m 03s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78ea17e2c83543f597360a74a2cb53db : SUCCESS in 1h 16m 47s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0416a761653947deb6fe943fd2aa71ce : SUCCESS in 1h 21m 18s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/49dc0166f90740218b7e588278de3640 : SUCCESS in 1h 59m 43s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/84584bdccef74988a95b0cd3de3718d6 : SUCCESS in 1h 49m 50s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/858c11885e304a919accd8d28f0d0f6a : SUCCESS in 29m 02s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2e766d686af040c5969f51c6e9a02fa9 : SUCCESS in 23m 38s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1a4f27cf6c354577a7a57f8c3f4c173b : SUCCESS in 28m 11s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/64b5c806652e46b98f4837e6a92b8541 : SUCCESS in 25m 18s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/915c6c43da664807a51a6823c0d211f0 : SUCCESS in 26m 57s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3d50545f2f1b4ab59e4acd022e25ac7a : SUCCESS in 30m 24s","accounts_in_message":[],"_revision_number":5},{"id":"b7a31a7b08dc4a57dcdbb709c391d30aacbd11fd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-07 12:34:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"2ace0789f0c41bb31e1fa4eea245025214f4119c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-07 12:35:05.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f21c5451f56348cfb51f432e50cb5fdd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6327571bc8f54c4884c31bb89152024d : SUCCESS in 43s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"22b6df4193bf723dcbc6d5ce4efff326e4117e86","revisions":{"bdb9cbec5a1ef0b368e51adbd0f42633eea0bf03":{"kind":"REWORK","_number":1,"created":"2023-09-11 16:14:19.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/55/894555/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/55/894555/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/55/894555/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/55/894555/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/55/894555/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/55/894555/1"}}},"commit":{"parents":[{"commit":"18e62bd573c9f83274e7be574cd13876109c5381","subject":"Add libpython mapping for debian bookworm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/18e62bd573c9f83274e7be574cd13876109c5381"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-11 15:56:35.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-11 16:13:24.000000000","tz":60},"subject":"Ensure systemd-resolved is present in debian container images","message":"Ensure systemd-resolved is present in debian container images\n\nIt is not present by default in the rootfs built by\ndebootstrap for debian bookworm\n\nChange-Id: Ie7200d5c01948c885c3dd4e8103c8f0a65e26108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/bdb9cbec5a1ef0b368e51adbd0f42633eea0bf03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/bdb9cbec5a1ef0b368e51adbd0f42633eea0bf03"}]},"branch":"refs/heads/master"},"6816279140302649c1956ccada2a420262ece360":{"kind":"REWORK","_number":2,"created":"2023-09-11 18:30:02.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/55/894555/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/55/894555/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/55/894555/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/55/894555/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/55/894555/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/55/894555/2"}}},"commit":{"parents":[{"commit":"18e62bd573c9f83274e7be574cd13876109c5381","subject":"Add libpython mapping for debian bookworm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/18e62bd573c9f83274e7be574cd13876109c5381"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-11 15:56:35.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-11 18:29:19.000000000","tz":60},"subject":"Ensure systemd-resolved is present in debian container images","message":"Ensure systemd-resolved is present in debian container images\n\nIt is not present by default in the rootfs built by\ndebootstrap for debian bookworm\n\nChange-Id: Ie7200d5c01948c885c3dd4e8103c8f0a65e26108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6816279140302649c1956ccada2a420262ece360"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6816279140302649c1956ccada2a420262ece360"}]},"branch":"refs/heads/master"},"c266b293d04f4b0432ccc42e82a401225e244c48":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-09-27 16:28:41.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/55/894555/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/55/894555/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/55/894555/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/55/894555/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/55/894555/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/55/894555/3"}}},"commit":{"parents":[{"commit":"2272de8f0c5977819ed38582c30cff68d6da7d5b","subject":"Fix linters issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2272de8f0c5977819ed38582c30cff68d6da7d5b"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-11 15:56:35.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-27 16:28:41.000000000","tz":0},"subject":"Ensure systemd-resolved is present in debian container images","message":"Ensure systemd-resolved is present in debian container images\n\nIt is not present by default in the rootfs built by\ndebootstrap for debian bookworm\n\nChange-Id: Ie7200d5c01948c885c3dd4e8103c8f0a65e26108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c266b293d04f4b0432ccc42e82a401225e244c48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c266b293d04f4b0432ccc42e82a401225e244c48"}]},"branch":"refs/heads/master","description":"Rebase"},"bb2a2f53c0bc97dcf5879ff430d5cf00320e5741":{"kind":"REWORK","_number":4,"created":"2023-09-27 16:29:20.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/55/894555/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/55/894555/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/55/894555/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/55/894555/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/55/894555/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/55/894555/4"}}},"commit":{"parents":[{"commit":"2272de8f0c5977819ed38582c30cff68d6da7d5b","subject":"Fix linters issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2272de8f0c5977819ed38582c30cff68d6da7d5b"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-11 15:56:35.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-27 16:29:03.000000000","tz":0},"subject":"Ensure systemd-resolved is present in debian container images","message":"Ensure systemd-resolved is present in debian container images\n\nIt is not present by default in the rootfs built by\ndebootstrap for debian bookworm\n\nChange-Id: Ie7200d5c01948c885c3dd4e8103c8f0a65e26108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/bb2a2f53c0bc97dcf5879ff430d5cf00320e5741"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/bb2a2f53c0bc97dcf5879ff430d5cf00320e5741"}]},"branch":"refs/heads/master"},"22b6df4193bf723dcbc6d5ce4efff326e4117e86":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-09-28 09:19:18.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/55/894555/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/55/894555/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/55/894555/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/55/894555/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/55/894555/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/55/894555/5"}}},"commit":{"parents":[{"commit":"42cfa88bb56f1cb29bb07f499a58ff9d6631f6cc","subject":"Fix linter error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/42cfa88bb56f1cb29bb07f499a58ff9d6631f6cc"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-11 15:56:35.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-28 09:19:18.000000000","tz":0},"subject":"Ensure systemd-resolved is present in debian container images","message":"Ensure systemd-resolved is present in debian container images\n\nIt is not present by default in the rootfs built by\ndebootstrap for debian bookworm\n\nChange-Id: Ie7200d5c01948c885c3dd4e8103c8f0a65e26108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/22b6df4193bf723dcbc6d5ce4efff326e4117e86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/22b6df4193bf723dcbc6d5ce4efff326e4117e86"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
