)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~894556","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I0c62340e7868948d9c55c96559ddafadf8cb7db1","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"osa-bookworm","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-10-02 06:50:04.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2023-10-09 12:19:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0c62340e7868948d9c55c96559ddafadf8cb7db1","subject":"Sync additional apt config from the host to the container base image","status":"MERGED","created":"2023-09-11 16:14:19.000000000","updated":"2023-10-09 12:20:15.000000000","submitted":"2023-10-09 12:19:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"894556-osa-bookworm","meta_rev_id":"4a040618c623e39a25fa7ad8013b7c175e5f4aa3","_number":894556,"virtual_id_number":894556,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-09 12:19:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2023-10-02 14:28:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2023-10-02 06:50:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-07 21:17:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-11 20:39:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-29 08:11:48.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2023-10-02 14:28:05.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"5f8148909772c972c0b12a0f3adb67f4014f3589","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-09-11 16:14:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4cf5f0dba4bb0dc13299da4329a31e86798f7370","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-09-11 18:30:02.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"f7d84cf75abff072024733203d62eba223490cfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 20:39:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bfed7752f3184d5e82bc5184166b0893\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4fdf117c3d74420a8813c3d907865c99 : SUCCESS in 13m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66a4b673159a4c1cbe6e4b6edbde2031 : SUCCESS in 4m 22s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/38a2a2471cbf4aedb917cb7df72367ec : SUCCESS in 1h 40m 29s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1e9c0cb847864caea8e47c96df6d256c : SUCCESS in 1h 11m 29s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/311250c1899e4b18b1221eba3d498659 : SUCCESS in 1h 30m 11s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/d9b5a12d4ff84b85b66d07314853d8fc : SUCCESS in 1h 17m 17s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9be5ea8f10dc4ea28d6498ed937e2c86 : SUCCESS in 1h 20m 09s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0974a073a131403580210acb464d179e : SUCCESS in 1h 28m 45s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d5df50c0d9ce4f4cb1841854a945b289 : SUCCESS in 1h 20m 08s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7548b82c2fc2446ca0a6d8d7c44beabd : SUCCESS in 2h 05m 57s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1ca7667d91b245b392976934ae388281 : SUCCESS in 1h 44m 03s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0abb2e854574d65be0495ed9a1ea8d6 : FAILURE in 33m 51s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf24b4a4a2194fba8ef4e0476cbe77a1 : SUCCESS in 33m 01s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f215a19a7e0840609323a9a0cf3ecf51 : FAILURE in 29m 14s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/3aab2dcbc5844bf38bae47c96bdc2f5e : SUCCESS in 25m 23s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/1ab0988789bd479cafc77b87af840760 : SUCCESS in 25m 18s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c0e980fe53944d238304655ae544a96d : SUCCESS in 29m 49s","accounts_in_message":[],"_revision_number":2},{"id":"04a20c53c2e595511b6ea6ee23423e7d7958262e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-09-12 08:00:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ee6907d934333c684d8b6c0961db89dbab4e2bfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 09:59:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cf615fa6fa840a899d4776bab8df276\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e467d5055e44264ba51060716d7ebff : SUCCESS in 14m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13cd902a3269499db05375afbe9ed3fc : SUCCESS in 5m 06s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/b725a64d05574677a0c682523c4948ff : SUCCESS in 1h 28m 28s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5623e0a21b434ca2b6172347e7903d42 : SUCCESS in 1h 30m 15s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/37b7c4ecf0db4571bfab36d708fd07eb : SUCCESS in 1h 32m 20s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/96c2502f65644ca880000a694da3942d : SUCCESS in 1h 10m 24s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/200c95f0d73f46deb74d8d908e061249 : SUCCESS in 1h 10m 08s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d9af2670248343ecaa9b247660878a51 : SUCCESS in 1h 09m 57s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/888f177e799e4405ad1ff1e1023e1160 : SUCCESS in 1h 18m 04s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/52aeab100cf1406ca828241c66dff43d : SUCCESS in 1h 56m 01s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3988ecc0198f40a0a375f35c984bd2d6 : SUCCESS in 1h 39m 50s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9559f96dab7549359e4e98285e5f61d6 : SUCCESS in 28m 03s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4f4f95fb4d8c45db825a3649fc2bfb9b : SUCCESS in 32m 45s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c9089664bb543e48c6a822f521014c0 : SUCCESS in 27m 25s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/fe3fed6a90ee4a698943055392b8a3e9 : SUCCESS in 26m 30s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/ddc1066b14894910aab8b27638634640 : SUCCESS in 26m 18s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/df4ce8eff93348c191d6149339f83a50 : SUCCESS in 30m 28s","accounts_in_message":[],"_revision_number":2},{"id":"50cec86824477d3429679243319a0ba05e44cde2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-09-27 16:29:30.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"83556cdd1ad3097b2f8b214c03ec3c123a9a2a48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 18:31:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8a09ee3c9b249628ca1e10710445bf4\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7c7876974a046e48fecb8b79a533159 : SUCCESS in 14m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/119dfb26c2d84c4fa5c60c61ea51e9c3 : SUCCESS in 4m 33s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/c3ec16b4ab3b4c829b464275fdbc6ced : FAILURE in 1h 39m 56s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1b993275c3064645a97985ae4e4649af : SUCCESS in 1h 37m 42s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bdf9d88944544c5ba033e57f9ce2b0cb : SUCCESS in 1h 46m 29s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/54ee44895d5a4697a7b7ba8d78dafc1f : SUCCESS in 1h 08m 04s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a9e4bfc0f6554097bdb4489d686fe652 : SUCCESS in 1h 20m 47s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/04761a3434c544e78c842e15a2c39c01 : SUCCESS in 1h 18m 13s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/313acb3154534f5a86f5ca484cf61519 : SUCCESS in 1h 29m 56s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/155fc8f4391c430cbdffa19f29b74be6 : SUCCESS in 1h 52m 17s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b47e5c7c58a1495fa203e539495cf961 : SUCCESS in 1h 50m 05s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce77aed4d4f24a5d9a494c0dde9aac00 : SUCCESS in 27m 18s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/18ceed893ecc4068ab32fbe7d1a054bf : SUCCESS in 31m 43s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae4a057b8b584e6ba45b8a0af43e56dd : SUCCESS in 33m 59s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/adf64dda67b640b7a376c6fc77829902 : SUCCESS in 28m 15s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/52fe40b545e84c1196831fe3f347a3d3 : SUCCESS in 37m 17s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1e42bc5513ab465799909a5ad0cfc558 : SUCCESS in 31m 47s","accounts_in_message":[],"_revision_number":3},{"id":"0e371729f04b6c16230d6f9218c688b6f56897ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-09-28 09:19:29.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"2fecd05a897d419fe3b797fd1dd177e3b2baaa67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 11:53:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14eb363210914ff3912acb8dcb1af575\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1e84f07aeb4547b7804c766eb235ba7c : SUCCESS in 15m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd54e6807d22458182c29681f4724a8c : SUCCESS in 5m 15s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/6462ff70a8844c9b8097ad2fb2ab12c9 : SUCCESS in 1h 35m 34s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e5fb506526bc44fb891a50fba43f0808 : SUCCESS in 1h 39m 08s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ec7f6b4c066b43bd9a93b522fcaba30d : SUCCESS in 1h 45m 02s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/24c4a28626e940d881c459bd9ce11ed0 : SUCCESS in 1h 20m 43s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/70c24f29671f4aedbc8e776c9a775040 : SUCCESS in 1h 26m 38s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bd79e3b522d849e9b9ac1932a2e0a026 : SUCCESS in 1h 15m 53s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e382f73e6c4b4ed0858c21b7293e0f2b : SUCCESS in 1h 18m 48s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fd1188fac20445808ddc8a7b37126e86 : SUCCESS in 2h 27m 33s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ac5c6212cdc849d7af36737eb4ca0cfd : SUCCESS in 1h 43m 53s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/58abe2d3751648c7bc2cb5ad1809500e : SUCCESS in 29m 40s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/84659448afbb48dfb5eaaeb89b319b0c : SUCCESS in 37m 03s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7386112e3a9a4486a9ba050a36ad893b : SUCCESS in 31m 44s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/0efa7a0138164b1791e5c934b0ce8ded : SUCCESS in 32m 41s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/5f68ed479f2e4504840cec2e6af06e2f : SUCCESS in 29m 20s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f658ca88764d4ca184e5cc2f8bbcc8f6 : SUCCESS in 29m 51s","accounts_in_message":[],"_revision_number":4},{"id":"95a1c0563847803a86f24bbb488e61eb8b1c0d7c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-09-29 08:11:48.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9d0c108aea9ad3723eb06b08af369764e1943095","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-09-29 12:43:58.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"e83d40e2fed33790d8103d25c3c3f5516a8ac61e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 14:58:56.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4caa2d09c7374abd919133e615576773\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8b149fbd08e34407a4eeef6e9799a369 : SUCCESS in 15m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09a2c43cd97646ee8defa9fb9d362228 : SUCCESS in 4m 38s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/8943862d089344399b8f16de1ebfd503 : SUCCESS in 1h 32m 58s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1ab3c44749374d07890ba273c22e231f : SUCCESS in 1h 34m 39s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/afe6397cfec14be2be3f674c0a1227b2 : SUCCESS in 1h 37m 49s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/0d29c6bb034044aa9337c7efe01e4cc8 : SUCCESS in 1h 04m 44s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2e090b51cce84f3682424845687f68ae : SUCCESS in 1h 19m 04s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eac29f5bc9974d3ca3fab3d0e36faca7 : SUCCESS in 1h 20m 05s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fb44cb8a6a29423594370986b0d1e628 : SUCCESS in 1h 27m 42s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a6d9c4a35d84448084e3aba94aa56c86 : SUCCESS in 2h 08m 34s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ddfb2a88484f41fea00f6d4cfdfa3e13 : SUCCESS in 1h 53m 11s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/278d0dacc31044fb82811c7f70cdb9ec : SUCCESS in 27m 11s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/24f9725099354b6eb4fd118cf32c4a54 : SUCCESS in 27m 59s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c5878fb756d482089afaf8b4e414bc7 : SUCCESS in 32m 10s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/9acd8bc66170405a9e624fbdff4a9278 : SUCCESS in 30m 03s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/da999ff056b445adbffb642ea48bd9ba : SUCCESS in 28m 15s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5eec688c06ea40b0a1579f4cb586cba9 : SUCCESS in 30m 43s","accounts_in_message":[],"_revision_number":5},{"id":"48d7282fe17de99d275b7d0a797217ce0b06378d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-02 06:50:04.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"f21ea04d21ab3cf69618e15e01666dec4a1e4f78","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2023-10-02 14:28:05.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"5f0a75b9eb2ac1997986ee27b54c4674177c6512","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-10-07 21:17:31.000000000","message":"Patch Set 5: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"184c1a8e556bf3e7ab2830fb044ac721e3ce1b23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-07 21:19:30.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"8b702bd8f54459c3082757e2437b482318c09b33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-07 23:36:00.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d373f2d70851472d83a5210b46bbd672\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b85ebbab06f34b5d8814ed9b1ea2adb9 : SUCCESS in 11m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49decbed8d224ba6a6725cfead6c9187 : SUCCESS in 4m 58s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0921f6e76f984554985dfde913851acb : SUCCESS in 1h 12m 46s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/00e9b2ec1c2341f68a84fe161f799143 : SUCCESS in 1h 11m 03s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/e23d47ace1df41d8bdab9f993dca12a2 : FAILURE in 23m 27s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c1cefb93f404ad8bf64d2762f492532 : SUCCESS in 1h 01m 22s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a8d5c8b065a741659196944cc317a54e : SUCCESS in 59m 41s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc6b88e6766446b6b9f0ff0ddfb85017 : SUCCESS in 1h 06m 41s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/00551a604ad9427e90b318e7a07c5349 : SUCCESS in 2h 10m 24s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/62c437cbda824af2a74648e034ee15c8 : SUCCESS in 1h 15m 56s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e64f794b6aa44e26b72695c4eaba97ae : SUCCESS in 25m 18s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c272b732f8b849aa8ccff1ba6c15c1e9 : SUCCESS in 26m 24s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce379ae9da184a9bb0fa6644b35f64a6 : SUCCESS in 25m 29s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/284e68bbdd3544edbb7dfee28d391bfd : SUCCESS in 28m 29s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/53b737d4dbc64bd49909e1d5d68d70e2 : SUCCESS in 28m 31s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9022d6e3802a46118655cfddef006cfe : SUCCESS in 30m 53s","accounts_in_message":[],"_revision_number":5},{"id":"bb128462622f211038c2e7e0ae3f75bb32a47abf","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-09 07:52:29.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ceacbeb581fa414596046bf57f9cdfd2950c9fad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 10:05:23.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5c3b3ef606346d8a3e0cc75d8e4b69f\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/54c4d37a02f643e6835a4cdeec4a5704 : SUCCESS in 18m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5ee71ca7b114bafaf347388a6a93acc : SUCCESS in 4m 46s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/3f35b61b8b4e46eca3158cfc200ccc69 : SUCCESS in 1h 25m 00s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a2df41768ac74a24a7e0e3ffe981f9bc : SUCCESS in 1h 25m 07s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ef2b1644864e4fbb9991d4756c17eb0f : SUCCESS in 1h 35m 01s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/f9856b6857294e51bb1aad42e927cc54 : SUCCESS in 1h 05m 51s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1f5634aa0bfe4e51a15b4dfd6d845a70 : SUCCESS in 1h 18m 50s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/15487d18ccb24229b904c6ef47c0178a : SUCCESS in 1h 19m 53s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e1e9f1a753d54c0491113c2a539f81f9 : SUCCESS in 1h 20m 55s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6f54226947444c9ba6e589914dcfa0fb : SUCCESS in 2h 06m 33s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1e91330f767f4a4dab9a8742eea4b992 : SUCCESS in 1h 45m 17s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/04785fcd815d4a508541a2c5aff96f78 : SUCCESS in 27m 26s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a72455a628743e8a30910e5081daf26 : SUCCESS in 31m 29s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4a314cb7fe774e29a611b02a5db2fe68 : SUCCESS in 33m 44s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/19752ea8f5da448984c489c2cf334539 : SUCCESS in 31m 51s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/973ac1042cc7489b98d84279dcdb3f83 : SUCCESS in 37m 08s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2b113f205e4d41ea8f657a6b1a6f5e2a : SUCCESS in 32m 52s","accounts_in_message":[],"_revision_number":5},{"id":"ec66c44446ef8d5bfe9e2569993f5ef056f3b337","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 10:07:31.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"692a77ac5fea65eec499a74ca3cae354c89b8027","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 12:19:17.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ebc13f46f8c4d49ad2903912165e59b\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a5f2ce63973a4ce09367c807e416e230 : SUCCESS in 11m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58a78fefd22a4e75808ccae4589b0309 : SUCCESS in 5m 43s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/adc5f62013674ca28e0febd39b2e851f : SUCCESS in 1h 40m 34s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/faa86c6b1d914ee58092578decfac831 : SUCCESS in 1h 40m 12s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/4858eebb6fb74516959363ca808c4e54 : SUCCESS in 1h 21m 19s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b273808800324fb2a4cb932c00be3e37 : SUCCESS in 1h 25m 47s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b8555c3cd6c4a0db3b40a1f31dc3da9 : SUCCESS in 1h 26m 43s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59e7ab47e4bf42779b118c6e0679d327 : SUCCESS in 1h 22m 51s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d45dfa9f61274ea3a53811ebcd8564e5 : SUCCESS in 2h 07m 28s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/32b94cc8db794f5189b5bafc9d3bc7a4 : SUCCESS in 1h 53m 57s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d6e2408b5e334d3f9e52442a3a8ca843 : SUCCESS in 29m 26s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/abeb0099e42f477e90f3b55ac5b92d8e : SUCCESS in 33m 22s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c0b16fdea5ca4cbb9c40adcb19631120 : SUCCESS in 28m 02s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/c959888b5cdc43069d5190a41d071660 : SUCCESS in 25m 30s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/e84f5c6ae61248aca871fbac8f8d2dd1 : SUCCESS in 27m 43s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d3aea6d2a775403594d147018e59ecd6 : SUCCESS in 34m 37s","accounts_in_message":[],"_revision_number":5},{"id":"14cc40d8064df32d8fa804278b41e8f22c547ab3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 12:19:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"4a040618c623e39a25fa7ad8013b7c175e5f4aa3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 12:20:15.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a971f6505b78452db66866cf1a2f661f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/791441ef2fd840b7bf91bc7c496dde9f : SUCCESS in 38s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"2506f0080a34cf48069954e26dc5a2fab23a310f","revisions":{"4a8bc6de280b5f0839abb3fb2aa4516f6c0bc6e3":{"kind":"REWORK","_number":1,"created":"2023-09-11 16:14:19.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/56/894556/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/56/894556/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/56/894556/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/56/894556/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/56/894556/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/56/894556/1"}}},"commit":{"parents":[{"commit":"bdb9cbec5a1ef0b368e51adbd0f42633eea0bf03","subject":"Ensure systemd-resolved is present in debian container images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/bdb9cbec5a1ef0b368e51adbd0f42633eea0bf03"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-11 16:02:19.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-11 16:13:26.000000000","tz":60},"subject":"Sync additional apt config from the host to the container base image","message":"Sync additional apt config from the host to the container base image\n\nDebian bookworm needs the sources.list.d and mirrors directory\nsyncing to the container image to result in a working apt config.\n\nChange-Id: I0c62340e7868948d9c55c96559ddafadf8cb7db1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/4a8bc6de280b5f0839abb3fb2aa4516f6c0bc6e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/4a8bc6de280b5f0839abb3fb2aa4516f6c0bc6e3"}]},"branch":"refs/heads/master"},"02297473ef32c3a49b74e676889f039da920c56d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-09-11 18:30:02.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/56/894556/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/56/894556/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/56/894556/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/56/894556/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/56/894556/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/56/894556/2"}}},"commit":{"parents":[{"commit":"6816279140302649c1956ccada2a420262ece360","subject":"Ensure systemd-resolved is present in debian container images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6816279140302649c1956ccada2a420262ece360"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-11 16:02:19.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-11 18:29:34.000000000","tz":60},"subject":"Sync additional apt config from the host to the container base image","message":"Sync additional apt config from the host to the container base image\n\nDebian bookworm needs the sources.list.d and mirrors directory\nsyncing to the container image to result in a working apt config.\n\nChange-Id: I0c62340e7868948d9c55c96559ddafadf8cb7db1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/02297473ef32c3a49b74e676889f039da920c56d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/02297473ef32c3a49b74e676889f039da920c56d"}]},"branch":"refs/heads/master"},"f435f144aa58ec73ef7b21925f435e03f435b300":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-09-27 16:29:30.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/56/894556/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/56/894556/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/56/894556/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/56/894556/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/56/894556/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/56/894556/3"}}},"commit":{"parents":[{"commit":"bb2a2f53c0bc97dcf5879ff430d5cf00320e5741","subject":"Ensure systemd-resolved is present in debian container images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/bb2a2f53c0bc97dcf5879ff430d5cf00320e5741"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-11 16:02:19.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-27 16:29:30.000000000","tz":0},"subject":"Sync additional apt config from the host to the container base image","message":"Sync additional apt config from the host to the container base image\n\nDebian bookworm needs the sources.list.d and mirrors directory\nsyncing to the container image to result in a working apt config.\n\nChange-Id: I0c62340e7868948d9c55c96559ddafadf8cb7db1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f435f144aa58ec73ef7b21925f435e03f435b300"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f435f144aa58ec73ef7b21925f435e03f435b300"}]},"branch":"refs/heads/master","description":"Rebase"},"b6390ed2d3c4b4a7684957f93a30f1b5a49be420":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-09-28 09:19:29.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/56/894556/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/56/894556/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/56/894556/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/56/894556/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/56/894556/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/56/894556/4"}}},"commit":{"parents":[{"commit":"22b6df4193bf723dcbc6d5ce4efff326e4117e86","subject":"Ensure systemd-resolved is present in debian container images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/22b6df4193bf723dcbc6d5ce4efff326e4117e86"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-11 16:02:19.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-28 09:19:29.000000000","tz":0},"subject":"Sync additional apt config from the host to the container base image","message":"Sync additional apt config from the host to the container base image\n\nDebian bookworm needs the sources.list.d and mirrors directory\nsyncing to the container image to result in a working apt config.\n\nChange-Id: I0c62340e7868948d9c55c96559ddafadf8cb7db1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b6390ed2d3c4b4a7684957f93a30f1b5a49be420"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b6390ed2d3c4b4a7684957f93a30f1b5a49be420"}]},"branch":"refs/heads/master","description":"Rebase"},"2506f0080a34cf48069954e26dc5a2fab23a310f":{"kind":"REWORK","_number":5,"created":"2023-09-29 12:43:58.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/56/894556/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/56/894556/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/56/894556/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/56/894556/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/56/894556/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/56/894556/5"}}},"commit":{"parents":[{"commit":"22b6df4193bf723dcbc6d5ce4efff326e4117e86","subject":"Ensure systemd-resolved is present in debian container images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/22b6df4193bf723dcbc6d5ce4efff326e4117e86"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-11 16:02:19.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-29 12:43:58.000000000","tz":0},"subject":"Sync additional apt config from the host to the container base image","message":"Sync additional apt config from the host to the container base image\n\nDebian bookworm needs the sources.list.d and mirrors directory\nsyncing to the container image to result in a working apt config.\n\nChange-Id: I0c62340e7868948d9c55c96559ddafadf8cb7db1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2506f0080a34cf48069954e26dc5a2fab23a310f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2506f0080a34cf48069954e26dc5a2fab23a310f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
